)]}'
{"id":"openstack%2Fneutron~877985","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~If362fab20ac03f8b62471b60c031f9349171ce93","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/2011724","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-23 16:55:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If362fab20ac03f8b62471b60c031f9349171ce93","subject":"[OVN] Explicitly define the fixed IPs for the metadata port","status":"MERGED","created":"2023-03-20 16:37:19.000000000","updated":"2023-03-23 16:57:55.000000000","submitted":"2023-03-23 16:55:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877985-bug/2011724","meta_rev_id":"7d386e6cfa4d8f2e7443b819c0b2800f97b7961c","_number":877985,"virtual_id_number":877985,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-23 16:55:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-03-22 10:12:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-03-22 14:14:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-03-22 14:14:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-20 17:02:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-20 18:37:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-22 10:12:41.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-03-22 14:14:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"41d76952d92f3f4d6810ed1aa4129c8125b04b63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-20 16:37:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d21eb33f044be0f68146c3e070ba5e4e89900e68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 17:02:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf7511e1cfec4823943372a233627cf3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47a9f8c9ed394bf3893c775aff96a015 : SUCCESS in 20m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7379f0f9e044078bbb14abac6a61b92 : SUCCESS in 23m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2e2476fee4e967ad59c2a8443ecb5cba7b1bd27f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 18:37:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad96a9eb502f4f47826f85eee3d33adf\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/7155dc44e60a4e3783034a1cfe17a435 : SUCCESS in 52m 17s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/8518fbcb4f49468b967d6ceb45f53657 : SUCCESS in 1h 56m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af3f3ccf6ebc4cf88c495058f4daae1a : SUCCESS in 32m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fc60a33338847f1ad015c76e06919eb : SUCCESS in 16m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68154b472d17438dbd938c3974faab03 : SUCCESS in 35m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0623e72b7014ef295ca93ac342f36f2 : SUCCESS in 29m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaa2157d427642999e281d0977c01d20 : SUCCESS in 33m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eff255c1aed14b74b851322880735767 : SUCCESS in 10m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9b3393ff74694db4b43f42509a6c870e : FAILURE in 42m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/90445062637a4871951dabb500047867 : SUCCESS in 1h 00m 57s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/1a456b4a3560465d9a4474bb1bc5d8df : SUCCESS in 23m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b4e86564275f40a8a0af130cfc666799 : SUCCESS in 1h 27m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/15b399d9426941ee978c374645300bcc : SUCCESS in 1h 21m 20s","accounts_in_message":[],"_revision_number":1},{"id":"bf7cacd98a64c91b598d7e9aeb01181c6eb3fc9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-21 17:07:04.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"87e9f84b5364ec03248a30fa3cde04ffa9676064","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 17:47:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5074b5555344782a1d725d048b9f92d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a471de7220445968b5cc241cbf04248 : SUCCESS in 26m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc84b4b5e1d0496580d799fc00d71002 : SUCCESS in 39m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"30b477c524a0e8a0ff910452960eb50c7a2ff81d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 19:19:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f30fa55ec78f4de8b29584b085290e02\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/d4fa7cc7f8784f6aa08a04955e2a4999 : SUCCESS in 59m 56s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/b76aacec561a499db969db6cb7845b03 : SUCCESS in 2h 04m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d720b8336a2b4ac9897d36a6dd897d9b : SUCCESS in 44m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c9b22355b3141f59960add9725ca124 : SUCCESS in 13m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cd3966a8f704bb4a53db03313b35154 : SUCCESS in 28m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc872bacdc12485283c809faa9d330f4 : SUCCESS in 28m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf2cf3ffe54b4461ae6d91595f2407d4 : SUCCESS in 29m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f366c9c6c84726965c28a3033bd992 : SUCCESS in 9m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/597e39f7f9ce454da6dabca13c9e5540 : SUCCESS in 44m 12s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/724593827ae04c1c9c0c65203f27af3c : SUCCESS in 1h 09m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/88461642e23d4418b6cfb2742b824321 : SUCCESS in 35m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ac6ac54faf1a484eb1a3e2576b846e09 : SUCCESS in 1h 43m 03s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d35af913ec2a4b4e8f7878e638e4e586 : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":2},{"id":"ad5b0a3ec04a5f633e522ad6436d3869385582be","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-03-22 10:12:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"607929f4ff35963b459eec43a6d601ecc3aa7f2e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-22 14:14:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2baccad1d2b018c262a639e2dcfcabbd53384631","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 15:04:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ac2fd76111a1cf1ef61279cc0137ea23cbf6eabc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:55:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42b7791d695f4fa3b58f2bfd853257b1\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/06c65d5323694e3584aab5a9ba222ba3 : SUCCESS in 53m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de4cec4fbfde4ceea98e782657700e94 : SUCCESS in 11m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb454d7d3ef1405d99033873826c5390 : SUCCESS in 33m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fd81371561d45768e57b796c451d95b : SUCCESS in 28m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff864363054f481fafe303d85e44abbe : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c5b19f92e0604e4da2db7666c5c65b7e : SUCCESS in 52m 29s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/194ef1d3ee6e48a7a895fbf02695ba44 : SUCCESS in 1h 03m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/42f354ad63994773b16408a408e4d58a : SUCCESS in 32m 36s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/cb3f35fffe7d4c85a995a96ffad620fd : SUCCESS in 1h 32m 52s","accounts_in_message":[],"_revision_number":2},{"id":"0e3e2bc9a064e95745522be6581911bb9d29e58a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:55:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7d386e6cfa4d8f2e7443b819c0b2800f97b7961c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:57:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f8d02b6cd2747768aee4e31d5228890\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7d994b6d66f498199d5cfd33deaedf1 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"512e1d64e3e9a5f12335e8aca9fde15eb186bf8c","revisions":{"44c4f69cfa799eb554b75d881bab246ccbee3476":{"kind":"REWORK","_number":1,"created":"2023-03-20 16:37:19.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/85/877985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/877985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/877985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/877985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/877985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/877985/1"}}},"commit":{"parents":[{"commit":"d159a7a62351d25ab085887eeed3acd9e6e19dd8","subject":"Merge \"[OVS] Allow custom ethertype traffic in the ingress table\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d159a7a62351d25ab085887eeed3acd9e6e19dd8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:29:39.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-20 16:37:09.000000000","tz":60},"subject":"[OVN] Explicitly define the fixed IPs for the metadata port","message":"[OVN] Explicitly define the fixed IPs for the metadata port\n\nThe metadata port fixed IPs depend on the subnets \"enabled_dhcp\" flag.\nIf the subnet has this flag disabled, the metadata port doesn\u0027t receive\nan IP on the subnet CIDR.\n\nThe method ``create_metadata_port`` should explicitly define what fixed\nIPs should request the metadata port during the creating depending on\nthe subnets \"enabled_dhcp\" flag.\n\nCloses-Bug: #2011724\nChange-Id: If362fab20ac03f8b62471b60c031f9349171ce93\n(cherry picked from commit 9704dca84ea3ad21ecf9730eea03692daeddf382)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44c4f69cfa799eb554b75d881bab246ccbee3476"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44c4f69cfa799eb554b75d881bab246ccbee3476"}]},"branch":"refs/heads/stable/wallaby"},"512e1d64e3e9a5f12335e8aca9fde15eb186bf8c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-21 17:07:04.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/85/877985/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/877985/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/877985/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/877985/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/877985/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/877985/2"}}},"commit":{"parents":[{"commit":"791ac95ecda9323d27fb4a5a2ed82715db2eced9","subject":"Fix default value for MTUs, when not provided","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/791ac95ecda9323d27fb4a5a2ed82715db2eced9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-15 05:29:39.000000000","tz":60},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-21 17:07:04.000000000","tz":0},"subject":"[OVN] Explicitly define the fixed IPs for the metadata port","message":"[OVN] Explicitly define the fixed IPs for the metadata port\n\nThe metadata port fixed IPs depend on the subnets \"enabled_dhcp\" flag.\nIf the subnet has this flag disabled, the metadata port doesn\u0027t receive\nan IP on the subnet CIDR.\n\nThe method ``create_metadata_port`` should explicitly define what fixed\nIPs should request the metadata port during the creating depending on\nthe subnets \"enabled_dhcp\" flag.\n\nCloses-Bug: #2011724\nChange-Id: If362fab20ac03f8b62471b60c031f9349171ce93\n(cherry picked from commit 9704dca84ea3ad21ecf9730eea03692daeddf382)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/512e1d64e3e9a5f12335e8aca9fde15eb186bf8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/512e1d64e3e9a5f12335e8aca9fde15eb186bf8c"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
