)]}'
{"id":"openstack%2Fneutron~878248","triplet_id":"openstack%2Fneutron~master~Iee72be61d00c11a938a0b1b31e75bdda788d5c57","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-08-22 11:18:06.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"36313":{"account":{"_account_id":36313,"name":"Adil Ishaq","display_name":"Adil Ishaq","email":"adilishaq.cc@hotmail.com","username":"Adilishaq","status":"All about AI cyber research for Climate Change with Peaceful Reachable Approach."},"last_update":"2023-10-04 07:56:57.000000000","reason":"Change was abandoned"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-08-10 03:53:44.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2023-10-04 07:56:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iee72be61d00c11a938a0b1b31e75bdda788d5c57","subject":"fix: add log message for periodic_sync_routers_task fullsync","status":"ABANDONED","created":"2023-03-22 19:26:34.000000000","updated":"2023-10-04 07:56:57.000000000","total_comment_count":8,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"07e182ca819010fd2b95028d913ec56423fcce2e","_number":878248,"virtual_id_number":878248,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":1,"date":"2023-08-30 16:19:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-22 13:07:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"disliked":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2023-08-22 11:18:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-08-22 10:49:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-22 10:49:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-31 08:29:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-22 19:44:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-03-22 20:08:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-22 20:30:23.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-22 21:39:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-23 09:05:36.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2023-03-23 09:16:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-03-23 09:21:26.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-03-23 18:27:10.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-06-14 11:56:40.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32823,"name":"Hannes von Haugwitz","display_name":"hvhaugwitz","email":"hannes@vonhaugwitz.com","username":"hvhaugwitz"},"state":"CC"},{"updated":"2023-08-10 03:53:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-08-31 08:29:12.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2023-10-04 07:56:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"6d55a39007f6b08958ddd86ceedd0f66e5753b76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2023-03-22 19:26:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aefbd706b9bcd8c6428552b269b4fe8a1deb23df","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-22 19:44:57.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8c4e90a5cdbbc453475baa358d52b6db214c0edb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 20:08:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/983e446de5614afeae89419ca9e036c7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b4882d1833f4bdb8200486baa4223d3 : SUCCESS in 39m 38s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f8da628bb5c4cdc9197c2f657df2116 : SUCCESS in 39m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6b769df20270453f84196b1c58804f0f : SUCCESS in 29m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9d2a60bfc22a4236137f9f24847d2b09f02b2d1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-22 20:30:23.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/48/878248/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"2e34a8e92ee5042fa930f36458e289b81b8982b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 21:39:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b7d4685b8384e61b603f08d14065afc\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/602512cb440545829fa29ac4dabf53bd : SUCCESS in 1h 42m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d83ab359ea734db9977e0173c9ce8c9b : SUCCESS in 1h 37m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f458888232ec423ebd060a4a4adab3a7 : SUCCESS in 1h 24m 18s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c75034ec7c414316aae8a30286ef8e36 : SUCCESS in 2h 11m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8d4c11e6cbe48588f73de4045d60d64 : SUCCESS in 33m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/061bc635139b4322be6d377e300244cc : SUCCESS in 14m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6da9af6709bf44efaa368843eb3762b9 : SUCCESS in 26m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6173d65c6c5842a7a04646b0237df9ad : SUCCESS in 27m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7040d6df76054c919aab3c64e003ce60 : SUCCESS in 29m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/891a6c0cf1bc46b29e2d8f1613ac6487 : SUCCESS in 11m 37s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/ba8d3be434f24648b47b57e4cd5e4be0 : SUCCESS in 23m 48s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/115facfada824c6a9000505c2e5e4bce : SUCCESS in 1h 32m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/caeaee971b13485f9383181441dd9719 : SUCCESS in 1h 48m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cdc18a0fd96e4545a3ee758974a7a8d0 : SUCCESS in 55m 07s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8f5e6989342b464da5eca0d887a84244 : FAILURE in 21m 55s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/09bf697943704a25b14d0d5008f449a5 : SUCCESS in 1h 02m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/88ff7e40babe4d76b8e8e65e74187a86 : SUCCESS in 1h 49m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e7e150c94f80466e9c11e94ccd3423ed : SUCCESS in 1h 36m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ffb3844bc2d44e5f9485158f9e98cb4a : SUCCESS in 1h 58m 48s","accounts_in_message":[],"_revision_number":1},{"id":"db657e2ab188c36b507518714f37d9716b679948","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-03-23 09:05:36.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e94aa51ecba2ca7684d378ba4194459d94ef70fe","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-23 09:16:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fead764f11e72a5cfa5a5c4543d33237e46137e4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-23 09:17:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1141a6553b6284c2265116ac8ab1b25eb43bd10b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-23 09:21:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed7c3c0b64eeb71f17539c8834082a3c98b776e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2023-03-24 14:53:30.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e665499f101c0181c53ae2e8a261da62d341b87f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 15:16:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04559a8036564d849bf670550be8422e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3758198d7f9246708597b98d094e6629 : SUCCESS in 18m 32s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac8f4d01ceea4036a0d8ff8fa8ac01d5 : SUCCESS in 19m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/05a7aa022496404ca6ace50eaa39dcb8 : SUCCESS in 21m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"92673c606932e11e4738293f059e603280808678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 17:06:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/49db4316bbbe469394e9873c1a9ff362\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e102eefa39f04ce68fe40c28bc2458d6 : SUCCESS in 2h 09m 17s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e1205b01f3984a769a39f24670263642 : SUCCESS in 1h 51m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4a712c2d681349a98a1ee0257301cf4b : SUCCESS in 1h 45m 05s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d2e0a44874b74a738dff0f715174a56a : FAILURE in 1h 48m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d82f4ae848741259e0c2b81bea4b633 : SUCCESS in 38m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48a0ab0aa50349ff87ec191acd66345e : SUCCESS in 11m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bd3dab34094484c8e310cb179b5374f : SUCCESS in 27m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac2c3909db464f549fc6b3a0984288f6 : SUCCESS in 25m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2640569de73346ada31de137db998237 : SUCCESS in 27m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f5f00ee618d45748d34da6cff2b7b77 : SUCCESS in 10m 53s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/ef9f5fcea8c84d5e878bb0f059f97e40 : SUCCESS in 31m 03s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d3cb2961482417d802eb0367c38518f : SUCCESS in 1h 03m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/edbce0d179d248ae8359c5375ad95782 : SUCCESS in 1h 48m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9568eddc98ab4c2fa656b2948ce49db8 : SUCCESS in 56m 21s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/981024d994654e7489e36d929ac95f84 : FAILURE in 9m 30s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e45d41dfe23a455b99b247f63cf0f445 : SUCCESS in 1h 01m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3feb27d62e044e39a014a06b5669ef46 : SUCCESS in 1h 47m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/99d2834b07a14c71adf498da73dd83cd : SUCCESS in 1h 31m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bd52364d309b46549afc1198dd2e0295 : SUCCESS in 1h 23m 21s","accounts_in_message":[],"_revision_number":2},{"id":"569b4051bdfeeae4f682f81d16e4b86b7a82a917","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-03-24 19:41:05.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"849bffc6c3d2c110f84c6521ceafe20d7c180ac2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 20:14:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cc3da44f005490a870f26d8c5b3340d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae75e7acea984d1982600b48e36a678b : SUCCESS in 17m 31s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dc0ac97cf45344b4a212beed49626b72 : SUCCESS in 32m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c67eb1215b8e42bdacd47b84b75e2da2 : SUCCESS in 17m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"573b7cd64d28d93cd0dbd70acb851ace67fd5b12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 21:33:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be5ea42b17cf471fb6ec03b304fa5bc0\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/6adc565b81a245bc818c88a5a89d860a : SUCCESS in 1h 16m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5ec22fea2508488ba1c04f3d82a5367b : SUCCESS in 1h 19m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/df623687d1f742a2bb2f1fa03cc7b528 : SUCCESS in 1h 09m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/87c34986760e4451bc2f0740c60e09d1 : SUCCESS in 1h 17m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a23ef6dc5854ad485cb678c073e92d3 : SUCCESS in 37m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/348e2d88a59a481895946e8471401434 : SUCCESS in 12m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/093277a790cf4e47a0ba7250b6ec7a6d : SUCCESS in 25m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/578228f3076941e8b997ec463a756d56 : SUCCESS in 27m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0279127a21c3494e9a88d6da0ab07cd5 : SUCCESS in 25m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40390f9b740a41f291bfa659c451643c : SUCCESS in 10m 09s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/18ec62abd6334311b422cdc2803b6a13 : SUCCESS in 30m 27s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/14a512dcd7994e16a33d0bffee51b435 : SUCCESS in 1h 00m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/abf8bcb28d9e44d39caee29030d01336 : SUCCESS in 1h 46m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/71099c35fb6f4230816ef6cab969c983 : SUCCESS in 1h 03m 23s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8112471952104ca6943c117671a60232 : FAILURE in 20m 41s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2b5c7c47037b4dcd9ff09c4ee285d85c : SUCCESS in 57m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ee6d08d93cd54339998d241e1db25d9f : SUCCESS in 1h 29m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6dfc3433b61e4902ace48818fd514eae : SUCCESS in 1h 26m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/490f55f2db824a94993a902f3d8652ee : SUCCESS in 1h 42m 55s","accounts_in_message":[],"_revision_number":2},{"id":"f58786158ea85f5d478b24dea445e242c3c41558","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-25 01:18:12.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/48/878248/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"831041805f17c0cdefd3bec0aafd0ba08d4ac868","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-10 03:53:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"02e2156a982f77174cac72fdc1f1193118edb2fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36313,"name":"Adil Ishaq","display_name":"Adil Ishaq","email":"adilishaq.cc@hotmail.com","username":"Adilishaq","status":"All about AI cyber research for Climate Change with Peaceful Reachable Approach."},"date":"2023-08-22 10:49:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"fbd953ed5b67f1f8968cd2a72b848ac820e679c4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-22 11:18:06.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6b71d81d29623728d5cac02f75c059f3b6659445","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 11:40:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f0d71e405654d4cbe542c58e8576334\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/966d36a1b6e6479c9bb2f4915b496130 : SUCCESS in 23m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be9ed5bdf6b046ad9202b2a06b6141a9 : SUCCESS in 21m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0a779366ef6d4c89ac3faf94a394f0eb : SUCCESS in 48m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a86e62466a44c52aaea92922fcff58f : SUCCESS in 21m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d6395edbe4309ff70accdb68a6d87b9c8f09ff7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 13:07:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5207ccdd7abf49b8a1593fbcf25fd2ee\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9a39b36e74864471a1369f4d37e7efa2 : SUCCESS in 1h 46m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/857fe58cd35948c7b82514a0afe99716 : SUCCESS in 1h 29m 15s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/be3373e6a4b041d7ad7fa44202b660a3 : SUCCESS in 1h 39m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebe7290d511e4ae0941e87e8b81229cd : SUCCESS in 43m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a171ffe97b24364b1d1eea032a96622 : SUCCESS in 10m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7c091cc29e443ae837b8398bfa119e2 : SUCCESS in 38m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e673141241ce45f681fbddce75a6177b : SUCCESS in 28m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1960bdffaff14cdd8c36e5c170fc17b3 : SUCCESS in 27m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9137ad7609d48e0a97e86a476f36d38 : SUCCESS in 27m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a944fcd9251444deba6c6de41f080429 : SUCCESS in 10m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/573d0e2e3bc349b1a73e73c3965731ca : FAILURE in 58m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6bd325e287064667b3e6167e538bedc1 : SUCCESS in 2h 11m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/34747f4d3fe048f28829fa02a5dadbe5 : SUCCESS in 44m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bd0ee63cc5da400ea06a9a92fba89aa1 : SUCCESS in 1h 02m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e3ff6d1428194dbdaba680b148265457 : SUCCESS in 1h 33m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9ff1426019f04f5bb2d201bd824621f5 : SUCCESS in 1h 27m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/92dfffc3dbf245af8044b9302e7c05f6 : SUCCESS in 1h 31m 25s","accounts_in_message":[],"_revision_number":3},{"id":"8f7bdaa2209e5161fee18de0e6000a1604f0d131","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-30 16:19:15.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/48/878248/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"276136077e4974012b58be7a281406333c69ff20","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-08-31 08:29:12.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"07e182ca819010fd2b95028d913ec56423fcce2e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-04 07:56:57.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7cbc1dab0c2dfbd96d3341d81948c1b44976d711","revisions":{"8bb000b7f9fde0ecc0f962b8bf94b15f0703a2e1":{"kind":"REWORK","_number":1,"created":"2023-03-22 19:26:34.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/48/878248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/878248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/878248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/878248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/878248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/878248/1"}}},"commit":{"parents":[{"commit":"11ace380febcac6f92d1586271a20d385227509b","subject":"Merge \"Open the 2023.2 (Bobcat) DB branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11ace380febcac6f92d1586271a20d385227509b"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2023-03-22 19:23:30.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2023-03-22 19:26:32.000000000","tz":-240},"subject":"fix: add log message for periodic_sync_routers_task fullsync","message":"fix: add log message for periodic_sync_routers_task fullsync\n\nThere is no way to know if a Neutron agent has fully completed\na periodic_sync_routers_task process and therefore it is ready\nand all routers are up.\n\nBy adding some INFO level messages, the operator can know the\ncurrent state of the process.\n\nChange-Id: Iee72be61d00c11a938a0b1b31e75bdda788d5c57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bb000b7f9fde0ecc0f962b8bf94b15f0703a2e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bb000b7f9fde0ecc0f962b8bf94b15f0703a2e1"}]},"branch":"refs/heads/master"},"77bdb59fc8404f6c62a62562888c06a66e750591":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-24 14:53:30.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/48/878248/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/878248/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/878248/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/878248/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/878248/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/878248/2"}}},"commit":{"parents":[{"commit":"11ace380febcac6f92d1586271a20d385227509b","subject":"Merge \"Open the 2023.2 (Bobcat) DB branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11ace380febcac6f92d1586271a20d385227509b"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2023-03-22 19:23:30.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2023-03-24 14:53:30.000000000","tz":0},"subject":"fix: add log message for periodic_sync_routers_task fullsync","message":"fix: add log message for periodic_sync_routers_task fullsync\n\nThere is no way to know if a Neutron L3 agent has fully completed\na periodic_sync_routers_task process and therefore it is ready\nand all routers are up.\n\nBy adding some INFO level messages, the operator can know the\ncurrent state of the process.\n\nRelated-bug: #2011422\nChange-Id: Iee72be61d00c11a938a0b1b31e75bdda788d5c57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77bdb59fc8404f6c62a62562888c06a66e750591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77bdb59fc8404f6c62a62562888c06a66e750591"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7cbc1dab0c2dfbd96d3341d81948c1b44976d711":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-08-22 10:49:47.000000000","uploader":{"_account_id":36313,"name":"Adil Ishaq","display_name":"Adil Ishaq","email":"adilishaq.cc@hotmail.com","username":"Adilishaq","status":"All about AI cyber research for Climate Change with Peaceful Reachable Approach."},"ref":"refs/changes/48/878248/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/48/878248/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/878248/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/878248/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/48/878248/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/48/878248/3"}}},"commit":{"parents":[{"commit":"977461f92b7b00a7ed518a7492622eac9b12323f","subject":"Merge \"Add max limit to agent_down_time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/977461f92b7b00a7ed518a7492622eac9b12323f"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2023-03-22 19:23:30.000000000","tz":-240},"committer":{"name":"Adil Ishaq","email":"adilishaq.cc@hotmail.com","date":"2023-08-22 10:49:47.000000000","tz":0},"subject":"fix: add log message for periodic_sync_routers_task fullsync","message":"fix: add log message for periodic_sync_routers_task fullsync\n\nThere is no way to know if a Neutron L3 agent has fully completed\na periodic_sync_routers_task process and therefore it is ready\nand all routers are up.\n\nBy adding some INFO level messages, the operator can know the\ncurrent state of the process.\n\nRelated-bug: #2011422\nChange-Id: Iee72be61d00c11a938a0b1b31e75bdda788d5c57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7cbc1dab0c2dfbd96d3341d81948c1b44976d711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7cbc1dab0c2dfbd96d3341d81948c1b44976d711"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
