)]}'
{"id":"openstack%2Fneutron~878445","triplet_id":"openstack%2Fneutron~stable%2Fxena~I60f495747f87d27d9c5668d9c91f64ecc159bd69","project":"openstack/neutron","branch":"stable/xena","topic":"bug/2003455","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-03-28 18:45:05.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-03-27 08:44:32.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-03-28 18:45:05.000000000","reason":"Change was submitted"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2023-03-24 15:58:30.000000000","reason":"\u003cGERRIT_ACCOUNT_23567\u003e replied on the change","reason_account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}}},"hashtags":[],"change_id":"I60f495747f87d27d9c5668d9c91f64ecc159bd69","subject":"Revert \"Ensure vlan network traffic is not centralized\"","status":"MERGED","created":"2023-03-24 15:51:28.000000000","updated":"2023-03-28 18:46:30.000000000","submitted":"2023-03-28 18:45:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"revert_of":877255,"submission_id":"878445-bug/2003455","meta_rev_id":"a1b5dcdd894bbaa82668e3812f4ced76d786d994","_number":878445,"virtual_id_number":878445,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-28 18:45:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-03-27 08:44:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-03-28 16:28:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-24 15:58:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-03-28 16:28:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-24 15:51:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-03-24 15:51:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-03-24 15:51:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2023-03-24 15:51:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-28 16:28:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"e1a28606d72038b2c783c0f8fc156c797b78b932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-24 15:51:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc6463336f1052845b9d183d7b89abd1f81211b7","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-03-24 15:58:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2aaa0e33ffb14cfafca5234300ee7a122ce04894","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 16:17:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63def1218eb546c58c1616adbeaf4cbc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24f086a2261c405f9620d8e65ec37c09 : SUCCESS in 23m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba4cf7e60bf24c0bbf0f8029c907549d : SUCCESS in 24m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0dedb26f3cf64b7eaed199b55027df300e862a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 17:43:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f334fa06e3ea4bbaa4c0d36b2b64c83a\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/1bb10f6866f64f9e8972647b64bc1357 : SUCCESS in 43m 23s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/60b8a0bcb8fa46eea2b038efe9be68fd : SUCCESS in 1h 06m 24s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/b2e4b9fc8d5f4e28a758d5c02534e32c : SUCCESS in 32m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/916c2caaeb754836aa7a638e4170db9d : SUCCESS in 39m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56dd8a7721064ebc96f155852aca0d63 : SUCCESS in 13m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5280349cbe245158c06a76a4d80aebb : SUCCESS in 28m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1be76df685434c64a4ebfd298fc09234 : SUCCESS in 24m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d14b8d6cb34444a5b5ddb9034841c388 : SUCCESS in 26m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/300438db9c2c4ffca1de393a9d527fb5 : SUCCESS in 9m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6868f9bf74b549d59cbabab428b65142 : SUCCESS in 12m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/258553856944457d93f66cbb3a212d43 : SUCCESS in 51m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/957c71a245464a20be76a54a94bc9a7a : SUCCESS in 1h 43m 31s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af5a7262375342b78ce6013accfd8165 : SUCCESS in 54m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/27d8de1f0050435d83dbb36685a34013 : SUCCESS in 1h 18m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"acd35f740b931737fdc0bb9683cbe7dd5b112cbc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-27 08:26:30.000000000","message":"Topic set to bug/2003455","accounts_in_message":[],"_revision_number":1},{"id":"2c4b1bfed90337dbb2fce1f726dbfbecdeb22107","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-03-27 08:44:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0275c224b711c4e48f1093393ce7f109711d1bd2","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-03-27 08:44:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"20a5810684693045a62b8d73f8674b29f739fc86","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-28 16:28:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"368ce040074ce1e7d03229997fad2a12383a4963","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 16:29:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57ee74436788cef40fe74c811c9e0616a79a30a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 18:45:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cade1f37db744d6fb9dae393aaac666e\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/e95d94e0940b48909a1bae265557482e : SUCCESS in 43m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d6f421279dd45bca5ba623a4571be90 : SUCCESS in 14m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cad9564637234965a3fba8f35fe7ef31 : SUCCESS in 35m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86518ba8c3474d7c820a84d38169f766 : SUCCESS in 24m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0a17a69f1de49bea0c2b49083fcc81d : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bc9985746af4346ad3a540bfa355fbc : SUCCESS in 12m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/da42259c10a048688389d782abee84d7 : SUCCESS in 39m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/acc2f7e56aa34c62ad31f7bfdfcd99b9 : SUCCESS in 1h 27m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3aa15532dbfb4ce98081555a3e83ffd2 : SUCCESS in 52m 59s","accounts_in_message":[],"_revision_number":1},{"id":"92d6c49bd59c98424860bdbcdeb39a96d1de4418","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 18:45:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a1b5dcdd894bbaa82668e3812f4ced76d786d994","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 18:46:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/725a67ca5023414b8d8d10a872631560\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d659e77dc3ff4d2bb3377da0e7f4cffb : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97d6c45dff6f4590a2d706e450d6bfde : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed44efb443f4ee52f7a259df9397ef8b9c1c745a","revisions":{"ed44efb443f4ee52f7a259df9397ef8b9c1c745a":{"kind":"REWORK","_number":1,"created":"2023-03-24 15:51:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/45/878445/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/878445/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/878445/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/878445/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/878445/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/878445/1"}}},"commit":{"parents":[{"commit":"fe0ffa1a2ea147835a1c55cb4c18cbb1c497c7e6","subject":"Ensure vlan network traffic is not centralized","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe0ffa1a2ea147835a1c55cb4c18cbb1c497c7e6"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-24 15:51:28.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-03-24 15:51:28.000000000","tz":0},"subject":"Revert \"Ensure vlan network traffic is not centralized\"","message":"Revert \"Ensure vlan network traffic is not centralized\"\n\nThis reverts commit fe0ffa1a2ea147835a1c55cb4c18cbb1c497c7e6.\n\nReason for revert:\nAs part of the reverted commit, the redirect-type\u003dbridged flag was\nenabled by default. However this have the side effect of also\ndecentralizing N/S traffic for geneve tenant networks, breaking the\nVM connectivity on them when it must be centralized, i.e., when no\nFIPs are associated to the VMs.\n\nA new fix will be provided ASAP.\n\nChange-Id: I60f495747f87d27d9c5668d9c91f64ecc159bd69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed44efb443f4ee52f7a259df9397ef8b9c1c745a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed44efb443f4ee52f7a259df9397ef8b9c1c745a"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
