)]}'
{"id":"openstack%2Fneutron~878549","triplet_id":"openstack%2Fneutron~master~I07fa45401788da6b963830e72a7b3a3cd54662e1","project":"openstack/neutron","branch":"master","topic":"wait_for_sb_pb","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-25 15:08:04.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-04-26 07:35:17.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-04-26 08:34:12.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-04-12 19:18:33.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-04-26 08:34:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I07fa45401788da6b963830e72a7b3a3cd54662e1","subject":"Fix intermittent failures in finding metada port in SB DB","status":"MERGED","created":"2023-03-24 17:03:23.000000000","updated":"2023-04-26 08:35:53.000000000","submitted":"2023-04-26 08:34:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"878549-wait_for_sb_pb","meta_rev_id":"316a2eab953ebf8018db4e7d986717c661f0bcf5","_number":878549,"virtual_id_number":878549,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-04-14 05:08:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-26 08:34:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-04-25 15:08:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-26 07:35:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-26 07:35:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-24 17:04:31.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-03-24 17:04:31.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-03-24 17:11:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-03-24 17:24:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-24 17:46:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-25 07:05:14.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-27 12:33:48.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"}],"messages":[{"id":"180448a6ac20eaed421e4fdb2d3c7b2accf22954","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-03-24 17:03:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5109b78a41c1af1003bea589de2a768f7c9023c6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-24 17:11:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2af00e7cf056e7e91c1db35d6e0887d9379ef7ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 17:24:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03f67ce9badc409f86f5b1339dcd16fc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05710ca1e85242bda1f980d385912039 : SUCCESS in 19m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c603bffb029deb5aadde11d1d1d34d20d24336be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 17:46:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/313f0bcd68874cbfaec014e032af010e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f88ee50a0f640c8b24f0c3de1b1b734 : SUCCESS in 14m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/350494ea7334476987a1f2c9b2679abd : SUCCESS in 29m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102445e42fad4662bf0e6bba9d0f0bc1 : SUCCESS in 10m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e8121ff80a584ef395863c78053ff9f6 : FAILURE in 41m 31s","accounts_in_message":[],"_revision_number":1},{"id":"768d9778f3658212352c720da2b57e92cf00b3c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-03-25 03:35:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a529b4212c10accef3e16be4a00638346e4cc2f5","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-03-25 03:40:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"266497498fbd74eb29248d90d44a64551f0586e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-25 04:07:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0d76d99b8764e99bd21b8a38ff10504\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26d8e68836074c55beb56744c0ebc004 : SUCCESS in 30m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6900c6a276dec858ad04eff5cebfc2b3c5e4ac38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-25 04:26:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8b11472f53c4658ad306d0295da7bb8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16d9db373e4c4f13914d078d58d92ad2 : SUCCESS in 12m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6b6d530c0354db08c355155e1be2e43 : SUCCESS in 23m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1e984607b154718a3725d7c85a32a21 : SUCCESS in 10m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/972cc60a78864ba78eb0b4ef81ff1c02 : SUCCESS in 48m 36s","accounts_in_message":[],"_revision_number":2},{"id":"d2db35a3ac23965337ed95317e87f7340d744cf4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-25 07:05:14.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/49/878549/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"f263fc64e15e076e21ecd126ba9296506f6a4d8b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-25 08:58:13.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/49/878549/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"ae02999a843f1c66423ce3a08ca499dea8400024","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-03-27 12:33:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e6903ed334ebab1d57dcab750bf8b8ced3ec36e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-04-05 02:51:35.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"27f5dc04160302df18b43e758f0700d27f02fbb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 03:22:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75874155b6c248a0b0665a29c3184aba\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a2b9c3a212f4692960b6bfcd6fbb15c : SUCCESS in 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2cd745414379978023066844d6b1cc70b6e24266","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 03:49:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1253f137e6342c8bf157a141a20692f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/774fc34b496d45b8be7b6dae660e9e2e : SUCCESS in 12m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ab4c4d7969b422fb10cf7b2e044db74 : SUCCESS in 28m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d38bcd7c20a34f9caeb52db1c00a0105 : SUCCESS in 10m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ccc8c61e453346d18cf2fa60ae67f038 : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":3},{"id":"ed662c56a52be5a11ee5ad3cf189393c89fbc6e3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-05 06:19:05.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/49/878549/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"70276e7d48b5389be0a9f1872a5ba21b622df346","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-12 07:06:06.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5576a3a2a6cc22328f1c120d78873196538969c6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-04-12 13:43:09.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"770d9c72dfa27e62348d2f96f961855d13b10175","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-04-12 15:02:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"3b07184d871ec08b6e450beaa751da31679b2b4d","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-04-12 15:05:42.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"70bdd2c56593cf8077dd0807efa12f74db14d7c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 15:25:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cf73c29e3a7475ba467556671ae4b33\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/49d47e1ddb4c448cb59e2fcf4fb2752f : SUCCESS in 22m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d511e43ba46f7309bcc3a03ed738bf91477f1b93","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-12 16:09:35.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c5b8a57d0c410ae3ba8201c7daf7467fa2534b3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 16:11:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d83faedcbad34870bfda1a462d60158f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d26a8024f053479a9e62fb3e7ca17576 : SUCCESS in 13m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac084e30b851413a9ea04856ff1a48de : SUCCESS in 26m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/626c3da9129c46edb6b1739b12b2206b : SUCCESS in 10m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/efbfebbd31cb40988bea9f2b1cdd26e7 : SUCCESS in 1h 02m 58s","accounts_in_message":[],"_revision_number":4},{"id":"f07db4f69286bef27b015621a342570f41296309","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-12 16:31:14.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/49/878549/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"226bd92909d5b03fc62196b83e861444514a6db3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-04-12 19:15:11.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"335532a32e1360eb965062236cb0de751679136a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-04-12 19:18:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ec14b3f8a669a1b8aaeeacbbc3ce3c01037bcb40","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-04-12 19:51:11.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"629db925db46a2406d1ae5062842c80dfe89f34f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 19:56:49.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/473a579b883540e89756d7b191a978c2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fca9ea2e15d2486f8df77e70e06ff863 : SUCCESS in 39m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1ea83701e8b7410f690c70f8ef4c7cfd37c123c6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-12 20:28:19.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/49/878549/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"5297360dd74128d97056d29fc47c9086902f116d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 21:01:38.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9fb5166f65443d396c22b770e7e8709\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b39ec0b517684157a534b0eeb8654a45 : FAILURE in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ffefa31c59d42db8751702a84f29a64 : SUCCESS in 28m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e900176330414dba88914c98153f4335 : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/09e1cf8be9e14c84aeb258e496bf554d : FAILURE in 1h 35m 37s","accounts_in_message":[],"_revision_number":5},{"id":"5fd6e5b1eedc817ed276bfa8f404af749e3519b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-04-14 04:33:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c62a57f125b4c471d4caada6946aafc7ffbbcc80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 04:56:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faf1a0e6cd23426ea59173d2c1f3174f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/184391353d1f490186b47cbb56e2e39f : SUCCESS in 21m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e81e709427562a243269a88668bc48e5206420ca","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-14 05:08:55.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/49/878549/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"4e910449d54aa385be2b1b5936a6e4f623c5845e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 05:34:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/158b4c8c4d2644739220de6e08fe31be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe6a440ef7854741a02c09042c79928c : SUCCESS in 12m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/663c35e1ac48443e9a7c68530899d41a : SUCCESS in 22m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80e17413adec47478eeccf81bdfc015a : SUCCESS in 11m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c303f51ab384d7aaf47e0f61b267a6c : SUCCESS in 55m 17s","accounts_in_message":[],"_revision_number":6},{"id":"125203117508433e1254199ea8ceed1a0796281c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-25 15:08:04.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"43e9379a0921be08a60b49ce2e720beb8e6f50b8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-04-25 17:24:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2d9ee6e7a8dbddd475382deb6b8bed97c71c7a0e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-26 07:35:17.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"f94127c1ba8fb43c00b69ad65566c72b82171346","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 07:35:40.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7e1b13179c4fb39e9db58da48bfcb388ed5e656d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 08:34:12.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef9e99454dc44c0b93b944cae43dd7c2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98f20c01a78b4d2d884d425e49011fcb : SUCCESS in 13m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f96487dcf3844f8eab7eed6d59a8d3e0 : SUCCESS in 10m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/997d4b9740f4428baa263c180f68e992 : SUCCESS in 57m 43s","accounts_in_message":[],"_revision_number":6},{"id":"585b6a75c8c401e1d9b622583b6df0cb2c420c9f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 08:34:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"316a2eab953ebf8018db4e7d986717c661f0bcf5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 08:35:53.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5e35bbd3bf641f68f5c2eb469b0c2b5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3540205d6fa4e838dcb9e7a4ebe75ce : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9d5163301310a8c3cb72e18cbaed89afbc07f674","revisions":{"8a79aebc6519ab2f15d2be9da1d680630eab7371":{"kind":"REWORK","_number":1,"created":"2023-03-24 17:03:23.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/49/878549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/878549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/878549/1"}}},"commit":{"parents":[{"commit":"1646e5b28d21392874384ccf0358e20d7f0710c1","subject":"Add neutron-lib project to the SQLAlchemy master branch CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1646e5b28d21392874384ccf0358e20d7f0710c1"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-24 16:51:06.000000000","tz":-300},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-24 16:51:06.000000000","tz":-300},"subject":"Fix intermittent failures in finding metada port in SB DB","message":"Fix intermittent failures in finding metada port in SB DB\n\nThe test_agent_metadata_port_ip_update_event can sometimes fail\nwith finding the port in SB DB Port_Binding table [1]. This is\ndue to a race condition between when the port is commited to NB DB\n,reflected in SB DB and calling lsp_bind on SB DB.\nThis patch simply adds check for the metadata port in SB DB with\n5s timeout before attempting to bind that logical port.\n\nCloses-Bug: #2012754\nChange-Id: I07fa45401788da6b963830e72a7b3a3cd54662e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a79aebc6519ab2f15d2be9da1d680630eab7371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a79aebc6519ab2f15d2be9da1d680630eab7371"}]},"branch":"refs/heads/master"},"38fde84b1942e9132f414cc65051affde41960d3":{"kind":"REWORK","_number":2,"created":"2023-03-25 03:35:38.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/49/878549/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/878549/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/878549/2"}}},"commit":{"parents":[{"commit":"1646e5b28d21392874384ccf0358e20d7f0710c1","subject":"Add neutron-lib project to the SQLAlchemy master branch CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1646e5b28d21392874384ccf0358e20d7f0710c1"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-24 16:51:06.000000000","tz":-300},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-25 03:31:40.000000000","tz":-300},"subject":"Fix intermittent failures in finding metada port in SB DB","message":"Fix intermittent failures in finding metada port in SB DB\n\nThe test_agent_metadata_port_ip_update_event can sometimes fail\nwith finding the port in SB DB Port_Binding table [1]. This is\ndue to a race condition between when the port is commited to NB DB\n,reflected in SB DB and calling lsp_bind on SB DB.\nThis patch simply adds check for the metadata port in SB DB with\n5s timeout before attempting to bind that logical port.\n\nCloses-Bug: #2012754\nChange-Id: I07fa45401788da6b963830e72a7b3a3cd54662e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38fde84b1942e9132f414cc65051affde41960d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38fde84b1942e9132f414cc65051affde41960d3"}]},"branch":"refs/heads/master"},"8e2db5f406321b898389d45a1362201e16c88036":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-04-05 02:51:35.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/49/878549/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/878549/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/878549/3"}}},"commit":{"parents":[{"commit":"1646e5b28d21392874384ccf0358e20d7f0710c1","subject":"Add neutron-lib project to the SQLAlchemy master branch CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1646e5b28d21392874384ccf0358e20d7f0710c1"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-24 16:51:06.000000000","tz":-300},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-04-05 02:51:35.000000000","tz":0},"subject":"Fix intermittent failures in finding metada port in SB DB","message":"Fix intermittent failures in finding metada port in SB DB\n\nThe test_agent_metadata_port_ip_update_event can sometimes fail\nwith finding the port in SB DB Port_Binding table [1]. This is\ndue to a race condition between when the port is commited to NB DB\n,reflected in SB DB and calling lsp_bind on SB DB.\nThis patch simply adds check for the metadata port in SB DB with\n5s timeout before attempting to bind that logical port.\n\n[1] shorturl.at/zPVX1\n\nCloses-Bug: #2012754\nChange-Id: I07fa45401788da6b963830e72a7b3a3cd54662e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e2db5f406321b898389d45a1362201e16c88036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8e2db5f406321b898389d45a1362201e16c88036"}]},"branch":"refs/heads/master","description":"Edit commit message"},"78da96cbce9b2073b5a487195ebba09b454ad60b":{"kind":"REWORK","_number":4,"created":"2023-04-12 15:02:19.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/49/878549/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/878549/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/878549/4"}}},"commit":{"parents":[{"commit":"1646e5b28d21392874384ccf0358e20d7f0710c1","subject":"Add neutron-lib project to the SQLAlchemy master branch CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1646e5b28d21392874384ccf0358e20d7f0710c1"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-24 16:51:06.000000000","tz":-300},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-04-12 15:01:42.000000000","tz":-300},"subject":"Fix intermittent failures in finding metada port in SB DB","message":"Fix intermittent failures in finding metada port in SB DB\n\nThe test_agent_metadata_port_ip_update_event can sometimes fail\nwith finding the port in SB DB Port_Binding table. This is\ndue to a race condition between when the port is commited to NB DB\n,reflected in SB DB and calling lsp_bind on SB DB.\nThis patch simply adds check for the metadata port in SB DB with\n5s timeout before attempting to bind that logical port.\n\nCloses-Bug: #2012754\nChange-Id: I07fa45401788da6b963830e72a7b3a3cd54662e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78da96cbce9b2073b5a487195ebba09b454ad60b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78da96cbce9b2073b5a487195ebba09b454ad60b"}]},"branch":"refs/heads/master"},"c58efe1753dc8910f374a4cdeb224f563fb9be33":{"kind":"REWORK","_number":5,"created":"2023-04-12 19:15:11.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/49/878549/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/878549/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/878549/5"}}},"commit":{"parents":[{"commit":"1646e5b28d21392874384ccf0358e20d7f0710c1","subject":"Add neutron-lib project to the SQLAlchemy master branch CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1646e5b28d21392874384ccf0358e20d7f0710c1"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-24 16:51:06.000000000","tz":-300},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-04-12 19:13:54.000000000","tz":-300},"subject":"Fix intermittent failures in finding metada port in SB DB","message":"Fix intermittent failures in finding metada port in SB DB\n\nThe test_agent_metadata_port_ip_update_event can sometimes fail\nwith finding the port in SB DB Port_Binding table. This is\ndue to a race condition between when the port is commited to NB DB\n,reflected in SB DB and calling lsp_bind on SB DB.\nThis patch simply adds metadata port create event check before\ncalling lsp_bind on the metadata port.\n\nCloses-Bug: #2012754\nChange-Id: I07fa45401788da6b963830e72a7b3a3cd54662e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c58efe1753dc8910f374a4cdeb224f563fb9be33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c58efe1753dc8910f374a4cdeb224f563fb9be33"}]},"branch":"refs/heads/master"},"9d5163301310a8c3cb72e18cbaed89afbc07f674":{"kind":"REWORK","_number":6,"created":"2023-04-14 04:33:27.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/49/878549/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/878549/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/878549/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/878549/6"}}},"commit":{"parents":[{"commit":"1646e5b28d21392874384ccf0358e20d7f0710c1","subject":"Add neutron-lib project to the SQLAlchemy master branch CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1646e5b28d21392874384ccf0358e20d7f0710c1"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-24 16:51:06.000000000","tz":-300},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-04-14 04:33:14.000000000","tz":-300},"subject":"Fix intermittent failures in finding metada port in SB DB","message":"Fix intermittent failures in finding metada port in SB DB\n\nThe test_agent_metadata_port_ip_update_event can sometimes fail\nwith finding the port in SB DB Port_Binding table. This is\ndue to a race condition between when the port is commited to NB DB\n,reflected in SB DB and calling lsp_bind on SB DB.\nThis patch simply adds metadata port create event check before\ncalling lsp_bind on the metadata port.\n\nCloses-Bug: #2012754\nChange-Id: I07fa45401788da6b963830e72a7b3a3cd54662e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d5163301310a8c3cb72e18cbaed89afbc07f674"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d5163301310a8c3cb72e18cbaed89afbc07f674"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
