)]}'
{"id":"openstack%2Fneutron~879029","triplet_id":"openstack%2Fneutron~master~I13d4a76572af050ee07fbb2b9fe2abd60d337955","project":"openstack/neutron","branch":"master","topic":"bug/2012510","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-03 20:56:25.000000000","reason":"Change was submitted"},"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2023-03-30 09:50:46.000000000","reason":"\u003cGERRIT_ACCOUNT_7730\u003e replied on the change","reason_account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-04-03 20:56:25.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-04-03 20:56:25.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-04-03 14:27:12.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2023-04-03 20:56:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I13d4a76572af050ee07fbb2b9fe2abd60d337955","subject":"Add debug information to ``MacvtapAgentTestCase.test_get_all_devices``","status":"MERGED","created":"2023-03-30 08:14:22.000000000","updated":"2023-04-03 20:57:32.000000000","submitted":"2023-04-03 20:56:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879029-bug/2012510","meta_rev_id":"f9b9cdad95a06f1b2ecdb13a366cb5dfa74b80fb","_number":879029,"virtual_id_number":879029,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":1,"date":"2023-04-03 14:09:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-03 20:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"recommended":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"all":[{"value":1,"date":"2023-04-03 14:34:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-03 19:53:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-04-03 14:27:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-03 19:53:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-30 08:45:54.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2023-03-30 08:54:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-30 09:37:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-30 11:22:09.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-30 13:38:26.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-04-03 16:14:22.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2023-04-03 16:14:22.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-04-03 16:14:22.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-04-03 19:53:27.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"9422efe328f961373f065e118e6259ba75db8fef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-30 08:14:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da4f817920e5e765aef410c0dfbd4272096986c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 08:54:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bd20b38001d429ba46ac47dc4a3fc77\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d4186607759f410a927e264911257175 : SUCCESS in 37m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c18526fd34d5b7b2c66bbe2e1516f76b6c5fc66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 09:37:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3b740d4f88e41cbb7491e223247950b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d903cf71b8d446d3b0619fc0cb732936 : SUCCESS in 13m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfce219e837842b184765ee04f338f00 : SUCCESS in 23m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57df14ff7b3c470cb1f81dc0d52d5f41 : SUCCESS in 11m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e792987349034d51aca1d8dcd0af990e : SUCCESS in 1h 20m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c9eb3f56475cd4940158487a6b0ce6718fa8b170","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-03-30 09:50:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5966e2e34f3b75941e2061c2278174c46f839550","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-30 11:04:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"882ab5ce18ccca9609976f242f24be89190f5dc7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-30 11:22:09.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/29/879029/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"dd781f7c943ce0ae67694b89114673e57274946a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 11:50:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aea63f137e04762afd5f062cb9786f6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d2539a21e9243b2a8b8454625078e20 : SUCCESS in 43m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"68d0c8834072c7010f5b0d82ba1bc2a824b79218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 12:08:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1953128af8c54286a3d4a6740dbf259a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68f12f99918e42e0bd329f5194477706 : SUCCESS in 13m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c202420347b4e0abb7c2bd7598d2e2a : SUCCESS in 25m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/923781025be14d7b9b0ea2e94263d4b1 : SUCCESS in 9m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/75ca278edf6c49bf86e207b4ca4abed6 : SUCCESS in 58m 00s","accounts_in_message":[],"_revision_number":2},{"id":"46dc6e93f288ae9c1b265586b25a2b08807f1e65","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-03-30 13:08:56.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2023/Arista_Tempest_Testrefs/changes/29/879029/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"c4bc44a1c8e73d84189163203af511063c881d77","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-03-30 13:38:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"72ac45f66820312c7503ebfb4e6445999bd07592","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-03-31 07:25:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91c90f2fa9edd2d086c410d6ceeb23ae3b153245","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-03 10:53:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"72036e11bdcbfcaf0d5eb0d3006a45d5f8e7f05e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-03 11:33:05.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/29/879029/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"0a967b88065818caa114bcb42e05a34f11c89589","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 11:58:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b197a26726c4df9ab44ab6d71393386\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0b5a9f7954044ccbd35765b794c30af : TIMED_OUT in 1h 01m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ca764bbe466e5fea406dfe4c9092e0f1e44eefa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 12:02:03.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8909a456571d44eabf0a97eb0f971800\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58ae6df9efb849fba483994c0b0756e4 : FAILURE in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f69f5ca83fec48fc9086ed96deae1ee9 : SUCCESS in 30m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aabf47f683e42c6823bde6985ca848b : SUCCESS in 9m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/12af5e2fef2a4d8da86e6c8b4f2ea80d : FAILURE in 1h 02m 41s","accounts_in_message":[],"_revision_number":3},{"id":"e30362081ab7327b04f63b34e7d01fd74a6e81fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-03 13:06:17.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"00d03c704cfc2da3c6f5955d42388726841386e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 13:34:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2cb6c7f77634250af5be79c6901f8d6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b43b0e4516f44df089e117200f34fbf8 : SUCCESS in 26m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e18f66ebfc688ca877c6880d4ad02cc12244cca9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 14:01:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/894a5544daa749fd842f73dacb631100\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/730910885936432aa66ca75922d6c9d1 : SUCCESS in 13m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4de0e99389ae4813a4e7c63422124990 : SUCCESS in 28m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/566aec863f02438891d872972a4b8ae1 : SUCCESS in 13m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/47d9342d503046e0a1235ad6337ba208 : SUCCESS in 53m 53s","accounts_in_message":[],"_revision_number":4},{"id":"4ff149273de09a7d27e46a815aa59d45a082d7ce","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-03 14:09:10.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/29/879029/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"f95c23b85be81e2fa8cee70e4716ad3f907e8138","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-04-03 14:27:12.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1cc61116669632b07a0a048a414a9abc58f6ac4a","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-04-03 14:34:23.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"776a4794c20b93a0eab84cfc6f098ff07062d787","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-03 16:11:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dec91660c6f3731025329967e439049d2b272aac","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-03 16:13:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"03ae830f5df44b8bace5e180120cacf72ca481ab","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-04-03 19:53:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"844e1177a2b5da38555cf69bd8317a253eca977b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 19:54:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6cf943b8c380e2a3dfcefdea165657f3bbd50c17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 20:56:25.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bca067677274fb0aea685330b187cdc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5afa64e824b343d48674e6728ce71926 : SUCCESS in 15m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f168dc2bc2140f39dd915694516a329 : SUCCESS in 27m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caac9f83b5a34a40914e2e4d0ac69f1f : SUCCESS in 9m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b5434c65fc14eb5bd984b4ed314c54c : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":4},{"id":"49f080f0bf3cfe9c537759212c4e224edc6cb58a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 20:56:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f9b9cdad95a06f1b2ecdb13a366cb5dfa74b80fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 20:57:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a05d1ec1f2d34d1eb1cbc64f6f5c8a42\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adf8901348ee43ac94cc04bb23b9aeb5 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"775d5de9f15dfd0be2bbc2996d2404c0f027244d","revisions":{"44b525fb1683e821d00304dcdd25e3232327a6b2":{"kind":"REWORK","_number":1,"created":"2023-03-30 08:14:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/29/879029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/879029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/879029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/879029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/879029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/879029/1"}}},"commit":{"parents":[{"commit":"b73399fa746d951a99fdf29950a1c0a801e941a2","subject":"Merge \"Revert \"Ensure vlan network traffic is not centralized\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b73399fa746d951a99fdf29950a1c0a801e941a2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-29 18:16:35.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-29 18:16:35.000000000","tz":120},"subject":"Add debug information to ``MacvtapAgentTestCase.test_get_all_devices``","message":"Add debug information to ``MacvtapAgentTestCase.test_get_all_devices``\n\nRelated-Bug: #2012510\nChange-Id: I13d4a76572af050ee07fbb2b9fe2abd60d337955\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44b525fb1683e821d00304dcdd25e3232327a6b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44b525fb1683e821d00304dcdd25e3232327a6b2"}]},"branch":"refs/heads/master"},"4cf9e854320563cb7c027edf938d5bc0228e9ccf":{"kind":"REWORK","_number":2,"created":"2023-03-30 11:04:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/29/879029/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/879029/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/879029/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/879029/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/879029/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/879029/2"}}},"commit":{"parents":[{"commit":"b73399fa746d951a99fdf29950a1c0a801e941a2","subject":"Merge \"Revert \"Ensure vlan network traffic is not centralized\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b73399fa746d951a99fdf29950a1c0a801e941a2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-29 18:16:35.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-29 21:07:12.000000000","tz":120},"subject":"Add debug information to ``MacvtapAgentTestCase.test_get_all_devices``","message":"Add debug information to ``MacvtapAgentTestCase.test_get_all_devices``\n\nThis patch is also adding a 1 second sleep between the MacVTap creation\nand the device MAC retrieval.\n\nRelated-Bug: #2012510\nChange-Id: I13d4a76572af050ee07fbb2b9fe2abd60d337955\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cf9e854320563cb7c027edf938d5bc0228e9ccf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4cf9e854320563cb7c027edf938d5bc0228e9ccf"}]},"branch":"refs/heads/master"},"23aede0b72de18efeee9b1f88fb9e8f7aa4537b3":{"kind":"REWORK","_number":3,"created":"2023-04-03 10:53:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/29/879029/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/879029/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/879029/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/879029/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/879029/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/879029/3"}}},"commit":{"parents":[{"commit":"b73399fa746d951a99fdf29950a1c0a801e941a2","subject":"Merge \"Revert \"Ensure vlan network traffic is not centralized\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b73399fa746d951a99fdf29950a1c0a801e941a2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-29 18:16:35.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-31 02:20:52.000000000","tz":120},"subject":"Add debug information to ``MacvtapAgentTestCase.test_get_all_devices``","message":"Add debug information to ``MacvtapAgentTestCase.test_get_all_devices``\n\nThis patch is also adding an active wait checking both the MacVTap\ndevice MAC address and the list of MAC addresses returned by the\nMacVTAP manager.\n\nRelated-Bug: #2012510\nChange-Id: I13d4a76572af050ee07fbb2b9fe2abd60d337955\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23aede0b72de18efeee9b1f88fb9e8f7aa4537b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23aede0b72de18efeee9b1f88fb9e8f7aa4537b3"}]},"branch":"refs/heads/master"},"775d5de9f15dfd0be2bbc2996d2404c0f027244d":{"kind":"REWORK","_number":4,"created":"2023-04-03 13:06:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/29/879029/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/29/879029/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/879029/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/879029/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/29/879029/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/29/879029/4"}}},"commit":{"parents":[{"commit":"b73399fa746d951a99fdf29950a1c0a801e941a2","subject":"Merge \"Revert \"Ensure vlan network traffic is not centralized\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b73399fa746d951a99fdf29950a1c0a801e941a2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-03-29 18:16:35.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-04-03 13:06:16.000000000","tz":0},"subject":"Add debug information to ``MacvtapAgentTestCase.test_get_all_devices``","message":"Add debug information to ``MacvtapAgentTestCase.test_get_all_devices``\n\nThis patch is also adding an active wait checking both the MacVTap\ndevice MAC address and the list of MAC addresses returned by the\nMacVTAP manager.\n\nRelated-Bug: #2012510\nChange-Id: I13d4a76572af050ee07fbb2b9fe2abd60d337955\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/775d5de9f15dfd0be2bbc2996d2404c0f027244d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/775d5de9f15dfd0be2bbc2996d2404c0f027244d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
