)]}'
{"id":"openstack%2Fneutron~879094","triplet_id":"openstack%2Fneutron~stable%2Fxena~I7a4620eebfddce9df371a748ede4f4e10742103b","project":"openstack/neutron","branch":"stable/xena","hashtags":[],"change_id":"I7a4620eebfddce9df371a748ede4f4e10742103b","subject":"DNM: testing grenade fix","status":"ABANDONED","created":"2023-03-31 05:04:25.000000000","updated":"2023-04-13 04:13:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":879116,"cherry_pick_of_patch_set":1,"meta_rev_id":"d56cddf9a40a4169adff7759d30b9dc4c11ebb1a","_number":879094,"virtual_id_number":879094,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-31 06:45:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-31 06:45:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ccff58e9852056d8105c418c53827bb260aa1a7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-31 05:04:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8850ec871681e56abef8d3eddeed33f7e42036d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 06:45:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ac3b6f8d5bb41e1b7fa90e1b0bbaa20\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/d831711fbfd24ea995c70aa98c7d1c37 : SUCCESS in 49m 11s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/6951b1dd04db425fbbb3210af194b83b : SUCCESS in 1h 01m 16s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/c5088014bb3b4590b7a4c0d530db7f42 : SUCCESS in 1h 09m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/5f04944e737549f1961e197dd5b19298 : SUCCESS in 55m 04s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/3138c1f4668c4510a5d8879d98e9b05c : SUCCESS in 59m 55s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/ea255a7fd9e64a3bb3ebf3bc99f423e5 : SUCCESS in 32m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37fe052be46e42f296791a694757355e : SUCCESS in 13m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd08bb4eb5ce4f75bf3e4e75a6008d3f : SUCCESS in 28m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1acbf8d1f404acfbebaa2a3a7a6cbe0 : SUCCESS in 10m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c59ed714eb44b4db784d5364d828fc7 : SUCCESS in 55m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7696a95bf5db4315920ca2a3c0b95a25 : SUCCESS in 1h 38m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4ceb9341ce684ea695fc3968ca7b50e2 : SUCCESS in 1h 15m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4e054acd21b4226bd370c7eacca43c5 : SUCCESS in 1h 13m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/112b7e5f5b96478f996794d226487d38 : SUCCESS in 1h 04m 39s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2173025b917042a6a767bfe233cb80fd : SUCCESS in 1h 15m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6527ec3551704fccb59c310c3ab7c584 : SUCCESS in 57m 03s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c68e9b975f64d89ae1a5c928b1ef5db : SUCCESS in 57m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/575b22154ab24a4fb7c9712e16fa362e : SUCCESS in 1h 08m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d56cddf9a40a4169adff7759d30b9dc4c11ebb1a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-04-13 04:13:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8f86076ced05600ad34b3d437c1bc90101f8c41","revisions":{"a8f86076ced05600ad34b3d437c1bc90101f8c41":{"kind":"REWORK","_number":1,"created":"2023-03-31 05:04:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/94/879094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/94/879094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/879094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/879094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/94/879094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/94/879094/1"}}},"commit":{"parents":[{"commit":"fa7c3d4649e9541e430f1ecc4f0775ee770a3749","subject":"Merge \"Revert \"Ensure vlan network traffic is not centralized\"\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa7c3d4649e9541e430f1ecc4f0775ee770a3749"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-03-31 02:32:00.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-03-31 05:04:25.000000000","tz":0},"subject":"DNM: testing grenade fix","message":"DNM: testing grenade fix\n\nDepends-On: https://review.opendev.org/c/openstack/grenade/+/879126/1\nChange-Id: I7a4620eebfddce9df371a748ede4f4e10742103b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8f86076ced05600ad34b3d437c1bc90101f8c41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a8f86076ced05600ad34b3d437c1bc90101f8c41"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
