)]}'
{"id":"openstack%2Fneutron~879295","triplet_id":"openstack%2Fneutron~stable%2F2023.1~I25e5ee2cf8daee52221a640faa7ac09679742707","project":"openstack/neutron","branch":"stable/2023.1","topic":"redirect-type","attention_set":{},"removed_from_attention_set":{"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2023-04-04 09:54:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I25e5ee2cf8daee52221a640faa7ac09679742707","subject":"Ensure redirect-type\u003dbridged not used for geneve networks","status":"MERGED","created":"2023-04-03 17:38:36.000000000","updated":"2023-04-04 09:56:17.000000000","submitted":"2023-04-04 09:54:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879295-redirect-type","meta_rev_id":"a8f1bdaf606b05539df6f4373e6d0f03896d9158","_number":879295,"virtual_id_number":879295,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-04 09:54:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-04-04 07:53:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-04-04 08:14:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-04-04 08:12:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-04-04 08:14:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-04-04 08:12:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-03 17:58:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-03 19:22:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-04 07:53:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-04-04 08:12:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-04-04 08:14:42.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"0df8805d0a41e56cbbd181f406b36a0bf7b6ed12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-03 17:38:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aaffd4cb3e934909b7776596eb9d484d09a4c931","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 17:58:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ceff952c3c3485aa9f9b11ae72ea3e0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5a8e00d9c6c4709ba3bcd1d1f385301 : SUCCESS in 18m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89465e030db14fc9ba61ef4aa955ddc7 : SUCCESS in 17m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb0cea89500d655379249412c142a781706d16c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 19:22:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3547073a8ef4ca5b3b034da0fd85804\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1e3876be9c6645ab8248b2953403faa2 : SUCCESS in 1h 30m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dfb9581b0cd44b50be195f780ddf1ca7 : SUCCESS in 39m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56bf24546bef469c930ff7129232174f : SUCCESS in 41m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/181491a43a6545588f10b827104bd122 : SUCCESS in 13m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0604386528fd4a52828561c1dcbffcc0 : SUCCESS in 29m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c529cd8475ae4cf68c787aee798f0d1d : SUCCESS in 23m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eeb125be5094afcb12daa4cc9bbc456 : SUCCESS in 11m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26b5b3c2cafb462ca34d874d36d7e0aa : SUCCESS in 10m 27s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/78db29c32a6b46ddbe69136afdfa4b24 : SUCCESS in 28m 25s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/23bca5a3c297422aabb7e2a25f3c2834 : SUCCESS in 58m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2101235528345cdb68f77eed0b426e8 : SUCCESS in 1h 30m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/009bb928a5a34e2bbd09dce5b6b2b7c1 : SUCCESS in 53m 59s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/71f14a58083e4bf0b3519be45f1dd280 : SUCCESS in 1h 12m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/41b7afb8885a4ce38b09b37e4696823e : SUCCESS in 1h 41m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/45ae2395935241b2a5f5ad1edd530e9f : SUCCESS in 1h 21m 45s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/806f2a3d422b40d5adcb0e307d92b831 : SUCCESS in 50m 53s","accounts_in_message":[],"_revision_number":1},{"id":"8e0a954421a8dcc1c83e1126273c1956026b0b37","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-04 07:53:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1ffa66a85e3b308a747a56a5c29946fa47926533","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-04 08:12:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"979a91233a7646bf30c68c22f6a71e06b8753d36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 08:13:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b6bade02071d95738ad561ee5605623a405d4b52","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-04-04 08:14:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"47839031b3e798e640dcbe08d790c9e167b24b01","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-04-04 08:16:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"892348602ed5ba76e48dd9a07bae1cc1d6ed45ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 09:54:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b563882d43e147d0a49bfab8484522cb\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/29bdaab9d17a47bc81ad4e517d8b4b88 : SUCCESS in 1h 40m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a4bfba0082c45c8ae2b1b3885918501 : SUCCESS in 13m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/978f8a5b3a3744958f0b716ca9cd7e28 : SUCCESS in 38m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32b072011c424aaa9da01f8d659e2f83 : SUCCESS in 29m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8940069b53a4b8598ff116b302bca71 : SUCCESS in 15m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6948e4cc39434452b2d698c9cde06992 : SUCCESS in 12m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bcba9423ed7c466a9a8abb2fe5f04638 : SUCCESS in 48m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a57af8169a854c4ab0e499ccf99bed21 : SUCCESS in 1h 21m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5201807e09984f04bc5a93c526058265 : SUCCESS in 1h 14m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9f3ff655c8524cb89c40bb0b24adfffc : SUCCESS in 1h 17m 30s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/2fb2537a7db24f9b86f4aacc1d5c6f8d : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5e742e4f593ef3c614aa6f5870b7fc6daf88f026","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 09:54:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a8f1bdaf606b05539df6f4373e6d0f03896d9158","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 09:56:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce802e6fb5a846bf889909e53f08ed26\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e2c821f0b91458d9c1cf3de4c248e6e : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/800cbf11361d4ca2829f97a43661c533 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da7b17993316aae5c9e9bc8af0140835381be7f0","revisions":{"da7b17993316aae5c9e9bc8af0140835381be7f0":{"kind":"REWORK","_number":1,"created":"2023-04-03 17:38:36.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/95/879295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/879295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/879295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/879295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/879295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/879295/1"}}},"commit":{"parents":[{"commit":"13934d38164e7fffe545e8e42a2ce99ef10d8133","subject":"Merge \"[OVN] OVN agent should register \"Chassis_Private\" by default\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13934d38164e7fffe545e8e42a2ce99ef10d8133"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-23 15:45:18.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-04-03 17:37:37.000000000","tz":120},"subject":"Ensure redirect-type\u003dbridged not used for geneve networks","message":"Ensure redirect-type\u003dbridged not used for geneve networks\n\nAs part of [1] the redirect-type\u003dbridged flag was enabled by default.\nHowever this have the side effect of also decentralizing N/S traffic\nfor geneve tenant networks, breaking the VM connectivity on them when\nit must be centralized, i.e., when no FIPs are associated to the VMs.\n\nThis patch differentiates and only enable that flag when the networks\nconected through that router gateway port are of VLAN/FLAT type.\n\nIn addition, to avoid MTU issues for the VLAN networks if there are\nalso geneve networks connected to the same router, we re-take the\napproach on [2] to ensure the traffic is centralized but not tunneled\n\n[1] https://review.opendev.org/c/openstack/neutron/+/875644\n[2] https://review.opendev.org/c/openstack/neutron/+/875676\n\nCloses-Bug: #2012712\n\nConflicts:\n    neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_maintenance.py\n\nChange-Id: I25e5ee2cf8daee52221a640faa7ac09679742707\n(cherry picked from commit 0ec04dd638da8cc9f4d5ebb21a09ea5ccb05623c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da7b17993316aae5c9e9bc8af0140835381be7f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/da7b17993316aae5c9e9bc8af0140835381be7f0"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
