)]}'
{"id":"openstack%2Fneutron~879296","triplet_id":"openstack%2Fneutron~stable%2Fzed~I25e5ee2cf8daee52221a640faa7ac09679742707","project":"openstack/neutron","branch":"stable/zed","topic":"redirect-type","attention_set":{},"removed_from_attention_set":{"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2023-04-04 10:14:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I25e5ee2cf8daee52221a640faa7ac09679742707","subject":"Ensure redirect-type\u003dbridged not used for geneve networks","status":"MERGED","created":"2023-04-03 17:45:41.000000000","updated":"2023-04-04 10:16:35.000000000","submitted":"2023-04-04 10:14:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879296-redirect-type","meta_rev_id":"7cea916cebcfa78e9b131c404d089a4f5c266f40","_number":879296,"virtual_id_number":879296,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-04 10:14:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-04-04 07:55:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-04-04 08:17:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-04-04 08:13:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-04-04 08:17:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-04-04 08:13:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-03 18:09:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-03 19:44:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-04 07:55:10.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-04-04 08:13:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-04-04 08:17:06.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"7a947a82949c52fe71d51d7bbd3ef6011ad17d02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-03 17:45:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97f1ea2a431d7a3cd423e75af7bfa242f671d6ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 18:09:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62e816a3b7b3424c92fee34c08ebe3bb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f50c279dc5094539bdb61e016d31adb3 : FAILURE in 18m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87be81a5b6c44a6ba33c45dbe587c598 : FAILURE in 22m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2970d753fa84cb9a7b5b84f68154c9cf2efa062a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 19:44:22.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0372fd6536a1444880f5a09db6226310\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/6835b5cc0c1b41d797ce466f96565a82 : SUCCESS in 1h 40m 25s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/54a0f7b1972a40a2aaf3e0af00aefc59 : SUCCESS in 38m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3560abffe653485d97fcc49ab793161b : FAILURE in 29m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edbcb398dffc40ccb4f62c5afc459158 : FAILURE in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cdc127245344db1877f0e669250f883 : FAILURE in 25m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b94b186c793544a485afe67556452105 : FAILURE in 25m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53ffe6f4e01a4d47bf7554a9771c33ae : FAILURE in 22m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0b8674697b5485a95b7995cea438139 : SUCCESS in 10m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34db6c52fdd745098b267c5fea5f32ca : SUCCESS in 8m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cea8d9fc597b49f1843460b2bb2da558 : SUCCESS in 53m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/237add70583d4a05b122d3816480301a : SUCCESS in 1h 57m 03s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f9b46936c94434db5571bc9d20b491b : SUCCESS in 1h 02m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a8fef2962570492f81681cf72c8b1576 : SUCCESS in 1h 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84d4b96b2498804e393cd1dafedbea138eb6e03d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-04-04 05:35:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c3d8d4c3bcfc7339f79b96863053a4d71c12b891","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 05:55:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e2c4457225f42cd84bfc3a29c2bac56\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c11f47756bdc4966877a706c7a68d4f6 : SUCCESS in 18m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2f63da389d64fe0be38984d4b055f59 : SUCCESS in 18m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1f146bef60459927244e3b9fa7d50169b2adef18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 07:23:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03d2fcc576e2463abcd944ee1f1e6473\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/4053fae773f74abeb547d43f8fc9915e : SUCCESS in 1h 42m 39s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/df07375becf24a798c2b17248ca5150a : SUCCESS in 31m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5231c3dd864944739872a4046c118781 : SUCCESS in 40m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55a4ee3a866e4b69bb92eef82afe6b69 : SUCCESS in 13m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/570829e4791344058ace9ac4afa76405 : SUCCESS in 20m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60bb2a4582c14e099118b7be49da32a5 : SUCCESS in 26m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa3380fda0714ec0b43dcc66aed642e7 : SUCCESS in 23m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60996b0d45734038a4f0f8cde56a3cde : SUCCESS in 10m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/294e9e7d5b2446b2a8d221e87a97d5a1 : SUCCESS in 8m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cb88c7472ee7469dbb9b05d0ada23409 : SUCCESS in 53m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/03f5a0acef9b40c6b71325eb1167c844 : SUCCESS in 1h 36m 49s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c40efdcb24b24b1c8343998894ea6f4f : SUCCESS in 1h 06m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/61780b48589644e988263f56cddbe58c : SUCCESS in 1h 30m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf5e3128ff1c2b1ddab2a8eb7b22fc2a6a5c4b93","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-04 07:55:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"374e5fa1c1ff292b07581e1e761cbd1617c7bafb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-04 08:13:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"92aec1a716c0cf3c58249f6cd8d9a8a6eb034509","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 08:13:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"553ac5bf6ca4c507ce037bc2461ce9efdac86557","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-04-04 08:17:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"97b2f6ba3dd5399bf9d753aac1cec4990e2add94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 10:14:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f722b70e007e4f48bb47245fe4ea6c26\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/12040e823e044e119d351f8dca05c143 : SUCCESS in 1h 59m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e76e235861c548cc82280142d074bb24 : SUCCESS in 13m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fec14aa36104b9a9b26139d4f83378a : SUCCESS in 24m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69c204c18950497e9a5ec859e7062705 : SUCCESS in 25m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4805754e9616402086aa763da454aa7b : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce8971558e8a429d93392a1d37f8ef7a : SUCCESS in 13m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa6609a0f902489bb3c35d552e8ebb3b : SUCCESS in 51m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1607b5d2e8b647dd84604f4c1c688468 : SUCCESS in 1h 18m 26s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b30994f16b504ca0b417819f7da5c451 : SUCCESS in 55m 42s","accounts_in_message":[],"_revision_number":2},{"id":"bf3bc0bbe99dcd1c1babd67a5e7e6e53b605417f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 10:14:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7cea916cebcfa78e9b131c404d089a4f5c266f40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 10:16:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d4a66968ab94a76b131a5b11e9ab5b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b62b136a2384692b7b62df32c9f4d00 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2187c0f35ef4ead837eea2e35db042d : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ec1a0e27dcacd3a7ad54d16099e89c5c06547db0","revisions":{"dd2412b93c40977f83e2d3ea625eeb771c1dce5d":{"kind":"REWORK","_number":1,"created":"2023-04-03 17:45:41.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/96/879296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/879296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/879296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/879296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/879296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/879296/1"}}},"commit":{"parents":[{"commit":"dc578adf17653157fd14a6f0136e01a98b32eaad","subject":"Merge \"Make path for tcpdump release-specific\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc578adf17653157fd14a6f0136e01a98b32eaad"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-23 15:45:18.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-04-03 17:45:05.000000000","tz":120},"subject":"Ensure redirect-type\u003dbridged not used for geneve networks","message":"Ensure redirect-type\u003dbridged not used for geneve networks\n\nAs part of [1] the redirect-type\u003dbridged flag was enabled by default.\nHowever this have the side effect of also decentralizing N/S traffic\nfor geneve tenant networks, breaking the VM connectivity on them when\nit must be centralized, i.e., when no FIPs are associated to the VMs.\n\nThis patch differentiates and only enable that flag when the networks\nconected through that router gateway port are of VLAN/FLAT type.\n\nIn addition, to avoid MTU issues for the VLAN networks if there are\nalso geneve networks connected to the same router, we re-take the\napproach on [2] to ensure the traffic is centralized but not tunneled\n\n[1] https://review.opendev.org/c/openstack/neutron/+/875644\n[2] https://review.opendev.org/c/openstack/neutron/+/875676\n\nCloses-Bug: #2012712\n\nConflicts:\n    neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n    neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_maintenance.py\n\nChange-Id: I25e5ee2cf8daee52221a640faa7ac09679742707\n(cherry picked from commit 0ec04dd638da8cc9f4d5ebb21a09ea5ccb05623c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd2412b93c40977f83e2d3ea625eeb771c1dce5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd2412b93c40977f83e2d3ea625eeb771c1dce5d"}]},"branch":"refs/heads/stable/zed"},"ec1a0e27dcacd3a7ad54d16099e89c5c06547db0":{"kind":"REWORK","_number":2,"created":"2023-04-04 05:35:04.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/96/879296/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/879296/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/879296/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/879296/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/879296/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/879296/2"}}},"commit":{"parents":[{"commit":"dc578adf17653157fd14a6f0136e01a98b32eaad","subject":"Merge \"Make path for tcpdump release-specific\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc578adf17653157fd14a6f0136e01a98b32eaad"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-03-23 15:45:18.000000000","tz":60},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-04-04 05:34:49.000000000","tz":120},"subject":"Ensure redirect-type\u003dbridged not used for geneve networks","message":"Ensure redirect-type\u003dbridged not used for geneve networks\n\nAs part of [1] the redirect-type\u003dbridged flag was enabled by default.\nHowever this have the side effect of also decentralizing N/S traffic\nfor geneve tenant networks, breaking the VM connectivity on them when\nit must be centralized, i.e., when no FIPs are associated to the VMs.\n\nThis patch differentiates and only enable that flag when the networks\nconected through that router gateway port are of VLAN/FLAT type.\n\nIn addition, to avoid MTU issues for the VLAN networks if there are\nalso geneve networks connected to the same router, we re-take the\napproach on [2] to ensure the traffic is centralized but not tunneled\n\n[1] https://review.opendev.org/c/openstack/neutron/+/875644\n[2] https://review.opendev.org/c/openstack/neutron/+/875676\n\nCloses-Bug: #2012712\n\nConflicts:\n    neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n    neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_maintenance.py\n\nChange-Id: I25e5ee2cf8daee52221a640faa7ac09679742707\n(cherry picked from commit 0ec04dd638da8cc9f4d5ebb21a09ea5ccb05623c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec1a0e27dcacd3a7ad54d16099e89c5c06547db0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec1a0e27dcacd3a7ad54d16099e89c5c06547db0"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
