)]}'
{"id":"openstack%2Fneutron~879908","triplet_id":"openstack%2Fneutron~master~I5f29b0feaa89b31bc41ea67ea1c6231fd3f7fa05","project":"openstack/neutron","branch":"master","topic":"ipam-register-opts","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-04-11 08:40:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5f29b0feaa89b31bc41ea67ea1c6231fd3f7fa05","subject":"Fix NoSuchOptError error in Ipam unit tests","status":"MERGED","created":"2023-04-07 18:16:12.000000000","updated":"2023-04-11 08:41:44.000000000","submitted":"2023-04-11 08:40:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879908-ipam-register-opts","meta_rev_id":"7dab26dbc2a084e4c05d072757ad3af7800b9bac","_number":879908,"virtual_id_number":879908,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-04-07 18:53:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-11 08:40:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-11 08:06:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-04-10 12:50:24.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-11 08:06:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-07 18:51:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-07 18:53:55.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-10 12:50:24.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2023-04-11 08:06:58.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"abd0223ae22a7583e74c559b624d15a795e519a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-04-07 18:16:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"548a2bca2e5698e7b75c9177a7008dfd891d9dc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 18:51:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6ae64de3da24c4687795917f746f07c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a9f3c2a61f34787a6063b7200ed5784 : SUCCESS in 33m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a2b1f1e5b254b2db47a39003329d7a0 : SUCCESS in 13m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e19316d1190746d48d9ce6a503a84c29 : SUCCESS in 26m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3207db42ffb54c61ae9a16e544f367a7 : SUCCESS in 22m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ee5d771d8004862add73ea5822c9a3a : SUCCESS in 23m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ce2e82162b04b35bca76ad5d0e3c953 : SUCCESS in 10m 13s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/17e4b8ae31754e739959aff9d0813808 : SUCCESS in 28m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e1c69a3746943f7638c86e669914ad40ffdd6e5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-07 18:53:55.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/08/879908/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c289e4811b3ebbbb50769b8036ee088f82bc0e0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 19:01:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dd99a0068ae4d0c82ba8afedd1cfeca\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c20e0740e394f3b9be83e1b9d752efd : SUCCESS in 20m 36s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32cca37a04134db3ba726c9778e7848b : SUCCESS in 44m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c55126272f9a4b0796f4be0622a67686 : SUCCESS in 19m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"76f86104c495968f8c0fda2288d699952828da40","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-04-10 12:50:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c021e71dfd469605f382ae2230e5c908a262e453","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-11 08:06:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"771d98ad94f8e240326d5d1939c1d985850083f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 08:07:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6758f98e8483e9431d8d3922dc5cb76703252884","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 08:40:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/457a21f90be14e0db2d685c22a399602\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68e26ba3fbbc494187bc4c83eb7675b9 : SUCCESS in 12m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f727f1b4f35342019273216540cac999 : SUCCESS in 20m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/574f9346eb2a44eead289002375fa9db : SUCCESS in 25m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db38649d2b0a4c28bf4e56f07e100cf2 : SUCCESS in 25m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf0405e9e61a4616bbe3943e4c6c3b03 : SUCCESS in 11m 36s","accounts_in_message":[],"_revision_number":1},{"id":"053895a269ceb4d17dacc8f9295cea882428dd0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 08:40:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7dab26dbc2a084e4c05d072757ad3af7800b9bac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 08:41:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c7f0b36355d4fff80955d0d13f7961e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcd8d348235240708fd7f0b8d505eb16 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a06fe7cfd9863230d29461ae8587a586f1c6d159","revisions":{"a06fe7cfd9863230d29461ae8587a586f1c6d159":{"kind":"REWORK","_number":1,"created":"2023-04-07 18:16:12.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/08/879908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/879908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/879908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/879908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/879908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/879908/1"}}},"commit":{"parents":[{"commit":"208421910d2bb3c71b0947254d5eca1326c184d0","subject":"Merge \"Fix concurrent port binding activate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/208421910d2bb3c71b0947254d5eca1326c184d0"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-04-07 17:53:42.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-04-07 17:53:42.000000000","tz":-240},"subject":"Fix NoSuchOptError error in Ipam unit tests","message":"Fix NoSuchOptError error in Ipam unit tests\n\nSometimes running the ML2 plugin unit tests generates an\nerror in unit/db/test_ipam_pluggable_backend.py\n\noslo_config.cfg.NoSuchOptError: no such option ipam_driver in group [DEFAULT]\n\nThe specific test I see triggering it is:\n\nneutron.tests.unit.plugins.ml2.test_plugin.\nTestML2PluggableIPAM.test_delete_subnet_deallocates_slaac_correctly\n\nProblem is that setUp() in the UseIpamMixin class isn\u0027t\nregistering the core options before trying to override\nipam_driver.\n\nTrivialfix\n\nChange-Id: I5f29b0feaa89b31bc41ea67ea1c6231fd3f7fa05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a06fe7cfd9863230d29461ae8587a586f1c6d159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a06fe7cfd9863230d29461ae8587a586f1c6d159"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
