)]}'
{"id":"openstack%2Fneutron~880161","triplet_id":"openstack%2Fneutron~master~I59bf1e53d4b63e31494890fcb313380698d6f88d","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-04-13 12:39:44.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}},"hashtags":[],"change_id":"I59bf1e53d4b63e31494890fcb313380698d6f88d","subject":"[CI][fullstack/functional] Report slowest tests","status":"MERGED","created":"2023-04-12 15:09:16.000000000","updated":"2023-04-13 16:41:50.000000000","submitted":"2023-04-13 16:40:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880161","meta_rev_id":"20f3c654c420c39bc96745a1a05b380ec49d8b89","_number":880161,"virtual_id_number":880161,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-13 16:40:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-04-12 16:05:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-04-13 09:48:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-04-13 09:48:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-12 15:54:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-12 16:05:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-04-12 17:43:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 09:48:07.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"be3cb67ec45fafb474f3ac6e81398f1df86d1681","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-04-12 15:09:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cd4fc64606911f9fb2a32a6f1eaa5dc9ad9300d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 15:54:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f8b907ae80e4de9a346b75f1c5956b5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c36609fbd4747cbac296f8da5d2a059 : SUCCESS in 41m 37s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0426abcb23be4edda62a04714bc3ac80 : SUCCESS in 42m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dee74811a6634334affaa5372bbec840 : SUCCESS in 41m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30bbfcb701ee6a69101299c4cd651d643ee7d447","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-12 16:05:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba1741efc59b086b72f8faaaeb17a49aa42a3980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 17:43:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3192c6f03fcb42cbb9a4fbe9a5eb91f7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88f0838e429b4ace83890c8e74ce1949 : SUCCESS in 35m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a390cff610e449cdbd7b2cdd7635cf7c : SUCCESS in 14m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8531bf1a2a8b4330b4136148b37a054e : SUCCESS in 27m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11c12d0777fc473481140b76f93d322a : SUCCESS in 28m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65f8236789234e2b8ed91655c3c502bc : SUCCESS in 27m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f2030e8165d4f7e8a0cb073524a81dd : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed7389c3cddf4800ba9f95bd8f5361df : SUCCESS in 10m 31s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/fa2308c2a17e4891849cf21ace9c5d09 : SUCCESS in 20m 13s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff2ebdb69c56497c988f59f38f39126f : SUCCESS in 1h 08m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6b1d41fc233492a9f666e3a5e1f99f9 : TIMED_OUT in 2h 32m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b44e5218311bfdba33c42d0e5be64c5a8f3b1901","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-04-13 05:18:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30436addb8e69e86d04630bab6dbf0d6966a33ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 05:49:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cddf18f801d948339ca2ebd44ed51e06\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5046399886a04bc3b567eae3d3bc747c : SUCCESS in 30m 11s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f87158a9eae4409a922fe068d333389 : SUCCESS in 19m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6c1cbe6f56ed4484970590b84daf92e2 : SUCCESS in 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b50b51434b8a7704b7c48c18546d0a6c4771d009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 07:50:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ff6efe7b858435c9161b6807ed2fa9d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d08eca944314c72bc4e8826914705cf : SUCCESS in 37m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04b5c3261c4942ceb5870186a07fc3cc : SUCCESS in 11m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9333f7410fda4d89aa01c85f22f786d6 : SUCCESS in 24m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e57ced54ac5401dbce14752259c3f49 : SUCCESS in 27m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ae7ab5425504367b26b21f7e0f89fe6 : SUCCESS in 21m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f3b1c025b548c8a96b167ab41745ea : SUCCESS in 11m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b9012bb35614616a3350af9185c10df : SUCCESS in 10m 27s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/d085dd338fd84f5b93707e772bb16843 : SUCCESS in 18m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a3991535664547f6bbef4a0497d9774b : SUCCESS in 1h 19m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/23e48b099d2547d4b184c6201a943b1f : SUCCESS in 2h 30m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1ea6f4de41051398b98a186c4d860b36f6607dcc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-13 09:48:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8e0d348bea3f908f6e26638ed033dc3d87d309ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 09:48:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f26ae29991d9d6f29de851795cc169623aaa7782","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 11:51:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b33d6ba4e69499bafeacce856eca1c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd111cda841242aa831c7c70a520b6bb : SUCCESS in 14m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23993dc6c6c64e6c8b34d14ccb1dbb39 : SUCCESS in 24m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f1f8064979e4702bebff3c6b1f4b61c : SUCCESS in 28m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01d3680e0d5244d6adfbec824a5da323 : SUCCESS in 25m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44a32be11ed4b688bb2d99b9870cafb : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e537cb149cd4a9ea29cc429ce955275 : SUCCESS in 14m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b00bcbd27b1642f9922fc2dabe635ac6 : SUCCESS in 1h 04m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ecff029dfc4944ee879e136ad3e87f5b : FAILURE in 1h 59m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7f9d31e7257e52f83ff67a9845255c52e42de20a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-04-13 12:39:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"571dd5a34c3d935ca8d734e3c8c883bfa70022c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 13:31:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd8c7902ab064f4b85818a517d409825\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13ae251d363a4c258712a506a48a5016 : SUCCESS in 48m 30s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6c7288cc880143229e82bb6364308a3e : SUCCESS in 47m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f8c8221be9aa4d5d964ea795dfc73e04 : SUCCESS in 21m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0c4701d13b099b3dd3254d47fd517b63056e4d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 14:41:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a51cef377c740a79e427b64a3bf6257\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0964d7b845cf43fd898e6000b9b03beb : SUCCESS in 51m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09e3f6f151d4c0dba770d5ceb2fdce5 : SUCCESS in 14m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/421407f25aec420bad59fb4617745efb : SUCCESS in 28m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da636a8437734d90914165004d677711 : SUCCESS in 30m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9dcff236e7140efbccd224d2f140864 : SUCCESS in 25m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1bb98b189804959a9fce80a49906702 : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8440a12b47f45e5ac8fce3245077677 : SUCCESS in 14m 15s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/b4f0e6096ea64a2fb84f50f27c17bd5c : SUCCESS in 25m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e94f6d677bc4ab085937de7d778c3d5 : SUCCESS in 43m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa73dc87e5bb4c0d9d7f722d5df73ba7 : SUCCESS in 1h 30m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9d5df67046c569321a548207f6a7b309c2e8daba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 14:41:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2dddffbdc434bd60558f2b498a12b80fe8ad932e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 16:40:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0b8b3fd776940eeabaecc23fcd8dfe4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d8e8f9d7d7a468da3e6e4ff995d896c : SUCCESS in 11m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2ea07615d80461ab699b7a70bc3b93c : SUCCESS in 25m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfd06dab1acc4a1383f64fdbb36e7114 : SUCCESS in 26m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52c7ada5aa3a4381a2cd6eab288d3d2e : SUCCESS in 22m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4413b6a0ba664f66b62f79118ab9f016 : SUCCESS in 12m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d311a22a76524bcebadaf30cd8d64f87 : SUCCESS in 14m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/26c8d2213b2f452c9598527c040a880d : SUCCESS in 1h 05m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5c2d86932b5b450ebcb21371639d6801 : SUCCESS in 1h 52m 18s","accounts_in_message":[],"_revision_number":1},{"id":"df5af7cfb8b43742aeb1747808358e8e83964fd5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 16:40:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"20f3c654c420c39bc96745a1a05b380ec49d8b89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 16:41:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ce6459cc13c443a9b9f9e8e8dafd11b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/068001c4a62a4603aace5ee8fe15f40a : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90b774a46ad2453388fd98dffdd4dfc3 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e374b82d8f71a31c3f22ef98d5ef7b00461edc49","revisions":{"e374b82d8f71a31c3f22ef98d5ef7b00461edc49":{"kind":"REWORK","_number":1,"created":"2023-04-12 15:09:16.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/61/880161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/61/880161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/880161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/880161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/61/880161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/61/880161/1"}}},"commit":{"parents":[{"commit":"fe9c321fd880c3a2fd3c669ce017d0e625306325","subject":"Merge \"Pin OVS_BRANCH to working commit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fe9c321fd880c3a2fd3c669ce017d0e625306325"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-04-12 15:04:40.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-04-12 15:07:32.000000000","tz":330},"subject":"[CI][fullstack/functional] Report slowest tests","message":"[CI][fullstack/functional] Report slowest tests\n\nWould be good to see the slowest tests in the\nreport after test finishes.\n\nChange-Id: I59bf1e53d4b63e31494890fcb313380698d6f88d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e374b82d8f71a31c3f22ef98d5ef7b00461edc49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e374b82d8f71a31c3f22ef98d5ef7b00461edc49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
