)]}'
{"id":"openstack%2Fneutron~880340","triplet_id":"openstack%2Fneutron~stable%2F2023.1~I5d7eb73606428841caa24ce1e38d1bebd5db0a9b","project":"openstack/neutron","branch":"stable/2023.1","topic":"bug/2016144","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-14 12:02:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5d7eb73606428841caa24ce1e38d1bebd5db0a9b","subject":"Do not check the context object in ``TestMeteringPlugin``","status":"MERGED","created":"2023-04-13 16:28:04.000000000","updated":"2023-04-14 12:03:43.000000000","submitted":"2023-04-14 12:02:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880340-bug/2016144","meta_rev_id":"bb9b8843a9b3fbdf9a59597d3faafce8d33ce935","_number":880340,"virtual_id_number":880340,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-14 12:02:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-04-14 10:14:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-04-14 11:12:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-04-14 11:12:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 16:49:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-13 17:13:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-14 10:14:11.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-04-14 11:12:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"fad8a66722c366e2012eb28d880a675970ff5c91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-13 16:28:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3565232ef3a1fe933d7c7acee54b75e5e3a32361","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 16:49:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e098ed868be840bab37492a3a5d1a0fe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bbbd0abf7254914bd8824c85177c482 : SUCCESS in 19m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dacd32e9ce8a4beaa727fa66804b0456 : SUCCESS in 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"001dd104f3047e18d51a01cc9643c595eaa2e3f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 17:13:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac8e55b10499490cb5ec791145edd2c3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c9972e167cb4ec7a08506086259a12b : SUCCESS in 36m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9753a51f403740249807d07487c9a3d8 : SUCCESS in 12m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5fab39e4c874b38bce0e252902d8148 : SUCCESS in 25m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02b8d0d7d1bd48469a8fb6be1290d41d : SUCCESS in 29m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e000a601df4045209cd56afaf8583397 : SUCCESS in 9m 58s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/8221a03f72f6459b9877d9106f6c3011 : SUCCESS in 22m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94b9874c9270b288d449f14b7b15b4fdbb4a2439","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-14 10:14:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"11ba559e32d501ca8d1c4a7bb5779fec9de9ae94","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-14 11:12:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2eafde045a76b1d91d7b8a5c102e2bf7b03d37b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 11:13:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d2c3f1a479c7846d25f682d893987859f9b34e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 12:02:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fc2d3cd1d634508b821b16bb02a2c81\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5bc8dc9bec44010852ffe86181c962a : SUCCESS in 11m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/951b6dc9c74d4a0c96cf2cb0b5e2e617 : SUCCESS in 24m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1533a0b2ab2477d9eed40e5259c0054 : SUCCESS in 25m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15fd4bee44db49a89e7dbba154811abe : SUCCESS in 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"56cb2141e694470d6181d7f5315a0ba7132b6a80","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 12:02:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bb9b8843a9b3fbdf9a59597d3faafce8d33ce935","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 12:03:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/200993ffa9024fad9345a551a3321278\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ab52b0041df4a6f98fb48d78db32943 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6efae0b110af0ea6ee92a9ef774cd702db3b834e","revisions":{"6efae0b110af0ea6ee92a9ef774cd702db3b834e":{"kind":"REWORK","_number":1,"created":"2023-04-13 16:28:04.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/880340/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/880340/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/880340/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/880340/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/880340/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/880340/1"}}},"commit":{"parents":[{"commit":"12809a6c67f9457e30495e8588845f5799c03100","subject":"Merge \"Fix concurrent port binding activate\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12809a6c67f9457e30495e8588845f5799c03100"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-04-13 15:54:47.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-04-13 16:27:49.000000000","tz":120},"subject":"Do not check the context object in ``TestMeteringPlugin``","message":"Do not check the context object in ``TestMeteringPlugin``\n\nSince [1], the admin context is not ``Context`` object created\nbut the new one created in ``ContextBase.elevated``. This method\ncopies the object itself (creating a new one), adds the \"admin\"\nrole (if not present) and returns this new object.\n\nBecause this method always returns a new object, it is not possible\nto check the ``Context`` object used to make the function call, as\nlong as the test doesn\u0027t store the elevated context used in the call\n(apart from the reference stored in the mock object).\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/880143\n\nCloses-Bug: #2016144\nChange-Id: I5d7eb73606428841caa24ce1e38d1bebd5db0a9b\n(cherry picked from commit c7ef8249419ecbefdc4a604a98d4b23554c78dae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6efae0b110af0ea6ee92a9ef774cd702db3b834e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6efae0b110af0ea6ee92a9ef774cd702db3b834e"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
