)]}'
{"id":"openstack%2Fneutron~880539","triplet_id":"openstack%2Fneutron~master~I551a5095429ca949e6886a98020c4ddd072de862","project":"openstack/neutron","branch":"master","topic":"neutron-ci-improvements","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-17 09:30:50.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-04-17 13:57:32.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-04-17 12:32:18.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-04-17 13:57:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I551a5095429ca949e6886a98020c4ddd072de862","subject":"Run periodic jobs in experimental queue too","status":"MERGED","created":"2023-04-14 16:02:05.000000000","updated":"2023-04-17 14:00:08.000000000","submitted":"2023-04-17 13:57:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880539-neutron-ci-improvements","meta_rev_id":"f284b0124c8e0a65bd71c142adf78dc3ba390369","_number":880539,"virtual_id_number":880539,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-17 13:57:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-04-17 09:30:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-04-17 12:32:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-04-17 12:32:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 16:06:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-04-14 16:33:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-14 16:52:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-17 09:30:50.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-04-17 10:41:11.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-04-17 10:41:11.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"8ffe0a75190cbf718ac563c9fb024c3bdc57da44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-04-14 16:02:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5f89a906f7333d88d70c08a1970bc1418893caf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-14 16:06:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b87c599b1c3abbfce6ad815c7e56b963c51e0e3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-14 16:21:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"125f4eea5341839256a5e7eb38f098129bb2f6bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 16:33:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe8df3db35ba401caea37cc0511ef138\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/193392c378504bb28ce7eba603408014 : SUCCESS in 29m 52s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b4298f8c4574e248492d59ec56c15a6 : SUCCESS in 18m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bec211fd6f5841e9b0bc3b2d3f6f4d83 : SUCCESS in 18m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f9428415f9bc73010ef0f88dd52be8d89842397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 16:52:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ccc928ad80845fdbd4b76b077e6555e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbdfae976f0c4341b4a3181467f04746 : SUCCESS in 44m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f13a49c2c8b4e96b681432b7c8ea78c : SUCCESS in 15m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bec5654b212e4aeaa3336fa82f3be1f0 : SUCCESS in 30m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bb1ce3396154a0ea2749e9f8877208d : SUCCESS in 32m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a435fc7b35ff4ccd987b899dec086a53 : SUCCESS in 27m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/242d69e802264afb9b2614c50a51a263 : SUCCESS in 10m 45s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/d12b3163e8934c11a552742633a84fc9 : SUCCESS in 23m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d754e917c927bfa9d03db0b4f3db1cfe16a8f340","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 17:53:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ecd549c053c48e6861d3a96e8111ce2\n\n- tox-py311 https://zuul.opendev.org/t/openstack/build/56e85b8dd3c94e7f89e23a78dffdec3b : SUCCESS in 23m 26s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/9490edab02a84d35ab0390687519f8be : SUCCESS in 54m 32s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/fab37efef1454a6cb30ecf600fad2fc8 : SUCCESS in 1h 12m 28s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/a01e922b633340b680265a500af1506d : SUCCESS in 49m 32s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/674b18aca9cd48dba074cf2ec8b8987c : SUCCESS in 1h 05m 44s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/d087592416fa41caa2ae95db6af1e77c : SUCCESS in 1h 39m 12s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/525bd55f77464e2ebcd03c378b361bb8 : SUCCESS in 29m 17s\n- openstack-tox-py39-with-oslo-master https://zuul.opendev.org/t/openstack/build/8297fc345aa44575b1f541625e288ed0 : SUCCESS in 25m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1b87b07bf128585b8f1f4b89e51ed8bd9beb56f8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-04-17 06:27:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35d7f5acd9af64ffc9b43cde57f43a61012ba523","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-17 09:30:50.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4da9e35b064882cd08d20f996e61cd61b90698f2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-17 12:32:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"42b1b3d937e960a84b8543b7acec627ca09b8b91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 12:32:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"31578d4ad3d4dfabf488fb8047a47d7f9efdbf24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:57:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59ad8dcca4204d82bf5904b7340169f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d38d29968704dffaa5995f2a1a46be9 : SUCCESS in 9m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/088338dde9d047ac8903dabb7ee866f3 : SUCCESS in 24m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/117486f64949407bb2700f5b515917fc : SUCCESS in 23m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58d4177a65a24dcca5ebb96592132072 : SUCCESS in 24m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb67002669bc4d5d843c9ced54f0a5b2 : SUCCESS in 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d4623d58c28a9252fa60009b6a1bf8ef333863b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:57:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f284b0124c8e0a65bd71c142adf78dc3ba390369","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 14:00:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e6630083fb641cf9e121f17aca1c187\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1d6beaa349e4640a75ad2dd9df2439a : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2364327bd1e4d26214e11c2f3f4627959a6a3e88","revisions":{"2364327bd1e4d26214e11c2f3f4627959a6a3e88":{"kind":"REWORK","_number":1,"created":"2023-04-14 16:02:05.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/39/880539/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/880539/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/880539/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/880539/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/880539/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/880539/1"}}},"commit":{"parents":[{"commit":"cda5107e6582bce5f0b0630991acd84f5df4e8b2","subject":"Merge \"[sqlalchemy-20] Add reader context to ``get_ports_on_host_by_subnet``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cda5107e6582bce5f0b0630991acd84f5df4e8b2"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-04-14 15:54:56.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-04-14 16:00:53.000000000","tz":330},"subject":"Run periodic jobs in experimental queue too","message":"Run periodic jobs in experimental queue too\n\nIt was done in past with [1] but since then\nlist diverged again.\n\nTo avoid same happening again used yaml\naliases for periodic jobs. So from now\nany job added to periodic queue will\nalso be added in experimental queue\nautomatically.\n\nneutron-experimental-jobs template will\nbe used for experimental only jobs.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/833172\n\nChange-Id: I551a5095429ca949e6886a98020c4ddd072de862\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2364327bd1e4d26214e11c2f3f4627959a6a3e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2364327bd1e4d26214e11c2f3f4627959a6a3e88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
