)]}'
{"id":"openstack%2Fneutron~880656","triplet_id":"openstack%2Fneutron~master~Ibd3e35955bd9030fb52b8f4700f115bbf20cc999","project":"openstack/neutron","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-05-02 10:29:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibd3e35955bd9030fb52b8f4700f115bbf20cc999","subject":"Fix servicetype unit tests module","status":"MERGED","created":"2023-04-17 15:31:29.000000000","updated":"2023-05-02 10:29:04.000000000","submitted":"2023-05-02 10:29:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880656-secure-rbac","meta_rev_id":"a911007c4ab53063d786383456c9fe319935823c","_number":880656,"virtual_id_number":880656,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-04-17 21:54:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-02 10:29:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-04-17 16:13:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-18 11:40:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-04-18 08:35:50.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2023-04-18 08:35:50.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-17 16:10:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-17 16:13:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-04-17 16:15:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-17 21:54:27.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-18 08:35:50.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2023-04-18 11:40:54.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"e247ff6e26d5f7f8219c89d012bf8d5dd9850bad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-17 15:31:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5633c6c5ba25865c24720b381b8010edd2dd94a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 16:10:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38ed24ba82be44738fba75db2e28675e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5c6cfad54384deca7bdd162cb7aebf4 : SUCCESS in 35m 33s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e7de4fe5bd54625bfa962800411dde0 : SUCCESS in 37m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/12618d8bdbc64f09840dc02d356e258a : SUCCESS in 34m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b29fbd6c0cab32dc603b15004a9b01c155e669a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-17 16:13:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2e81a02a578a8dbb7827eead8a8a9e471240f538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 16:15:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a229b3f51a9a45baa435c1ea6070ee65\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aff2c91faa5a4e2a903bad375906aa07 : SUCCESS in 41m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab7a9a29780f4b5faf2f4ad72edc6001 : SUCCESS in 13m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4ceccbfa3cb46faa43eba3c1e4dacf0 : SUCCESS in 22m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fa6dc94076547e4b7d73f78d5073591 : SUCCESS in 20m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8372b79a9f3f4b7b92b42755672647d0 : SUCCESS in 24m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce9f19cb6754e2993677080353811ec : SUCCESS in 10m 14s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/7f2225d349c94175bf62c19ef6fe658b : SUCCESS in 21m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd1afa9771beee06b025af89de122a9318674364","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-17 21:54:27.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/56/880656/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"585ad3ebc8e50dfb6ade2d29696fc66da23a0b53","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-04-18 08:35:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3add23c6a40601be94730026536bf0e2534ccd22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 11:04:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06a285bba297d4bf5313048083bb30cf7bac57ad","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-04-18 11:40:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b537caa27b29578e4cd613c1d37380637a26d554","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 13:48:52.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/880638, https://review.opendev.org/880461 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"2f8265b5103ad153751a166fe5a6c8e40c6dc9e0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-25 09:21:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aaac4fa58e8edbd11a7dc082eb00348f6412a5aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 10:00:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96ce3124f4784e37a2891f87dd4d3c01\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc81c9b041a74903b064bf5240761092 : SUCCESS in 25m 06s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/587608841437465ab93db38b6606c42d : SUCCESS in 37m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/decf8bad211047d283ec0ab4c30aa02b : SUCCESS in 23m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6aac4ac5fe8b4287bcbbfc960873de0b502e2b17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 10:08:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/82516dde14c7414c9b3e16c6b89e45ba\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19ef5ccb50bb4e66a39e68a6498415ce : SUCCESS in 44m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d46310dc0024088a1cb1fc6320aaed3 : SUCCESS in 15m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfc0181905ae4cdd8f32b8aa9cdda635 : SUCCESS in 44m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/088053a1ccc64da2a47e65c275d8d0a8 : TIMED_OUT in 40m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a1e5ac653ef4995adbd796340185d2c : SUCCESS in 28m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee3c9ce500084f49ab73a1ec22674831 : SUCCESS in 11m 15s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/9098e4d6e52245c192499aa042af58e0 : SUCCESS in 19m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb8c57c2e27fddeda3ba138df9958bdb8ce2fae1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-04-25 10:30:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f81203657851193bc4d92890974f5d413ff9db4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 11:08:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfb30cc6892b43f386513f54bb747130\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07d2c03cd49a48c3b1a56257c6c31ca7 : SUCCESS in 30m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ba1fe8885e04be382c4b9584c72ebee : SUCCESS in 13m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35b9fe6baa344b40bca21610b0342b30 : SUCCESS in 26m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52a9e98886c34e8c92fcf2116a871a6e : SUCCESS in 31m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2912aaf215c4afcbb0b5950c2c79934 : SUCCESS in 23m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e95f6be19bec4c95aa1f48a3904d70d8 : SUCCESS in 10m 59s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/c70e7085f2b54004996c8f2438be9876 : SUCCESS in 23m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbfdbd3ab2750b6f24742c6402c981a2dfdb764c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 11:20:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc57bbd813bf4825b70dff87d63501db\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f4284becf2948efa75ba4db38fc5ca8 : SUCCESS in 47m 51s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bcd5be97c5534deb94d7f5caf753893f : SUCCESS in 43m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/74a8155066dd4f05b9aa049b3b65c464 : SUCCESS in 47m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3b6fb4feadd3fc4631de1124cb4f3a6607f85d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 06:51:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5ae29881fa9fae87f562cdae9f9c521d663891b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 10:29:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0065187750044a82b88c8d8e30222201\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7345ef3201a241e6a1b393256c9ecce6 : SUCCESS in 10m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e57a52bccf34f6cbe2fb7120ed3abfa : SUCCESS in 37m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/984649f41cae42009fbaab9518189356 : SUCCESS in 27m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f73d1027905b4fb6b3c5778861929e8f : SUCCESS in 37m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fef551f140024497b296e19786bc1635 : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a911007c4ab53063d786383456c9fe319935823c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 10:29:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5db17654b99f1ebabc87b65a44c1b02133558fd9","revisions":{"5db17654b99f1ebabc87b65a44c1b02133558fd9":{"kind":"REWORK","_number":1,"created":"2023-04-17 15:31:29.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/56/880656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/56/880656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/880656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/880656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/56/880656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/56/880656/1"}}},"commit":{"parents":[{"commit":"f92d6fa72aa5965493375aab156153013b20db75","subject":"Fix segments unit tests module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f92d6fa72aa5965493375aab156153013b20db75"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-17 15:29:46.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-17 15:31:20.000000000","tz":120},"subject":"Fix servicetype unit tests module","message":"Fix servicetype unit tests module\n\nThis module needs to ensure that common config options\n(core_plugin option especially) is registered.\nOtherwise this test module can\u0027t be successfully run\nwithout other tests as it was then failing due to\nunregistered config option.\n\nTrivial-fix\n\nChange-Id: Ibd3e35955bd9030fb52b8f4700f115bbf20cc999\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5db17654b99f1ebabc87b65a44c1b02133558fd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5db17654b99f1ebabc87b65a44c1b02133558fd9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
