)]}'
{"id":"openstack%2Fneutron~880934","triplet_id":"openstack%2Fneutron~master~I699dd30d58e3b02713776393c497cf9fa3793185","project":"openstack/neutron","branch":"master","topic":"bug/2006603","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-26 13:33:29.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-04-26 18:08:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I699dd30d58e3b02713776393c497cf9fa3793185","subject":"[OVN] Add update event to ``OVSInterfaceEvent`` class","status":"MERGED","created":"2023-04-20 10:09:27.000000000","updated":"2023-04-26 18:09:58.000000000","submitted":"2023-04-26 18:08:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880934-bug/2006603","meta_rev_id":"7e458d5c0b394663a012bcef7dcd9a178aea1614","_number":880934,"virtual_id_number":880934,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-04-20 10:48:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-26 18:08:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-26 06:59:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-04-26 06:42:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-26 06:59:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 10:48:47.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 10:57:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-20 12:19:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-25 15:19:52.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2023-04-26 06:42:15.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-04-26 06:59:34.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"925bb990e83acdd1aefd9df3c785e320c5df282b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-20 10:09:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96e8bf72748b3dd28a7424c65e4fa2ab2a8411b6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-20 10:48:47.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/34/880934/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ca70fff7c1309a837f57b84ff120ad6b2ad437ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 10:57:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f27e6452f8b34ac28634341a5920cf7d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/004b5758068d4c47816e5c4304a70ed3 : SUCCESS in 19m 34s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a65f821ab944e409faf62a1459b5956 : SUCCESS in 19m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/04ca6dce3c294abca9b00a86cd9350d7 : SUCCESS in 45m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"388fb0ead7f2de8af17e53b219bf91f27a5f4738","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 12:19:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/229d653d3f5f4faf8a6c1b820d58880e\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/984b16f9e72d45c1a9b004903cb65755 : SUCCESS in 2h 08m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d61d8598ea5b4cd39c57ed1818971b16 : SUCCESS in 34m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f65bb964d917429192e8691afbf8f224 : SUCCESS in 12m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea9d01e8102d4276af5ef6902d161331 : SUCCESS in 26m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2a1aa409a7d4c298b9a6a1a542d6818 : SUCCESS in 26m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb912490767947c18c929ac4e8e140ce : SUCCESS in 24m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6943b7a819a47a28c3476b62540e3aa : SUCCESS in 10m 02s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/09851abfb5ae40c6b2c0845c69f91ca3 : SUCCESS in 21m 49s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ed7b93d02b4f41f997035f6b6888c748 : FAILURE in 48m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8e442a97f1b2481eaa044df63c947aeb : SUCCESS in 1h 16m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c739bffc03624bcd93bb416514e9bb98 : SUCCESS in 1h 42m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d2e7706e88174c5da3146ac63dca01d6 : SUCCESS in 1h 18m 22s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/d5898b8cd5d44de2a8d65f92c58f84d4 : SUCCESS in 1h 15m 29s","accounts_in_message":[],"_revision_number":1},{"id":"322b869660fad8413eea170a1ce8f4dc7c38ce5b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-25 08:40:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1dfed91f8062b0a0189abf5090bc8d606799ebe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 09:53:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9ade0e68fa44e349eaa55e869488410\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13df1e4377274587b9c11e6fd7b166b1 : SUCCESS in 45m 47s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b1701b06a2e473fabdb737b755e7ba5 : TIMED_OUT in 1h 01m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bdd22e49286d410381cb171ac025dec0 : SUCCESS in 1h 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e5a45e9308fb1608865fc1f85b0f4df71b500b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 10:28:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d81772a135d949dfa11ee2cc170e7998\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/26ad54e9b7aa4537a0c3b3f09ce3c5a5 : FAILURE in 11m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c88edd6943f141a6b5542ea200320a8d : SUCCESS in 40m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8046558d9fcf4df086d342d29e319221 : SUCCESS in 12m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10b5fc11fcea48bb80fb3822a274c256 : SUCCESS in 24m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b156b8f5e2bd4403b90b4baa3eb20267 : SUCCESS in 33m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/302f1727544f4c9eadb845c481ffb213 : SUCCESS in 21m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edadafecf84444a9bec05e020081ca5f : SUCCESS in 13m 32s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/de75becd8ef34edabe00129858fb383c : SUCCESS in 21m 28s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/668bc536e85244238fbfa4eb5424005f : SUCCESS in 47m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cf995115629e486bab463505a3bdcb37 : SUCCESS in 1h 00m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/292bfee01b5b4c938bf305d9e19dbc6c : SUCCESS in 1h 46m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/703d75c23fd44756af423206e83ab74a : SUCCESS in 1h 16m 33s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/bbcfc1ee94744b0caf7fbb6c6101fd5c : FAILURE in 17m 36s","accounts_in_message":[],"_revision_number":1},{"id":"e2bd51136ae6129743ee9065c9dab625154fbc8f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-25 17:09:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec54d81de446b86a4bc4d848401a7e312c394212","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 18:01:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eb45bb1190249ea8d0074c4b0a9726e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf9add7a3d1b4f49aa5e505a73f2464d : SUCCESS in 26m 40s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b0ec248bbb14b59a98670add57f8939 : SUCCESS in 49m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9fd2b722aee54ada807c53b72c5541c4 : SUCCESS in 27m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"75ee0f1a8539d4cfd32b512cd2dc62dbc8164104","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 19:04:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da70a64342e9416db32cad7f334bdfcd\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a5f712e0d72140148b74700e7722c920 : SUCCESS in 1h 45m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b84a2b935b7149cca79d709ff62c1714 : SUCCESS in 38m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99a5643d11064f9e8fc306b37676f06c : SUCCESS in 12m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae79532f13aa46d08b0b0de7796c6262 : SUCCESS in 23m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/148c0a86fe4a46399287076c627c44c4 : SUCCESS in 21m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e812c46adab44709935e663ded067390 : SUCCESS in 29m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91ba4e5e35424bd9a25214293eb5c56f : SUCCESS in 10m 26s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/f65024ef78ce4f3f857094c17dd367a9 : SUCCESS in 22m 56s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b2e169d264804739ba5d9ca544e0a2f7 : SUCCESS in 52m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d5996edaa25a41a89987123b82c81d55 : SUCCESS in 1h 05m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cd409992dca94f9fa934f67951cc2dd5 : SUCCESS in 1h 49m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/da790f2eca64410293d486fe7f72205e : SUCCESS in 1h 10m 44s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ee55a7a14e9a419ab2065a4296555bc3 : SUCCESS in 1h 11m 12s","accounts_in_message":[],"_revision_number":1},{"id":"99c7bdf13fa31a33bf827d1bd7cc4bf8f4b4b79a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-26 06:42:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"200c1252668dac96caba768e2c7f7257347c3daf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-26 06:59:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"217b79e601febc97b8b6d062a5bc0da482e35968","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 07:00:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1bb835f618d757ea8641418ea52ea6c4d2ebd7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 08:23:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/026fc7aea0e14e57bee1655cfd955c9b\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/95a86b7d47d8438b9015468aca32976f : FAILURE in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae345eeab01d43cda8581323d60b4f4a : SUCCESS in 12m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e955525ecea4b2eb06d817919d2d50d : SUCCESS in 22m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d257748a73d742b5ac03a422fc1d88d0 : SUCCESS in 24m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19270321cc4243bf9f35e166e28f242c : SUCCESS in 21m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24d7f6195b93414f9afee773051f7499 : SUCCESS in 12m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b03519aaf444d5da8d740380addcc44 : SUCCESS in 58m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1914d0405e554735a9df1e1348b5bf95 : SUCCESS in 1h 18m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4c4773eea248499cbbd9d0d84f6c024d : SUCCESS in 1h 17m 42s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/e650f9f567af4ea7ba6973d486df322d : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":1},{"id":"63ef5c2e2f60fb5f7de4542c2bdadeb69ce1389e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-26 13:33:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"05c9824a63f1f9c36e163b04746e1fe2d7a144c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 14:20:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a84f0be4b0dc45c7897f940398c888bb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfe4a6e0cd204067a99e2b1cd5187c7a : SUCCESS in 24m 30s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3813cecc4e834d10a276d63fe53c70ad : SUCCESS in 25m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9e33fa4b735e42fb811daec3fc020717 : SUCCESS in 45m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"275b17bb54b4bf248ebfecb3492f7479488640c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 15:40:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/707f0ded1ee04998a778bb230ac2814e\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fe70ce204ade4eeda37f0f88df4b44ad : SUCCESS in 2h 06m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/805cfbffb71e4eea95387e0104223b33 : SUCCESS in 33m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f281cbdebd994343a9cd4e83721562a8 : SUCCESS in 13m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c312cb33d844d04af5869e4bec8d761 : SUCCESS in 23m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6d2b8e4156b41ff9ff1d2710f77cba7 : SUCCESS in 27m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab3ea183cb3f4e42ac0cd87fa82cf5c5 : SUCCESS in 25m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d97958300e814176b598e92584a5ae42 : SUCCESS in 11m 06s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/47a6b5bffd414cac9b2dff15bd1ce90f : SUCCESS in 23m 06s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/81ac434babdb47b7b4c7bd99737b6dbd : SUCCESS in 45m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/75f316efb41b463d919ab05550a25b90 : SUCCESS in 1h 11m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f13d1427cbf34bf5b153f80d5064e989 : SUCCESS in 1h 54m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ed8be6e3a24d448dac6de26e525bd141 : SUCCESS in 1h 26m 55s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/c5741bec70d64e32952440fb387d19f7 : SUCCESS in 1h 20m 48s","accounts_in_message":[],"_revision_number":1},{"id":"badb49f6989c61e8e24e25385581f02262324b77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 15:41:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7128cd74b383a37c64c1d37f3f30a9c931ee418d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 18:08:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/904c19b13abf4217b4f4050455b40911\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/43ca0f95dda042f3b9833d5f3404942f : SUCCESS in 2h 25m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d5519b299fb4609848a6375e3e1f88d : SUCCESS in 13m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/826c537a364841c5a98530766c59dda8 : SUCCESS in 23m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef8233704c484b9c9f870c52ec01b09e : SUCCESS in 27m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26dfa5a95f8248ff9b0bdbee2f198049 : SUCCESS in 24m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe5ed1ba662042abbc3715e5bd72d14c : SUCCESS in 11m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d6874a2b86249838f688dcf231f00f3 : SUCCESS in 1h 11m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/078be6ed47da4a0bbb79d017d1782e92 : SUCCESS in 1h 10m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e90f0958bcc340e59b5b026b1e45ff5f : SUCCESS in 1h 16m 28s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/d87801be93194872a2219172fedb64be : SUCCESS in 51m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ec9249608cc3a8184caad8c5a4b24268fd2ce9eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 18:08:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7e458d5c0b394663a012bcef7dcd9a178aea1614","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 18:09:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8868e84345f46ad908125c092f593fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42838ae7a1934b6ca8bd3ee16ce4293d : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34ea8988d6cdc3d676968d57b57525a2515d6128","revisions":{"34ea8988d6cdc3d676968d57b57525a2515d6128":{"kind":"REWORK","_number":1,"created":"2023-04-20 10:09:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/34/880934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/880934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/880934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/880934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/880934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/880934/1"}}},"commit":{"parents":[{"commit":"18bc5b82a1d313a27411a9a88488f1cc306f0fee","subject":"Merge \"OVN: Always try and create a metadata port on subnets\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18bc5b82a1d313a27411a9a88488f1cc306f0fee"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-04-19 18:42:33.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-04-19 18:43:02.000000000","tz":120},"subject":"[OVN] Add update event to ``OVSInterfaceEvent`` class","message":"[OVN] Add update event to ``OVSInterfaceEvent`` class\n\nIf the Interface \"iface-id\" is added in an update event, the event\nhandler will also register that in the HWOL extension.\n\nRelated-Bug: #2006603\nChange-Id: I699dd30d58e3b02713776393c497cf9fa3793185\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/34ea8988d6cdc3d676968d57b57525a2515d6128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/34ea8988d6cdc3d676968d57b57525a2515d6128"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
