)]}'
{"id":"openstack%2Fneutron~880943","triplet_id":"openstack%2Fneutron~master~I9499678fe244187977576cece101e28aa359a0d7","project":"openstack/neutron","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-05-02 10:29:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9499678fe244187977576cece101e28aa359a0d7","subject":"Remove creation of DHCP port in the OVN metadata unit test","status":"MERGED","created":"2023-04-20 11:16:44.000000000","updated":"2023-05-02 10:31:40.000000000","submitted":"2023-05-02 10:29:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880943-secure-rbac","meta_rev_id":"4166702635feeeaa747477cb594f44bb94a08054","_number":880943,"virtual_id_number":880943,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-04-20 13:27:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-02 10:29:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-04-20 11:21:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-23 22:42:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-23 22:42:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 11:21:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-04-20 11:57:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 13:27:06.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-23 22:42:19.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"c3ae99659b8940739686af9e67d297e351c47d56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-04-20 11:16:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6471449ddeb59a149ad256208bea339b2e277353","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-20 11:21:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"84fe1784efab18aaf083d79fd4a29939fb4c0d82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 11:57:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cf3df9d5cfe4a299653b8dc215d51c0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2242572f3ba45018834029a54b1488c : SUCCESS in 37m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fe3a6eb885c459bbba8ce0ae93afe27 : SUCCESS in 12m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37d4ade721224fb2b03f531538c9f484 : SUCCESS in 26m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99c6598ef3eb4d1297d73734096b6497 : SUCCESS in 23m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67cfaa767cad428fab8226f03d85b9bc : SUCCESS in 26m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c1a3f24ab146499d797c5d227bf04b : SUCCESS in 11m 37s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/721e64ec468442aa82b7894abf0842d6 : SUCCESS in 24m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f31f939266a8bd2da4fa2923ba42d8f5e6226d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 12:08:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cdb4e6498af4d9e88f7f95867395ef7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5fed6b91158e48ea92fe78c79c932e32 : SUCCESS in 19m 32s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4057b3e97bda422c8eee3e91684eea6c : SUCCESS in 49m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f9650d81b5e243b8a0073956be529028 : SUCCESS in 23m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a3ffd192481e15576b5c8ed6bb8f5e6f321c3cf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-04-20 13:27:06.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2023/Arista_Tempest_Testrefs/changes/43/880943/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"78a46328c57b975295eb924a6bc547be7c81dca5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-04-23 22:42:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"91a63171dd2237a02d47c4e114a08a52b8671614","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-25 09:21:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb52fef5f0f76af1ab885a9ce8d5650f0dfa40e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 09:57:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab12f5a2c5f947ddbc4f27b024b2a94b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f3da096f1f8411282f2d7fd32fd325a : SUCCESS in 35m 06s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/adf768a1a3f14928b21fac74e0e613a8 : NODE_FAILURE Node request 300-0021054763 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e6ea5efa6bad4cc5a81c4d0b87ead86c : NODE_FAILURE Node request 300-0021054764 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32bd345d4c62e5bc58425b75fa22b072551ce3ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 10:17:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bb6e97a65844835a3f5c87d907a4f1c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c9d9e3e99494b81be7c3175ac7fd5c2 : SUCCESS in 54m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fafe48bbebf461eb093e8d39969c840 : SUCCESS in 17m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5da49629fde044dab7bcb6464a9fe4e3 : SUCCESS in 26m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e00174f12684f28a33f66766746068f : SUCCESS in 29m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d9f1e3498d946b18d5769e36639b422 : SUCCESS in 25m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9419805a4cdc4a86a726c8d7b7dc2d6f : SUCCESS in 10m 26s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/1d7d189cc1f14633bc74fa5fa6067da5 : SUCCESS in 30m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ef2ccad2cae47da5a23447f773eee09876cf58b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 06:51:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef325f65f41d25c71e9df2ae495246c8fb5475a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 10:29:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/918cf9485ca9486e8ae613d9ad5a1571\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51d920c0d177474bb0a7e0d559dbbd10 : SUCCESS in 13m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f59148b8ed347f988e33bee3eea58cb : SUCCESS in 25m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba1ecd1af0c9404e8273d4cd4c2242fd : SUCCESS in 29m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2fa6cf5a1c2544c297f51e945dcf443d : SUCCESS in 22m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db3efd72c5ea412e95c3246edd08c07d : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":1},{"id":"693c6efb2a88bbee269b6023e5b8bdcd396b4fd6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 10:29:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4166702635feeeaa747477cb594f44bb94a08054","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 10:31:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/592911270477403bb72c021ce652f2f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6900fe1ee3b4936b3410cd22a1b35ea : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05ba4257dec5e80565aedb49685d8784424be461","revisions":{"05ba4257dec5e80565aedb49685d8784424be461":{"kind":"REWORK","_number":1,"created":"2023-04-20 11:16:44.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/43/880943/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/43/880943/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/880943/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/880943/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/880943/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/43/880943/1"}}},"commit":{"parents":[{"commit":"5db17654b99f1ebabc87b65a44c1b02133558fd9","subject":"Fix servicetype unit tests module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5db17654b99f1ebabc87b65a44c1b02133558fd9"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-20 11:13:47.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-04-20 11:13:47.000000000","tz":120},"subject":"Remove creation of DHCP port in the OVN metadata unit test","message":"Remove creation of DHCP port in the OVN metadata unit test\n\nIt\u0027s not needed to create such port manually as it will be always\ncreated by neutron when subnet is created.\n\nCo-Authored-By: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\n\nTrivial-Fix\n\nChange-Id: I9499678fe244187977576cece101e28aa359a0d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05ba4257dec5e80565aedb49685d8784424be461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05ba4257dec5e80565aedb49685d8784424be461"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
