)]}'
{"id":"openstack%2Fneutron~880966","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I2dddc296826e5ab5e057c32a554e353577cc36e8","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1953165","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2023-04-24 01:54:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2dddc296826e5ab5e057c32a554e353577cc36e8","subject":"Change flag check order in wait_until_address_ready()","status":"MERGED","created":"2023-04-20 12:52:12.000000000","updated":"2023-04-24 01:55:50.000000000","submitted":"2023-04-24 01:54:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880966-bug/1953165","meta_rev_id":"7c890e3933ee6627a16e2c33018756b52700f201","_number":880966,"virtual_id_number":880966,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-24 01:54:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-04-21 11:38:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-04-24 00:17:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-04-24 00:17:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 12:52:12.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-04-20 13:20:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-20 15:41:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-21 11:38:26.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-04-24 00:17:53.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"b10e37f19ddc00ff0638057076022b5943a9cde4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-04-20 12:52:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b1174bab69c1d8f05160af0bc976a2fdc97af5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 13:20:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47527f17a98b4716a1226bf0ae3ccd9c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49cfc8b4430e4f729b6f2269d21e837b : SUCCESS in 23m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/644df3329c9044169837053670e75d03 : SUCCESS in 21m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b34e107df6a89a27290d9754b5c24b4a8fd8cedf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 15:41:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1af217d76e284dedbb4da453242c045f\n\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/9acf9e6fbe8b409188b2bff507373356 : SUCCESS in 2h 09m 10s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/273b8a969406402197a9ba5237b7a2d1 : SUCCESS in 1h 59m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/7ecdd3620b3f4c0295e7d5bf1ae0afff : SUCCESS in 1h 39m 36s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/6426f1825fb54739bc7b8dafad768fa0 : SUCCESS in 2h 45m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c533ea6c923f4aea93f4c3857a8af45e : SUCCESS in 39m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d4c8fe21e1f4f97963cee6016e4de17 : SUCCESS in 12m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f959820ae6da4ec3a7bc8ddcd7b2e027 : SUCCESS in 32m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74d1b26ddfc94912a459ccb083ebda9b : SUCCESS in 32m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1ecbc765aba435d94fd230342125129 : SUCCESS in 32m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ca6c2e861a4b21a90e2197531384fe : SUCCESS in 8m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8fb19b50def446f3b86b9b1a2dbec0fa : SUCCESS in 53m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1da44735679445fb9955858bb0ca71c2 : SUCCESS in 1h 17m 17s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/b76316f019bb4b4a97863c177e133c96 : SUCCESS in 1h 15m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d2cb79208de64dc2901b51e74a75585a : SUCCESS in 1h 10m 03s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/252c20203cf844ed988f36bdb96b779a : SUCCESS in 1h 01m 37s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6fae3247217c4faaaadbb374be239052 : SUCCESS in 1h 23m 04s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0b438ced5f484f2fb2ae52deb961c142 : SUCCESS in 1h 11m 49s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f64c1b5b62644933989aa56f3dae7bce : SUCCESS in 1h 34m 14s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c9e1fd790d44a48a2e3ad2c4add078e : SUCCESS in 56m 35s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7c672866669c4f6990c867d95a84e2a3 : SUCCESS in 1h 02m 45s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/83672fb6e702468184329c1c8c6deed3 : SUCCESS in 40m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/35138b2ea22b4a839c72adcb5bfe7437 : SUCCESS in 1h 27m 07s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8ce21416d86647928f943bcd8d76e646 : SUCCESS in 1h 24m 12s","accounts_in_message":[],"_revision_number":1},{"id":"edd3c0ba7367682b8126a30b7eb952533cdc13fb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-04-21 11:38:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"15faa4edc4f8e7d916615636c294e03c1783fb85","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-04-24 00:17:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b03bb26ed611fc89451a975cfbad5b89692b115c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 00:18:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca22e811f0bf20e276942d95a16e5e44e79c7666","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 01:54:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcf9cc3b40a146a8b84e10925440e7a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d42bf6108d34a4093f88549a52d96e3 : SUCCESS in 13m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b93dcb257641481ab113b46558e1f9a7 : SUCCESS in 29m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9403156fe8cf47d4be2a69f6290b90cd : SUCCESS in 28m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46d7f54e93d543a28cb02dd6b0c00d71 : SUCCESS in 9m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/345033317f064729b7c54dee2e5e8ce6 : SUCCESS in 52m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a0f726dfe804b149d3ef517360a1ada : SUCCESS in 1h 28m 49s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e7f54b62a0094444af1e0efbb4f134b0 : SUCCESS in 1h 19m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/75245f2611b84fa59fcbda965a295505 : SUCCESS in 1h 03m 12s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3a2818a543514ee58b087ae54f985fb0 : SUCCESS in 1h 31m 39s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca5f1bdef5fe405cac4f6b39e18d8b21 : SUCCESS in 52m 36s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3a90c39233a94bc5b0e17770f90cfaed : SUCCESS in 45m 24s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/34bcaaba6dc74cf7b17dabb3778957cb : SUCCESS in 41m 02s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/d618f5b63c8b46f68629f1e8512881c1 : SUCCESS in 1h 18m 19s","accounts_in_message":[],"_revision_number":1},{"id":"679ea69545f9b950b7c8b1ab04715e5fa2dc14b9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 01:54:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7c890e3933ee6627a16e2c33018756b52700f201","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 01:55:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5e92eb182c349f09b09114d4ceaec24\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c637ac626d842b8be75ebb7d48b687e : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41e13d6055d4a47cbd2ff01cfdbbfbe0087bd6e9","revisions":{"41e13d6055d4a47cbd2ff01cfdbbfbe0087bd6e9":{"kind":"REWORK","_number":1,"created":"2023-04-20 12:52:12.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/66/880966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/880966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/880966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/880966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/880966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/880966/1"}}},"commit":{"parents":[{"commit":"541d969a7d05281a567c98362ca3d5cfd66257f5","subject":"Merge \"Fix concurrent port binding activate\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/541d969a7d05281a567c98362ca3d5cfd66257f5"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-01-20 21:59:26.000000000","tz":-300},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-04-20 12:51:11.000000000","tz":120},"subject":"Change flag check order in wait_until_address_ready()","message":"Change flag check order in wait_until_address_ready()\n\nWhen DAD fails on an IPv6 address, both the \u0027dadfailed\u0027\nand \u0027tentative\u0027 flags will be set. So change the code\nto check for \u0027dadfailed\u0027 first, just to be explicit.\n\nAdded better unit testing to cover more cases as well.\n\nTrivialfix\n\nChange-Id: I2dddc296826e5ab5e057c32a554e353577cc36e8\n(cherry picked from commit 1d22cfa99a529933d436a449f1971a96c95f2b66)\n(cherry picked from commit 0c18181ee9afc09b209f3cae0eec098095ae0ca9)\n(cherry picked from commit e1bc6bb273b4840cf5c20db029d4984c16548d5e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41e13d6055d4a47cbd2ff01cfdbbfbe0087bd6e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41e13d6055d4a47cbd2ff01cfdbbfbe0087bd6e9"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
