)]}'
{"id":"openstack%2Fneutron~880967","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I6b544c5528cb22e5e8846fc47dfb8b05f70f975c","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1953165","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2023-05-04 18:52:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6b544c5528cb22e5e8846fc47dfb8b05f70f975c","subject":"Suppress IPv6 metadata DAD failure and delete address","status":"MERGED","created":"2023-04-20 12:52:37.000000000","updated":"2023-05-04 18:54:22.000000000","submitted":"2023-05-04 18:52:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880967-bug/1953165","meta_rev_id":"04b5c53fc84a31774b4aebe39fc9fcf950503576","_number":880967,"virtual_id_number":880967,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-04 18:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2023-05-02 18:15:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-05-04 13:10:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2023-05-04 13:10:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 13:18:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-20 15:54:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-02 18:15:38.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-05-04 13:10:23.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"ea7f34b5d0d5dfb6230d86b6a8d241015781c36d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-04-20 12:52:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fec9249bacd8dc1298b6c295999ccba896c9dab3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 13:18:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adb3082702e64ec5932d4a1e511c35d6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d17e91af1f1d44f6865129fceebdc43f : FAILURE in 20m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46edf672c13142c09f282b0324949aee : FAILURE in 21m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3969e3a27e7055eb54dfe2f9860cae71d292a38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 15:54:10.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/864cc4e1cc2943d2a43fbcabfb0ce8f4\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/3ae43d978bbf4eb29bc02b4e668ff30b : SUCCESS in 52m 04s\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/9e4bc2a31d2b422cbe6988bed988d829 : SUCCESS in 2h 50m 42s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/6778ffdd41c94e31ac00d9631121a4da : SUCCESS in 2h 34m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/69d5654b1eb549058067c65dccb284af : SUCCESS in 1h 37m 03s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/ad88459e39624aa6b8507d6e8ec9f440 : SUCCESS in 2h 10m 18s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/a502dd73034a4d2dbfd68dcce7c1975b : SUCCESS in 35m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62a9059c919b46a5b4caee0091aefb88 : FAILURE in 38m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a42a45d041f54158901e694230c11307 : FAILURE in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02790ae5f4b347b58050e38baca61761 : FAILURE in 36m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e63fdfa36fd543c09b09c40939b80f27 : FAILURE in 28m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a4b5cc0d5d44ffbaad9d6d0461f1752 : FAILURE in 23m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9adf9774a1324d4b98083c78eeaf43e6 : SUCCESS in 11m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d2786ea9b0f4d38ad824b09c7774bda : SUCCESS in 11m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0e7dea84cf5a45628e65fce04c47e3e6 : SUCCESS in 41m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a316dc03d5ec4fe6997cee35ce7066ba : TIMED_OUT in 2h 33m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/985d40a677bc47e982cb887bc7927087 : SUCCESS in 1h 33m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/508f9ae30c944dad819b49acb6a9661b : SUCCESS in 1h 09m 20s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c242de8f808441dd857945e7644a91e1 : SUCCESS in 1h 03m 51s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a98c702549e64aebaa77cabe607d5347 : SUCCESS in 1h 09m 01s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/22657824949249d2bc1c623063c1243f : SUCCESS in 1h 10m 35s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6caf0bc9dbe4fb2aef7b74f5a5ea101 : SUCCESS in 1h 39m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a4e6297329e48ba80967f71e567bb0f : SUCCESS in 1h 00m 54s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/143b0da09dd449ea8c71c866bdaa361d : SUCCESS in 1h 00m 42s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/5d9ec06f30264087b70400fff57af776 : SUCCESS in 35m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f105c8fe19b841dba09ac61cd8f3435d : SUCCESS in 1h 22m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/c9b4460c92814277ad14eee0af09250b : SUCCESS in 1h 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"440cca7c8b939052bd1c5a79d59c7ce9281b4bf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-05-02 13:41:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bd32193103e733cffc33c029e1bf1303a43b0829","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 14:14:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34def85b296d4251964fa6a8be9ecf09\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7326c381f3d849eeb03e91410c65f6db : SUCCESS in 27m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91f78af7347045ffa07b7902150d7a1d : SUCCESS in 26m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b093369c3723bdb050b33ab821155a691d68710f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 16:50:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2f61cd58ad94063827af6506580281d\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/dde0eeb4b0fb403a8c43c2b31d0f40a7 : SUCCESS in 1h 02m 55s\n- neutron-tempest-plugin-scenario-linuxbridge-wallaby https://zuul.opendev.org/t/openstack/build/c06e323cd4af4e178052994ae50385ae : SUCCESS in 2h 44m 17s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/c444ca16dbe84c3d850a132f4dfd3797 : SUCCESS in 2h 06m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/564557dbbe814e74ab769c2b1956fe3e : SUCCESS in 1h 46m 49s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/1f017a4954a14eb8886ee7c284d636e9 : SUCCESS in 2h 09m 39s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/e498464c6da34e8195cfc0f3f05c35e7 : SUCCESS in 45m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1db95050977416fa59b726c99a8a9eb : SUCCESS in 39m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42ac731096ad4823956caceae3143d58 : SUCCESS in 13m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fec91271bfb47c5b7a234329d834ddc : SUCCESS in 30m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a869ba59aa6749c19bc82324d63894ef : SUCCESS in 30m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9dd768def5f34ce3ba3bd1b193d2319a : SUCCESS in 30m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a715e39a2d9423695187abc2e6d9533 : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e65c7de7a27840abba1cdf76ff299ef4 : SUCCESS in 13m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/10d7b1d314e34b31848a275cde013399 : SUCCESS in 56m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/58896b0c658a4ed0a560cacc5b4b97b3 : SUCCESS in 1h 18m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/cde7fbecfe4a4496835160a5a7fa4439 : SUCCESS in 1h 34m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3afb91ed76d54f7ba1b89b471bc6f5ff : SUCCESS in 1h 15m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/221072d1446a4d3497e5805c38fc7b48 : SUCCESS in 1h 00m 17s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/39bde37bf8f74710afc658d6925afc3b : SUCCESS in 1h 09m 33s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/448b18f67900453aa133b8678ca49ece : SUCCESS in 1h 59m 06s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/20024fd082f94cbf9a439f8888fce39a : SUCCESS in 1h 46m 16s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b7e0ed0999c47fcaa81383236bca845 : SUCCESS in 1h 44m 19s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/eb6189296c40410aa5e27635804858ef : SUCCESS in 1h 12m 48s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c162dd0dfba64222abeba81b12799011 : SUCCESS in 45m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ddb6996f77fb49e992144e6ee5fe1203 : SUCCESS in 1h 42m 57s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1f5fbe99e7cd4e0592859c5808dcf707 : SUCCESS in 1h 24m 27s","accounts_in_message":[],"_revision_number":2},{"id":"96fb286598c121c80f063f89882fdb1bd4110b25","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-05-02 18:15:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"000c42653d3036855b60122d43ae894c7e5d0a6c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-05-04 13:10:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"547b2ac9c3407a5aa2597ef8d56e2023510ce114","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 13:11:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7bc0b7d45c50aa99b7165e1810564b6d85373d40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 18:52:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bbc5483337a4c6ba3e80383d0973a3c\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/e2a83187bffb44a88d750ab7ba82a7e5 : SUCCESS in 1h 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5aca8e13f0d4077a3c53dbaea1be6c6 : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6961686afc3430d82df630d1b3e9c91 : SUCCESS in 32m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90890b64050d4ff9ae6e75e39ff08eb4 : SUCCESS in 34m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d80c097a9014412e873c319c286e387a : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a370b048a0f40d2b8c7980730533b6b : SUCCESS in 13m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d61987ee5aa74bb4af9d0d549e6015f5 : SUCCESS in 1h 02m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cd60c86cc3a84412990d484ea829daa4 : SUCCESS in 1h 19m 16s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c82615180e2d4937b0815ea41aca95e9 : SUCCESS in 1h 20m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d61f2868344401eb100226fae13376e : SUCCESS in 1h 35m 13s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7678f0f7f89e4ba4a187c4a33f896c0c : SUCCESS in 2h 01m 56s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/859cd0bd9c0c4fdd9ea196f2c6d98626 : SUCCESS in 1h 40m 12s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/97026e0b86f5490b8c10d124c5986e18 : SUCCESS in 1h 13m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e651e61748a84746bfad03d3196f24b7 : SUCCESS in 39m 46s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e27fdc1b081a40eb834686a16c21c765 : SUCCESS in 1h 40m 15s","accounts_in_message":[],"_revision_number":2},{"id":"1e1bd3649c3fc99be486e5624bd1c649e0d9a752","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 18:52:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"04b5c53fc84a31774b4aebe39fc9fcf950503576","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 18:54:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1baff149b1441c9ba20217985ef2d51\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a40ceb7b598d4ee5ad6dbe2b9780b2e7 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2784601147184835b8a082cb3bf7aa93 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f53cff4a9c57bb39db8baf3f4a41ade085af98b4","revisions":{"32a150d4bf71d879696ff308c29647a4c84fa810":{"kind":"REWORK","_number":1,"created":"2023-04-20 12:52:37.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/67/880967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/880967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/880967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/880967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/880967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/880967/1"}}},"commit":{"parents":[{"commit":"41e13d6055d4a47cbd2ff01cfdbbfbe0087bd6e9","subject":"Change flag check order in wait_until_address_ready()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41e13d6055d4a47cbd2ff01cfdbbfbe0087bd6e9"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-03-06 12:04:01.000000000","tz":60},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-04-20 12:51:14.000000000","tz":120},"subject":"Suppress IPv6 metadata DAD failure and delete address","message":"Suppress IPv6 metadata DAD failure and delete address\n\nIPv4 DAD is non-existent in Linux or its failure is silent, so we\nnever needed to catch and ignore it. On the other hand IPv6 DAD\nfailure is explicit, hence comes this change.\n\nThis of course leaves the metadata service dead on hosts where\nduplicate address detection failed. But if we catch the\nDADFailed exception and delete the address, at least other\nfunctions of the dhcp-agent should not be affected.\n\nWith this the IPv6 isolated metadata service is not redundant, which\nis the best we can do without a redesign.\n\nAlso document the promised service level of isolated metadata.\n\nAdded additional tests for the metadata driver as well.\n\nConflicts:\n    neutron/tests/unit/agent/metadata/test_driver.py\n        conflict with f430cd00725f8303f5313cb7784c9aed4b982e62\n\nChange-Id: I6b544c5528cb22e5e8846fc47dfb8b05f70f975c\nPartial-Bug: #1953165\n(cherry picked from commit 2aee961ab6942ab59aeacdc93d918c8c19023041)\n(cherry picked from commit 071255f098e0e73fd5220f83cbbc8ac1c421f3ab)\n(cherry picked from commit 5337c3ca7fa2b1d078940823bfc07bc275ca5824)\n(cherry picked from commit 707a09bfa83f3298bdd082b4733d5db6f6470654)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32a150d4bf71d879696ff308c29647a4c84fa810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32a150d4bf71d879696ff308c29647a4c84fa810"}]},"branch":"refs/heads/stable/wallaby"},"f53cff4a9c57bb39db8baf3f4a41ade085af98b4":{"kind":"REWORK","_number":2,"created":"2023-05-02 13:41:02.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/67/880967/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/880967/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/880967/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/880967/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/880967/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/880967/2"}}},"commit":{"parents":[{"commit":"a4fa664d919dc13cc5f865114e19b45583951215","subject":"Honor debug mode in keepalived-state-change script logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4fa664d919dc13cc5f865114e19b45583951215"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-03-06 12:04:01.000000000","tz":60},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2023-05-02 13:40:00.000000000","tz":120},"subject":"Suppress IPv6 metadata DAD failure and delete address","message":"Suppress IPv6 metadata DAD failure and delete address\n\nIPv4 DAD is non-existent in Linux or its failure is silent, so we\nnever needed to catch and ignore it. On the other hand IPv6 DAD\nfailure is explicit, hence comes this change.\n\nThis of course leaves the metadata service dead on hosts where\nduplicate address detection failed. But if we catch the\nDADFailed exception and delete the address, at least other\nfunctions of the dhcp-agent should not be affected.\n\nWith this the IPv6 isolated metadata service is not redundant, which\nis the best we can do without a redesign.\n\nAlso document the promised service level of isolated metadata.\n\nAdded additional tests for the metadata driver as well.\n\nChange-Id: I6b544c5528cb22e5e8846fc47dfb8b05f70f975c\nPartial-Bug: #1953165\n(cherry picked from commit 2aee961ab6942ab59aeacdc93d918c8c19023041)\n(cherry picked from commit 071255f098e0e73fd5220f83cbbc8ac1c421f3ab)\n(cherry picked from commit 1c615281f7632f3f1cf4bd37eefe90c50c6dfe25)\n(cherry picked from commit defb6018f3a395094cc85a03b93a2a0b43d2f6ff)\n(cherry picked from commit 1d674825ebbe5fcab6c8fef7d03b5cf9b332b743)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f53cff4a9c57bb39db8baf3f4a41ade085af98b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f53cff4a9c57bb39db8baf3f4a41ade085af98b4"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
