)]}'
{"id":"openstack%2Fneutron~882121","triplet_id":"openstack%2Fneutron~stable%2Fyoga~Ia5380bc78cc1d0136e11cc4692069279419e285e","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/2018000","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-05-04 16:09:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia5380bc78cc1d0136e11cc4692069279419e285e","subject":"Avoid retrieving ports if network list is empty","status":"MERGED","created":"2023-05-03 07:27:35.000000000","updated":"2023-05-04 16:11:04.000000000","submitted":"2023-05-04 16:09:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882121-bug/2018000","meta_rev_id":"cf1e369d202c7b5659784138d8d0443de1dc1a23","_number":882121,"virtual_id_number":882121,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-04 16:09:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2023-05-03 20:53:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-05-04 12:08:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2023-05-04 12:08:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-03 07:53:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-03 09:51:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-03 20:53:19.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-05-04 12:08:42.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"bc09fb1e4f6e3bf226844e3ef08b6e4c41761d8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-03 07:27:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56ef3b19e325dd689e689d3c1b9fc399c4e15818","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 07:53:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fcf6d45988d45bc95645067dd70bd87\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4897c9dbb2484922ab008418be500636 : SUCCESS in 24m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0042878f2c89417b972bfce97657737b : NODE_FAILURE Node request 300-0021120958 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2bf00cda8be0af6f7677589afb22186b6f523dcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 09:51:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/22744fb11097434b8426e334d45fe625\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/e0ec6e81dba84bb69329bf8003eb8b5b : SUCCESS in 1h 13m 28s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/e1b17016c58742b6b0e82273286714cf : SUCCESS in 2h 18m 38s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/ccec57487ee84537a304cf24511d4b8d : SUCCESS in 1h 23m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/88cfc6fa0b7247d8b9f76e33d68907e8 : SUCCESS in 1h 38m 08s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/a796450da0ae4adcab6ad08251f66dfb : SUCCESS in 1h 31m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/063dd3f8084a4311a770a38813cc19c9 : SUCCESS in 39m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/596c16a10db24a49b1171d9ace1cd237 : SUCCESS in 12m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d1e520bf1994ddc9ffa175eb2e7dd39 : SUCCESS in 35m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c628de87dd9e4797b879067db1e12bf5 : SUCCESS in 29m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/736f7089e1964982826394fa68c61f24 : SUCCESS in 8m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b53f6a79f5cf402280418ca1fdb8cb75 : SUCCESS in 1h 00m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a74845b5af4e415fb71ee350b6425b68 : SUCCESS in 1h 37m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c60da5d15c374b9badc8fc802ff54cec : SUCCESS in 1h 19m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9627881b09634b729194e6787fc2882d : SUCCESS in 57m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ec2c6eb0953c41489d03cdcac9f5af7e : SUCCESS in 1h 07m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8ed85eaa8c6247e1bb48a0efed2d4634 : SUCCESS in 1h 16m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d19ce4c059d045a79a726a892f2de1f5 : FAILURE in 1h 22m 15s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0feb62b835b84dbd8240417deb7f2eec : SUCCESS in 1h 22m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/65cd6354e4484ee49573e02c98b529ba : SUCCESS in 1h 39m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"553d87a16739782f2cec61c3989859bb44e79d1c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-03 13:08:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c0d2a3c86a097a2133754e31b2d425d69624951","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 14:26:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05d7937776c34056a7b1a9f31d7f5951\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/371c19b127dd4e8084f8fa1a89d61700 : FAILURE in 1h 14m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c51bf8e4a50145df985b337281c34ca9 : FAILURE in 1h 16m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd6679ef95124d7080f52137ed7d61b7bad282b5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-03 14:47:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4bac2bc23910dfa8bf92756435bf6be5dce5f54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 15:13:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef06d55bc1e74db7bab995546f72371a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/774e25163b4745d4b35281ec8df589bd : SUCCESS in 17m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15a60e5f88164658b0c55490a93defad : SUCCESS in 25m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c516ff34f5ef75bb7f3e2d75a2ad1d0141f29a50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 15:17:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98f8808f2eb342c0b4446e9422d3544b\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/bbd6934897aa427ea86d4d614711fb7d : SUCCESS in 1h 04m 11s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/0c18bb206951424fa4faf18bbc29eb84 : SUCCESS in 1h 25m 03s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/d041e3aeb4d14593a8c4da8aa350afc5 : SUCCESS in 2h 02m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/241a0963ab27412da195b0ff083d74bb : SUCCESS in 1h 37m 15s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/ecc0dff77c48449a9cbee858173caeb3 : SUCCESS in 1h 31m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/209f29e52de448039e0002ed01d6fa5b : SUCCESS in 35m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83ba2a303ff6485cbdc5b21306119abf : SUCCESS in 13m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9518a158c30d4825b2dab458a6772b4a : SUCCESS in 29m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4144a312ffdc4f5186242d33fc658d1f : SUCCESS in 25m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8604164500e146f18ec7d1413e4cb35f : SUCCESS in 9m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/319ed3cef8cb4e7aa9207d7d985e7344 : SUCCESS in 44m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bcba222a61544ce7aeda09937155e4ed : SUCCESS in 1h 23m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e1d1e4f3ccba4d9aafe8ea07c7c49266 : SUCCESS in 1h 04m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6cb3645b2fb94df591cfaafae05053e1 : SUCCESS in 1h 05m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2f7b3ea3ec594dc6923a4cea0baee94e : SUCCESS in 50m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7df173d3339b4699a2004359d57fff34 : SUCCESS in 1h 20m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ef66eeeb2a98481397daf9a8860e151a : SUCCESS in 1h 35m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/102332843f3543eca31207952d453ef1 : SUCCESS in 1h 17m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0acc4ea1436444f1857f958d37a231d2 : SUCCESS in 1h 09m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"76cdfc0ac0f0ffb37dd2a834d7273083f05267f8","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-05-03 20:53:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a41bf018b70e4192d770e579a82f6e536de4b1fa","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-05-04 12:08:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"08698ce4cb2d9bb1bd911cf0ce7fe6ed34f51096","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 12:09:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ec45c5b7b0196d3fd38d62d1bfd92ad3eecc5c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 16:09:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a2dd4be07b14c40a3df1f1d28d3ed8c\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/2e02677109ba45e78e0eae3b36c07ef5 : SUCCESS in 1h 18m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02f5d3699da946a6b050224a3934e381 : SUCCESS in 13m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c962f9367c145138ca7cfd1ec472d2a : SUCCESS in 26m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51f5dc1d50c244ff9f4a22803dd0d4fa : SUCCESS in 31m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f588713b16429298affe922c99fde7 : SUCCESS in 13m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c748877b2c6469f864d776e57e52ee9 : SUCCESS in 49m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/51355d1c589f436ea54b26d4cb061112 : SUCCESS in 1h 29m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7b4bc2256fc643fa8dcb8c94215b4902 : SUCCESS in 1h 11m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4d3cb46cc6f4057bb15be07e6383326 : SUCCESS in 57m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/99f6894620c4452f8f920ac00a75541e : SUCCESS in 54m 12s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d97f1422f16a4c7e9045e1214ed96328 : SUCCESS in 1h 17m 07s","accounts_in_message":[],"_revision_number":1},{"id":"8c0d0c62526cd7bfa8117ad6e3ef0b40cebab34f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 16:09:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cf1e369d202c7b5659784138d8d0443de1dc1a23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 16:11:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b55aab7978a4fa180f5d45aea270a01\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c373a67553bd4934af4e34e0c08740f2 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"995c465539a9ea336b8623b4849f41f58f8318ef","revisions":{"995c465539a9ea336b8623b4849f41f58f8318ef":{"kind":"REWORK","_number":1,"created":"2023-05-03 07:27:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/21/882121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/21/882121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/21/882121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/21/882121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/21/882121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/21/882121/1"}}},"commit":{"parents":[{"commit":"28dd08a40e593227f88271d062b82938a6aca7bb","subject":"Merge \"OVN: Always try and create a metadata port on subnets\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28dd08a40e593227f88271d062b82938a6aca7bb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-04-28 14:34:22.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-04-29 11:08:55.000000000","tz":120},"subject":"Avoid retrieving ports if network list is empty","message":"Avoid retrieving ports if network list is empty\n\nIn ``QoSPlugin._get_ports_with_policy``, if the network IDs list is\nempty, the SQLAlchemy throws the following warning:\n  SELECT statement has a cartesian product between FROM element(s)\n  \"subports_1\", \"ports\", \"portnumaaffinitypolicies_1\",\n  \"testportextensions_1\", \"portuplinkstatuspropagation_1\",\n  \"portdataplanestatuses_1\", \"standardattributes_2\",\n  \"portdeviceprofiles_1\", \"ml2_port_bindings_1\",\n  \"portsecuritybindings_1\", \"portdnses_1\",\n  \"securitygroupportbindings_1\", \"qos_network_policy_bindings_1\",\n  \"qos_port_policy_bindings_1\", \"trunks_1\",\n  \"standardattributes_1\" and FROM element \"networks\".  Apply\n  join condition(s) between each element to resolve.\n\nThis patch avoids this query by checking the network IDs list. If the\nlist is empty, the expected port list will be too. This is also a\nsmall optimization because we are skipping the port query.\n\nThis patch is also applying the same logic to the second query in this\nmethod.\n\nCloses-Bug: #2018000\nChange-Id: Ia5380bc78cc1d0136e11cc4692069279419e285e\n(cherry picked from commit 9ac59e4b4a3b40861308c71bbacc379dacb2bd20)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/995c465539a9ea336b8623b4849f41f58f8318ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/995c465539a9ea336b8623b4849f41f58f8318ef"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
