)]}'
{"id":"openstack%2Fneutron~882689","triplet_id":"openstack%2Fneutron~stable%2Fzed~I24cef85ed68c893a740445707f88296d763c8de8","project":"openstack/neutron","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-05-11 18:00:15.000000000","reason":"Change was submitted"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2023-05-11 18:00:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I24cef85ed68c893a740445707f88296d763c8de8","subject":"[OVN] Update ovn meter when neutron server reloads","status":"MERGED","created":"2023-05-09 10:46:49.000000000","updated":"2023-05-11 18:01:45.000000000","submitted":"2023-05-11 18:00:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882689","meta_rev_id":"b8cf6bcade32853c503a90f942f74eb790145bf8","_number":882689,"virtual_id_number":882689,"owner":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-11 18:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-05-10 08:13:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-05-11 14:11:46.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-05-11 14:11:46.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-09 11:13:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-09 12:41:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-10 08:13:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-05-10 08:13:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-05-11 14:11:46.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"}],"messages":[{"id":"b2b066988f9ef18c7503f33afb57ab453e470f90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2023-05-09 10:46:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ca51d0af5fb74e86bcea03c0cbeb4b1f3ac7adb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 11:13:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b82d76707474e5c8d20ca08c46eb73f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1190d1b19b9345e9a8a54ccfe81bbbae : SUCCESS in 22m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/83cae1a0e1774ea5a922d6945f8bf6a1 : SUCCESS in 25m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5804eedad98596e402372af882793c0d586668c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 12:41:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f76397aa4b0403eae316511605c7cf5\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/e095b70d2a674f899f32f4bf7f02c59b : SUCCESS in 1h 53m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e977f6af850647a98ec735c5cf9cfbe3 : SUCCESS in 38m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b558697416c414b8a9cfa12a1d48f76 : SUCCESS in 14m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b2ab07dbf744ea89f5abca047311d35 : SUCCESS in 29m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7e8307c14e84c479ba5940c4147ee54 : SUCCESS in 27m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/729af24033f3454ebc93c747b0f879c3 : SUCCESS in 26m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae39d39ff0141cdbda79e738625a191 : SUCCESS in 11m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/551be7c6ed3c411f8d1aaedd2d10eb6a : SUCCESS in 54m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d70a1cd8fb074d6db6a500969a986e6f : SUCCESS in 1h 38m 47s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00a0fdcdcba64fa1b333c198de4cac66 : SUCCESS in 1h 04m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e438a8f93d2e4abf9eb7794aa7d3ac0e : SUCCESS in 58m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f421e89dd7df1103af3b53d628258d70b2564dc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2023-05-10 08:02:56.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e2bcbb6ec2620b07ce276140c5bb4583dce3fc05","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-10 08:13:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ede499010dd108104b94f8df402ae138614ebb58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 08:24:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9463b1e9bdb479da8cf35a7e1cf41a8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49fe442cbe2646c297b785ec681775ae : SUCCESS in 19m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a07d72355d814529b9856627a395c24b : SUCCESS in 20m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"818cde6dba1cc6fd39818562aa06c1d91d85a5bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 10:08:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f5b6b261d654d22835e4626a21443f2\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/1001d76dc3cb4a279974a45e6e752eb0 : SUCCESS in 2h 04m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4a37053a9f4478cbce850ddad6f6ca2 : SUCCESS in 37m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1105165dc43141489a2e5717fe2fc9c4 : SUCCESS in 14m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbbc03428ede43a28bb7243f9c02d39b : SUCCESS in 25m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1950977875f42f5908dc7ae6cb3807e : SUCCESS in 25m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4381c63dbd1340548fd316741b9dc483 : SUCCESS in 24m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02f0193f16544fa49fc873d10cd10e80 : SUCCESS in 7m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d0cb28440c84c099aaa11245a576b7b : SUCCESS in 45m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3cefd7a9844a44d6860df2c060757b61 : SUCCESS in 1h 32m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/516492edfb3147afabfbeca325ef480b : SUCCESS in 1h 02m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2e6c2acb11b341ceb34d45bb6dfeb9db : SUCCESS in 1h 15m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58a51f4a896391b579bcc4f191847f26f6833093","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-05-11 14:11:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fbe53e71c04bdef85dfc32383415f96fed6fcd35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 14:12:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"db58d5745314d5a574a5e764265f665671d3bbd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 18:00:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/394740b7fda242af8370518abbcf80ab\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/2646f1a8a45749ef8ca2641a6e15ec97 : SUCCESS in 1h 47m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d77a94f0cf65465bb4b209269a1bb7b3 : SUCCESS in 11m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbc9d581e44f462ca18d59bd2a3ea0be : SUCCESS in 26m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dca59d227144e0bb34c6852a373bf36 : SUCCESS in 26m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aad4d4c672f043ce8787ee09141410de : SUCCESS in 10m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/35f4089715214f239b681d38caefa202 : SUCCESS in 56m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1495b4b4cc2042459d59224db16e7299 : SUCCESS in 1h 48m 19s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8cde485772774b8d833c4e9d7f258207 : SUCCESS in 1h 07m 01s","accounts_in_message":[],"_revision_number":2},{"id":"20c1f9e7a02b2a08f2fa3d5dbc29f07e97e47a88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 18:00:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b8cf6bcade32853c503a90f942f74eb790145bf8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 18:01:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4c1c690b906421b984e3bfe03063372\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee649038a91f467f863b748253ff6a28 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1864dd868459384e8f19ba413929a8c196877370","revisions":{"bb9136322fa214221e83183d1a4d07458a553aab":{"kind":"REWORK","_number":1,"created":"2023-05-09 10:46:49.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/89/882689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/882689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/882689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/882689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/882689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/882689/1"}}},"commit":{"parents":[{"commit":"baee022ec586a0bf4ba2c14011c6f33a7a137afd","subject":"Merge \"[S-RBAC] Allow network owners to get ports from that network\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/baee022ec586a0bf4ba2c14011c6f33a7a137afd"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2023-04-20 14:39:12.000000000","tz":120},"committer":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2023-05-09 10:46:37.000000000","tz":120},"subject":"[OVN] Update ovn meter when neutron server reloads","message":"[OVN] Update ovn meter when neutron server reloads\n\nUp until now, we needed to remove all logging objects to see the\nmeter-band properties being changed after a server restart. Now we check\nfor inconsistencies between the neutron configuration and the OVN\nmeter-band object after a restart. The function create_ovn_fair_meter is\nnow located in the ovn_driver instead of the log_driver so as to be able\nto call it from the maintenance task.\n\nCloses-bug: #2017145\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: I24cef85ed68c893a740445707f88296d763c8de8\n(cherry picked from commit c3602ac19b62b77d3cb763746e124881d4c061e4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb9136322fa214221e83183d1a4d07458a553aab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb9136322fa214221e83183d1a4d07458a553aab"}]},"branch":"refs/heads/stable/zed"},"1864dd868459384e8f19ba413929a8c196877370":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-05-10 08:02:56.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/89/882689/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/882689/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/882689/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/882689/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/882689/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/882689/2"}}},"commit":{"parents":[{"commit":"baee022ec586a0bf4ba2c14011c6f33a7a137afd","subject":"Merge \"[S-RBAC] Allow network owners to get ports from that network\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/baee022ec586a0bf4ba2c14011c6f33a7a137afd"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2023-04-20 14:39:12.000000000","tz":120},"committer":{"name":"Elvira García Ruiz","email":"egarciar@redhat.com","date":"2023-05-10 08:02:56.000000000","tz":0},"subject":"[OVN] Update ovn meter when neutron server reloads","message":"[OVN] Update ovn meter when neutron server reloads\n\nUp until now, we needed to remove all logging objects to see the\nmeter-band properties being changed after a server restart. Now we check\nfor inconsistencies between the neutron configuration and the OVN\nmeter-band object after a restart. The function create_ovn_fair_meter is\nnow located in the ovn_driver instead of the log_driver so as to be able\nto call it from the maintenance task.\n\nConflicts:\n    neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/maintenance.py\n    neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n    neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_client.py\n\nCloses-bug: #2017145\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: I24cef85ed68c893a740445707f88296d763c8de8\n(cherry picked from commit c3602ac19b62b77d3cb763746e124881d4c061e4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1864dd868459384e8f19ba413929a8c196877370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1864dd868459384e8f19ba413929a8c196877370"}]},"branch":"refs/heads/stable/zed","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
