)]}'
{"id":"openstack%2Fneutron~882949","triplet_id":"openstack%2Fneutron~master~I5a581f7ace6c25ea40a19c23868e591ae97fce1a","project":"openstack/neutron","branch":"master","topic":"bp/keystone-v3","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-05-18 13:24:13.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}},"hashtags":[],"change_id":"I5a581f7ace6c25ea40a19c23868e591ae97fce1a","subject":"Replace \"tenant_id\" with \"project_id\" in address scope","status":"MERGED","created":"2023-05-11 13:17:17.000000000","updated":"2023-05-18 18:05:49.000000000","submitted":"2023-05-18 18:04:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882949-bp/keystone-v3","meta_rev_id":"78380cafaf6bfc2a8e0bd98db464615d42b1ae0b","_number":882949,"virtual_id_number":882949,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-05-11 16:05:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-18 18:04:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-17 10:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-05-17 10:56:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-05-17 10:56:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-11 13:40:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-11 15:37:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-11 16:05:26.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-17 10:20:40.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-05-17 10:56:00.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"73806863e19584b3c1421fd21e2e3d3236c8c5df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-11 13:17:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f32c71941ae56fd9585a3cf24db22145260fa640","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 13:40:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d6115f30feb45b3883faa1a47e58c60\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fdddd2bd260493fbb27948ecc8222c9 : SUCCESS in 22m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd81120687624499b956f0690dc901bd : SUCCESS in 21m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1ca9f94171cf4b4bbfc01b026f719e1d : SUCCESS in 20m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30da1372f333e8c4a65da2d825bb902aaf0f79d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 15:37:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/71902fcf528448dc8001d71e8f200529\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/cd8b0aaeb7c44544af38029d8e9fbed6 : SUCCESS in 2h 08m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ea5746b496e949d1841034492bb4f787 : SUCCESS in 2h 01m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b32b368ff1f94baab21ff656ccc14ad7 : SUCCESS in 2h 05m 29s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9991aecb866f40679cec7af511115f89 : SUCCESS in 1h 55m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b2a2c411dcd94080ae34c82147598f4e : SUCCESS in 2h 18m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6947b833eea64b2bb75e97ab4fb5f6c1 : SUCCESS in 35m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0d2e9855fcc4540bd4cffd830e7cfb5 : SUCCESS in 39m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dfecf8dccde4f71b2a6d836edd3ff33 : SUCCESS in 13m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/291e7201c8e543c6adf5234f519d8731 : SUCCESS in 26m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ab36869399841f3b1923669a4e4a4ec : SUCCESS in 24m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c4a2229b6464cec9fd0b8641faac554 : SUCCESS in 27m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c641fd731d1425683d70ad90a7b9b4c : SUCCESS in 10m 54s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/9c5cb129b26d43ac9a988a64d7b11d0a : SUCCESS in 24m 54s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1619d89ae95849779000239daeca0802 : SUCCESS in 1h 09m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/74ee497da1554863b1a13172db7ea51c : SUCCESS in 25m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2817b8e4c6454aed910f913b0617a75f : FAILURE in 38m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/953949d601a847c280900a4134d16499 : SUCCESS in 58m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6d79b3ee6da640709744c0126c96699f : SUCCESS in 1h 34m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7cb5ba5a129d457f999fb1b896bd9cf3 : SUCCESS in 1h 21m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/15ef6b605b694118bbea2c7c8f6810e3 : SUCCESS in 1h 12m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a9a5e12e790f4e4f90bcc421561176e9 : SUCCESS in 1h 43m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cd57ed4c8b944620b28e788f706157b7 : FAILURE in 10m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c34786166a4b497993760bebfb5062c9 : SUCCESS in 1h 11m 54s","accounts_in_message":[],"_revision_number":1},{"id":"256ea74b22c5d1c02fb0045856f0ed63789cbd78","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-11 15:46:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5950c9fa34ec50bc70bb5db87a298aadaa28a3d4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-11 16:05:26.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/49/882949/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9d0fe66642642e3249a2c87128bcdce6793b947a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 16:08:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60ac34cbe39a48f7aa888af1c002c2df\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec5ad79e247047898763f33bd974beb6 : SUCCESS in 20m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87b7c144a0ff40e78d6dcc0b31bb30d5 : SUCCESS in 21m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bb8c5368bbb6449b9cc57f7f0941b437 : SUCCESS in 20m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1a6e10ea5519254386ed9b7bfb1a1d59b35c2d7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 18:08:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce1e0264550b4c8d868b8e5644dba52b\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d2ba39b4034a4fc3bd7ecd6f899f7ed1 : SUCCESS in 2h 20m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bbc4613f8cac40e89bd0e61e5bcc21d9 : SUCCESS in 1h 37m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7b5df50c733e4b1cb6d19dd5e87f5e1e : SUCCESS in 1h 47m 46s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6b6b910d39e44d378f25ad8b4647df37 : SUCCESS in 1h 44m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3713bce83c1d42b3bcb915074605dd01 : SUCCESS in 1h 58m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/63cfb1a0302d483db85b7f4fbc13f9b4 : SUCCESS in 36m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a5b5083e350484a9c5061ba7e1a19dc : SUCCESS in 31m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/675d56ccf4374df5a3da4d133101284c : SUCCESS in 15m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ef162e0a33e4ae781440a528fd9c9f8 : SUCCESS in 29m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a68299111a2447eade00a34aa2d455f : SUCCESS in 31m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4e4695908874f0780c450cd4ec7a51a : SUCCESS in 31m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53487066754644949122bc1ac44312e8 : SUCCESS in 10m 22s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/b75ac31c5edd48d6a817f5de216f7f54 : SUCCESS in 26m 30s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/535c588ae430453ab0076f39d871181c : SUCCESS in 55m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0bae6af5f80248bead6ade1d85556ace : SUCCESS in 26m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/255b965ba2fb40a88d90b8ceaa66f3e6 : SUCCESS in 41m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/77457ff0473d4c6cb0e2d77bdb2fdc8b : SUCCESS in 50m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9da4ae5133524fc6acee55ba8f828961 : SUCCESS in 1h 29m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2f001ac5a87e4453a99aa1c198848660 : SUCCESS in 1h 18m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bd56c702c8d1428da6804b00b2e35908 : SUCCESS in 1h 07m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/202f671535e1469f855527e20bebe6ad : SUCCESS in 1h 50m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/232cb4971c544307831a819dc7acb05b : SUCCESS in 1h 39m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5c62431025464c3692f53e9917894a21 : SUCCESS in 1h 10m 48s","accounts_in_message":[],"_revision_number":1},{"id":"541c894bc80e6d78cdea476cbdad5d1320b3637f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-17 10:20:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5470192bb2c6eeb6e66fb915a8e79b904fd11a5a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-17 10:56:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8afc7b900429489ec0e8121f19ad8598bf2cef6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 10:56:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84a6d9f59971364efaa816b3e09fef5f0c96908e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 13:19:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4842e8c9122f4c97a054b9102c889e39\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e259dc4985254d8ab43dea5d791f74c2 : SUCCESS in 2h 22m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/34893494d12f4e5eb1e3a567da446991 : SUCCESS in 1h 45m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8561a2040b164f2e9754655f8cafb1c1 : SUCCESS in 1h 29m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b4598216f18f4436a33e9a9f66469016 : SUCCESS in 1h 52m 40s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6e7e9aae3e1648a9a0f8ccb8e0a723a1 : SUCCESS in 1h 29m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62ab46110f9e4aa9b9e7ff41b2786def : SUCCESS in 12m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c97b798232b4a7d8920c29313a58a07 : SUCCESS in 27m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44e01bffb50e40ba8bfadec3401b3ebc : SUCCESS in 33m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6be3a66ce2424e35b709ba90a699325d : SUCCESS in 28m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f85f246b0de14a11ad9e6d0b6044ee9b : SUCCESS in 11m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e9220f6554d4108817272e8c6fc2682 : SUCCESS in 1h 01m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/35ca67ef30914347bc2dc44bc8d38057 : SUCCESS in 28m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4f780969c55e4cb5afd18d55da531623 : SUCCESS in 1h 24m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4c60969b97d49e48e03f4c92720ddd3 : FAILURE in 36m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aee2a608d0ba474ab22e26eba17e57a4 : SUCCESS in 56m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5103156abb24471482676d60b6ea1e62 : SUCCESS in 1h 04m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/43a6c126a23a4640a9024e1a7e22e30e : SUCCESS in 1h 23m 02s","accounts_in_message":[],"_revision_number":1},{"id":"4c13f726771c19fdd255d25f052d569bb7180a6a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-18 07:38:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9225002c338410d4b2d3ddaaf6e39b2b2293313","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 08:28:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5d91f43621f4240b74f73a6d3ac7e2a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f565866b9e8d49ae93093eb6f5436449 : SUCCESS in 48m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66a892f356c24d22a8cb27cdfc46288b : SUCCESS in 20m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/baa6864745134c8c8d2bb369d781d774 : SUCCESS in 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e2de04aebbad9453783beb1327988fa9db22708","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 10:21:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a01d65cbfc924757aa21f2debfae4057\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/52f48a728bc54f52a76bc3cd027bcdc8 : SUCCESS in 2h 19m 50s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6abd24754e1546288711235c0070e488 : SUCCESS in 1h 19m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a11ce7a078ed4ce6a63a7642b92e250e : SUCCESS in 2h 03m 24s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/03de641501b8442da3704100a1f0d3a1 : SUCCESS in 2h 19m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c346c517f25942ae8f83fc7c912bd1fb : SUCCESS in 1h 55m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/18894633b74245e6b6a502a638aa62cc : SUCCESS in 44m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4b2552d779f43cdbc1808422e70d47d : SUCCESS in 37m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a002f569ea54bacb89f4f503d28de01 : SUCCESS in 12m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f4f5682f412430d85a64e60b3f5006c : SUCCESS in 29m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3de102dd80874df8bc3267e50bd7efdd : SUCCESS in 26m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dff63bb34834489eb296944c7fa9b5a4 : SUCCESS in 29m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6746a8becd524e8f9becfa235f14bf42 : SUCCESS in 11m 04s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/aa642cf5026448a498703e34ba772c11 : SUCCESS in 24m 37s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/027bcf60f701406e871bdbb5f6fc2b08 : SUCCESS in 47m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/504874ca087e49408f7d4d3cfcd060c9 : SUCCESS in 27m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/670dc002b5294b3092ada12a48fd5fe3 : SUCCESS in 1h 07m 33s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e379312653d54556bd3be5bf488713be : SUCCESS in 56m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/48033f28479b4bcdb087bea04b551634 : SUCCESS in 1h 25m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b97ed76aa40c448cba925cdcaf1e24ac : SUCCESS in 1h 24m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5a1c58f7270c4d7dba2858f3a97a27da : SUCCESS in 1h 13m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fcaa6ca7748f47af90f54722b0e8e7f5 : SUCCESS in 1h 45m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/db133ded69914220a07f3be195e2e948 : SUCCESS in 1h 55m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a837bd0431274207b4cc0693239426f5 : SUCCESS in 1h 10m 09s","accounts_in_message":[],"_revision_number":1},{"id":"466d799c559c98a4e8f90d5cb009a7cf4c0c6cc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 10:22:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f0a1d3a956665efdb6a430dbe501194c317ceab0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 12:31:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7033ba6370e74cb39be3aae2357799e8\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f3a791016a494f5ca987d44b0e5beb88 : SUCCESS in 1h 29m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a7c478fc3b4845199e3fd228f6fca337 : SUCCESS in 1h 15m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/571abd9d3821435987f3a8a6d5c71566 : SUCCESS in 1h 59m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5d16a31ad50e43f5beca532b8eacc2f4 : SUCCESS in 1h 32m 42s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/75d85d30aaa943caa6624d11a642e602 : SUCCESS in 1h 33m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7d4f4a20a79456f927ca59dc058b1ef : SUCCESS in 13m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a72e6984347b4894b41aea420aa0c1c1 : SUCCESS in 27m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3a1d9a7a83a4b6597aefd2214ca1895 : SUCCESS in 32m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9624cfe74b604defa38817a6a3186262 : SUCCESS in 27m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efc290cc9fab402390ef0b71d75d2304 : SUCCESS in 10m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f63b2858c68342deb0635a519cc67058 : SUCCESS in 1h 01m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1df4afb732e14cc38c5d2c799a04d806 : SUCCESS in 44m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/901a0165bd274eb4be465e31a66e5437 : SUCCESS in 1h 32m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2db3679d04eb43c28e4be1fca8275ff4 : SUCCESS in 1h 00m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/713e0ccdeb104e17a21caaf4e24ab27b : FAILURE in 44m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/aed06fc46bc94c48b58d38b44d32d7bb : SUCCESS in 1h 15m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6545f847a3a64b0abff5e98a0e0d277f : SUCCESS in 1h 19m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a7bebfe8df3301e1965bf8eacff0d6f29ca4c778","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-18 13:24:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"338f84e88ec52fa748a22ccdf8693507ea51880e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 13:46:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1b2dc1ffb494f3ca92bc1fa1a6a5af4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1631649b4df433da392d44833bc37b9 : SUCCESS in 21m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2cb1bbd1051941fcb06dbf78d3fb0acf : SUCCESS in 21m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/99ae622a652343d596dfadce4fc6ad01 : SUCCESS in 20m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02f4af1b6386c20d947398e1cbc67ed80e323117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 15:49:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/422c73265bcb4ebd906d12ed5d6fc61d\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/6d2da5a67958494589ab291f3ffe5b90 : SUCCESS in 1h 52m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/af025012b5cf4fa1872d620e624a5bd0 : SUCCESS in 1h 21m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/da3b266655ef45aeb2f9690ba455e14a : SUCCESS in 1h 27m 51s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5e1e81accf184767833a3419807aa78f : SUCCESS in 1h 21m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ee4bc5764bf34fd198b519abbc1a8ef5 : SUCCESS in 2h 17m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7c56962c80ef49e39df1edda68daa3a0 : SUCCESS in 31m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec38159b3ea34262902e9526544ee4f4 : SUCCESS in 38m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65c52f39a3614202bddc7359c6b9963a : SUCCESS in 13m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33ba427d3fca407f9013a886274b5683 : SUCCESS in 28m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27fa9db21f25479cb5afde9ad99c5d29 : SUCCESS in 27m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfb0a95c7de94d3b8a24297e58235a17 : SUCCESS in 29m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76d23809fd914e7a816b0e591adab138 : SUCCESS in 9m 51s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a6c3950f70d8435cad9e3deceea86e09 : SUCCESS in 24m 12s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee15c8859b7f49a29b1d2f1da94989b8 : SUCCESS in 54m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/71673e71d35540c0a69a399275be22c1 : SUCCESS in 32m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5ddae341d5524b23a0d0bb58e4a290e7 : SUCCESS in 59m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ce4307b12bc14fb2b6fe7ac9e727e7bc : SUCCESS in 55m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9549f22752a3488f9a756c5148235924 : SUCCESS in 1h 21m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c57174deef454c7fa07cfebe38ca7942 : SUCCESS in 1h 27m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d24c1d1413984568a901db37d005aba3 : SUCCESS in 1h 03m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f3d9ce7de3a845968a6c3159558d32ba : SUCCESS in 1h 46m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/56fa868275454df588ed6e593c5ca717 : SUCCESS in 1h 59m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/366444f29c484ce2821b02c657f1b87a : SUCCESS in 1h 08m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ff91c528bfd1222c817fa4d25e5e4a620c372367","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 15:49:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f262796e48b0db20cd5936fc1007bd21caf6c6d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 18:04:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9e3b06ff488406d8cb717de42113a7a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e049a4f6499643a8bc37457c4a718b29 : SUCCESS in 2h 14m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1b30567f8d6940558599e579a7fa3e88 : SUCCESS in 1h 44m 49s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/358e38b82307473c8e5293f5940b7664 : SUCCESS in 1h 51m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a11162d526d24d318b6132076e4c3667 : SUCCESS in 1h 23m 47s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e274841fe1e24b0d8155ab6169728357 : SUCCESS in 1h 26m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/094066732fea4933b4c1069d395a6ab3 : SUCCESS in 14m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24b8dbb696b54790a2b202594f626904 : SUCCESS in 33m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96e317354433402e8fb52bf0a744715d : SUCCESS in 24m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aadee80727474d0bad3f5d94b9e90f68 : SUCCESS in 25m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f52992e9046f4272b493a0637fa40b0e : SUCCESS in 10m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/addb097f7c144092b7f392b68e386f2a : SUCCESS in 1h 05m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ae063462f07492da31d72104d4ada68 : SUCCESS in 28m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/61c92ed7c3224a0c8be9e45903b43b2e : SUCCESS in 1h 22m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b05a0a9aa104558b255abb75df64389 : SUCCESS in 55m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1f41d6b56e664c5ea9781a44b359aa8e : SUCCESS in 1h 04m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/706c5cb2247f42998b54b01dc34d71e8 : SUCCESS in 1h 13m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6344a8632d714331a15d984834f30387 : SUCCESS in 1h 21m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6c326792920a82bfd080e957acfc4cc98c0ace95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 18:04:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"78380cafaf6bfc2a8e0bd98db464615d42b1ae0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 18:05:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55633ad724134cb6921315890f5b66af\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd1444c18ad94ba2851f78d478ae4267 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"363c690529fe1958d3d7710a165a7ae1a5c81ca0","revisions":{"363c690529fe1958d3d7710a165a7ae1a5c81ca0":{"kind":"REWORK","_number":1,"created":"2023-05-11 13:17:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/49/882949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/882949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/882949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/882949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/882949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/882949/1"}}},"commit":{"parents":[{"commit":"47d4ec4e99d5aae62656c88206eb6a77f70d4a8b","subject":"Merge \"docs: Deindent code blocks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/47d4ec4e99d5aae62656c88206eb6a77f70d4a8b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-10 08:49:37.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-10 09:02:18.000000000","tz":120},"subject":"Replace \"tenant_id\" with \"project_id\" in address scope","message":"Replace \"tenant_id\" with \"project_id\" in address scope\n\nThis is part of the remaining technical debt of the specs\nhttps://specs.openstack.org/openstack/neutron-specs/specs/newton/moving-to-keystone-v3.html\n\nBlueprint: https://blueprints.launchpad.net/neutron/+spec/keystone-v3\n\nChange-Id: I5a581f7ace6c25ea40a19c23868e591ae97fce1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/363c690529fe1958d3d7710a165a7ae1a5c81ca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/363c690529fe1958d3d7710a165a7ae1a5c81ca0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
