)]}'
{"id":"openstack%2Fneutron~883120","triplet_id":"openstack%2Fneutron~master~Ibd75c3526a3a8c4138c5f266448299786ccbf045","project":"openstack/neutron","branch":"master","topic":"bug/2015847","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-05-22 15:44:10.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-05-22 19:49:29.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-05-22 19:49:29.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-05-22 19:49:29.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-05-22 12:51:25.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"32238":{"account":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"last_update":"2023-05-22 19:49:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibd75c3526a3a8c4138c5f266448299786ccbf045","subject":"Revert \"Use a writer context for the online alembic migrations\"","status":"MERGED","created":"2023-05-16 08:54:06.000000000","updated":"2023-05-22 19:50:45.000000000","submitted":"2023-05-22 19:49:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"revert_of":880867,"submission_id":"883120-bug/2015847","meta_rev_id":"1f26965bda92994dcaad590c0d8657925cbfc26c","_number":883120,"virtual_id_number":883120,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-22 19:49:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-05-22 13:30:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-22 13:29:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-05-22 13:29:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-05-22 13:30:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-16 08:54:06.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"CC"},{"updated":"2023-05-16 08:54:06.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-05-16 08:54:06.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-05-16 08:54:06.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-05-16 08:54:06.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-16 08:54:06.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-05-16 08:54:06.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2023-05-16 08:54:06.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-22 09:55:31.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"}],"messages":[{"id":"1f417a535252be51872319ce46c4ff1af823e234","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-16 08:54:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ce1ee00376f3f3dded0c271c0d9a0a3973de6c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 09:17:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8c7fa2a695a437aaa8f279d31637d09\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7261fa5f49c4508b5d0a8aac93c8573 : SUCCESS in 19m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e233006bc7d9442a8c8c4cbbb5b7368d : SUCCESS in 22m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d31e6c23bebb41ca874046295091fe2d : SUCCESS in 19m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ddfbbbacbb0a5c9836b21ecd95b87a5e3885219d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-16 09:47:10.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/20/883120/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"99f4791de694c7431e46ed8e4263005f8e2270b0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-16 09:58:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7cc29268d7df3604c8a3db471db1d9e2d665a572","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 11:15:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89bbd3eb5d6c4bdabc29cab12ecc62fc\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/896248311e6c4ed8b751c41586a412c1 : SUCCESS in 2h 19m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2cf4d8e4ea734216a0227b9a0cf222d1 : SUCCESS in 2h 07m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/455af8829a5a4f7e8a061655d152c557 : SUCCESS in 1h 31m 37s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e84d42fa525a42ac8dc67a10ee0caa4e : SUCCESS in 1h 23m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8971f25dab0e44e4b0e6081c906b92ee : SUCCESS in 1h 55m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8f97edd727df401eabeb36d37b9a010e : SUCCESS in 40m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/421f6dbdd9ed4f2bb35fbe9d22a2cf5b : SUCCESS in 49m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d619761f571461296f337f011c8215a : SUCCESS in 12m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38b145afe4994eb493c5fe72c3e6b9a4 : SUCCESS in 37m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa2a490c22ae417a8b00362814cc1492 : SUCCESS in 30m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b4d82b67693449280e997ee79975b69 : SUCCESS in 31m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc78f30ffe624cdaa6de7f4f2fbd8a08 : SUCCESS in 12m 10s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/224c99b1f65f441ab5abf17e78595b33 : SUCCESS in 28m 08s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b0b3ad4d3b04e0bb6dd11b60a0bc4ee : SUCCESS in 51m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2fa783af3b224607b73a85f2f33f15c1 : SUCCESS in 1h 33m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b302e2f25be4b3889338a0cea4904c3 : SUCCESS in 51m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d379ba86f6ad4bffbbe63a5aa926f44a : SUCCESS in 53m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fa9d25bcc19b40acac85f9f09af3bef5 : SUCCESS in 1h 35m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8ae7b84ebb6b438392cf7cb8734b6788 : SUCCESS in 1h 32m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5ab35902462f4284b4a1a75657b735d4 : SUCCESS in 1h 09m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fc1e56b655194b1f803bc5ec9ba57614 : SUCCESS in 1h 37m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e857d83f14ce4823897bcd3428084f56 : SUCCESS in 1h 48m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4b383f04402b47fe982958ba614cff1a : SUCCESS in 1h 21m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e2d5772285257e8ddf64c1a7e1aa60d049e55b12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 11:15:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"35824b56cacd8a7c54d3c16278e730fcda3642ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 13:31:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00211c0f66eb4bcaab0bd12b108beeca\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9b6e03f97c7b4fc180a088dc00d8fa55 : SUCCESS in 2h 14m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dfb7e800c11a49008ea9e2edddc04359 : SUCCESS in 1h 20m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/20f1807cff124001bcb6099201d0ce24 : SUCCESS in 1h 26m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ea772b88c45045c2aa10704d4a774618 : SUCCESS in 1h 56m 44s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/833d1705379b4fe18f468e12c7dc7976 : SUCCESS in 1h 52m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ff47c1301cc4f86b0f171908444d07c : SUCCESS in 16m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebcc4cf4e852421d8a0fd36523ac53f8 : SUCCESS in 30m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c128296476054c1085deb4c87f31f4fb : SUCCESS in 28m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cb8ee1123744cd9bf33c3ba6a6d8693 : SUCCESS in 31m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31ffa08152b04a14b90d540cb963d197 : SUCCESS in 13m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5ca4c5bea2648f9b3a352c6565789a9 : SUCCESS in 45m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/948aafa897d948b79acdcd45c38215d9 : FAILURE in 1h 14m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/532691da90164fdb8d40c3219df5dac9 : SUCCESS in 1h 13m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fbb9b0becddc44d3b2a9207abab18d19 : SUCCESS in 47m 30s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1805f054d480435cb84358fec1d6f1d1 : SUCCESS in 59m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a0d68eb0009a416f9361b804276a8328 : SUCCESS in 59m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5c239a732bf54ce79414eda1fb79758c : SUCCESS in 1h 22m 48s","accounts_in_message":[],"_revision_number":1},{"id":"981e9b04e2412df0ca7d98c1d887846ab0d74285","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-16 14:42:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fee26d02644849ecb5a228779d7b7a53a3ca38ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 15:27:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a442943e8ff4679a6927408f5a16068\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a439db0a3174adbb5e47f32ec5de827 : SUCCESS in 21m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9bce12f28c7749c490ba5b20d48d137a : SUCCESS in 42m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d2616cfa781a4f5fa8e76cc8b4371e9b : SUCCESS in 21m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"222bffa2536a6ec990028e67a8b23cd332bdccd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 16:52:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1b9253d34c6411a83d0c40bd00f1422\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/7fe74be4869c47ab80fb108e53d96145 : SUCCESS in 2h 00m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a80a6bbba50741be9de4eab736b57650 : SUCCESS in 2h 08m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/545c44f380d246b48ef068fa00208874 : FAILURE in 34m 06s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/21e843c26ffd47768a73c114c6e96a83 : FAILURE in 58m 19s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c7351828ab984552a51f9229518b6317 : SUCCESS in 2h 03m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9fa4ed45415c447a9936c86ec9a52caa : SUCCESS in 51m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bdadefa4ce84f20b6d422fbfc2f060b : SUCCESS in 38m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aaf4743590b40c8b8fb9b5c027aabac : SUCCESS in 13m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72623de903d2435bb39027f40771e0ca : SUCCESS in 22m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ba903cc112a419c9573dafe43ab168a : SUCCESS in 25m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9cffe78f062421b974cf71d4b8c6bb9 : SUCCESS in 28m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c368ca34456745f6bfcd2b9c0e3ee6ef : SUCCESS in 11m 04s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/23b954ceae714ae58ac11f7ea0eb3065 : SUCCESS in 21m 33s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/52e135f8639c45ca9fb4351f11d61815 : SUCCESS in 48m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3da62363a4d34aca826272b3177e9887 : FAILURE in 1h 37m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b289851878694c34879e02654e288e18 : SUCCESS in 57m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/066e892c03c84aa1981154dd627efa94 : SUCCESS in 1h 00m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/789c5c55040b4ce1ac73be561c7f473e : SUCCESS in 1h 22m 55s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bcf16502483840b5b2d2f2e0a39ba3ca : SUCCESS in 1h 06m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2bcd3941f4fb45559b7788955ea0e85f : SUCCESS in 1h 04m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/153dd9de3f2d403a8c440b5e67e8a0bd : SUCCESS in 1h 44m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/69f44c6a15224fefaacd79d59871cd02 : SUCCESS in 1h 52m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ff002a248e294bdda58aa550c35d7477 : FAILURE in 57m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c3c90785afb18afc657c49466e575163bd88a35f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-17 04:56:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df6afeb5439c9e1bc8b1f01054382534777dbf1a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-17 04:57:40.000000000","message":"Patch Set 1: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"b041168f92af3e453ca648bb387cd595797ad11f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-17 04:58:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c55ad2eab4dc2751002bcf8a00a38a1b80b3f585","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 05:26:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de2d03fe62424bb5bb530bcfe5517fbb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2a1969da92642e583269c1b42d6bfcc : SUCCESS in 20m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a6287561fa74ae3a4c59db543c11900 : SUCCESS in 27m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bcf23dbdffc949c0940ecd8930b3fe19 : SUCCESS in 20m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6cf488356aba63a05dd5497c3d8f34206ae1996","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 07:28:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9af4c66e4b344abebf95d8b8490cd8e8\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/029ab7455c9944ad9587727994890ce9 : SUCCESS in 2h 26m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6492e34cdbb0428f85e62a110492e05e : SUCCESS in 1h 54m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/81e581f559f54f5aba26280706192406 : SUCCESS in 2h 28m 04s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f8d47785165e4fa0a7283237fa38301d : SUCCESS in 2h 12m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/06a6d6c24aec4f84bcf2ae181fc5d548 : SUCCESS in 2h 07m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e80150e820c401e8fea55340b9344f1 : TIMED_OUT in 1h 03m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f573739a421d4d209a6ae48f21f21959 : SUCCESS in 43m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e283fa2ffba845218938d56a21d51170 : SUCCESS in 11m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95891fef82724ffbb74ae70603307247 : SUCCESS in 24m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9eaeaa15da1b47b7864239f69a1d8b2d : SUCCESS in 25m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1328456d8d234e87bbe5c0d52b94395f : SUCCESS in 29m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/362a3207e6ae4883ae4a5bdc74f6575c : SUCCESS in 10m 23s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a37616607a7542ee989bffce31bc4933 : SUCCESS in 25m 49s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3c1913f259745d4bf8fb7b36c158a55 : SUCCESS in 56m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b79fd40a31a548e2bd25c8de1ea89e3c : FAILURE in 1h 31m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/54c4fe96089e4e26a244127c25135b9e : SUCCESS in 49m 42s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0d7a35abd0d74432881829a47f03c7e6 : SUCCESS in 50m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8c4045b8820049b3b4999e87bde743cb : SUCCESS in 1h 39m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eb5e93958f8a412ca4d40dc7e10dbb50 : SUCCESS in 1h 14m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0de8948a2e5e47c995ee0b0e7f24ceb2 : SUCCESS in 1h 06m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/83a3d43c7bfc48858ee766c80a479ca3 : SUCCESS in 1h 36m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0474b65763ea4b5c8bde945f0a12879f : SUCCESS in 1h 40m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f62e2a698faa46189ca543a6c3a543a0 : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":2},{"id":"744fc23f6da1dc09cbef24d805978f1cf3267c17","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-17 10:59:26.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/20/883120/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"738563c95d453289b249e74d581e897969d19190","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-17 15:51:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"da53c007bb593048d4cd3a35bd9abcf35f028377","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-18 09:21:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a1a6e2e6be8997c253abf108608e82f120957f25","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-18 09:21:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"123fa812f020bfc0bdc37817e42d8f71fef52c7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 09:32:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71b8b6006a61420aac52db28d08d8342\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a434d46d62046fcbe4faa452b8cb2cc : FAILURE in 3m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3645d3ff0b2343fb93685ef25d181846 : FAILURE in 9m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/86f37dcf96cb4c259273c577b2035829 : FAILURE in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2cf4a57683fe3030b2c3cadb675c930d0b632ee4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 10:26:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d874142420864a319db8c7eb9b87cd21\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b9f9dcb777a947678c323825e92bb3a1 : FAILURE in 8m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7a2aab3354254ef0bdeb02bc9e2974b5 : FAILURE in 11m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/da13eb538e3a494484386364c848beb0 : FAILURE in 10m 44s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/440074429ce641cab8848e2f539dc079 : FAILURE in 11m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/049205bcb0df415cab3b67ec8b19b7a2 : FAILURE in 10m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/49cbb1d8cd5e490793c983ba573ded46 : FAILURE in 10m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e39ea9e6d4f84d16b01d04714a4ae06c : FAILURE in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc5817d37fa44b088cabf8d4306f8621 : FAILURE in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/985f80d14c1f4a1792bfcc3cf65c9d0c : FAILURE in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3fd4601ffa441c6b139b31639ab73e7 : FAILURE in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64d91c42e7de4dc1bf40f14c117cdc87 : FAILURE in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c74b64f30d63478f95723f901bdc72fb : FAILURE in 4m 04s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/df640ae91b104bf4a377b797df96f9eb : FAILURE in 4m 19s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0d361552725d4467ad789032f34ac953 : FAILURE in 23m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8708a621ec824db58c45eb3fd814b115 : FAILURE in 12m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/15a33afef4064482859144c4d7f97e85 : FAILURE in 48m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/25e124b8899642a39d823005a262707b : FAILURE in 57m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7d406511bdbe461d9918c5649f081a44 : FAILURE in 21m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5bc1e507ef0743d49868c0b93e7b4ba6 : FAILURE in 25m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/19ef9b459e454cff967c557492d0c8dd : FAILURE in 25m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ece47f938a864df494e59f71c31f591e : FAILURE in 12m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e65b9eff35bf42b6bd9bf1fd603cdceb : FAILURE in 10m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9a9f63ea72cd43ee8b3bdef2a1649b9d : FAILURE in 24m 13s","accounts_in_message":[],"_revision_number":3},{"id":"b5de6dde86f2de0f45066507f5a53ce7d574b667","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-18 11:30:48.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/20/883120/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"25aba282e582a81b8cab37b5e9b4af1b3380563c","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2023-05-22 09:55:31.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c6a213bd4ea18d1058534e8c6790ed0e55195d3d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-22 12:51:25.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7d561b00f9b330d9345b6c90a27bee8b4e49eccc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 13:01:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31e69f51d063499f8d1a7d143803b32f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b60b5772bf3b4510a58a4501e75d263e : FAILURE in 3m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f7d12f6d226243609681ea86de34306f : FAILURE in 7m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/42467b02508b455f80cc775b63d4af9d : FAILURE in 8m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b7dc7a5867657a8e370e0da6daadb12de0220c76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-22 13:29:55.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ff7e86ca465ccd0ff162f17a148f7860e676fb9e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-22 13:30:00.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3232236e50487cbcf59e4e77cb4158c3f148d089","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 13:39:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd143c86fc8c4984a24a26360a0edb5e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0af19804bb974401a6c36cae2013bc35 : FAILURE in 6m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/baab4f1aa7dc4154a437152dce62e3b7 : FAILURE in 6m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d2ce570167ba4f618a684a5cbaf2ae9c : FAILURE in 6m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f4095f67228683ca09f0627de553537c6ae6e7b4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-22 13:44:08.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/20/883120/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"f00f6f29a6962ddb4c68057425566dbcf59f168b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 14:58:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e5ee72a15ba4b79b9964dd2cd5526f2\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f594161b4ac1429aa1d5692728403582 : FAILURE in 10m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f77fab91330b4e0285c10583d537c4f0 : FAILURE in 11m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/36d6b11f5f0f4427b9b91eaacb650cd4 : FAILURE in 7m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3d22daf821924df58cdefbedaecb1a79 : FAILURE in 11m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0ca2a85c7fa244149bb9a5a0441a9318 : FAILURE in 11m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/118b1db066ad4d14a7f441bf1b53bced : FAILURE in 8m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbb312411b8649d6ac440ccb6d344ca5 : FAILURE in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e855a89881a44de984d3df5315d3836 : FAILURE in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19733f3f8cdc40748e31171b0b283eb9 : FAILURE in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0444bb30012444fd9c60d9109d97a886 : FAILURE in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cf478dbe00e4fe9abfec93af00bc84f : FAILURE in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3918fc1a25fb46e9a70c35a6d4d15905 : FAILURE in 4m 49s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/c715eedc849a44f4bfda0f90252cc9ac : FAILURE in 3m 39s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/21ab19210c3b4e53bc763a8ffb694c1e : FAILURE in 19m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b2d14c81e9f8473bb7e71bdda37f7632 : FAILURE in 19m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c7b8064b3f544e8a89bf499e2922e33 : FAILURE in 58m 43s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/24cb1c400b354df482d370b45bb90811 : FAILURE in 52m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0e363e799a2d4fc6bade77f00546d7f5 : FAILURE in 21m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f117031f268943b8a4b7f384bba4fbcc : FAILURE in 25m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d8a91bfa05ee41c7be3b6b104fc0f74c : FAILURE in 11m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8ee368fceabc4a43824032402f947a88 : FAILURE in 8m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/702305ec63854ac09e373315f4a8b337 : FAILURE in 10m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d22797abf02a496ab247f50bad94c2b1 : FAILURE in 20m 48s","accounts_in_message":[],"_revision_number":4},{"id":"4c4926845905afcd8b7d7626cbcb2cb0200448f5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-22 15:44:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4fa6a9dd07d359ed98b0f0c778426dfafd8cff65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 16:31:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/144e23d17a96469bb10b8e5000f282e6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4f798fecfab4df680b1c620863a6f9b : SUCCESS in 20m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/199abea9a0df4050b94cfb6390bfd357 : SUCCESS in 23m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6ece33361c114df99382dfbd2698267f : SUCCESS in 43m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7e0ee7b2dea173cba800b509fc06801542e8e1f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 17:53:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd34aca94df64c198063c8c424c66d95\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9e9a34e9fff44840b103efeab752619e : SUCCESS in 1h 30m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/32e1fe604bc246089d47ea2dd5443458 : SUCCESS in 1h 32m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e614c8b7a6024cca89de3619ec173a9e : SUCCESS in 1h 50m 49s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/02057261c92c4580b99ee089804f53d1 : SUCCESS in 1h 25m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3f9ff75449094ea48ed16c50e5e0cefb : SUCCESS in 1h 27m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/721001fcc9be44329c3bf0ee2b6bdf05 : SUCCESS in 35m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07f285bb878a47e29a8475a0e0143f07 : SUCCESS in 43m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62573760991c416ab7fb164d5c6461d7 : SUCCESS in 11m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02014b0018df46629cceece8b6abb124 : SUCCESS in 32m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ad6f229c0ec4857ad854c8f312cef55 : SUCCESS in 26m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecf7bd63951d4b24b670f1edaa0c4e6e : SUCCESS in 31m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b9290aaea940ceb3731a5deb6026a0 : SUCCESS in 10m 26s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/881317f8b70246dbaa63cdc0469beb7f : SUCCESS in 30m 51s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f4ff6fe975d415c814f4c98b52c2bf3 : SUCCESS in 57m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2066070bd70447f28695e7d718c92771 : SUCCESS in 1h 47m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bdec3ab9e8e744278b50b9813c6187a8 : SUCCESS in 57m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/92beb748930c47bc8f184ff6988f75cd : SUCCESS in 1h 02m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5531ff6b051540b58cea8e2f25665a88 : SUCCESS in 1h 22m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a9e275d885a94b80931a86ce7b595f3e : SUCCESS in 1h 33m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5ddd525d98864d62bbfe26c2763950bb : SUCCESS in 1h 18m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5724c15eb0984a03bd462830bc3cfdc2 : SUCCESS in 1h 57m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5e8d7afec0e0433592e38b2d50142fff : SUCCESS in 2h 00m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cd6a1be907d14b77beccca47edf7903c : SUCCESS in 1h 24m 55s","accounts_in_message":[],"_revision_number":4},{"id":"902623d1bc72961ff1f3b7cbb034c794abb0eb3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 17:53:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"164e909247080b34088d35adc6a71473d8e98af8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 19:49:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82b0c2b8764b481eac8a20ad9984b8f8\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/565126cfdfd44ae18708a0ea87a4be36 : SUCCESS in 1h 41m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/318a2ae091fd48dc8eba7025143f8f11 : SUCCESS in 1h 13m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0c7da1a03e0b4dcbaab9a34d56407dfd : SUCCESS in 1h 23m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9efd6f40cc2441d7b55ad9e69cdd874b : SUCCESS in 1h 30m 30s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5c0bad9047844246882d51e9faa3cdbe : SUCCESS in 1h 15m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a57da3c48d64f638b5f7bbf9dfac942 : SUCCESS in 11m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e73babe0ff5246fb90036d1743856e2a : SUCCESS in 29m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6da211287c44ad6bcf18d8bca1d3b2d : SUCCESS in 33m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e53172da1eac4a1fa17aa9583582e54a : SUCCESS in 28m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/865fc817610b44539eb1048037f28150 : SUCCESS in 11m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8b063d3e0b5543ae94baadd3941531b1 : SUCCESS in 1h 17m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/856ec02dddac4451ba1cc180a51c0a74 : SUCCESS in 1h 39m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7cbb081b9fd74854a8a5ef1fc1f76299 : SUCCESS in 1h 29m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3aed2e15b1cf4001ab3761bf322c80ac : SUCCESS in 1h 04m 56s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7a2600d2511f4f54934f7d2ccde10bbe : SUCCESS in 52m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6ffe4333ecdc466ea2808516d174f13c : SUCCESS in 1h 03m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/63f65dfa08ec45f093102dbcaa3c24b3 : SUCCESS in 1h 35m 28s","accounts_in_message":[],"_revision_number":4},{"id":"697a7ec38d54f241c56f15a2d15b088eabea17ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 19:49:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"1f26965bda92994dcaad590c0d8657925cbfc26c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 19:50:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/909fa451ea344b129855425f16801388\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167f0544e9b74b3e92f279e3b148734e : SUCCESS in 52s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"452973c0dba6c2f1dfce02bd7dce514c8350a846","revisions":{"edca58c7f2a17aa3e1ac591d0631a202fd1b1626":{"kind":"REWORK","_number":1,"created":"2023-05-16 08:54:06.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/20/883120/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/20/883120/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/883120/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/883120/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/883120/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/20/883120/1"}}},"commit":{"parents":[{"commit":"82029c2c514aaca5b9d45af46d2f7d945ff3099b","subject":"Use a writer context for the online alembic migrations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82029c2c514aaca5b9d45af46d2f7d945ff3099b"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-05-16 08:54:06.000000000","tz":0},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-05-16 08:54:06.000000000","tz":0},"subject":"Revert \"Use a writer context for the online alembic migrations\"","message":"Revert \"Use a writer context for the online alembic migrations\"\n\nThis reverts commit 82029c2c514aaca5b9d45af46d2f7d945ff3099b.\n\nReason for revert: fullstack tests broken with this.\n\nChange-Id: Ibd75c3526a3a8c4138c5f266448299786ccbf045\nRelated-Bug: #2019802\nRelated-Bug: #2015847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/edca58c7f2a17aa3e1ac591d0631a202fd1b1626"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/edca58c7f2a17aa3e1ac591d0631a202fd1b1626"}]},"branch":"refs/heads/master"},"929472e05be9a2a3ab61153669f8c6b0266c1583":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-05-17 04:58:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/20/883120/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/20/883120/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/883120/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/883120/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/883120/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/20/883120/2"}}},"commit":{"parents":[{"commit":"7b9e4a9659b930354577145e1d70782711a58ca0","subject":"Merge \"Add scope ID to the \"GROUP BY\" clause in ``get_scoped_floating_ips``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7b9e4a9659b930354577145e1d70782711a58ca0"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-05-16 08:54:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-05-17 04:58:21.000000000","tz":0},"subject":"Revert \"Use a writer context for the online alembic migrations\"","message":"Revert \"Use a writer context for the online alembic migrations\"\n\nThis reverts commit 82029c2c514aaca5b9d45af46d2f7d945ff3099b.\n\nReason for revert: fullstack tests broken with this.\n\nChange-Id: Ibd75c3526a3a8c4138c5f266448299786ccbf045\nRelated-Bug: #2019802\nRelated-Bug: #2015847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/929472e05be9a2a3ab61153669f8c6b0266c1583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/929472e05be9a2a3ab61153669f8c6b0266c1583"}]},"branch":"refs/heads/master","description":"Rebase"},"a5f6de82c0df843bb810d0c70885cc478e822956":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-05-18 09:21:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/20/883120/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/20/883120/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/883120/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/883120/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/883120/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/20/883120/3"}}},"commit":{"parents":[{"commit":"32e8a7611e6f6fcb6d450f70363c4ee26050fb8a","subject":"Bump neutron-lib to 3.6.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32e8a7611e6f6fcb6d450f70363c4ee26050fb8a"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-05-16 08:54:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-05-18 09:21:16.000000000","tz":0},"subject":"Revert \"Use a writer context for the online alembic migrations\"","message":"Revert \"Use a writer context for the online alembic migrations\"\n\nThis reverts commit 82029c2c514aaca5b9d45af46d2f7d945ff3099b.\n\nReason for revert: fullstack tests broken with this.\n\nChange-Id: Ibd75c3526a3a8c4138c5f266448299786ccbf045\nRelated-Bug: #2019802\nRelated-Bug: #2015847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5f6de82c0df843bb810d0c70885cc478e822956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a5f6de82c0df843bb810d0c70885cc478e822956"}]},"branch":"refs/heads/master","description":"Rebase"},"452973c0dba6c2f1dfce02bd7dce514c8350a846":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-05-22 13:29:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/20/883120/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/20/883120/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/883120/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/883120/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/20/883120/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/20/883120/4"}}},"commit":{"parents":[{"commit":"f42d86cc0d672433d4940cbc05a44d9736d1ba54","subject":"Bump neutron-lib to 3.6.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f42d86cc0d672433d4940cbc05a44d9736d1ba54"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-05-16 08:54:06.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-05-22 13:29:55.000000000","tz":0},"subject":"Revert \"Use a writer context for the online alembic migrations\"","message":"Revert \"Use a writer context for the online alembic migrations\"\n\nThis reverts commit 82029c2c514aaca5b9d45af46d2f7d945ff3099b.\n\nReason for revert: fullstack tests broken with this.\n\nChange-Id: Ibd75c3526a3a8c4138c5f266448299786ccbf045\nRelated-Bug: #2019802\nRelated-Bug: #2015847\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452973c0dba6c2f1dfce02bd7dce514c8350a846"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/452973c0dba6c2f1dfce02bd7dce514c8350a846"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
