)]}'
{"id":"openstack%2Fneutron~883235","triplet_id":"openstack%2Fneutron~master~I53e11f558a6cd84a02041758badddfe87a10c95c","project":"openstack/neutron","branch":"master","topic":"bug/2022360","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-06-05 16:14:51.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-06-23 20:24:14.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-06-06 13:14:26.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"9642":{"account":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"last_update":"2023-06-30 21:34:34.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2023-05-25 13:00:10.000000000","reason":"\u003cGERRIT_ACCOUNT_11583\u003e replied on the change","reason_account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}}},"hashtags":[],"change_id":"I53e11f558a6cd84a02041758badddfe87a10c95c","subject":"Do not query neutron-rpc for sg rules upon sg deletion","status":"MERGED","created":"2023-05-16 08:29:03.000000000","updated":"2023-06-30 21:35:57.000000000","submitted":"2023-06-30 21:34:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883235-bug/2022360","meta_rev_id":"977fde41a61baf55171a2c66937b36cfc685298e","_number":883235,"virtual_id_number":883235,"owner":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":1,"date":"2023-06-05 14:10:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-30 21:34:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-06-05 16:24:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-06-06 10:01:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-22 14:33:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-06-06 13:14:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-24 06:04:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-22 14:33:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"reviewers":{"REVIEWER":[{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-16 09:08:48.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-16 09:27:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-16 10:58:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-16 14:58:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-05-16 16:39:34.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-05-25 12:52:28.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2023-06-01 12:48:46.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-06-06 10:01:22.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2023-06-06 13:14:26.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-06-22 14:33:27.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-06-24 06:04:15.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"5be71b42e8c49788a81ee5610c72a2631fc4bd44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"date":"2023-05-16 08:29:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af6e801b72499dc39911797a235868260af99e07","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-16 09:08:48.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/35/883235/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"226d1aaeed74a2da939034d0248acb5c276b78a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 09:27:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b295239cd7024fa3b3efa87016265e24\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7e2b766e70547a1a20802fbe737d6a4 : SUCCESS in 20m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a8ddfd25cee4b4eb60c5f71d7d4bbb2 : SUCCESS in 55m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/173df08e44624cb9b540e9b10ff0b500 : SUCCESS in 55m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee4b19014cf71db5134bab18515004b903782f11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 10:58:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23e0a7ad59734904b493487c384611dd\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/2983ee88b3bc4a97939e11ca142ae05e : SUCCESS in 2h 20m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fd7ca3f2504246fbbf715f78c2e17a8a : SUCCESS in 2h 12m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d31e8308b6ba445b9d424634d757547e : SUCCESS in 1h 35m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/918199c6f99944a090f44591811a834c : SUCCESS in 2h 02m 53s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/198c151e20a447c1bbc7dfd697459df9 : SUCCESS in 2h 24m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/70ae1540b4534a34930ac969b0844278 : SUCCESS in 53m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/515c643baeb24b5da9f83d42191ddf3b : SUCCESS in 41m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfc098e07f1e4d54a6d80b91fc334696 : SUCCESS in 11m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab073fae0d324fad8933545c6c156203 : SUCCESS in 25m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b609532e0c44051bde11aa95843c94d : SUCCESS in 26m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58bbf2afd1ba49a88987150122649cfa : SUCCESS in 30m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4165bcc7316247089b33e6a4897a532a : SUCCESS in 10m 50s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/ea7ac05f9d324fb9b74198fc2e5a9012 : SUCCESS in 21m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7679f54d1ecd4278ac52ad6672fde2ef : SUCCESS in 1h 10m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a5c4ace38c214b22a62ea7262865db55 : SUCCESS in 29m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b892f6e1f51c4c6183a42d791bf7fcd9 : SUCCESS in 45m 02s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4bc55f9642ab454f84e79bb4ba10d274 : SUCCESS in 56m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f8544067005c4a00b517900695fc3236 : SUCCESS in 1h 47m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c738b2ae24df4bb3876e989bc8383481 : SUCCESS in 1h 17m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8fa6c2894192425d99f4e2e59e7747b0 : SUCCESS in 1h 11m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7fbf6642e19e4cd5a9b04b0d2cb681fc : SUCCESS in 1h 56m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fe798341b7664c219fd2814689b46024 : SUCCESS in 1h 58m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/52d40fab782b4831aa5889b8543876d1 : SUCCESS in 1h 35m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1a4583c179a7353f9cdb3b4c978cb876bd0b4b19","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-16 14:58:18.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9bb02bc3e0336f75356d950c0db115f69fa024c","author":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"date":"2023-05-16 15:15:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93aaa10bd4561980d6934828ffa5e0827eabf0b3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-24 10:24:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c058a84bacb011e13a11ffb360e696f206c55f8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-24 15:55:20.000000000","message":"Patch Set 1: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9b423898ccffd553e6ace902f38d0cd2a3520f06","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2023-05-25 13:00:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"383e65d76e4236cbd63b8a3ec0bcb8f7e51df6b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"date":"2023-05-26 08:43:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3a95916083bef78a5f026d687fe6a0b224dd2ab5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 09:06:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14c7bd4e94144c08b0fbd31cb3e7e762\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c3a37a6b5cc4b88b309d85e6794138c : SUCCESS in 19m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4dce6c64b0f544caa726b7a544ebe810 : SUCCESS in 21m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a1eb450b0ebd43cba5ed664a23773aa4 : SUCCESS in 20m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cba7224412cd112ffa7d86952d78cbe91aca5ad8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-26 09:27:03.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/35/883235/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"0efa2a10d220489b62262748fe66761756281553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 10:57:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba76fa9b1be44c0884f7ebfde187b4bc\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/27b0c0eb2ac34443b1c44e4e981427a1 : SUCCESS in 1h 38m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/34ba5664c33e4bb1a416404d3bf07887 : SUCCESS in 1h 12m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0bffd844bc0342638d593c915d63f7ef : SUCCESS in 1h 26m 33s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/829c038bc382455ca7b9d69e32ec221c : SUCCESS in 1h 27m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9b508f427f504462b7e7edfcba7adccf : SUCCESS in 1h 25m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b4b56902527e423ea61f2f11d92b8277 : SUCCESS in 34m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c014de89bd5c4e3881b6841f5039643c : SUCCESS in 37m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/094ec212a20e4a86b9869e898af6d612 : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d320ee41b8c342219b628913fbde048c : SUCCESS in 23m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/473488e2757e4e66b9046fae2dc4031a : SUCCESS in 25m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d979cf59ee0414eab5847497a5bbfe2 : SUCCESS in 21m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5796ef63e1c949cc9576fe28277b17ba : SUCCESS in 12m 34s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/324f5ca517b24f5d9e2a782da0000705 : SUCCESS in 29m 29s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8974514c3adf4bd3ab9349f56c38fa06 : SUCCESS in 58m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/24147f97110e4fbc9b6ff560e0aea31c : SUCCESS in 1h 32m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f2a3cafa626469b8cccdaf799ca578f : SUCCESS in 56m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a5a32930a8d745d797f09983b5b05da0 : SUCCESS in 46m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e99ed188c8ba4fdbbacbf493ae7f4923 : SUCCESS in 1h 28m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/13bbc313693f421d83e5ce3a44885f8e : SUCCESS in 1h 22m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/67f627d7de184de3945606c8ab60eee2 : SUCCESS in 1h 08m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/451a6dc824094c17894aee21d3bf82e8 : SUCCESS in 2h 06m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8554ca612a434b9db3a62ccf6ff51356 : SUCCESS in 2h 01m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3961fc75222e4be589c3951014723760 : SUCCESS in 1h 24m 15s","accounts_in_message":[],"_revision_number":2},{"id":"523e2b5964a6f0d293b87762f1746acbd0c943d7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-29 15:32:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a44b375cb2d679aeaecd9a677341859812979d3e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-06-01 12:48:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d808f60bde0be12edccb44f1370b669476d8f3a3","author":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"date":"2023-06-02 12:21:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ccc5ee6b221ea995c14ea02a1fc2801014e16d5c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-02 13:46:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6563bf532f763797930b37b8d585653d98e85b5e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-06-02 13:48:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"835b8e3b4df043f710db1d6d3d3f3086b01cc414","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"date":"2023-06-05 13:02:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"29b4661ea4ba7ff23506ff6bd16a94c6134447f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 13:29:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/248ca792cf274ff0af0f39bfded1363a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e580d44fb02544dcae6a437cb60999dc : SUCCESS in 26m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09626365099a4dcc8672a64ad2b38961 : SUCCESS in 21m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/de4f83ffd7b74e148c1bf32dde989fa8 : SUCCESS in 19m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a8c76d6088e2ec9b4b4beda8102ac5b887f7e727","author":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"date":"2023-06-05 13:46:02.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"11abe8920c54ed88e632389c28392b921bcdf7b7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-05 14:10:39.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/35/883235/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"30571f7a085a1b18ebe6e8239dbf42364d819b52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 15:36:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/485c65d0b18e4a6daae157d8dcda47b2\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/5e35089f2986461a973183b56b93f27c : SUCCESS in 1h 49m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7067a28c3b0945bfa0a0f20f645979ae : SUCCESS in 1h 58m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/01c0d882bd5f4842a8d27b58ebe99f06 : SUCCESS in 2h 16m 34s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7deee3ba4132464ca25c0fb5509b3e88 : SUCCESS in 1h 35m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/720cb78a60f84f518f551c8f0110b5fb : SUCCESS in 1h 39m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f3312beaffc642e693e921f447d9db69 : SUCCESS in 34m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a16c401bc14f414ca8bea75dcd4cba69 : SUCCESS in 36m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb759dd0622f4fcbbb5f82ae37c98a2d : SUCCESS in 10m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d61c37faa224141a5b8b155306df015 : SUCCESS in 27m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5dbc1432e5d4b738ed15a01616aa14c : SUCCESS in 27m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a8370df35a24c709780fff0acdb8c08 : SUCCESS in 25m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f3137c6996b4f43b8ab6a1847f0bb92 : SUCCESS in 11m 15s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/418a91562c254758ba70268726e91998 : SUCCESS in 21m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7aa54d5ddd5649a3b6f9e827d0227f32 : SUCCESS in 59m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f31c6baf20d446118f8e093978b18c28 : TIMED_OUT in 2h 32m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a1b9a15f30b4fe2b24a9f95809d84f5 : SUCCESS in 48m 52s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e7f31ea3d301463ebbe3319af4b00bfe : SUCCESS in 50m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fe19b724ad04460eb3f4c419fd6c7018 : SUCCESS in 1h 34m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aa904edbc1ef43b781131419289635c6 : SUCCESS in 1h 16m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8cda168b6ec449b3aa0f2ad8ea6cdf7b : SUCCESS in 1h 03m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ec1995de9bec4742a360ada48c4e856e : SUCCESS in 1h 32m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/49c5365703a34473a6d5d49e1a6e7895 : SUCCESS in 2h 08m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3b94aa0c07d0456283b2bdd357124fb8 : SUCCESS in 1h 18m 09s","accounts_in_message":[],"_revision_number":3},{"id":"0f1578e9671953f9bef08aeefa63c5a8957422c5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-05 16:14:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"702994ef4aef5f01de199828451c2db68e644afc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-05 16:15:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fb5e205be32b7ee69b1bde9a0ec78475b31d9a85","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-05 16:16:09.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d33f7ded0255784ace49340229d44ccf4574cd80","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-05 16:24:24.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4886cd14225665a31493e3e2c3844bdaef36722d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 17:10:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a796646ba55c4fcaa08954d51ad69451\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fcdc198c7034bfa90f86f1578a0b20d : SUCCESS in 20m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/161ae7ce89d44bb69a1a26914eaf6199 : SUCCESS in 19m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e65f3ead786f4cf19dfb7f003edf31b1 : SUCCESS in 53m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8641ab3a0ce7ba4c9303d0347bf0a922307ada94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 18:37:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1be624e3527347ffaa758151806a6237\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/fbe2e514b2d84604a92f1ba326952089 : SUCCESS in 1h 53m 36s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6320ce3dc2da4e839f1127b20b10d1b6 : SUCCESS in 1h 13m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d42cff4c5a5844d18da4db54c9f430b5 : SUCCESS in 2h 20m 27s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/85fe0bbccc3f4b6a91204c1064216006 : SUCCESS in 1h 39m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f94cf2e324b642daae6ac04288cc5a4e : SUCCESS in 1h 18m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a7b1278093f14d5e86dc059dec64feec : SUCCESS in 47m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/938e9a7f2a684d7899893de95d90f8ec : SUCCESS in 41m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4bc7f90d7554c488b166e0183b95d0d : SUCCESS in 11m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d464c5a7d3764f4fa8e078d22d69b56e : SUCCESS in 29m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72d1cabb81964814a668b0c1f58eb53e : SUCCESS in 25m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4154f3855a5544df918c34a3b721a0fe : SUCCESS in 35m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b5835479fcb4384bd85eb6dc81f4c91 : SUCCESS in 10m 30s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/006dc4c5b5fa456aa513d3ab3a705b52 : SUCCESS in 22m 39s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f233535131c7432fb551658c565fbf0f : SUCCESS in 51m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/444b8805f6e74fac8c2e8382d5555811 : SUCCESS in 1h 52m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7dc006767e734ecba88bce4038410de4 : SUCCESS in 58m 16s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/96defba3d8cf4c73b031f4e6db63e293 : SUCCESS in 1h 03m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0032f1803bc8476b87e27cc2252f7a5e : SUCCESS in 1h 33m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1624be7985f340c78a73c67fa1255b9e : SUCCESS in 1h 24m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c9f0614b9b1d440ebcc0bdd88b422702 : SUCCESS in 1h 01m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1b0bdceca2904503aa18d9003ccd373a : SUCCESS in 1h 22m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/85a8d123322a4765ab4b0ef3a8d2c3c6 : SUCCESS in 1h 20m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/96845b69d2644efca871e1ec963e8ad6 : SUCCESS in 1h 23m 28s","accounts_in_message":[],"_revision_number":3},{"id":"e22e86314a7587cc75871f08c53f1ce11c472e6e","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-06-06 10:01:22.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3a859bee792c60786ae391d2a78d6a40fde2fe74","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-06-06 13:14:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"63c1f7e161459549b735de52534722c5aa0ebbbc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-06-22 14:33:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"fc07a160fabb7ca0d9847d862b3f08de12a37e96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 14:33:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c217cef44b1d67b9dd0ed6675b38f309bbcc733b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 18:45:24.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/493ee1374a3a49aa872d4de2f74cc3e6\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f9e175aa75db46c29b71e9f863cb236b : SUCCESS in 1h 59m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0b465c220eb44e499ac622d37e5fb17e : SUCCESS in 1h 21m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b9fd677dcb284d5da9477c1a3d268b95 : SUCCESS in 1h 28m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0c7449607abb4d42b665fe947c30de5b : SUCCESS in 1h 32m 01s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bd217fd3115641feb5222e90fed78f02 : SUCCESS in 1h 21m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/646fd0ffd52a43cda7d3a0ef2af87be5 : SUCCESS in 11m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2972935ca4934dc499d5560327733de4 : SUCCESS in 24m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9009bee719d44d3dac0313e536bd9ad6 : SUCCESS in 32m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/704619ed25ed4099b080169d67499aca : SUCCESS in 21m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f7132793cbd4364a755f6ac3121e35a : SUCCESS in 13m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3cb34a004bb49f0a8446b8a42a097b9 : SUCCESS in 57m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e415c28ff38e4cc185ec791af6845362 : SUCCESS in 2h 22m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bac27558978145a884d1a9c8b951ac59 : SUCCESS in 1h 44m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f43e30f43f454c6db18a30ca6e42c18f : SUCCESS in 58m 27s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a84034e77095432da279fe0e630cc896 : SUCCESS in 1h 12m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b9c830c421474d28a0b635c03898195c : TIMED_OUT in 3h 10m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/262a4de12e284250bf6923cb10a7ba9a : SUCCESS in 1h 18m 54s","accounts_in_message":[],"_revision_number":3},{"id":"8abb2204b8ea97aead930345ac88b3efec27a299","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-06-23 07:37:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5ea712440193087b502d13db905fd719ce6088c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 08:23:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4013dc0b8ce646bc809f838e1dd65fd3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/87b12c7ad38441699d2e4b4a56b1815b : SUCCESS in 23m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0dd73ae5c354fbeb7e51b6614a497cc : SUCCESS in 44m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/964a4428b76e4cbda58634dcc68a0519 : SUCCESS in 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7715126d19d73af9ef99ab2d300c9410563a03cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 09:45:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/26a12aa295a342618e46434fbaf6fb00\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/3faf70f1c5c5448fb171bdfadd6edda5 : SUCCESS in 1h 55m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3b941909353d44ef83d9e12908709055 : SUCCESS in 1h 34m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5ac9f42b6a5e43d390763fd69ecadd46 : SUCCESS in 1h 32m 36s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0e3548608de0440d8ddf654224e75b58 : SUCCESS in 1h 29m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b771e914d9654cc49b1869686dc0237d : SUCCESS in 1h 28m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ee2318e9bdad4645a54e3fa808919e7a : SUCCESS in 45m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2eeced66f29b4a1ebcb4a00285e94ef6 : SUCCESS in 37m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/037fe0c0f2854fb1809e2496ab11038f : SUCCESS in 10m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83ae51d109554e90a371132bab317525 : SUCCESS in 26m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f4a81b2ac3d44c18b8084730f1d6062 : SUCCESS in 24m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7272fc643f242d686634003afdce320 : SUCCESS in 27m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2315471d1c174f88a857fccce46c1c69 : SUCCESS in 9m 44s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/b1167e7e0d70495a8db1650ca4d7c7a6 : SUCCESS in 25m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3caa39bfeefc4357a072659837016099 : SUCCESS in 1h 02m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f7d2f4d8931b4149897bc628a193f957 : FAILURE in 1h 50m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0de25eef8c69405eade571677ab587ee : SUCCESS in 56m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e5f7ef74b3c34fbfa5737af90670080a : SUCCESS in 59m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/614e8f7c056c488c86b5401bae7e63f2 : SUCCESS in 1h 23m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/12564f5610524c79a0f3c6dbc9b222f0 : SUCCESS in 1h 25m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b03aa170b2134594abdca0cefcaf843a : SUCCESS in 55m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/777753b31bce48d0bc039083fa60a675 : SUCCESS in 2h 01m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/15a8e91157a940c48e4ee3d09cccb2a3 : SUCCESS in 1h 57m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/06c762f37ad248268b2861ad3200d2db : SUCCESS in 1h 34m 21s","accounts_in_message":[],"_revision_number":3},{"id":"ed6372a5113c29edc49cdbc3d8172ef922bf3634","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-06-23 20:24:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"efcea36ab5230afba6c2daf1eb076226472b6d77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 20:50:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c77497399924a5cae546e5dc2a2a05c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14fb00d829e64a71becadc070242e186 : SUCCESS in 20m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3dae0218503c4391ac9a2bd4691e81af : SUCCESS in 25m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4a5b1639a5994594a2444dc819ccfaa1 : NODE_FAILURE Node request 300-0021530333 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9ab8ddb67b8fa39a24545bb7ee8cf1f49bedbcdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 23:20:13.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf57546867cd44be8f8ffd640b26f593\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b68af0930faf47e88ff6d28f2b0f0b0b : SUCCESS in 1h 31m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/20bff5d9ebc04b798a5aa4d3493ca9ed : SUCCESS in 1h 56m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6abf380ee2564795bb6a8d479d81f27f : TIMED_OUT in 2h 50m 02s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/516cd34a1ff84948aa6a84f30749b995 : SUCCESS in 1h 15m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e9480546506844e9b44c7c4d94214f95 : SUCCESS in 2h 15m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b33832633e9b403bb74e0116c600afc6 : SUCCESS in 34m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f04e4a90edb4a118ea4e27b6bd67d6b : SUCCESS in 38m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/897cfb146d724368b737a2b9466a77d9 : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8218aa4521d4ad9bb691e25a6f1b9a8 : SUCCESS in 26m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06289f5bc15b495da6ad8e65963f01cc : SUCCESS in 24m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/256ae0ae489f4199a9fc809c285b25a3 : SUCCESS in 23m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f31eb1ee294879b561277746260c7f : SUCCESS in 10m 26s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a2c1cc8206af498a814be4e35283dd1f : SUCCESS in 32m 41s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fce03a156b834221ab94a4de677cf064 : SUCCESS in 55m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d34ad36a061c48c48e4da2a20f173d7c : SUCCESS in 1h 54m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9921bce5e3f946db81c1dcaa64f08df5 : SUCCESS in 49m 47s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/475c499b577f471a9d3d3f3b9279012b : SUCCESS in 1h 00m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/be2db5e1fb874e13abe46902fa6bd7bf : FAILURE in 1h 29m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bf89f145af2c427d852e936f9c9b5432 : SUCCESS in 1h 27m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1be4b45bc468455cbdc8196b0612d5aa : SUCCESS in 1h 11m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/afd305e690b54cd297028287096279e6 : SUCCESS in 1h 49m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7487ab66dd3842379ce113f7bc5233e8 : SUCCESS in 2h 03m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1566280dd74f49d9bf2adc6d423a6b19 : SUCCESS in 1h 24m 20s","accounts_in_message":[],"_revision_number":3},{"id":"4486f52ec08800f5ddd93234556b0798c2567c1c","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-06-24 06:04:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ad79265a4b51669bb20594ce76ef3b0edcd9f963","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-30 11:05:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f0a9ed952cd863e47cc29a0c28ee157ce6d40773","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-06-30 11:06:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f9800e5769d762878dc66e4844aaa3552835788a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 11:53:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a02f8b7c77a34858958903db138273b3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b12c19e95c44dc18e670c8b1d38574a : SUCCESS in 20m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/376751c3e64941c3bd7a1da5ea6bfd10 : SUCCESS in 47m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/25fee5467075407aa066e6070faffc9c : SUCCESS in 24m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3fdf088157cfe63b5535d51bff2ffce1c23e363c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 13:38:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6109aedbc6ad4a3e9902f4e647d3618c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a0fb502154a14c6385d61d73078f068b : SUCCESS in 1h 34m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2630306315024e6fb7eafcf94545bf54 : SUCCESS in 2h 09m 27s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4be9a9e0aeab48d6aceb26bb6692f086 : SUCCESS in 1h 56m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a51c0803c9d14521b89765faed23fe83 : SUCCESS in 2h 31m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2a8cfee2b16349e39eb08486e53eaea2 : SUCCESS in 38m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d67fb5587cb40ae87549c8c4ac5dcdc : SUCCESS in 39m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6924845bcda14263b166a4dccdceeae5 : SUCCESS in 11m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ff7c4fc6a94418ea085a68347135caa : SUCCESS in 27m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/614fd10983004ecfb452bcbd776cd4cc : SUCCESS in 24m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f0c529b6c034422937d3b655c757d73 : SUCCESS in 24m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a4650de0e347049e948f71be46968b : SUCCESS in 10m 49s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/137c6f4d86184f269c0253320a444319 : SUCCESS in 24m 26s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6a4fc8f1faec455bbc33af8d4c256bdb : SUCCESS in 1h 01m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f1f6f112d1bc49c899d363928d61db1c : SUCCESS in 2h 04m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b69f321e12234b1397767717b272a559 : SUCCESS in 42m 30s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/47ca59fc6d10410fbf74ca4a25e2f502 : SUCCESS in 1h 00m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1f9a42decbb344e4966c487ef60bd5f9 : SUCCESS in 1h 20m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0e79da09bef1465bbcd4645281d8e936 : SUCCESS in 1h 26m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/091a271c7c3c4f42bb514ff623445d5d : SUCCESS in 1h 04m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ea0defb981e74122a608fa73fc507fb3 : SUCCESS in 1h 58m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a2640a9955e6444e82858b94c3aceed1 : SUCCESS in 1h 41m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f7e5052b5753468197c78f37c522a5c5 : SUCCESS in 1h 24m 56s","accounts_in_message":[],"_revision_number":3},{"id":"f89e3946cfeb9b60cbc9636386afc9be7c6a33a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 13:39:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0fa21de121b674ffae53fabe9f8a0f519b8cb238","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 21:34:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d0abd2d6b034b61818d13f97cfa0a3f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2ae62b7dfd9f4ecf92d9ab1c072928f9 : SUCCESS in 1h 20m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/68f4cebd8e184414934b076e98d069c0 : SUCCESS in 1h 44m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2c1b0bf55672489cac33d39269958cd4 : SUCCESS in 2h 19m 51s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d4fc864cc0e0424eb8eaebbb991931c2 : SUCCESS in 2h 08m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae876e2d9d494b8fb57d218bc75bbd8f : SUCCESS in 13m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f74aa98c22448b4952c905a760d2223 : SUCCESS in 29m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58dd6043335e440cb50a3eb9d4558cfa : SUCCESS in 24m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5453063888214e08b9bfa63601ec31ca : SUCCESS in 22m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acba0ce901254f67a0dea8b5754effb0 : SUCCESS in 10m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/87d31ba3f10b4927ad3f991254872f26 : SUCCESS in 50m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/987a30e6b985472bb637e1a6a6a9d8e3 : SUCCESS in 2h 24m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/313dfba09b9f4d1190cbb3d415b88f40 : SUCCESS in 1h 35m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fbe0745c815c476dacb482d249d541b2 : SUCCESS in 55m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3f680d8a52b0431ba53c5aa5d4bdb22f : SUCCESS in 48m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c251388876b242148fe52ce53c491cd2 : SUCCESS in 57m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a06b44f4259b4e4f8626c6183ca60743 : SUCCESS in 1h 23m 47s","accounts_in_message":[],"_revision_number":3},{"id":"12af9262bac7a6fc6fd90d4a483bb8fa50e355ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 21:34:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"977fde41a61baf55171a2c66937b36cfc685298e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 21:35:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/969d09a02841430482e290cd24321279\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fd8119f9a4649708fff1f5b0a8b972c : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ce12b6ac195c83fad424ee35c5e92b1d12d040ae","revisions":{"d334b31cce0a57a4e89a1562322c51b335efe351":{"kind":"REWORK","_number":1,"created":"2023-05-16 08:29:03.000000000","uploader":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"ref":"refs/changes/35/883235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/35/883235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/883235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/883235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/883235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/35/883235/1"}}},"commit":{"parents":[{"commit":"a06b44e12da4ab8fb037b34593f33c4df7dbafa1","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a06b44e12da4ab8fb037b34593f33c4df7dbafa1"}]}],"author":{"name":"Guillaume Espanel","email":"guillaume.espanel.ext@ovhcloud.com","date":"2023-05-15 15:16:05.000000000","tz":120},"committer":{"name":"Guillaume Espanel","email":"guillaume.espanel.ext@ovhcloud.com","date":"2023-05-16 07:51:49.000000000","tz":120},"subject":"Do not query neutron-rpc for sg rules upon sg deletion","message":"Do not query neutron-rpc for sg rules upon sg deletion\n\nThis makes _clear_child_sg_rules look into the local resource cache\nfor sg rules that have to be deleted instead of querying the\nneutron-rpc server.\n\nDeleting a security group results in each agent of the region running\na bulk_pull query for all the rules in the security group against the\nneutron-rpc. This is incurs a load on neutron-rpc, rabbitmq and the db\nproportional to the number of agents and the number of security group\nrules and has a noticeable impact on larger infrastructures.\n\nHowever, when the cache is properly populated, we already have the\nrules we should remove from the cache and don\u0027t need to ask neutron\nserver for them.\n\nChange-Id: I53e11f558a6cd84a02041758badddfe87a10c95c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d334b31cce0a57a4e89a1562322c51b335efe351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d334b31cce0a57a4e89a1562322c51b335efe351"}]},"branch":"refs/heads/master"},"1c80a98ccec7ddd4352a5172b29b98484927e205":{"kind":"REWORK","_number":2,"created":"2023-05-26 08:43:22.000000000","uploader":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"ref":"refs/changes/35/883235/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/35/883235/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/883235/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/883235/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/883235/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/35/883235/2"}}},"commit":{"parents":[{"commit":"d242792c47b703a72ace065f35eed86eeb2918e3","subject":"Merge \"[sqlalchemy-20] Add the transaction context to the upgrade checks methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d242792c47b703a72ace065f35eed86eeb2918e3"}]}],"author":{"name":"Guillaume Espanel","email":"guillaume.espanel.ext@ovhcloud.com","date":"2023-05-15 15:16:05.000000000","tz":120},"committer":{"name":"Guillaume Espanel","email":"guillaume.espanel.ext@ovhcloud.com","date":"2023-05-26 08:42:12.000000000","tz":120},"subject":"Do not query neutron-rpc for sg rules upon sg deletion","message":"Do not query neutron-rpc for sg rules upon sg deletion\n\nThis makes _clear_child_sg_rules look into the local resource cache\nfor sg rules that have to be deleted instead of querying the\nneutron-rpc server.\n\nDeleting a security group results in each agent of the region running\na bulk_pull query for all the rules in the security group against the\nneutron-rpc. This is incurs a load on neutron-rpc, rabbitmq and the db\nproportional to the number of agents and the number of security group\nrules and has a noticeable impact on larger infrastructures.\n\nHowever, when the cache is properly populated, we already have the\nrules we should remove from the cache and don\u0027t need to ask neutron\nserver for them.\n\nChange-Id: I53e11f558a6cd84a02041758badddfe87a10c95c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c80a98ccec7ddd4352a5172b29b98484927e205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c80a98ccec7ddd4352a5172b29b98484927e205"}]},"branch":"refs/heads/master"},"ce12b6ac195c83fad424ee35c5e92b1d12d040ae":{"kind":"REWORK","_number":3,"created":"2023-06-05 13:02:38.000000000","uploader":{"_account_id":9642,"name":"Guillaume Espanel","email":"guillaume.espanel@gmail.com","username":"quatre"},"ref":"refs/changes/35/883235/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/35/883235/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/883235/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/883235/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/35/883235/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/35/883235/3"}}},"commit":{"parents":[{"commit":"d242792c47b703a72ace065f35eed86eeb2918e3","subject":"Merge \"[sqlalchemy-20] Add the transaction context to the upgrade checks methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d242792c47b703a72ace065f35eed86eeb2918e3"}]}],"author":{"name":"Guillaume Espanel","email":"guillaume.espanel.ext@ovhcloud.com","date":"2023-05-15 15:16:05.000000000","tz":120},"committer":{"name":"Guillaume Espanel","email":"guillaume.espanel@ovhcloud.com","date":"2023-06-05 12:53:18.000000000","tz":120},"subject":"Do not query neutron-rpc for sg rules upon sg deletion","message":"Do not query neutron-rpc for sg rules upon sg deletion\n\nThis makes _clear_child_sg_rules look into the local resource cache\nfor sg rules that have to be deleted instead of querying the\nneutron-rpc server.\n\nDeleting a security group results in each agent of the region running\na bulk_pull query for all the rules in the security group against the\nneutron-rpc. This is incurs a load on neutron-rpc, rabbitmq and the db\nproportional to the number of agents and the number of security group\nrules and has a noticeable impact on larger infrastructures.\n\nHowever, when the cache is properly populated, we already have the\nrules we should remove from the cache and don\u0027t need to ask neutron\nserver for them.\n\nCloses-Bug: #2022360\nChange-Id: I53e11f558a6cd84a02041758badddfe87a10c95c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce12b6ac195c83fad424ee35c5e92b1d12d040ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce12b6ac195c83fad424ee35c5e92b1d12d040ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
