)]}'
{"id":"openstack%2Fneutron~883239","triplet_id":"openstack%2Fneutron~master~Ia12fc0a3d4812a0aba816b49dec60a7dcfaf0623","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-05-17 17:09:17.000000000","reason":"Change was submitted"},"27654":{"account":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"last_update":"2023-05-17 17:09:17.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-05-17 17:09:17.000000000","reason":"Change was submitted"},"27623":{"account":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"last_update":"2023-05-16 09:31:34.000000000","reason":"\u003cGERRIT_ACCOUNT_34512\u003e replied on the change","reason_account":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"}},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2023-05-17 17:09:17.000000000","reason":"Change was submitted"},"13995":{"account":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"last_update":"2023-05-17 17:09:17.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-05-17 13:39:48.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-05-17 17:09:17.000000000","reason":"Change was submitted"},"34512":{"account":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"last_update":"2023-05-17 17:09:17.000000000","reason":"Change was submitted"},"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-05-17 17:09:17.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-05-17 17:09:17.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-05-17 17:09:17.000000000","reason":"Change was submitted"},"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2023-05-17 17:09:17.000000000","reason":"Change was submitted"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2023-05-17 10:28:15.000000000","reason":"\u003cGERRIT_ACCOUNT_5948\u003e replied on the change","reason_account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}}},"hashtags":[],"change_id":"Ia12fc0a3d4812a0aba816b49dec60a7dcfaf0623","subject":"Fix not working use_random_fully config option","status":"MERGED","created":"2023-05-16 09:11:08.000000000","updated":"2023-05-17 17:10:37.000000000","submitted":"2023-05-17 17:09:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883239","meta_rev_id":"6f8688938ebfff55cf858c20c62d2a825c5ab70e","_number":883239,"virtual_id_number":883239,"owner":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":1,"date":"2023-05-16 18:11:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-17 17:09:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":0,"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},{"value":2,"date":"2023-05-17 10:43:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-05-17 12:28:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-17 13:39:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-17 10:28:15.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-17 13:39:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":0,"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},{"value":1,"date":"2023-05-17 10:44:09.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-16 09:29:46.000000000","updated_by":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"reviewer":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"state":"REVIEWER"},{"updated":"2023-05-16 09:31:34.000000000","updated_by":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"reviewer":{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"},"state":"REMOVED"},{"updated":"2023-05-16 10:08:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-16 10:25:29.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-16 11:39:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-16 12:13:30.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2023-05-17 10:43:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-05-17 10:44:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-05-17 10:44:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-05-17 10:44:37.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"CC"},{"updated":"2023-05-17 10:44:37.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-05-17 10:44:37.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2023-05-17 10:44:37.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2023-05-17 10:44:37.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-05-17 10:44:37.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2023-05-17 10:44:37.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2023-05-17 10:44:37.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-05-17 10:44:37.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2023-05-17 12:28:04.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"REVIEWER"}],"messages":[{"id":"f43c842d81b52505d8fd8c606f679d8f0209f973","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"date":"2023-05-16 09:11:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c9dfc2a84eec7321ade1478959b8a5eaf0251a4","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"date":"2023-05-16 09:31:34.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_27623\u003e.","accounts_in_message":[{"_account_id":27623,"name":"Zuul","email":"test.openstack.test@gmail.com","username":"openstack.test.test"}],"_revision_number":1},{"id":"442315c1e852d8f718f9dc9375749809340e7219","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 10:08:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afbdd3bdb18b44ea86d1bb3ea91bb1e3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fd6d48124c14ec28fd5026d5126d04c : SUCCESS in 54m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3763bc1496b1435490cd7ca019f1a7c9 : SUCCESS in 28m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cfaf4ed683614103a401855b33790e48 : SUCCESS in 50m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ccda02dd2a2e9827bdca76cf7a8eec956aa05791","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-16 10:25:29.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/39/883239/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"a1f1c2d929156d67ead331a455ddeee2dff1cf0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 11:39:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/726679bb1e8540559d9d76d4d6f66923\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/2c49c0b5327044a7b186a9c9d54df89c : SUCCESS in 2h 25m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0a26ee64923647c19e27c9b328428689 : SUCCESS in 2h 03m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d83859a9797e4304893a11b057983f12 : SUCCESS in 2h 12m 43s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/90f056ffd5064b77bad56794b12579c8 : SUCCESS in 1h 20m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9b5edb1775f04e56802ea0667e4d97e0 : SUCCESS in 1h 56m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9435872de554514bf70048c02b558ea : SUCCESS in 44m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daa0314f074c46ccaa134a3d51705253 : SUCCESS in 12m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ee7bb2563f84e7d9d4582aace51b946 : SUCCESS in 26m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7de5dd9bacb4cdc90b11a2bd84145d8 : SUCCESS in 28m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b051af43f723410ea6e6d89e36460d3c : SUCCESS in 31m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255b6a472f2146d5b4440f72ba18380c : SUCCESS in 10m 04s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/e38210667d3346778a8817b4b35674cb : SUCCESS in 26m 35s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/be10955613f4486782a90d76af40f9b0 : SUCCESS in 59m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/394b9659835a470b9c03dfb6545487aa : SUCCESS in 28m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a73cea854fb4a3589da8e46257b3a50 : SUCCESS in 44m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f0307638ff7c404da3a2bb60040dfea5 : SUCCESS in 52m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0834e8f0b045412da814d2c1128a5b7b : SUCCESS in 1h 31m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f65e3ba9842a4d7a9af6b0ad6cfd779b : SUCCESS in 1h 28m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7edd0913793f4a328b274aa497fccd7d : SUCCESS in 1h 13m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3d98b0f9747c4897a39bd0bfb60e245d : SUCCESS in 1h 28m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/73f01861ee3b4d92bcb1ce2b54763185 : SUCCESS in 1h 32m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/43f69ccec69f42e7a999086d57709502 : SUCCESS in 1h 07m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5f5a311cdcd12f821f2d4ada0bef3e40189ae803","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-05-16 12:13:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"54c280c774bb2eab582c3897f3393b0dbd6908fc","author":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"date":"2023-05-16 13:11:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"452d3edd3cf73c359476c8ad975cb84c6ce3597e","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-05-16 13:36:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b39265d86ef0dc9c59cfca14cb519774b7b7ed36","author":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"date":"2023-05-16 14:24:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"92bc0137d71e0c99b3f9b4bf15fae124890d7d3c","author":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"date":"2023-05-16 14:29:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d96e6bfb4564cb62f1244efba00039351c974651","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"date":"2023-05-16 14:40:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b36de84b3986c3c8e49e897e7463caea3ee5a520","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-05-16 14:46:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"254bbfd75c403eda40e274d12702108be329a9d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"date":"2023-05-16 14:50:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5203130a9c57e3ecc10c2fa3749fcc2bec93d5bb","author":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"date":"2023-05-16 14:51:36.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"31b23a3b73928a114af90cca93078ed32d60b96e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 15:44:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cae8539178444306b69ec01586213cbb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/879183517fc34d04a1007eb678f04d78 : SUCCESS in 22m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c57f4efe1737431c8aac8c3c91e8f815 : SUCCESS in 21m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bbe4d9001e9f4285a19541e8e56ba4e8 : SUCCESS in 50m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6f60685ee35da4d30105da64be19e6d755da5967","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-16 16:11:35.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/39/883239/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"9d5b4535a97902e81e7b239e6cb6b414b5407be1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"date":"2023-05-16 16:24:37.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"be4fda6dfc05babb07a19306c6491c271dde0892","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-16 16:51:09.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/39/883239/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"95ba8db254788133386a879c31b6761e4230f6c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 17:34:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1c792d332574eae9178789a47f956e2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8879bd89a5f944fb90b805b3f8191fc1 : SUCCESS in 1h 09m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f12d98df695145558c67bf8f61c17125 : SUCCESS in 1h 07m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18ad7c3e5c6a4abca4fc54e1bf2345cb : SUCCESS in 1h 08m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"07f6e8aafebb2df276d908d665d87287f4ebfce9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-16 18:11:25.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/39/883239/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"d0714270babd655e8c6e00b480fd2486a4a1a613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 18:45:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/816f9491994f4bae9ecbab992eaae891\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/e0578da689174267acd24dfdc7f359b8 : SUCCESS in 2h 20m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c86b9e1e4e784b6f8f9d17618313d860 : SUCCESS in 1h 46m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c1f9ef7fec3a43fa92f6d0b5ccb6c180 : SUCCESS in 2h 08m 10s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e9767f70eada42e3824eaf3167c41f49 : SUCCESS in 1h 42m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d03f5cdcd8f84f9ab658a531f9b744b3 : SUCCESS in 1h 57m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d7833f11bb048a9825cef563ff24d3b : SUCCESS in 47m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33fbfe8760914357b29ef4cedfdba442 : SUCCESS in 13m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b96525639d8141b8b67cb28cf4680530 : SUCCESS in 24m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92ccd1ebfd574476a700d54aaae419da : SUCCESS in 23m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6b960b4f5ca4869a433194763da2f55 : SUCCESS in 29m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67f7ff398acb4895bfb8dab06d2e9140 : SUCCESS in 10m 55s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a5de635738f743adbdca78b740f651ee : SUCCESS in 27m 55s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d1e02cc28d94f149bc33ac310223fc6 : SUCCESS in 1h 02m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2663e877018b43f992a80fcebcacb189 : SUCCESS in 33m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1992e2b74aef472e9ccd98fb9aee5ea2 : SUCCESS in 49m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/90068beaed514fccb87e22f610905487 : SUCCESS in 58m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5966ccbfc00a4e12af1a104c25481723 : SUCCESS in 1h 27m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9c914efc89bb4c0092c901feb8d516ea : SUCCESS in 1h 21m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7fa4716764c9451fb95406752f91c3a8 : SUCCESS in 1h 13m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3671e2f3f55c4ac49e5b7aec92419dd0 : SUCCESS in 1h 28m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5b20b966bc624a7ca47d2b5949dab344 : SUCCESS in 1h 37m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/393afb4198f94d42b634fe4d73bcde70 : SUCCESS in 1h 14m 08s","accounts_in_message":[],"_revision_number":4},{"id":"5474d95e0a0ab77a59436b5eb75a8eb3135fb5b7","author":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"date":"2023-05-17 08:45:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"327f428ba29e991596d8551425892d0c0186c555","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2023-05-17 10:28:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c3fdd2ef4f84a65a61e2ce584e1ad8c7e7ba42f7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-17 10:43:59.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e60650ec3a99b60d3dd1b155352e6eee12d9e2e1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-17 10:44:09.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"292b5ccd31129fc47d3ffca1f71274c480455ee4","author":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"date":"2023-05-17 11:24:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a8afc5a6c819795deddd5a0b3774aeb02d1aa29d","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-05-17 12:28:04.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b594307ab01d3dfcdb8663f2affaf757c7ebebe0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-05-17 13:39:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6c22a1157f8148ef4ce8fa3b143ace1d6cb37904","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 13:40:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6f241259afe1b233f87bfa4ec9ac1eabe339941a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 17:09:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a3bd6b2a94e4676863baef3b146ee30\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/90d5a0c1d9674ce098744806db4deab2 : SUCCESS in 2h 01m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d511fa43b45c4dc6ae845c649eb0ddf9 : SUCCESS in 2h 07m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/857d49846f284deb9dee19d2dff76d57 : SUCCESS in 1h 28m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/359095a958044dcf86298a74d087a439 : SUCCESS in 1h 25m 48s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/16dfc63170874baf8e1feb8fd0e0bb9a : SUCCESS in 1h 53m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/711c431b38e645c292cbdb6f446661ee : SUCCESS in 13m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5debee0338d4ee79f4fab3c1d896c39 : SUCCESS in 29m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27f2cdf5c9e54e94b526da9bf7ac307f : SUCCESS in 32m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c4866463eb1416e825abf7fe2851dfd : SUCCESS in 33m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c196e5f71e16442ca8b1335b0ccac997 : SUCCESS in 11m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fce2b5d22347471e900c194e17a15d63 : SUCCESS in 1h 00m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3671808582c4ca19559858de4feb752 : SUCCESS in 35m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/366063f41bc642dda173fc7d35f83630 : SUCCESS in 1h 34m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f38e9cc539eb4f769c3e43628ea08728 : SUCCESS in 48m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1bedab25c20948799701bad762b325fc : SUCCESS in 50m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7ededa1ff04a4d979369681fbed1e41f : SUCCESS in 1h 05m 51s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2b56601eced64873861ec4935a8af9a1 : SUCCESS in 1h 15m 11s","accounts_in_message":[],"_revision_number":4},{"id":"7558f197610b2a2febbb6c42d32624eee4850115","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 17:09:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6f8688938ebfff55cf858c20c62d2a825c5ab70e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 17:10:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c59133aac7444d69d6ec172e6447b4b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85d536d3de7d453098c14d61b8dbf3f9 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a612346146db2f9e70a23af55eb7502655666940","revisions":{"0865890edf8d0055adef0f55ad0d276f277fe2e6":{"kind":"REWORK","_number":1,"created":"2023-05-16 09:11:08.000000000","uploader":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"ref":"refs/changes/39/883239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/883239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/883239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/883239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/883239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/883239/1"}}},"commit":{"parents":[{"commit":"01af4b2cda928d76d64ff828f597f3a4dc988199","subject":"Remove the neutron-debug tool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01af4b2cda928d76d64ff828f597f3a4dc988199"}]}],"author":{"name":"Alexander","email":"shishebarov.a@selectel.ru","date":"2023-05-15 16:16:20.000000000","tz":180},"committer":{"name":"Alexander","email":"shishebarov.a@selectel.ru","date":"2023-05-16 09:11:05.000000000","tz":180},"subject":"Fix not working use_random_fully config option","message":"Fix not working use_random_fully config option\n\nCloses-Bug: #2018599\nChange-Id: Ia12fc0a3d4812a0aba816b49dec60a7dcfaf0623\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0865890edf8d0055adef0f55ad0d276f277fe2e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0865890edf8d0055adef0f55ad0d276f277fe2e6"}]},"branch":"refs/heads/master"},"2aca8e286c97dd53866333e8aed7b2b56e33f82e":{"kind":"REWORK","_number":2,"created":"2023-05-16 14:40:30.000000000","uploader":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"ref":"refs/changes/39/883239/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/883239/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/883239/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/883239/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/883239/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/883239/2"}}},"commit":{"parents":[{"commit":"01af4b2cda928d76d64ff828f597f3a4dc988199","subject":"Remove the neutron-debug tool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01af4b2cda928d76d64ff828f597f3a4dc988199"}]}],"author":{"name":"Alexander","email":"shishebarov.a@selectel.ru","date":"2023-05-15 16:16:20.000000000","tz":180},"committer":{"name":"Alexander","email":"shishebarov.a@selectel.ru","date":"2023-05-16 14:40:21.000000000","tz":180},"subject":"Fix not working use_random_fully config option","message":"Fix not working use_random_fully config option\n\nCloses-Bug: #2018599\nChange-Id: Ia12fc0a3d4812a0aba816b49dec60a7dcfaf0623\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2aca8e286c97dd53866333e8aed7b2b56e33f82e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2aca8e286c97dd53866333e8aed7b2b56e33f82e"}]},"branch":"refs/heads/master"},"8daf3f04c8ac3487bbf47f003fb8b3cb9db3fda8":{"kind":"REWORK","_number":3,"created":"2023-05-16 14:50:40.000000000","uploader":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"ref":"refs/changes/39/883239/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/883239/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/883239/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/883239/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/883239/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/883239/3"}}},"commit":{"parents":[{"commit":"01af4b2cda928d76d64ff828f597f3a4dc988199","subject":"Remove the neutron-debug tool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01af4b2cda928d76d64ff828f597f3a4dc988199"}]}],"author":{"name":"Alexander","email":"shishebarov.a@selectel.ru","date":"2023-05-15 16:16:20.000000000","tz":180},"committer":{"name":"Alexander","email":"shishebarov.a@selectel.ru","date":"2023-05-16 14:50:29.000000000","tz":180},"subject":"Fix not working use_random_fully config option","message":"Fix not working use_random_fully config option\n\nCloses-Bug: #2018599\nChange-Id: Ia12fc0a3d4812a0aba816b49dec60a7dcfaf0623\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8daf3f04c8ac3487bbf47f003fb8b3cb9db3fda8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8daf3f04c8ac3487bbf47f003fb8b3cb9db3fda8"}]},"branch":"refs/heads/master"},"a612346146db2f9e70a23af55eb7502655666940":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-05-16 16:24:37.000000000","uploader":{"_account_id":34512,"name":"Alexander Shishebarov","email":"shishebarov.a@selectel.ru","username":"ashishebarov"},"ref":"refs/changes/39/883239/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/883239/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/883239/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/883239/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/883239/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/883239/4"}}},"commit":{"parents":[{"commit":"01af4b2cda928d76d64ff828f597f3a4dc988199","subject":"Remove the neutron-debug tool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01af4b2cda928d76d64ff828f597f3a4dc988199"}]}],"author":{"name":"Alexander","email":"shishebarov.a@selectel.ru","date":"2023-05-15 16:16:20.000000000","tz":180},"committer":{"name":"Alexander","email":"shishebarov.a@selectel.ru","date":"2023-05-16 16:23:50.000000000","tz":180},"subject":"Fix not working use_random_fully config option","message":"Fix not working use_random_fully config option\n\nFixed bug when config option  use_random_fully is\nset to False all routers accept one configured\nby l3 agent with iptables \"--random-fully\" option.\nAlso added storing of use iptables --random-fully\nconfig option to \"_random_fully\" class variable\nof IptablesManager to reduce checks of iptables\nversion by instances of this class.\n\nCloses-Bug: #2018599\n\nChange-Id: Ia12fc0a3d4812a0aba816b49dec60a7dcfaf0623\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a612346146db2f9e70a23af55eb7502655666940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a612346146db2f9e70a23af55eb7502655666940"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
