)]}'
{"id":"openstack%2Fneutron~883269","triplet_id":"openstack%2Fneutron~master~Ic3e06460ac8294bfa882991eb678878b238735d7","project":"openstack/neutron","branch":"master","topic":"bug/1983053","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-05-27 01:36:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic3e06460ac8294bfa882991eb678878b238735d7","subject":"Update api extension for default sg rules API","status":"MERGED","created":"2023-05-16 14:20:06.000000000","updated":"2023-05-27 01:37:21.000000000","submitted":"2023-05-27 01:36:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883269-bug/1983053","meta_rev_id":"ca8ac7fecb65ab463df35f228c9f3baed05601de","_number":883269,"virtual_id_number":883269,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-05-25 11:27:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-27 01:36:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-26 22:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-05-25 10:48:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-25 10:48:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-26 22:52:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-16 14:46:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-16 15:31:45.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-16 16:29:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-16 20:23:45.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2023-05-24 12:10:41.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-05-26 22:52:16.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"dc6e7fbb5af593f001bfb8552a3072ce3b4c1e83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-16 14:20:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d92d571332453625facbc7f385fd34a96f76ce0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 14:46:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c55422b0b84f4278bedd20e7a4203858\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd293c64b3014100be25a4f8911073ec : SUCCESS in 23m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e444c7eb118f45f7a1b1a404035449f3 : SUCCESS in 21m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e4e5f3248cc40528784075abacf4301 : SUCCESS in 22m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5455bee079789e7a033dfcd0c14f8b4c11c00c88","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-16 15:31:45.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/69/883269/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"51fdc62921f2b6bbec2bfc434fc868fa8e88204f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 16:29:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b075b09a80d4fab872b0a87160a551a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/0eaf765f70c44a8fa9472080a784a8c3 : FAILURE in 22m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8c6b359bbf584790ade501b17d1640f9 : FAILURE in 22m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5a43f6442d4f4952872a46345b85ba1a : FAILURE in 26m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3081ec05183b429ea33b8dd4480e77fb : FAILURE in 23m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/960a3d8e74734ee6bf66522cd5506327 : FAILURE in 26m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b84254dc655c48c8ab0e5681ede54cb5 : FAILURE in 27m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e3a5fffabd84c279857e31ec242f3bc : SUCCESS in 41m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3663ad69588b48fc86fc2f75ac2f86be : SUCCESS in 13m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e9b0b46f12b41c89022cf75ba6274f6 : SUCCESS in 25m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/688989a79aba468ca3d652da9631f09b : SUCCESS in 23m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e9270b5addd4bb89e96a2e408aacf1d : SUCCESS in 30m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4adbd599b4b41528526bc20a4238848 : SUCCESS in 10m 27s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/9f5588a91ef244439a2414cec4c6a3d0 : SUCCESS in 20m 14s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/44eea0fceea04d989b5bcb0ead017a1d : SUCCESS in 56m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7677ae87ea5c43b9ab18db0206c46432 : SUCCESS in 43m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b90b01fc523a4c28a1e5476f50a0d2ba : SUCCESS in 1h 03m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b7340cf9b6ed429a9b2f80074f7e43a0 : FAILURE in 30m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/401e776d4f2147f1b8fad6e69b7cca07 : SUCCESS in 1h 29m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7a9be80ae25b403fb4743529d9db444f : SUCCESS in 1h 39m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ec99b794165c4e21b31370feeeb6b91d : SUCCESS in 1h 09m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/260b2bec27504e6a9418eda06e2686fc : SUCCESS in 2h 03m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4ed109835da843b8abff1f7702ea8473 : SUCCESS in 2h 02m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/163b96ca98b649d3bcddb2eb448764e6 : SUCCESS in 1h 20m 47s","accounts_in_message":[],"_revision_number":1},{"id":"dd862a32a9c0f71ed74c9ef9e50f0795b905a9ef","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-05-16 20:23:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"30569d174372871886e1cfd6d909bba9065adcf4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-17 12:54:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"039a34ed8093bf585a30309b869d066c1f2bbc44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 13:27:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75782398303c4b98a655fb635e667bae\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d9304cb1c4f4b1b91304a15387f3d3f : SUCCESS in 31m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51e83516688e4a129914cd876f3baddc : SUCCESS in 23m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2f80693a92c649ae900b8802a1dabb08 : SUCCESS in 24m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d066dc3cff6fefafd87cab6b3371dcd05e32a6b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 15:26:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61f9117a985f49398eb162dbd7a0568b\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/1de2214bc29542739cc0d44c2341638c : SUCCESS in 2h 23m 15s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/49ada3642e254a868c71147620010d83 : SUCCESS in 1h 59m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/981f7704e0c84393867c26675510f8f1 : SUCCESS in 2h 08m 32s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cf81dd1f084d4de6a10d47a7204bc679 : SUCCESS in 2h 07m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9aaf5452fbf2432da372b932b6291c5a : SUCCESS in 2h 03m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/827806248b2f4dc4826033b7091fd6d4 : SUCCESS in 31m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a257289861bb4e4bbb8721796d5b1b6f : SUCCESS in 38m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1e24f6044f1439c8a02bf97f80a4145 : SUCCESS in 13m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d59684eb308347c3a7f6d57092df155a : SUCCESS in 29m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77417d56a9584bb0b3e7e9f7119ea75d : SUCCESS in 28m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c14e629f72a340c9a0d17412e397e70a : SUCCESS in 22m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f128044e3a47d391e69d371ab5eaf3 : SUCCESS in 10m 43s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/634e32b054ff4d3c9c1d42f7ad59f2cc : SUCCESS in 24m 26s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4adcd0b47374add91f34bff113ae9ee : SUCCESS in 52m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/649eb5f75330453eba8ba9fa8f99f867 : SUCCESS in 29m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f828f6603da14e3eb06df64b428297d6 : POST_FAILURE in 1h 10m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/04411309fcfe48c0939c7f2af6352b58 : SUCCESS in 1h 10m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8615661ee44344079ad9321f6d2525b5 : SUCCESS in 1h 26m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a9440820029f4292a431cb546e2bcba4 : SUCCESS in 1h 14m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/98dd3f288b314642a1c11afbd25dd022 : SUCCESS in 1h 01m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a112867fbdba48e39ab52b6d9750e459 : SUCCESS in 1h 32m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c547b0af20a242f9b7d13347d454be0c : SUCCESS in 1h 46m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d8ef3cb4648c49868f164ccef4b6677d : SUCCESS in 1h 11m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9edd345262c9a90990fb8183031fea230d3de3e4","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-18 17:13:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19ed1b5c7d6abad52808d5bbdb8c65b613186644","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 17:37:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c89607d4dc9847c7b2bc7c8a68bd4ec9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fc0fc3a803841faa20feaca18314cb9 : SUCCESS in 21m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f6b790a630c44dedb902408b5822159a : SUCCESS in 22m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d35dd61c7a5645beaa2e72adef507255 : SUCCESS in 21m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8dc8029ef7a924bed84b227f90ce5ed639c45a2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 19:38:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db2546c986c84cce8e4af459102d8421\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/98d43d3d71ce4ded978322507630d4f6 : SUCCESS in 2h 00m 20s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9d874aa7ef354091b6d1f081f968ce52 : SUCCESS in 1h 48m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/67c2ae8f0a224ce49fae24bc0f58d83a : SUCCESS in 2h 21m 47s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a1ecd8695fff46a7a8c9b17eb9907ffc : SUCCESS in 1h 39m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8a20b64e5cca446fbc0e30c5f2053a06 : SUCCESS in 2h 06m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/06feb5f8a0ac4a3ab497a51a9bb2a02f : SUCCESS in 37m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/918273b942f142daad6257b74f2081d3 : SUCCESS in 43m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05c3c432959d4a9083b0f80eab75c65f : SUCCESS in 15m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a506decde6914ca4b89635b076a485ee : SUCCESS in 31m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efc5bb9880984e0999855141678a95c7 : SUCCESS in 30m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8d132ba41f1425780152948fca82e39 : SUCCESS in 26m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0249cfddf7bd4903b557426c02afbee1 : SUCCESS in 12m 28s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/e0d33bdd41704039ba2df38bbecfcea7 : SUCCESS in 25m 01s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b6f73fb10d3454597e1a81c50702d4c : SUCCESS in 59m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/07c2f8ed86094060b477642d990ba21d : SUCCESS in 45m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b646cb318be3493fabe732b4e67e12b6 : SUCCESS in 47m 11s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2fd244125e974067aaf7be45227b23aa : FAILURE in 30m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c5010646fc614487b0be00d34e2bcece : SUCCESS in 1h 33m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/52ed63b080b94614a6eeb859e7130cdc : SUCCESS in 1h 43m 16s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fb98b17bde2946b4894682496dce6ef1 : SUCCESS in 1h 04m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/016e649a0c8a4b7ebb72c37d828eec64 : SUCCESS in 1h 51m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/33ffa08f29fb4c1db7ec252a8bc1e1bb : SUCCESS in 1h 55m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c37e9552ce9e448491e0c71a26c2589f : SUCCESS in 1h 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c8c907e753102d17b895d44e09fee76e9880adba","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-22 09:41:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d596e483c4d8f31cc28d2c4407f63173c2d81e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 10:38:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/385953d6747248ca8aaf417448996fab\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a91ca79a6b147869912c16a5016a660 : SUCCESS in 21m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33b708bcd23f4fa18b5bfbad6e409821 : SUCCESS in 53m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6f5e5008cb8048f98687982b6e53bbff : SUCCESS in 27m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e8476527abbe992b08d7bb3a9da222054d7dab1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 12:03:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d39e422295f434f992693fdba24d8ea\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/a60b67d3cd3a4750a233bba063614554 : SUCCESS in 1h 56m 09s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/54b376b2e05f47259202661656283377 : SUCCESS in 1h 48m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b95ad55ad183487986ab46b48b620114 : SUCCESS in 2h 20m 20s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d743ea72ea8e4e4cb5603c6e40b9065e : SUCCESS in 2h 01m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/19ff79753cce405f8fb462e24a8a94ed : SUCCESS in 1h 42m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b30e3025f54e4ff1b45a4e16e7a4b08c : SUCCESS in 39m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39e527ada4b14f118ebd26d654366c41 : SUCCESS in 40m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a243fac4459544db954ffc5d01455c17 : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f36b2599fdd4521b79acaa61a7851da : SUCCESS in 29m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d144b0df42fd4c7dacfac88ef7f51212 : SUCCESS in 25m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f28d8dbcb18141fd876718afc89bfcbe : SUCCESS in 27m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a05e4aab97244b8c8fbabecda5fbf231 : SUCCESS in 10m 39s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/88dc38b933fc4de7847f36d5615a7435 : SUCCESS in 23m 36s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b081adc964649f19f62d5085643f94d : SUCCESS in 50m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f9935e2a75e46ad96bfdec61bc1d664 : SUCCESS in 26m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/740d83ed722a44ceac15933eda10d901 : SUCCESS in 48m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1b77d7b1038d4315ad75a6a52d25eb38 : SUCCESS in 56m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/41cb6d68f00c4b038847b578ca836810 : SUCCESS in 1h 40m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3f3f88d872d9437dbdfd203fcc88079f : SUCCESS in 1h 22m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fd711c02264a4ec1bac51c35be0f6442 : SUCCESS in 1h 15m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c5e7623d2ea04aecb0f3339ff67bc91a : SUCCESS in 1h 43m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4041649da8614a81b29316c722049a79 : TIMED_OUT in 2h 13m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/85afdd376dc643ebb1cd842a6c3187bc : SUCCESS in 1h 07m 47s","accounts_in_message":[],"_revision_number":1},{"id":"337cf24b822b9f5197e7cf7a78bd3b75c8d3b7be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-22 13:44:37.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cecb94b55393c6e4e077a4bcc2295cfb936236c3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-22 14:23:16.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/69/883269/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"a1a66a9278214000a1627396b8594861dc6f611a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 14:59:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea38c1a3d6404da2b939c7e0ef19e232\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ce7fdb029304578b19e40c362212732 : SUCCESS in 25m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d6464350cf654b8193b8362c33657001 : SUCCESS in 1h 12m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/610e1194df1448dbb85cf769bf91b55f : SUCCESS in 25m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7027ab1f682d904840d806a2fc8661c7d9972d36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 16:00:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2080e90ba48d480f94e4fb03ac53ec54\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/0208e2d6a1d5486c8917394b3ba481c7 : SUCCESS in 2h 13m 45s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c45673681a8a4bba8cffa0a3306deb24 : SUCCESS in 1h 51m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/de53f5a3718042039f34d6c55cbb756f : SUCCESS in 2h 09m 58s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/77eaa22e9f374d4fbf5004f50bda6116 : SUCCESS in 1h 45m 45s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e2f1ea87172148e6bac3b2c284410c4b : SUCCESS in 2h 11m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2d6709c070dc484dbaada62b9db841ef : SUCCESS in 39m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d19f2b185fe433bb436f784e963dd5e : SUCCESS in 43m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c16060d913694b7d800927e8e4b4dc89 : SUCCESS in 8m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ca10310ae8b4c01ba5f02683e29207f : SUCCESS in 25m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4329200d38a54cae906eadedc30446ad : SUCCESS in 27m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74bd154c7af84ae48997aedcfee97262 : SUCCESS in 21m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/291a6714f27a47dc9e8435f02d93dd68 : SUCCESS in 10m 41s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/66157fd386634dd38b594ae218d03b0d : SUCCESS in 23m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3242b71935d4d2383a618b1bac61b39 : FAILURE in 46m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2eaad50d452c4f9db528d00e7ca3a046 : SUCCESS in 29m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bebc6bb1ab6d4ac59ceae93d0c644d58 : SUCCESS in 55m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/24d5185080d04f28982275af57f70d12 : SUCCESS in 49m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6b2516c3ec204d7ebbec733212320506 : SUCCESS in 1h 20m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a1d67f3ea47843d7855e7ffc8d55d87c : SUCCESS in 1h 09m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e9f6a5663cc34279854a451374823db5 : SUCCESS in 59m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/13eeefe24df84e3f9102029a22eb15b1 : SUCCESS in 1h 35m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cbb2c43b482448f591584c0496237c69 : SUCCESS in 1h 58m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/38cb92b994c94906b3bc08775b2d1377 : SUCCESS in 1h 12m 02s","accounts_in_message":[],"_revision_number":2},{"id":"b5ab0bbe8c6d8e25ab780681f268749c06ac5c9e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-23 08:46:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c659dbc5e96a57efcfd9cd309d16ee59576c6047","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 09:33:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bdd3a1e04ac4900a374eb7f71966c0f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/284355910f0745aba1b40202e1df9452 : SUCCESS in 24m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f91ec83df2c44743b080121d270a885d : SUCCESS in 21m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/498eaefb7887452aaa09f19f06057dd8 : SUCCESS in 46m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c106a1aea171bd33713bb31d6f6f55c47951c876","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 10:58:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e982177c78f54adda3a07cac7d846a69\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/cf5b39386f984b8187de7df5b4c3daa4 : SUCCESS in 1h 58m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/08cc6bdb43d747d7b956471e6769c023 : SUCCESS in 1h 41m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bcd3613e946e440ca299afc1a088e7ec : SUCCESS in 1h 32m 05s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/71ff4ad784a24d93b25b0a105ba11a7f : SUCCESS in 1h 50m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d043926b793b4eefa4cf51931f91da9e : SUCCESS in 1h 53m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aa40a29eee0742fc9afdbdbfe8fe972f : SUCCESS in 47m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3259bb77092a48019444080cdf189f54 : SUCCESS in 39m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fb4d77b3bd14d67bf67274305885942 : SUCCESS in 10m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aeb568b9ee014e9490e5d4c5024c5484 : SUCCESS in 23m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c023e8eee36e44f888f8635b34acb10a : SUCCESS in 30m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09a099ece8fd46ba879c5e3c64ebd8f7 : SUCCESS in 22m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa9a788e73a4acabb3a4f8e42bd69ad : SUCCESS in 11m 38s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/03c36e9c4f5a45d7b66fec5292f649fe : SUCCESS in 24m 48s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9cf994e6e2074f28a2cceeb23ef09aaf : SUCCESS in 52m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/905e534e22ca4ad2858d4ccc710de531 : SUCCESS in 1h 51m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d1c7c0055014fdea136b6c1a73a964f : SUCCESS in 57m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f422618e399f46798b3b83cf55bf8d1c : SUCCESS in 48m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/74e71df3dd934704950e0f148404309b : SUCCESS in 1h 27m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/25e866df7fc14d9aa6d7dc9e3c883ce6 : SUCCESS in 1h 21m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8a48333bf1074676b4ad24b4227d5df3 : SUCCESS in 1h 09m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/46b086314c1d47c29f62bf03024aa74b : SUCCESS in 2h 07m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c021a5d1880647ab809ab42d341b812f : SUCCESS in 1h 41m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/609ca7fe4a3e4aaf81434c34c17567b1 : SUCCESS in 1h 14m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f98327075d1ccb23305f2fd6791a8218dea42009","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-24 12:10:41.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f96059b5d7b06db78cdc846d2f553921ca847ae0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-05-25 10:48:48.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5c09765b3c05b83c0a1d97eb1c732021e6a8f557","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-05-25 11:27:12.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2023/Arista_Tempest_Testrefs/changes/69/883269/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"51bfe18d5ba148ed38f7c9da9cff87efcfa25875","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 11:47:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a2f300a431b4c2f8a4888ad387b112e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c58b4ca62947485f84bc15796324fa35 : SUCCESS in 23m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c79ed1bb866e4dc39b3e7e58e804ae5c : SUCCESS in 56m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c22dac880ea42b8a23d397dfed70f5a : SUCCESS in 20m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"35f1b67be1faf70dd1a4324cf0bf6228027d5066","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 13:09:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b9d769548f74f22953e89396ce0aef0\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/5bbec1ce83e64692bde00a8bfbf473c6 : SUCCESS in 2h 10m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/441bbec76cc046f69bf024a9260630eb : SUCCESS in 2h 17m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/250c1700371f4468ba4a626a79ebfd63 : SUCCESS in 1h 58m 23s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6f7bc1f0da9a4634a8adf1620f9a0de0 : SUCCESS in 1h 55m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2363295b1faf4de1a67c0a3286b86d4c : SUCCESS in 1h 50m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3b1bfb90055a404897c47b44944e2a39 : SUCCESS in 41m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74c112073a594072ae2877cf9b532662 : SUCCESS in 42m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17279041b06f43fca9a951dab77a9544 : SUCCESS in 10m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cefdfaf647ae44c0acac6863dd920301 : SUCCESS in 28m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b74973eb178244e9866d00a09406779a : SUCCESS in 22m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83f28b2848c6449a9f5780143ccb0426 : SUCCESS in 29m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df86f128ecd4f8bba941c1bda636de6 : SUCCESS in 11m 29s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/968851e9ec274d2b8b43823664ab4b62 : SUCCESS in 28m 24s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/20c51792aaa24ac1bb174332d37f5e61 : SUCCESS in 53m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/474e5bc8344d42a392591c215f4b681f : SUCCESS in 1h 35m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6bf2cc4ba66c4125ae7584d36161f1df : SUCCESS in 55m 11s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/679259cab93d49cbb3e75124e9a83ef6 : SUCCESS in 1h 02m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/202bcb422fff4259959e218294234604 : SUCCESS in 1h 28m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c4437f07d1324954b6658ce471958de0 : SUCCESS in 1h 23m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c38e0120c0cc40bf9dd53f0e6bd54ac7 : SUCCESS in 1h 11m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dc829b7a8c1f4799ba496c9ce8df760e : SUCCESS in 1h 28m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ef60eb6145e54e1a802613522ea60ee3 : SUCCESS in 1h 45m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cf53ca61d1864c9880b7a168d59b43ab : SUCCESS in 1h 28m 03s","accounts_in_message":[],"_revision_number":3},{"id":"130db8a10d73037b61aa352f72fc14d204f3a2dc","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-05-26 22:52:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2484d052727214568b0eadd1df1ebbe9ec6b6936","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 22:52:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cbb49b69a7b3e1f1fee9bdbc5e86a34c39ebe48b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 01:36:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6190b0fdfc3a4fb289aca19e6241d0af\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f95eeff5ebc544be9d980f5d552c44cb : SUCCESS in 1h 56m 44s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0e79dffb958b46c4a519428912799c9d : SUCCESS in 1h 26m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/baeef666d68049c99e139abbf4ce874e : SUCCESS in 1h 29m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3f67f931d1014281b16a49c6ebf2be80 : SUCCESS in 1h 45m 15s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0081dd12c5c7492a9c8560932c9cebd0 : SUCCESS in 1h 31m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70a2cf7c058045219a4f40d253b309df : SUCCESS in 7m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a1b5f7a51024320a5700b56a70310a6 : SUCCESS in 19m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7328b8cd133468d95d2a0fe9780076b : SUCCESS in 20m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6bf8ddb1da9498ca7f756c16c8dcff5 : SUCCESS in 19m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdd3eaaace2849aa9f2336957e56f37b : SUCCESS in 8m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c20fd606f3484b7c88bf7dec6ac6a1c9 : SUCCESS in 45m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b253fbf581ca43679997ffac82f7d67d : SUCCESS in 1h 34m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6cdbb7711b154c159ba39b9b4cdb4ebf : SUCCESS in 1h 49m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea13bfdbb889457ab8cef3f93b96b1ca : SUCCESS in 59m 10s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9659093b020f44f8a01482844f846659 : SUCCESS in 56m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fff6c839035f4ebfafef14c6ae000228 : SUCCESS in 1h 23m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3eb6f84cd8c8434faadfc7c8e1f21f26 : SUCCESS in 57m 08s","accounts_in_message":[],"_revision_number":3},{"id":"93d69df267b92b46f7aaacfac73cc316307d8105","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 01:36:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ca8ac7fecb65ab463df35f228c9f3baed05601de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-27 01:37:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4dfc52d2d8c4ebf8774dbc3049d7d01\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adfc0b2b322141aabf91d68b3b3b538f : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a72e97ddff1c2a13cc3dc7636b927bd28bdf2ae3","revisions":{"cfa4c2659267baa552ab7a75676c0d1fb90e1211":{"kind":"REWORK","_number":1,"created":"2023-05-16 14:20:06.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/69/883269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/883269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/883269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/883269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/883269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/883269/1"}}},"commit":{"parents":[{"commit":"ae4084c1735f63de7fc685ee51502f5d0d9db67c","subject":"Merge \"[S-RBAC] Get QoS rule types API available for READER role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae4084c1735f63de7fc685ee51502f5d0d9db67c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-16 14:17:58.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-16 14:17:58.000000000","tz":120},"subject":"Update api extension for default sg rules API","message":"Update api extension for default sg rules API\n\nThis patch adds two new attributes to the default SG rules API:\n* \"used_in_non_default_security_group\",\n* \"remote_address_group_id\"\n\nThose new attributes are descibed in the proposed update to the related\nspec in [1] and [2].\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/883267\n[2] https://review.opendev.org/c/openstack/neutron-specs/+/883268\n\nRelated-bug: #1983053\nChange-Id: Ic3e06460ac8294bfa882991eb678878b238735d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfa4c2659267baa552ab7a75676c0d1fb90e1211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfa4c2659267baa552ab7a75676c0d1fb90e1211"}]},"branch":"refs/heads/master"},"b80575d9c2bc4d1aa9bf4ab4c797697481504284":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-05-22 13:44:37.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/69/883269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/883269/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/883269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/883269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/883269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/883269/2"}}},"commit":{"parents":[{"commit":"e29fc916b3dd7d893547608d7e8f15887016e1eb","subject":"Merge \"Update pylint version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e29fc916b3dd7d893547608d7e8f15887016e1eb"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-16 14:17:58.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-22 09:44:08.000000000","tz":120},"subject":"Update api extension for default sg rules API","message":"Update api extension for default sg rules API\n\nThis patch adds two new attributes to the default SG rules API:\n* \"used_in_non_default_security_group\",\n* \"remote_address_group_id\"\n\nThose new attributes are descibed in the proposed update to the related\nspec in [1] and [2].\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/883267\n[2] https://review.opendev.org/c/openstack/neutron-specs/+/883268\n\nRelated-bug: #1983053\nChange-Id: Ic3e06460ac8294bfa882991eb678878b238735d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b80575d9c2bc4d1aa9bf4ab4c797697481504284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b80575d9c2bc4d1aa9bf4ab4c797697481504284"}]},"branch":"refs/heads/master"},"a72e97ddff1c2a13cc3dc7636b927bd28bdf2ae3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-05-25 10:48:48.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/69/883269/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/69/883269/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/883269/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/883269/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/69/883269/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/69/883269/3"}}},"commit":{"parents":[{"commit":"d242792c47b703a72ace065f35eed86eeb2918e3","subject":"Merge \"[sqlalchemy-20] Add the transaction context to the upgrade checks methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d242792c47b703a72ace065f35eed86eeb2918e3"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-16 14:17:58.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-05-25 10:48:30.000000000","tz":120},"subject":"Update api extension for default sg rules API","message":"Update api extension for default sg rules API\n\nThis patch adds two new attributes to the default SG rules API:\n* \"used_in_non_default_security_group\",\n* \"remote_address_group_id\"\n\nThose new attributes are descibed in the proposed update to the related\nspec in [1] and [2].\n\n[1] https://review.opendev.org/c/openstack/neutron-specs/+/883267\n[2] https://review.opendev.org/c/openstack/neutron-specs/+/883268\n\nRelated-bug: #1983053\nChange-Id: Ic3e06460ac8294bfa882991eb678878b238735d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a72e97ddff1c2a13cc3dc7636b927bd28bdf2ae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a72e97ddff1c2a13cc3dc7636b927bd28bdf2ae3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
