)]}'
{"id":"openstack%2Fneutron~883282","triplet_id":"openstack%2Fneutron~master~Id72ab1e521d7d4b6186577ca1c5a6bd1ad4f4648","project":"openstack/neutron","branch":"master","topic":"bug/2015065","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-12-04 04:53:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id72ab1e521d7d4b6186577ca1c5a6bd1ad4f4648","subject":"[DNM] check grenade","status":"ABANDONED","created":"2023-05-16 16:00:40.000000000","updated":"2023-12-04 04:53:45.000000000","total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a926c45a383371111accf894e355430de90ed36a","_number":883282,"virtual_id_number":883282,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-24 07:10:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-16 17:09:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-01 18:32:08.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"}],"messages":[{"id":"4e3f2bebda886d272ebafefaf4377ce4cc5dab4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-16 16:00:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b8c1da276c9a4d0e2750238c697802d11740fe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 17:09:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/646eac9eb8b54f2a836db568c51c87f3\n\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e64b3a1e811d41b1b20ea43b18ec4050 : SUCCESS in 44m 36s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/463ced7c87624a9a93dd7ee6817748a0 : SUCCESS in 45m 10s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/69a82f47aa58465892fadcf8789b464b : SUCCESS in 58m 53s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/72a8f8b7ba28466d9981cbd7c13236ab : SUCCESS in 57m 55s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/6894dea0c95446bead7e69da87763c2a : SUCCESS in 1h 01m 31s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/9ca799ce71184727ac2b31fbfe473ca8 : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":1},{"id":"75cdce42eccd155aab3fc969c33b98b2a3869df5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-16 17:52:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e583cd4d965b0e6ba560b2beb5ad45bb5abb5c00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 19:15:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d53a121dc80545b9b5a840f44b025a53\n\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/646a09baee2242d4a8f4a223073ea684 : SUCCESS in 47m 34s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/238ceeedf1c74c1eaa8df6083c962aa6 : SUCCESS in 1h 15m 27s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/85c96ee6c8104ebc856ca8b42a66a585 : SUCCESS in 59m 30s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/070da908619a49e28493b2c545dc5a17 : SUCCESS in 1h 08m 35s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/435f6c29451b46a1901d8eeaf0f546e0 : SUCCESS in 58m 31s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/1186bd44a56a48d686c94b230fcabe60 : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":1},{"id":"1a99779089c3a3b131685785a8d58de1503717f8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-17 03:56:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11fb5cb96f23869d6243b69130d2fd7744f76a8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 05:18:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27e0026b2c7048b5ada24131b4c82cff\n\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c0d160a32644ed1abdb94d9707d1b9e : SUCCESS in 1h 01m 48s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/e10f416d14374288b3d2064ef6df6681 : SUCCESS in 58m 58s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/270c863179b049a0aa871289a3bb5731 : SUCCESS in 1h 03m 51s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/9d27e74879874a33919a2da5ab876419 : SUCCESS in 1h 01m 10s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/da0f5e904a7a46a78d25da9a2d8b7f72 : SUCCESS in 42m 19s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/3968fcae2ec34d7394f0a92e33873cf7 : SUCCESS in 1h 08m 49s","accounts_in_message":[],"_revision_number":1},{"id":"07932bc469d2b8d03d674286a5b905231100cbd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-17 07:44:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6ed505e6ede3b7eaf51d85ae9174c32b08acd94b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 08:58:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98bd956acc3e49fabd955a1fcac64c2f\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/2e6902a79e2b406ab679944e63484f87 : SUCCESS in 1h 00m 06s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/5ae2bae86d3244f1a08a4edc70819e57 : SUCCESS in 57m 55s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/ef94bf8479dd4c4bb365b7684d1c3216 : SUCCESS in 59m 12s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/c414a9fc8fa440968fbb4b6984ffe64d : SUCCESS in 1h 08m 32s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/96c924c18bd64a06ade40e22b1e4cca3 : SUCCESS in 44m 46s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/dc14640d4b3d40558397576864eed3c1 : SUCCESS in 46m 01s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/51f57e2a5e7a4ea2b35802f2d98efec3 : SUCCESS in 46m 43s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/cf391da75ddb4ae997d87e4799b4909b : SUCCESS in 42m 22s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/50215a8af53b4dd3abf4e0da3545734d : SUCCESS in 52m 19s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/a50ae522221f467cb6e60ce96f6110b6 : SUCCESS in 48m 07s","accounts_in_message":[],"_revision_number":2},{"id":"c36dfa33cdeea592addf84c67a527cd237222668","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-17 09:31:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9eca5eb981ef62c444a3eab44484b85ac2e8b8dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 10:59:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f51d51fe9f94b6bbd534f9f9d68dc3e\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/866d2160656349fdaa6259fe665837d4 : SUCCESS in 57m 35s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/7038c58b00ee4251bbd976b597f54cdf : SUCCESS in 56m 51s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/758820f5c61749678e4f976e1b001d22 : SUCCESS in 39m 30s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/c10ce2d0cf9f4e9bb00b80df85b93d97 : SUCCESS in 54m 50s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/ef9fd99b1f114689aa3f83b5f6a17562 : SUCCESS in 57m 50s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/3e553f258bcf48a088eb4959cf952081 : SUCCESS in 57m 38s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/c9a37a85143448868f267cf5d045cf49 : SUCCESS in 43m 20s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/51f6752f0ef04178b0b36a1348f23954 : SUCCESS in 57m 10s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/18ccc22292fd4f92b958be025a303b39 : FAILURE in 43m 24s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/d0e96fd21d1944589aad5705efe25163 : SUCCESS in 56m 15s","accounts_in_message":[],"_revision_number":2},{"id":"779be1641faf8973fa454adbbd889a1e21c8d02a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-17 15:26:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b9101a06574eb2f589814a3434bf90a7c7684902","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-17 17:34:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3a2c3c3665e5a6d9d8b19472d48b7333d03e067c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 23:03:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/99665510038341b1894a109d597423f9\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/1e583548d3da419b9c5e292e4231ae88 : SUCCESS in 49m 14s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/78edf0b82f6c473ebc88d4e4111e19b0 : SUCCESS in 1h 04m 44s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/37276202b0bf4a209b694a972a327a1b : SUCCESS in 55m 36s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/02e90a51a0fe418fb528b5aed3d80e47 : SUCCESS in 48m 13s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/e4198e222e534cba8d90cba5559193db : SUCCESS in 52m 34s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/b0ac57c1994d49eaab6ba4a20c93b8a6 : TIMED_OUT in 3h 03m 30s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/ef79b3d711d04455bc86fcc43f724bdd : TIMED_OUT in 3h 02m 22s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/86595ef9f7e84e4dbfa3729a54719942 : TIMED_OUT in 3h 02m 22s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/7c6c78b162324fbaadbbde73b2c54296 : TIMED_OUT in 3h 02m 36s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/dd4aacecdf7c4540b4a212c2e3b31647 : TIMED_OUT in 3h 02m 17s","accounts_in_message":[],"_revision_number":4},{"id":"59187aab6266e4fec20d2cf12c04e06cb86809c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-18 04:16:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"136b367c303ae51d5336ca4cfc36071fb9f6be4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 04:47:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d13b2fe91994fe1ab2551bbddf91a1b\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/209c6ee0b09c415c94824fa215a04125 : FAILURE in 23m 46s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/aa2eef65fb3949b7a4bc5da6a00fefd9 : FAILURE in 28m 25s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/7151e31da0d74ff7b5aae6118c00b992 : FAILURE in 26m 53s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/b98fe3f3a73e457f9017bdc8241f546a : FAILURE in 22m 26s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/ad2dd823cc87418d817b10d6cece371c : FAILURE in 28m 49s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/67f906a1807e445f86b10a35e26be153 : FAILURE in 19m 16s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/af6c0d13959146c89d607bb382e0d9ba : FAILURE in 18m 35s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/e4ce0c3d7ec34318a58f8966fe6a2836 : FAILURE in 24m 16s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/cdc026e2537341d0a9991916cf93fa92 : FAILURE in 20m 01s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/e5a74d9fabf44cdfa6a43d5f2e53dce8 : FAILURE in 19m 52s","accounts_in_message":[],"_revision_number":5},{"id":"a8deabda8c79d34cd9e261da0affb166295a5dd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-18 06:01:44.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"907a4a1629af5d7b1d5f0e52562a38b50504dd03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 07:06:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddca471099534cf9974ba4e4cd371be8\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/25bc13ca651c4c928fed770422338a5a : SUCCESS in 46m 40s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/e492278af29743b8879345b64cd075ef : SUCCESS in 44m 38s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/9b63abbdfd934432828c122e2c894267 : SUCCESS in 44m 47s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/15a4b77c3e264ab3a96179d6c56f1319 : SUCCESS in 57m 00s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/8a3c3d6d40714749a3116cc30f51b66a : SUCCESS in 36m 55s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/2883acf905924253abb83ecf6f72a682 : FAILURE in 46m 41s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/86976705eaa7499db6bdd653139f0393 : SUCCESS in 38m 37s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/21e62b7782aa4d4b9f06060cf57a212e : SUCCESS in 1h 03m 48s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/32b533cd48364010919808665d6a45c3 : SUCCESS in 57m 16s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/73cf06c359264da6aa2ed3592e1389d7 : SUCCESS in 39m 00s","accounts_in_message":[],"_revision_number":6},{"id":"bc3d01da9e64b44895d5ba2c542711a80dddb52d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-18 07:06:28.000000000","message":"Topic dnm removed","accounts_in_message":[],"_revision_number":6},{"id":"9b8f88760ba6a229292dc1d5ed9399ec574844e8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-18 07:06:34.000000000","message":"Topic set to bug/2015065","accounts_in_message":[],"_revision_number":6},{"id":"38eaf936a7b7c8d3efe8de6fcfc717c12ad86573","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-18 07:07:35.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c0010f3842aa105da1f100eed4906b8fa0584ad4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 08:16:34.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/132844897a88418e9bd97d2cbac7b327\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/b5d61b1159f8466dafa3adde9ea19d33 : SUCCESS in 55m 37s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/1d78316510874189acf3dfda5d2bfb11 : SUCCESS in 1h 00m 22s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/182d7150dc1b4c72ba6d2cc84c0f1af6 : SUCCESS in 1h 00m 46s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/3b18fb7e0733448a9a1147bc35da6d1a : FAILURE in 41m 55s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/9818019e3a714b3cbd589cff07cb0abc : SUCCESS in 58m 31s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/f6c2c204f7854599a442efc7ea9fa09d : SUCCESS in 1h 01m 18s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/364380270df743e9aa90c6ea94c58883 : SUCCESS in 46m 28s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/96280ec4dede4aa698ae7e0787a8c5dc : SUCCESS in 51m 39s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/191163532ac341dfac36760be30efbe8 : SUCCESS in 44m 02s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/f245da8c82e9409fb91f10a8a59f3138 : SUCCESS in 1h 01m 04s","accounts_in_message":[],"_revision_number":6},{"id":"bd4bdde4f79d8e5b61ab72b2ef44f44bfe7ec767","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-18 09:08:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"77e0b8adcf74afbe465c0d045e9ea6378f923ba7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 10:36:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3907be7ff3743b9b2c89124ff077e3c\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/32bf68feac1d4735895af28077f89984 : SUCCESS in 56m 52s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/7111c0ccba0544f8bd84ef10129f02b7 : SUCCESS in 1h 04m 56s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/153e6a4d79824625a35abc7a15b064c8 : SUCCESS in 53m 38s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/d378dcb209f0479db76426fffd6dc8aa : SUCCESS in 56m 02s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/e3aed97d735444b0b8d65b486c538f08 : SUCCESS in 1h 06m 15s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/aac12d23325b4d7b8647932c0fdca394 : SUCCESS in 1h 07m 36s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/ee130efc21c64335adf17d61ca0aef1d : SUCCESS in 1h 18m 02s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/9ac954f5bdda44edbe420aac26c2e614 : SUCCESS in 58m 38s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/fa36dee6cbaa4d688007c2a01528c155 : SUCCESS in 1h 10m 47s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/30102377a7f44484b472c78c5b82501c : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":6},{"id":"9c97e736ac34dd7cf09617fb97cc0118315ece51","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-18 11:06:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3e793ad9ac0a6e0d662bdf4095d5c2c981c6ae76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 12:25:59.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed656e33d09b451c93ddc00f9953b991\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/1e98b6472a504cd5a7161671c847ba2e : SUCCESS in 57m 48s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/2d31810a8ea84b1ebcee1dc36fe716e2 : SUCCESS in 47m 54s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/581f7b56cf4d44a7b115ddd9bc55335d : SUCCESS in 57m 07s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/a6638ab975494d70ab09a6372bc8922b : SUCCESS in 1h 01m 20s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/c23b4a041f6c46fd934eeaf962d0cd6d : SUCCESS in 48m 42s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/58c7ea3e06904bb59b2a70b9ddcd229c : SUCCESS in 1h 06m 53s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/2daff9ae27144259aa8706e4037c4a84 : FAILURE in 31m 12s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/81a1ce9f5fb142a0b19cf110ec4c5f38 : SUCCESS in 47m 20s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/add3592a82cf41aeb37bcc760ea368f4 : SUCCESS in 1h 11m 33s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/5cb2c3f1040e4e52b66b3c6c9e60a43b : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":6},{"id":"5ee834e4592e1dcf64e174616b95cc8c599fd51a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-18 12:39:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"23c04ae28449b12d3bece2ace982aa2741cd7fad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 13:55:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e93fc2e69b2e4017b7b27a8131a3afe1\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/b895bfacfdc246fbbaf78f4a69dee8a3 : SUCCESS in 49m 59s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/273f5ac7ce644b689c1df2d9d8fe6b0b : SUCCESS in 1h 03m 42s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/7caa4d376c0641a9858f9c6e324c925b : SUCCESS in 56m 34s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/432ab0f802274ef69d0cf4b98eb5f7a2 : SUCCESS in 59m 19s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/66b982d1bc22438ca1da2ed5c3528fc1 : SUCCESS in 52m 21s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/15b1b75dbf744f828f3d8aaf2d7a4181 : SUCCESS in 46m 39s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/35beab6c2a1e486c9f18d07e92b84029 : SUCCESS in 1h 07m 11s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/2fe2360300aa406da48d0333800f45ff : SUCCESS in 58m 40s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/9058138a31304b028a90fa69ff1a184b : SUCCESS in 46m 15s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/6d1010fac56149b588d53b4a6309ee11 : SUCCESS in 55m 05s","accounts_in_message":[],"_revision_number":6},{"id":"4ca74b71efbd914224df0fd640483fb962daea51","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-18 14:01:01.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"63954dc389d9f3e8e45b3c54cdcc8a5e820d1242","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 15:27:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a20c78d7796f48bc82f0c8e483164ac5\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/1a776e12406442e495cbe637a9f99611 : SUCCESS in 1h 19m 55s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/ea897b9c931642d889934faaa66af5e1 : SUCCESS in 56m 32s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/85dead4d0b7843c9ba00cab581025dc1 : SUCCESS in 1h 00m 22s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/185a9bef08874c2dae635388a2f30104 : SUCCESS in 59m 10s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/548a07220eba4ba49c55e5bd4fc37fd3 : SUCCESS in 1h 05m 08s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/071dbf38abf14eb5ac28da75e3f478cd : SUCCESS in 1h 02m 49s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/7ce00e8293eb4fd49e40d5478b8c6b99 : SUCCESS in 1h 04m 39s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/50384ab4fc2e4925979e165b992b1917 : SUCCESS in 59m 25s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/6ab1fb716d794a02af25beaebb938edd : SUCCESS in 57m 31s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/309595d52f4747efa589ccc1ffec4191 : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":6},{"id":"5a8354f1c89bc99b9a63b2682e4a50c98d1eb03e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-18 15:43:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6cf89d9712d161d79ee86887f53e8e75c5a18b35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 16:59:40.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/51d0ec434e1443318cbf1354b3052602\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/36f66c5c0fe647d29c3343caa93f5c0e : SUCCESS in 46m 50s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/a43f495a2b7f4e088fac4f98c5c10c69 : SUCCESS in 47m 05s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/44a9332a2e364e66bde0e345c9710a26 : FAILURE in 54m 38s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/c7e67d3c3eb9401a8928419822c55c97 : SUCCESS in 54m 57s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/eb5eeda45e244b6498fc705a2cec21c2 : SUCCESS in 1h 07m 17s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/fd18980d7707468296005cc94380ded8 : SUCCESS in 52m 11s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/f0c0661d341e4bf18fd55845520e0e8f : SUCCESS in 53m 35s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/d1d3ddab71524fa8917e568c12a22b6c : SUCCESS in 53m 14s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/769c227f9da348da855bd81fe70932b4 : SUCCESS in 1h 03m 10s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/96c1ef7bdf184f739085c7f7bbdc790e : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":6},{"id":"831b8c115e46323e809591c6ea99f27e463405d6","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-18 17:06:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"53386fc9b77c5b8ae0356a62403bc8f773dc2ef0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 18:19:42.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27c668896f7e4bb2a16ffed9a2ec2daa\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/f6619bf6ed944195937a3e5af0e37d0e : SUCCESS in 1h 06m 41s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/baf66908a5144c498de1fcd83671e8a1 : SUCCESS in 46m 23s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/e1f98a1e15864a75a9b5257d944f8a62 : SUCCESS in 53m 30s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/3b166c6e34b5486ea421fcfded94781c : SUCCESS in 1h 04m 29s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/ca5c34799b0b45a5944cd95c9b3b1f9d : SUCCESS in 56m 49s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/ba67db2f81984b1daeee4874ef47b4ec : SUCCESS in 49m 13s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/ce0e2e7df63844a2af8e672c2cb2c0df : SUCCESS in 46m 12s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/48bd575c511a4e9190175bab39e5ded3 : SUCCESS in 1h 03m 54s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/89026e3708f642afbb24223800923053 : SUCCESS in 44m 13s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/5cb64a0cf00c4dcb93ed2826855c622a : SUCCESS in 1h 02m 10s","accounts_in_message":[],"_revision_number":6},{"id":"f18708eff53bb4692630d7c7dfc0c89d2c563f7b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-19 03:44:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"863d3b81b0cd4752b8886e89a3265e0d50448697","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 05:32:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2a3213978134d59965feb3ea10f0eff\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/fdaa83b43a08415c9c180ad0ce27ba9e : SUCCESS in 44m 33s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/95a26e31ceaf41128f900353d208df9e : SUCCESS in 57m 42s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/3160f08c023a45d59cdcbc2f287bea60 : SUCCESS in 42m 41s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/68421b8dc40b45cd8a1445e479d24eac : SUCCESS in 53m 45s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/776f88b515de4e2ba76d4f0d73b82499 : SUCCESS in 41m 26s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/735004e3161a4df592a193ab4e79fef4 : SUCCESS in 1h 03m 08s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/124d4f8bccb44bf49376b72dc4da56bf : SUCCESS in 1h 08m 39s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/cda9fb2262fb40ad9838f82f82525584 : SUCCESS in 50m 17s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/b8997bc20c384c25aac01c572720b6ff : SUCCESS in 1h 06m 32s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/8d57665c67f7468598026cd362559205 : SUCCESS in 46m 09s","accounts_in_message":[],"_revision_number":6},{"id":"b8dc162df7189d53d96fc95667f1d9ddececdf36","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-19 05:40:03.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ec885aee4c2e40011d49cdf783771431ee89a905","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 06:29:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/894b7e8309914ee38751d9a1371380a1\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/3ee6631838e24ce5aa3f1ad9c8565870 : SUCCESS in 45m 39s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/a6d61ae557fc4161bce0bbcaaaca6b0b : SUCCESS in 43m 18s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/688d791c15434bf8b8ebf06074898e87 : SUCCESS in 44m 16s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/ba23614616514adcb451c58517775a02 : SUCCESS in 48m 19s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/20c0fd38cb88406f99d6079e8405d97b : SUCCESS in 44m 47s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/1fe5ccaece434d01b423e7f141889620 : SUCCESS in 46m 59s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/6082989a75324f17a662a269bf08f25a : SUCCESS in 46m 39s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/0a565cd07d6e42ddb76ee0f78d3662ac : SUCCESS in 44m 04s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/2088bc8e2e84449dae9fa9dea06c1b35 : SUCCESS in 44m 38s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/374243fb90ab41798ae37fba225ff324 : SUCCESS in 35m 39s","accounts_in_message":[],"_revision_number":6},{"id":"a2424b8b5013797ceac4d475d30610aa7c5555cd","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-23 07:17:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c231e13e8309a73c6d40ef38c06c28e0a037f889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 08:31:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/569e3514459b40b492da42bce3be96a9\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/fa362bf604e648a5a119fbd1897ae94c : SUCCESS in 47m 30s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/35d5223c67e44c9595b3ba9794f71139 : SUCCESS in 46m 11s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/55ee98c9914c4381a431fedea9fd02ca : SUCCESS in 42m 01s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/121100bef3914ed5a65c69b97fa3ac1a : FAILURE in 42m 35s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/3a268f9527b6485693ddfc3077f75911 : SUCCESS in 1h 02m 07s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/e112e24ad2f642718310d043d1f274ef : FAILURE in 49m 09s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/bf5cd3f4050849b2b4e7345c28c8f824 : SUCCESS in 59m 03s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/12fc03a0b87c4a3aab05a2ad57d3bedf : SUCCESS in 49m 52s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/71411693b03b459ebeecf66d89af4fa1 : SUCCESS in 41m 02s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/80a604115979470fb81bc1efd086cd49 : SUCCESS in 1h 05m 44s","accounts_in_message":[],"_revision_number":6},{"id":"173720036926a0602bb0c3e4b6ed563bba3862bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-24 06:00:59.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"97558eec1b9b77613417fb5e020e187614271a8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 07:10:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cba0f4bc3094ec88b79d5b0acab8adf\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/fb4221a3d7834e878aa498c64eb5202f : SUCCESS in 47m 15s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/7d112e8391de46f288167d44726a025e : SUCCESS in 1h 01m 19s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/ec95864639a34b40a952b0bf9b20bb3b : SUCCESS in 41m 55s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/43f677e0d04842d2bea42184ba4ef23b : SUCCESS in 42m 19s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/5e7e63faf39144c5946de79d46df8383 : SUCCESS in 40m 15s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/3061db59d95e4c4ca66bf572b4e69af8 : SUCCESS in 57m 50s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/232973b4c0414850895ac76549c3957c : SUCCESS in 45m 24s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/6ccb0b49dd5d495ab85ef44c6efd2e13 : SUCCESS in 59m 29s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/108e48a2213540b5805c60cacfb57dc6 : SUCCESS in 1h 02m 39s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/4c01bcb46d4845619a4eb10ba4e19c2b : SUCCESS in 58m 18s","accounts_in_message":[],"_revision_number":7},{"id":"5ad9035e9fdbdb88470dd2a1dd4b98dd301721f7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-24 08:15:37.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"322c9f3539a1bebf7e04cd1917553909cc2b6853","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 09:26:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e9fcd6a18474ee68176837121838a7d\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/4a8f5bfdedd440d79dbfa0c08038ae64 : SUCCESS in 1h 00m 43s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/68a40cb3ea564db3ab4ee39aa04db78d : SUCCESS in 1h 00m 23s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/2e303326d4de444c8d65bd77825d4c91 : SUCCESS in 50m 19s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/ef0575841028420e991ae13224e81231 : SUCCESS in 51m 30s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/3dfc3e5353214046a456e00416db5629 : SUCCESS in 46m 22s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/5f548cbf73604cdd9bd6bcdcaa54b35e : SUCCESS in 53m 14s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/160dc6f3075f4405901414756e54636b : SUCCESS in 1h 05m 16s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/023fdaa1f817495f958aa277ebc1b871 : SUCCESS in 1h 01m 20s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/796e018e6ccf4b5db704eda75c7d2853 : SUCCESS in 1h 02m 34s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/ddc24e690c6a4e2aa4915fd118a0aab5 : SUCCESS in 1h 01m 08s","accounts_in_message":[],"_revision_number":7},{"id":"0288d8d2429b771aaead876035e6aa4e049a9f21","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-24 10:59:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"35c13895b40f12696276aa55999c215e981885fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 12:19:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/734a4d07c92d4892953a4cf8a7bbc716\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/2d3d8b522ec542fcb672b62967108dd1 : SUCCESS in 57m 35s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/e6629a3e644147408dd7013d86bcb78e : SUCCESS in 52m 02s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/79efc4f0bfc942dda3e060134898002e : SUCCESS in 55m 27s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/60a9bf9ecfad47e0b74436bbbbedd79f : SUCCESS in 50m 06s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/8191a1670bd34b33a1cf42fb521a4036 : SUCCESS in 48m 40s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/c47b6f02bebe41d7aa152750c1214fe3 : SUCCESS in 59m 02s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/254d484d9dcb400f9cd55e7114cef749 : SUCCESS in 1h 00m 54s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/3f15cad22dc2402b9d90ce674b601ab2 : SUCCESS in 59m 28s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/6136794909904253b5eca8cd92ee1cab : SUCCESS in 53m 13s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/858a4573a80d4e67a4e48086d9bc18a2 : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":7},{"id":"f924a0dfa89930825a3ba3fdaf1283c289edece7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-05-24 13:08:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c7b1415eba5bd83dee3f84b7a2f8c399c7d29686","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 14:19:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a565d7aa7edf4697821358517405e32f\n\n- neutron-ovs-grenade-multinode-1 https://zuul.opendev.org/t/openstack/build/287f2fcf00494afab06fb1631a732e2c : SUCCESS in 51m 46s\n- neutron-ovs-grenade-multinode-2 https://zuul.opendev.org/t/openstack/build/fba29cdc2aaa4f57a10c9a51b81408e1 : SUCCESS in 50m 16s\n- neutron-ovs-grenade-multinode-3 https://zuul.opendev.org/t/openstack/build/6b908a2de78e42e49925ec6fbe3fd876 : SUCCESS in 49m 46s\n- neutron-ovs-grenade-multinode-4 https://zuul.opendev.org/t/openstack/build/aa16cd992de64af9b619cff6186abc61 : SUCCESS in 50m 35s\n- neutron-ovs-grenade-multinode-5 https://zuul.opendev.org/t/openstack/build/79b0dcc4f6d14ed7aacf08d5c0fd36f4 : SUCCESS in 48m 51s\n- grenade-1 https://zuul.opendev.org/t/openstack/build/7be315a84e6e415780c0518bf2b5c4c1 : SUCCESS in 48m 41s\n- grenade-2 https://zuul.opendev.org/t/openstack/build/0cd717d262e44af2b2258ed2ce356852 : SUCCESS in 44m 25s\n- grenade-3 https://zuul.opendev.org/t/openstack/build/36e461c4296147cda93164048acc3b98 : SUCCESS in 56m 18s\n- grenade-4 https://zuul.opendev.org/t/openstack/build/318e3935e6834169bfe5dfda27d20b31 : SUCCESS in 48m 00s\n- grenade-5 https://zuul.opendev.org/t/openstack/build/1c2c8e8a672c41cf886ef10e05b651d6 : SUCCESS in 1h 05m 37s","accounts_in_message":[],"_revision_number":7},{"id":"03bff0cec197496bf31a08917220b58b351443ad","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-01 18:32:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a926c45a383371111accf894e355430de90ed36a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-12-04 04:53:45.000000000","message":"Abandoned\n\nwas for troubleshooting the issue","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2cea94de8b45cb214644a74db65ff90e248de12e","revisions":{"0de21795126e58a958a8c14bc63031e7888b2136":{"kind":"REWORK","_number":1,"created":"2023-05-16 16:00:40.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/82/883282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/883282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/883282/1"}}},"commit":{"parents":[{"commit":"ae4084c1735f63de7fc685ee51502f5d0d9db67c","subject":"Merge \"[S-RBAC] Get QoS rule types API available for READER role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae4084c1735f63de7fc685ee51502f5d0d9db67c"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-05-16 16:00:25.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-05-16 16:00:25.000000000","tz":330},"subject":"[DNM] check grenade","message":"[DNM] check grenade\n\nChange-Id: Id72ab1e521d7d4b6186577ca1c5a6bd1ad4f4648\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0de21795126e58a958a8c14bc63031e7888b2136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0de21795126e58a958a8c14bc63031e7888b2136"}]},"branch":"refs/heads/master"},"254e6aa25e345f5f5a322ee3b52c9b4bfb492a38":{"kind":"REWORK","_number":2,"created":"2023-05-17 07:44:01.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/82/883282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/883282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/883282/2"}}},"commit":{"parents":[{"commit":"ae4084c1735f63de7fc685ee51502f5d0d9db67c","subject":"Merge \"[S-RBAC] Get QoS rule types API available for READER role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae4084c1735f63de7fc685ee51502f5d0d9db67c"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-05-16 16:00:25.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-05-17 07:43:47.000000000","tz":330},"subject":"[DNM] check grenade","message":"[DNM] check grenade\n\nChange-Id: Id72ab1e521d7d4b6186577ca1c5a6bd1ad4f4648\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/254e6aa25e345f5f5a322ee3b52c9b4bfb492a38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/254e6aa25e345f5f5a322ee3b52c9b4bfb492a38"}]},"branch":"refs/heads/master"},"20e386bf67db5fe17295181173494e7549670c7b":{"kind":"REWORK","_number":3,"created":"2023-05-17 15:26:32.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/82/883282/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/883282/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/883282/3"}}},"commit":{"parents":[{"commit":"ae4084c1735f63de7fc685ee51502f5d0d9db67c","subject":"Merge \"[S-RBAC] Get QoS rule types API available for READER role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae4084c1735f63de7fc685ee51502f5d0d9db67c"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-05-16 16:00:25.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-05-17 15:25:59.000000000","tz":330},"subject":"[DNM] check grenade","message":"[DNM] check grenade\n\nChange-Id: Id72ab1e521d7d4b6186577ca1c5a6bd1ad4f4648\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20e386bf67db5fe17295181173494e7549670c7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20e386bf67db5fe17295181173494e7549670c7b"}]},"branch":"refs/heads/master"},"4ad9fa4bc496d3905d905beb9c73b218cf3ea499":{"kind":"REWORK","_number":4,"created":"2023-05-17 17:34:26.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/82/883282/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/883282/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/883282/4"}}},"commit":{"parents":[{"commit":"ae4084c1735f63de7fc685ee51502f5d0d9db67c","subject":"Merge \"[S-RBAC] Get QoS rule types API available for READER role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae4084c1735f63de7fc685ee51502f5d0d9db67c"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-05-16 16:00:25.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-05-17 17:34:03.000000000","tz":330},"subject":"[DNM] check grenade","message":"[DNM] check grenade\n\nChange-Id: Id72ab1e521d7d4b6186577ca1c5a6bd1ad4f4648\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ad9fa4bc496d3905d905beb9c73b218cf3ea499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ad9fa4bc496d3905d905beb9c73b218cf3ea499"}]},"branch":"refs/heads/master"},"628d47bc1edda51aba030f4227e01b3e9eccd9b5":{"kind":"REWORK","_number":5,"created":"2023-05-18 04:16:14.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/82/883282/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/883282/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/883282/5"}}},"commit":{"parents":[{"commit":"ae4084c1735f63de7fc685ee51502f5d0d9db67c","subject":"Merge \"[S-RBAC] Get QoS rule types API available for READER role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae4084c1735f63de7fc685ee51502f5d0d9db67c"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-05-16 16:00:25.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-05-18 04:15:58.000000000","tz":330},"subject":"[DNM] check grenade","message":"[DNM] check grenade\n\nChange-Id: Id72ab1e521d7d4b6186577ca1c5a6bd1ad4f4648\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/628d47bc1edda51aba030f4227e01b3e9eccd9b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/628d47bc1edda51aba030f4227e01b3e9eccd9b5"}]},"branch":"refs/heads/master"},"1d488f092b3ca29985ab7ded8ccc08338c50e75f":{"kind":"REWORK","_number":6,"created":"2023-05-18 06:01:44.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/82/883282/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/883282/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/883282/6"}}},"commit":{"parents":[{"commit":"ae4084c1735f63de7fc685ee51502f5d0d9db67c","subject":"Merge \"[S-RBAC] Get QoS rule types API available for READER role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae4084c1735f63de7fc685ee51502f5d0d9db67c"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-05-16 16:00:25.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-05-18 06:01:15.000000000","tz":330},"subject":"[DNM] check grenade","message":"[DNM] check grenade\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/883472\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/883473\nChange-Id: Id72ab1e521d7d4b6186577ca1c5a6bd1ad4f4648\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d488f092b3ca29985ab7ded8ccc08338c50e75f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d488f092b3ca29985ab7ded8ccc08338c50e75f"}]},"branch":"refs/heads/master"},"2cea94de8b45cb214644a74db65ff90e248de12e":{"kind":"REWORK","_number":7,"created":"2023-05-24 06:00:59.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/82/883282/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/883282/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/883282/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/883282/7"}}},"commit":{"parents":[{"commit":"0360eb8e127a9f162a987fe2fc8cfefdafdb9c32","subject":"Merge \"Revert \"Delete sg rule which remote is the deleted sg\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0360eb8e127a9f162a987fe2fc8cfefdafdb9c32"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-05-16 16:00:25.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-05-24 06:00:25.000000000","tz":330},"subject":"[DNM] check grenade","message":"[DNM] check grenade\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/883472\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/883473\nChange-Id: Id72ab1e521d7d4b6186577ca1c5a6bd1ad4f4648\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2cea94de8b45cb214644a74db65ff90e248de12e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2cea94de8b45cb214644a74db65ff90e248de12e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
