)]}'
{"id":"openstack%2Fneutron~883288","triplet_id":"openstack%2Fneutron~stable%2Fxena~I9e1e43975fb616b72f83489f34c573674a03a698","project":"openstack/neutron","branch":"stable/xena","topic":"bug/2019859","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-05-22 12:51:45.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-05-17 12:45:22.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-05-22 12:51:45.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-05-22 12:51:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9e1e43975fb616b72f83489f34c573674a03a698","subject":"[stable-only] Don\u0027t use \"sqlalchemy.select\" with SQLAlchemy\u003c1.4","status":"MERGED","created":"2023-05-16 16:52:09.000000000","updated":"2023-05-22 12:52:59.000000000","submitted":"2023-05-22 12:51:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883288-bug/2019859","meta_rev_id":"55ad3b94486b6252e8e516215b8ca7546e47d593","_number":883288,"virtual_id_number":883288,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-22 12:51:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-05-22 10:19:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-05-17 12:28:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":2,"date":"2023-05-16 19:35:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-05-22 10:19:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-16 16:52:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-05-16 16:52:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-05-16 16:52:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-05-16 17:15:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-16 19:35:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-05-16 19:47:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-17 12:28:10.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"}],"messages":[{"id":"fd19f4fcd8a039dde138d83483792357cec76e92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-16 16:52:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a55b1a7ef9b6900a08c0e4634bdb5d177357f32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 17:15:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea47124e2110439ea3bfccb206253acc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/771758fd732044e58a1040f119d74c47 : SUCCESS in 22m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c5ca0ba1e9a468ba9e6341ef4f6539c : SUCCESS in 18m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"744599bd2479862b0d894c1d5133cdd749339c6a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-05-16 19:35:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a21d783d74b989a89145c201d637408c2965fb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 19:47:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb7581960c2d4b24b9e8662f1edd1a59\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/35af84013a0e466bbf0186e969536286 : SUCCESS in 1h 05m 06s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/122ed1c68cab405883ccc263de655863 : SUCCESS in 1h 52m 46s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/8dac4105bcbd442187b10b4515ee011b : TIMED_OUT in 2h 49m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/3dd26388414046a59e79bb095ec29ea0 : SUCCESS in 1h 26m 21s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/0ef293ed244a49038f761183dcb63d40 : SUCCESS in 2h 36m 11s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/3355a8bd13394c1aa8ab2084cd111919 : SUCCESS in 41m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a48f8a3f8ef94fbbbd506bb93ed73ceb : SUCCESS in 38m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09790e6e05824130a2cbe7d9574dc450 : SUCCESS in 14m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06230e0894db461081e96a6942dffcf9 : SUCCESS in 26m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cab83d2dfe004d55a3e957d42024231a : SUCCESS in 23m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/904df8b4936d48b489d1d587fa7a7440 : SUCCESS in 24m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eb403a3d8be45d498df8201aab6f4ba : SUCCESS in 9m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6a46264166e34435bcc6e15564f8fae9 : SUCCESS in 50m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d59416098de43fcac630b9df43f82bf : SUCCESS in 1h 23m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1380dcb1f0b8406c8e563bb66aa41f88 : SUCCESS in 57m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e2840b5a5844e1092c5c689f2a91b4c : SUCCESS in 52m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a6f3b13a2f944d309d726e4d48e9ceee : SUCCESS in 1h 07m 26s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d1bac2e22ea74932a1851e8f1bd2cf8e : SUCCESS in 1h 01m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dc195e116bdf4c528ce80f53eba1bcfc : SUCCESS in 1h 11m 03s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87be7f05e0304ff88142e97ea84aa85b : SUCCESS in 48m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/54a0527df6ec403596ee702a75b0b46a : SUCCESS in 1h 03m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec13f9ab778b6c2f1fba666c1cb04b19b2f5f2a7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-17 04:53:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bfc4009b8167fd05ccfd7e2c002bf7670e3e8272","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 05:50:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd4ab789df9b4633974c9efb335b859e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03bac359dd6d4cda860b947053209fb1 : SUCCESS in 54m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d151321614fb4ecfadee783520e680c9 : SUCCESS in 55m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6ddc9d1a54ee1bf43ad5457bb56fd68d902f4d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 07:32:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/25f1906439f9460f952e9e492347808c\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/b57ea188fc984aec9a85be920bebba09 : SUCCESS in 46m 40s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/4734438bff8a46cba4de3c525947c3cd : FAILURE in 1h 54m 23s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/21caa00b81b9485a86d9abac46edd011 : SUCCESS in 1h 48m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/90fbc44d69ef427c9efcaf490aa20c08 : SUCCESS in 2h 01m 53s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/9f496621923148899444acb9c3bbb709 : SUCCESS in 2h 31m 58s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/4efbe9247f574137bd7ade4849d0c97c : TIMED_OUT in 1h 02m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2ac76729c904f29bba72257f07693a7 : SUCCESS in 35m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab3adeed732406495a8828b3a276dd9 : SUCCESS in 13m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e88f30596374352b19400501ac4104c : SUCCESS in 26m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8458789154ff481086b49cb1d7b570a5 : SUCCESS in 22m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/575476a4868b4a5cb3db7bb63d90cc34 : SUCCESS in 27m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fca01155824b464a9886ffb8d46746cd : SUCCESS in 8m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cfd37ed053794b3999cc5140379f7b1a : SUCCESS in 47m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/55da9d6b72484862bc7cb895b2a01db8 : SUCCESS in 1h 26m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9713ccfd9a104f1baeaa011c595b3490 : SUCCESS in 57m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e88cedb20a5740768835e95bf1a4a754 : SUCCESS in 1h 03m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/25290940c6eb47248e64405bd4b0c074 : SUCCESS in 52m 54s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ce061144b6084306aee35a4bc4ede26d : SUCCESS in 1h 07m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c5c342bab0f24b47ac9827f5a354c387 : SUCCESS in 1h 08m 47s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2eeb11df1e324bf7aef65c3f94d210a9 : SUCCESS in 44m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e2010f6b248b44f99a7c1af0a00c2a5e : SUCCESS in 59m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f9b6132aeb10cc4c1bffc58e2839ac776a4feb1c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-17 10:08:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38766a5a2f7dfb216ff00c99c400e47567aebaa4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 11:17:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b8830ee13db4b60a4e7ec9b05783e60\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c85e8e9a33c64868bfe92aa6ac7a4af1 : SUCCESS in 1h 06m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c1c644f7fb640f69fa7a60a4fcede8f : SUCCESS in 1h 07m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"038a58b73d69c8cf806bb6647d8c55c1858759fd","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2023-05-17 12:28:10.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"92a3d054561c5d070e82add99981061106ba6331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 12:34:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/751e26eca0d64d38bbce23f93a299ada\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/0487563b2b7d484c8fa56153e9ebccce : SUCCESS in 1h 01m 45s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/09708b7e7a064c79b081d79c8db437d5 : SUCCESS in 2h 24m 50s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/90decf4795b04629ba5c4d757c180d2b : SUCCESS in 2h 00m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/08e09e13b445484c9c7a57c37b1896f4 : SUCCESS in 1h 57m 42s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/fdecb2686eca4965a1eab21bb14f4d85 : SUCCESS in 2h 02m 02s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/fc8e2ee09a2a4350987732394767acde : FAILURE in 24m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2ac32f2ade54eda8570de1fab3b0295 : SUCCESS in 38m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a47d6f9d1da8490cb95f3a0ea918ff91 : SUCCESS in 12m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7176383bcb114be3bd32d8ccdf2fa57d : FAILURE in 23m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4b92aee85c74c69b6b04df7710aac34 : SUCCESS in 21m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/794bffec63594777b0fc32756e62e3d8 : SUCCESS in 26m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5699080e67774140824dc514a197809a : SUCCESS in 9m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/889cef9db8ea4048b2e647f8c5c42446 : SUCCESS in 58m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9ac52dc9ddeb4a9e98eb525ed55919af : SUCCESS in 1h 32m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fc9451bd3c524676a0031d70831fb601 : SUCCESS in 1h 00m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce48f27f5510456e979ef9a34a7fc73b : SUCCESS in 51m 10s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8954c899f3d840ef921ca455cd538d5e : SUCCESS in 50m 56s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/42899df1009b4a628d70d815a3f9e859 : SUCCESS in 58m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4c5b57addfda4c6eaccd0e7c39936cd5 : SUCCESS in 1h 01m 45s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edef823cfabf4075982480ffa5c39be0 : SUCCESS in 47m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dbf8fb8c28d947a2b92bd6b7b7925fe5 : SUCCESS in 1h 05m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"353e452b0a1eff020907865c6889eea69c3950f9","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-05-17 12:45:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a193aebca4af520a8355db9d78cf62c55260754","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 12:49:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8001cbd7fdb41f48ea778adef95cfb0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a421fa794ae8410289483d4602521531 : NODE_FAILURE Node request 300-0021232902 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80c0c0602fe34457969b6e92c509efea : NODE_FAILURE Node request 300-0021232903 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40b792d79f2b2b4466f873883039c594f9e898fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 15:25:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b7086228a894a17a71c71099e4592c1\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/00f343f8b90c4f6985b547a6001c64f2 : SUCCESS in 55m 23s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/67acd6075f1243929bbe1e29d011c0d5 : SUCCESS in 2h 24m 54s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/8a0f2b63a21a494195b6e76833e273ef : SUCCESS in 2h 35m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/44c27335c6734a008b80242847ef72a3 : SUCCESS in 1h 53m 25s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/1a26e29cb4f940828245b016490d85f3 : SUCCESS in 2h 17m 14s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/d451466fc5554cb3a0d2630d052aca20 : SUCCESS in 47m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04907216eb02448b8018d5dd787b4a7f : SUCCESS in 28m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89dff6e839094c3ab2a5af85f7eee1ec : SUCCESS in 13m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2acad0b2364442c8b2cab1825e9b6bb : SUCCESS in 21m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5919016bf5dd4a36973a39c19a64cceb : SUCCESS in 25m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13d3a1a2f39b4f2f89621dffc8452b2d : SUCCESS in 26m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea6149ef42b94ff2ba2a14cb2c2c7657 : SUCCESS in 8m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/02aaf753d087477d92a17c7582383cda : FAILURE in 42m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/61be75324a884cd3b79f825b05a5addd : SUCCESS in 1h 22m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cd2964d445864289a3a7319a63ce1367 : SUCCESS in 1h 11m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e6cf60e93db4f779d60416348c9624d : SUCCESS in 58m 33s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7c3453a402b041739b8812f6b8c143fe : SUCCESS in 1h 07m 21s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b183e6b941a34ec183bd2762eeb3b40a : SUCCESS in 1h 10m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/931001f609084cc692d0cb450ed8cfbe : FAILURE in 1h 12m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f781df94b5014ac9a0b1d9b490d664cc : SUCCESS in 58m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7aab57d4e18941f79ced66ec0704f017 : SUCCESS in 1h 13m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c1abf147ce4813e3071570f9923bbdf037c629f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-18 15:26:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f214081cecdb7b970013cccf0341cc3612723b81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 15:53:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/097a947041794b628ba3dead32901e05\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/156a48e27e8547579b9d7d3583063d1f : SUCCESS in 20m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b423e4e66a1456a990d2666102ae762 : SUCCESS in 22m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e7f81a774da141c40c4d009af1bf973c73a9508","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 18:22:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22db72a9d4c942539d3ff442d0eb6eea\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/ffd014a3bb2a4ce096faec6432f805f6 : SUCCESS in 1h 03m 06s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/efde7056f3504b01991a91d0c1f4af8b : SUCCESS in 2h 14m 03s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/5dd787b87eb94b93b790d014dd332db5 : SUCCESS in 2h 42m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/6df2d9b1848a48a9a1d8ce16b838f490 : SUCCESS in 1h 57m 52s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/d748fd88319946ad920ae6eb51fe76c9 : SUCCESS in 1h 36m 32s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/0c053fbbd27641678c32f2b0d46c525a : SUCCESS in 55m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1ebef485c8242f39f690b3ed9fb72d6 : SUCCESS in 33m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ac817fe0d4641ca91c7eafbaa7de32d : SUCCESS in 16m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c49d47b59cb74c12bf362bc4e33f27f4 : SUCCESS in 25m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a0a38f1aede4d6aaddef8f893720435 : SUCCESS in 29m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e6cc4ddf4974811bedad39fe2c3495f : SUCCESS in 26m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e30a6c3ad2a247eba3af80be65503b14 : SUCCESS in 8m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d61693cef2c4259a0dbbca67ff4b9d6 : SUCCESS in 50m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/91b96e20201e42bbbc908e2103746c4f : SUCCESS in 1h 15m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e2f7848163fb4fd2a6807225f1394778 : SUCCESS in 1h 14m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4ac87b11763e441cae09d9c0855e401f : SUCCESS in 51m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f41fd840bc4840ea88a6dacdef513efc : SUCCESS in 1h 03m 45s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0bc7d6659ae549b1b24f15f638166a7c : SUCCESS in 1h 03m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d961c37cb529433887582482bdf79e3e : SUCCESS in 1h 07m 12s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08b64892444a4beeb065422c4a504c11 : SUCCESS in 1h 09m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a055be5f44f1474da2bd5d65ad393e90 : SUCCESS in 1h 22m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8044ed0b9f999c7e61f9122fef3d3b4fb776f791","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-05-22 10:19:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"778e059e11e3b50077f86da249a1eafa6fad1a4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 10:20:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a34e6dfaee63d33f4d7576c3d6523a02a32c48bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 12:51:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6908005f371a404eb4b78bc81afb2704\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/3af3ed79f0514138953403c0fa0cc3bb : SUCCESS in 47m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34adedb061c8415d81d8c7aec6ad14bd : SUCCESS in 15m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e2f4339113742f6baca5ce23eab690d : SUCCESS in 28m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba89ffc0c83e487bbeeaa8a06acf8412 : SUCCESS in 23m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93d07d01993a4070a351faa0689cd080 : SUCCESS in 9m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b76b7de166944ca591427c97f4d5b76c : SUCCESS in 44m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/93c3d3abaeaf424ca4a36fd3d939e766 : SUCCESS in 1h 26m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9c5f6260350f4869a4a1fb31cb05f501 : SUCCESS in 1h 05m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/adc7184a3e364b22a1480f4bd845292e : SUCCESS in 1h 08m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5c42ed43ecc4226bcde3d1645d1dd49 : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c1627c7670bcf625304ed1c846018b4b61a1c629","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 12:51:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"55ad3b94486b6252e8e516215b8ca7546e47d593","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 12:52:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb0fed59633e49d4a79ce9c353908a36\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a4bd4ef2fe94d75a9ceb3bee745feb7 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4372dd097a23ff7f042b3dfc3bf14880861b6a8f","revisions":{"4372dd097a23ff7f042b3dfc3bf14880861b6a8f":{"kind":"REWORK","_number":1,"created":"2023-05-16 16:52:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/88/883288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/883288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/883288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/883288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/883288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/883288/1"}}},"commit":{"parents":[{"commit":"43ee9f3a9474f382fba9db063722d1c0dfee9b48","subject":"[OVN] Update ovn meter when neutron server reloads","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/43ee9f3a9474f382fba9db063722d1c0dfee9b48"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-16 16:48:25.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-16 16:48:25.000000000","tz":120},"subject":"[stable-only] Don\u0027t use \"sqlalchemy.select\" with SQLAlchemy\u003c1.4","message":"[stable-only] Don\u0027t use \"sqlalchemy.select\" with SQLAlchemy\u003c1.4\n\nThe \"sqlalchemy.select\" function accepts column arguments positionally\nsince 1.4.x [3]. However, the minimum version accepted in Xena branch\nis 1.3.23 [2]. This patch uses the implementation merged in Wallaby\nbranch [3] for the method \"get_duplicated_port_bindings\", that is\ncompatible with SQLAlchemy 1.3 and 1.4.\n\n[1]https://docs.sqlalchemy.org/en/20/core/selectable.html#sqlalchemy.sql.expression.select\n[2]https://github.com/openstack/requirements/blob/stable/xena/upper-constraints.txt#L157\n[3]https://review.opendev.org/c/openstack/neutron/+/877493\n\nCloses-Bug: #2019859\nChange-Id: I9e1e43975fb616b72f83489f34c573674a03a698\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4372dd097a23ff7f042b3dfc3bf14880861b6a8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4372dd097a23ff7f042b3dfc3bf14880861b6a8f"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
