)]}'
{"id":"openstack%2Fneutron~883713","triplet_id":"openstack%2Fneutron~stable%2Fzed~I9c49a307956ecfbf8bd2e866cefb21a212c38bd6","project":"openstack/neutron","branch":"stable/zed","topic":"bug/2016704-stable/zed","attention_set":{"30772":{"account":{"_account_id":30772,"name":"Ihtisham ul Haq","display_name":"Ihtisham ul Haq","email":"ihtisham.ul_haq@mail.schwarz","username":"ihtisham","status":"STACKIT (Schwarz IT)"},"last_update":"2023-05-30 14:33:08.000000000","reason":"\u003cGERRIT_ACCOUNT_7730\u003e replied on the change","reason_account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"}}},"removed_from_attention_set":{"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2023-05-30 14:30:29.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-05-30 12:48:36.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}}},"hashtags":[],"change_id":"I9c49a307956ecfbf8bd2e866cefb21a212c38bd6","subject":"Use explicit inner join for networks in port query","status":"MERGED","created":"2023-05-22 06:37:57.000000000","updated":"2023-05-30 14:33:08.000000000","submitted":"2023-05-30 14:30:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"883713-bug/2016704-stable/zed","cherry_pick_of_change":880621,"cherry_pick_of_patch_set":9,"meta_rev_id":"4ba44c7e0a33b256b4761b2cec7544890692f71c","_number":883713,"virtual_id_number":883713,"owner":{"_account_id":30772,"name":"Ihtisham ul Haq","display_name":"Ihtisham ul Haq","email":"ihtisham.ul_haq@mail.schwarz","username":"ihtisham","status":"STACKIT (Schwarz IT)"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-30 14:30:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-05-25 15:44:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2023-05-23 18:25:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":2,"date":"2023-05-30 12:48:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":1,"date":"2023-05-30 12:48:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-22 06:59:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-22 08:32:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-23 18:25:43.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2023-05-25 15:44:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-05-25 15:45:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"a936aaed0ccce26f2023758a62f5ceb01f181c3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30772,"name":"Ihtisham ul Haq","display_name":"Ihtisham ul Haq","email":"ihtisham.ul_haq@mail.schwarz","username":"ihtisham","status":"STACKIT (Schwarz IT)"},"date":"2023-05-22 06:37:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ba1c0e9f5142348aef557f005c91f3e3bb6029d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 06:59:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9e25a6237c348eeac67d7107b16aebf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7847d05fb8243489795d6846e5e4567 : SUCCESS in 18m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/86962f7143e44bc4891eac8d78c8445a : SUCCESS in 20m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c33f2348fdf4e82c3daf2f36d8fbdd1ddb45fa74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 08:32:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df25d699b62c4756b0424585547f8f3a\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/e6cc8803ccca492abc8645b7ce2a1c50 : SUCCESS in 1h 51m 46s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/53834d78463f4f28a5b7c228d7ddd3c8 : SUCCESS in 1h 13m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/7077424a1e0a4d57a0777a1a838a99d1 : SUCCESS in 1h 52m 51s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/5c9cc32bd4194123be53ff764db304d0 : SUCCESS in 1h 33m 55s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/231824ad93a6409495ecfbf90cfb9835 : SUCCESS in 37m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/288ddb3ee6b944bf87a0b98640ff22b6 : SUCCESS in 45m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9422a34851e84f0198246fca26ed49ea : SUCCESS in 10m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0532013be2d4f96832eb8c47fa34713 : SUCCESS in 25m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f28bd71ea2d4621986466956f31208d : SUCCESS in 25m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23eb6e2ff66b450d8ba50917ab3ae355 : SUCCESS in 29m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dae18d1034f4b5cb931530d608ce34c : SUCCESS in 9m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bcd41863ca2544ceae7e26c260ce6c2f : SUCCESS in 57m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/25ca6f850248480fb8183493fcddade2 : SUCCESS in 1h 38m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/308118b782fe4bb0958cbf03297b0aaf : SUCCESS in 1h 09m 25s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb40d45125b04aa89bb5930479168041 : SUCCESS in 1h 06m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eea72a9828714d8dbe65100e20aec537 : SUCCESS in 59m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7bf3633e56b341d484f395a2400e7fa3 : SUCCESS in 1h 26m 15s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/812908af3fff45398c1c0ecd3970f70a : SUCCESS in 1h 18m 01s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f96ff09ee4e54aa79f5b206d2b69523c : SUCCESS in 1h 10m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/384783dacbaa437daeb86d569358292f : SUCCESS in 1h 24m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"757e55081b3c9aae8b736c7a1bba21d6eb613b50","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-05-23 18:25:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ddefbedacea50ce9861b46bb10bd5ad52efd20e3","author":{"_account_id":30772,"name":"Ihtisham ul Haq","display_name":"Ihtisham ul Haq","email":"ihtisham.ul_haq@mail.schwarz","username":"ihtisham","status":"STACKIT (Schwarz IT)"},"date":"2023-05-24 07:17:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9d8a1d72b47f831e1cb0e2e39119d16fe3b19c9","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-05-24 07:26:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0e9e02adec0f6ab64c91917217236470dac6081","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-25 15:44:27.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"64add005c7163bcafebefabd8d8d53ba8f7583f2","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-05-30 12:48:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8de5a0428dfd5bed42f8988d1360d89c9ddef2c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 12:49:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ec6341ad37a241c77e5860f0aa46d6d554400a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 14:30:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7be83d94ede440a1a557a17da469cbc8\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/65ff7902f34944dda4b5a74a0d26a407 : SUCCESS in 1h 31m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97181ec6eca641e68cf0f65f3a7fb62b : SUCCESS in 13m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29f3e452c9b94b31b3028629f03a6d38 : SUCCESS in 26m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/426f2c79856f4fb3b5636494205c1cba : SUCCESS in 24m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9ceddc1477b43b48738f8bb6cd4a947 : SUCCESS in 9m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/829cb8a2c5a94337923a284445e00e53 : SUCCESS in 1h 00m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ecac307abce4e9a87d853ca746e6685 : SUCCESS in 1h 34m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bc99808ce0b0425aaf7829278693c239 : SUCCESS in 1h 34m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/39985ac15fc141d6b1cf3378948bff50 : SUCCESS in 1h 00m 16s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/40bfd1603e1a4ba5955fbeff3aff7afd : SUCCESS in 1h 00m 56s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a6a4dd75b8d447fa262381b107d6456 : SUCCESS in 1h 08m 06s","accounts_in_message":[],"_revision_number":1},{"id":"562119fe375c7865f0266474af60b97835dc3814","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 14:30:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d816f625afdf98546e217f2492655dd6968f9555","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 14:31:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a32f0d4ce734ca2a46b197de5c46640\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d7f8cbfc364ddfbb721119cdec1931 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1},{"id":"4ba44c7e0a33b256b4761b2cec7544890692f71c","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-05-30 14:33:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"004ed33ec1ab549444e518ed8681b51737b81b9c","revisions":{"004ed33ec1ab549444e518ed8681b51737b81b9c":{"kind":"REWORK","_number":1,"created":"2023-05-22 06:37:57.000000000","uploader":{"_account_id":30772,"name":"Ihtisham ul Haq","display_name":"Ihtisham ul Haq","email":"ihtisham.ul_haq@mail.schwarz","username":"ihtisham","status":"STACKIT (Schwarz IT)"},"ref":"refs/changes/13/883713/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/883713/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/883713/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/883713/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/883713/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/883713/1"}}},"commit":{"parents":[{"commit":"1529c0a29728955064a53d616a118bc8a7ca1b4c","subject":"[stable-only][ovn] Fix ovsdbapp db_set command for stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1529c0a29728955064a53d616a118bc8a7ca1b4c"}]}],"author":{"name":"Ihtisham ul Haq","email":"ihtisham.ul_haq@mail.schwarz","date":"2023-04-17 10:23:30.000000000","tz":120},"committer":{"name":"Ihtisham ul Haq","email":"ihtisham.ul_haq@mail.schwarz","date":"2023-05-22 06:37:57.000000000","tz":0},"subject":"Use explicit inner join for networks in port query","message":"Use explicit inner join for networks in port query\n\nThis improves the performance of the database when fetching a list of ports\nfor a project user. This change creates an inner join with the networks\nbelonging to the ports.\n\nPrevious SQL query:\nSELECT ports ...\nFROM network, ports ...\nWHERE ports.project_id \u003d \u003cproject\u003e\nOR ports.network_id \u003d networks.id\nAND networks.project_id \u003d \u003cproject\u003e\n\nCurrent SQL query:\nSELECT ports ...\nFROM ports\nINNER JOIN networks ON networks.id \u003d ports.network_id\nWHERE ports.project_id \u003d \u003cproject\u003e\nOR networks.project_id \u003d \u003cproject\u003e\n\nCloses-Bug: #2016704\nChange-Id: I9c49a307956ecfbf8bd2e866cefb21a212c38bd6\n(cherry picked from commit f23d7af8d7e36b9d52a8c8ccf64355e22f3adb9c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/004ed33ec1ab549444e518ed8681b51737b81b9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/004ed33ec1ab549444e518ed8681b51737b81b9c"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
