)]}'
{"id":"openstack%2Fneutron~884205","triplet_id":"openstack%2Fneutron~stable%2Fzed~I077fe87435f61bd29d5c1efc979c2adebca26181","project":"openstack/neutron","branch":"stable/zed","topic":"bug/2019449","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-05-25 09:15:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I077fe87435f61bd29d5c1efc979c2adebca26181","subject":"Revert \"Delete sg rule which remote is the deleted sg\"","status":"MERGED","created":"2023-05-24 14:13:21.000000000","updated":"2023-05-25 09:17:16.000000000","submitted":"2023-05-25 09:15:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884205-bug/2019449","meta_rev_id":"6f0a11f3225118a4ad405227d64ba8d754f43e2f","_number":884205,"virtual_id_number":884205,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-25 09:15:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2023-05-25 07:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-05-24 16:03:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2023-05-25 07:31:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-24 14:13:21.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-05-24 15:07:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-24 16:03:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-05-24 16:12:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-25 07:31:43.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"64edc3be35fb8d585a94158b3846d6bd2ef3cabe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-05-24 14:13:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"388fa9e54fe62c42536994b52c8db94c5c35b228","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 15:07:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2122027292b4113af81573a43d6d111\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5554312a880744a28e7265eee3322619 : SUCCESS in 52m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b12e2847b094e4fa2ce5b5d78ba5f80 : SUCCESS in 22m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7fd7ebc3176238e7901fa4c53fe5ee6892b0c3b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-05-24 16:03:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3aede1a5445196bd2dcf89baa4d472496d760d2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 16:12:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e66c394cf5e4a10ae531e4724a97572\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/1b460d52932c41ebad7c6c9037038305 : SUCCESS in 1h 55m 26s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/b36271f802f04a1597cf4875e27eea42 : SUCCESS in 1h 20m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/0b2a45f995a94395b688970a8eb23176 : SUCCESS in 1h 49m 47s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/ba56fde3fdf946d2b9f296aff15f88ce : SUCCESS in 1h 26m 35s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/4588c796940f4047bd1e57cf82e47126 : SUCCESS in 32m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19d8dc4d9114412f9b1a0d6a03b9b9d5 : SUCCESS in 33m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3db7e827423c4c3981e0c5314f2269d9 : SUCCESS in 12m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc82ee9ab1fc4c0d936694c7295ccfcd : SUCCESS in 23m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32ec9ae5f2e2493891be07fb59e5f758 : SUCCESS in 27m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4274bfe9b5ff47d19123623d78bb2602 : SUCCESS in 26m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fe3d09ac27f426dae557dce023b92a2 : SUCCESS in 10m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4fb53fca025a4074be75e89f739e70b1 : SUCCESS in 1h 07m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9371f241ce1b47a58ca5b0808b9accff : SUCCESS in 1h 52m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bf783540796e4f0887b44136461da80e : SUCCESS in 1h 05m 11s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6fa02ad916604007a0a0963bdb471843 : SUCCESS in 53m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f348db93602f408391f1b527943fa8a5 : SUCCESS in 1h 06m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/95941d93dbbb45b09578727e90ccab27 : SUCCESS in 1h 18m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1d077810a2044614a94d786775dcee22 : SUCCESS in 1h 29m 10s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3950f2305cbc4d8c9b8b34ea6cd223d8 : SUCCESS in 1h 00m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3f1be2204f384118af83731fe5838ba7 : SUCCESS in 1h 08m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4227c31d7bf2062dd73b90194c265a8d141d5622","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-05-25 07:31:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4287635c631797873c98a8239480ec666401d85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 07:32:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2aaed5436b5031ca127f61eb29f4d2c8509d529a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 09:15:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddfdd8ee7bd9430cb28f8d2575f472f1\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/5d65d5c5f8f14c4fb43e4d82e96fa97a : SUCCESS in 1h 38m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38854922512c4af1b522e8ec9d064749 : SUCCESS in 13m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/125bf30d756a4f9b848fe984fda4b3a1 : SUCCESS in 26m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/795d8db55bdc49f984c0816b672b72e1 : SUCCESS in 24m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7630a0e28d454a5ca0cc53a1e344eaff : SUCCESS in 9m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2cd5c66c197d4249b2e454704d5192e8 : SUCCESS in 54m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2fd2017601714e579868273350dd635b : SUCCESS in 1h 36m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/765ab7e294f24413982be1e875fe63e9 : SUCCESS in 1h 27m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c2efa75e447e4085a5bb8115478e92ef : SUCCESS in 50m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/acfb3186fab64425aa7a4b137513077c : SUCCESS in 1h 00m 35s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6f896b62365428b8a834a96ed7464d8 : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":1},{"id":"408da696ca40d0a1a68453fb1f2f7d2d6009b56c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 09:15:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6f0a11f3225118a4ad405227d64ba8d754f43e2f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 09:17:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2de31e0511e4e36ad0af7b1b3eed1d9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c2dd730ac3403db060c73bfb40fa81 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8bf6f7f03c04d90d6953d9cd1a286d82a01fc040","revisions":{"8bf6f7f03c04d90d6953d9cd1a286d82a01fc040":{"kind":"REWORK","_number":1,"created":"2023-05-24 14:13:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/05/884205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/05/884205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/884205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/884205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/884205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/05/884205/1"}}},"commit":{"parents":[{"commit":"8def3b694dd800da09e44f332c0ea11e802dbde4","subject":"[ovn] Avoid unwanted ACL_NOT_FOUND error when deleting log objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8def3b694dd800da09e44f332c0ea11e802dbde4"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-05-19 14:45:53.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-05-24 00:49:59.000000000","tz":120},"subject":"Revert \"Delete sg rule which remote is the deleted sg\"","message":"Revert \"Delete sg rule which remote is the deleted sg\"\n\nThis reverts commit 63584957203ec9f5ba165177978213c3909f81f0.\n\nReason for revert: This is generating a lot of\n\"SecurityGroupNotFound\" errors in neutron-server.log in\nthe tempest-integrated-networking job.\n\nCloses-Bug: #2019449\nRelated-Bug: #2008712\nChange-Id: I077fe87435f61bd29d5c1efc979c2adebca26181\n(cherry picked from commit ebc0658d5566ce527cb1104968d247db10edf3db)\n(cherry picked from commit d3fb57617d56acb312cef18d32881dd608967313)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bf6f7f03c04d90d6953d9cd1a286d82a01fc040"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bf6f7f03c04d90d6953d9cd1a286d82a01fc040"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
