)]}'
{"id":"openstack%2Fneutron~886005","triplet_id":"openstack%2Fneutron~master~Ic90432b5ddea8cf176de159ec7eaafd5fd7bdd6e","project":"openstack/neutron","branch":"master","topic":"bug/2023670-improve-hash-ring-logging","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-06-15 12:29:01.000000000","reason":"\u003cGERRIT_ACCOUNT_6773\u003e replied on the change","reason_account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2023-06-14 13:01:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-06-14 19:54:51.000000000","reason":"\u003cGERRIT_ACCOUNT_8655\u003e replied on the change","reason_account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}},"hashtags":[],"change_id":"Ic90432b5ddea8cf176de159ec7eaafd5fd7bdd6e","subject":"[OVN] Improve Hash Ring logs","status":"MERGED","created":"2023-06-13 15:04:06.000000000","updated":"2023-06-16 13:03:17.000000000","submitted":"2023-06-16 13:02:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"886005-bug/2023670-improve-hash-ring-logging","meta_rev_id":"54ef7fa5158ea25fa9330cd4e4a6630a99563cbb","_number":886005,"virtual_id_number":886005,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-06-14 09:46:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-16 13:02:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33983,"name":"Luigi Dino Tamagnone","display_name":"luigi","email":"ltamagno@redhat.com","username":"ltamagno"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"recommended":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2023-06-14 13:01:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2023-06-15 07:00:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-14 15:15:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33983,"name":"Luigi Dino Tamagnone","display_name":"luigi","email":"ltamagno@redhat.com","username":"ltamagno"},{"value":2,"date":"2023-06-14 19:54:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-16 10:02:14.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33983,"name":"Luigi Dino Tamagnone","display_name":"luigi","email":"ltamagno@redhat.com","username":"ltamagno"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33983,"name":"Luigi Dino Tamagnone","display_name":"luigi","email":"ltamagno@redhat.com","username":"ltamagno"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33983,"name":"Luigi Dino Tamagnone","display_name":"luigi","email":"ltamagno@redhat.com","username":"ltamagno"}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-13 15:52:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-13 16:15:46.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2023-06-13 16:37:10.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-06-13 17:14:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-13 17:16:22.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-13 21:49:38.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2023-06-14 15:15:35.000000000","updated_by":{"_account_id":33983,"name":"Luigi Dino Tamagnone","display_name":"luigi","email":"ltamagno@redhat.com","username":"ltamagno"},"reviewer":{"_account_id":33983,"name":"Luigi Dino Tamagnone","display_name":"luigi","email":"ltamagno@redhat.com","username":"ltamagno"},"state":"REVIEWER"},{"updated":"2023-06-14 19:54:51.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2023-06-15 07:00:30.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"8743424ab6c8e72fc003564f1755720fa7844160","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-06-13 15:04:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20992dc11b0e69ad0ff8728fb5ff9f9b7e291164","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 15:52:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35e1f830424b417f9c0c3abad40a7457\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/549b0beceb094e848f02d024bb225226 : SUCCESS in 18m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f384ba714184a1d86aa1c9972640ce9 : SUCCESS in 44m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/69a28dc02ccc407987f72bc3247ada99 : SUCCESS in 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"409242313f11f8a5092be6a093ffc8bb98de50c3","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-06-13 16:15:46.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d077087aebf5c60ee9e64082ea3ca8639bae26c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 17:14:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ada85c9a958a4b3bb394dd5613509f78\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/36216b03b65b4b379754741812fd98bf : SUCCESS in 2h 03m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ff43e353f24545d79b8d8460ca3eb339 : SUCCESS in 2h 01m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f079361306c143d9a97ee0c610cf95ae : SUCCESS in 1h 51m 10s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bd1e36a68e114d669f68aa4c35386f96 : SUCCESS in 1h 48m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a94ca963a4bd42fbbe35dce9568a1f45 : SUCCESS in 1h 44m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/771c241aee37447680fa27b0830e9e6f : SUCCESS in 33m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6754d8e262b4533b01ff419ef0393c7 : SUCCESS in 33m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b28a7ab7c51649658fff548351e6f8cc : SUCCESS in 10m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c161832acc194c07a7d0f0b58f341e3d : SUCCESS in 26m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2d5919a5c8c47fca6339d44a223d275 : SUCCESS in 22m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efcf3e58feac413b9b2aa0ab77c5729e : SUCCESS in 27m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2a209215b17496c8108c2e9af86420e : SUCCESS in 9m 21s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/2017ffb7d62043f3b088edca3f11ea84 : SUCCESS in 21m 26s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dc3004f7655047fa9e65d164e0812334 : SUCCESS in 1h 01m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c4a4222e1bb45fd88d151a5d41a02dc : SUCCESS in 1h 56m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bfb5a2b007d9495399c0357b0c6102af : SUCCESS in 1h 07m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8507ceac069c4586b07ecc387ad13533 : SUCCESS in 42m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8d73185dc73d411fb42e5e413e4c73cf : SUCCESS in 1h 18m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/abc05d18dfcd464788de94ff03ba6e69 : SUCCESS in 1h 25m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fa000c5f66b646de81c580fb66491fc9 : SUCCESS in 1h 02m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3d8ce9c7ab804326b5293bed025f5662 : SUCCESS in 1h 31m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/163476d6c34a4257bb8eccf1651e4b8e : SUCCESS in 1h 37m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ac5711af261a470399a66dd7c830f14b : SUCCESS in 1h 30m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f4eb8c611462b14d851011672170ef0fd600bb04","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-13 17:16:22.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/05/886005/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"24b4d1b465753baa418c0d3e821bf8674b3cd7f7","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-06-13 21:49:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ed1197b53ae428c451c7fa9fef9d402f92e32d55","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-06-14 09:00:19.000000000","message":"Patch Set 1: Workflow-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"371ecfc01d0acebe7bb516b0f5ba74886433977e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-06-14 09:04:42.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8b4af8477a2e89c3a52c2e1ccd5b8d1c2ef728b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-06-14 09:07:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2160cab4cf74ff2749c42e56ecf8a7c3e977165e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-14 09:46:33.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/05/886005/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"3e2c08b8f8d60cb25ef2fa5d8fb4c74705c803e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 10:00:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8f1c61dbc9d44ab87e3b4d86c8a134f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea977f6da09a4cdc931e52478bd9879b : SUCCESS in 52m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85666c951b0243578812993673535003 : SUCCESS in 23m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2b25a012bc4a46fa9d7d12ab762f6616 : SUCCESS in 52m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c028d5a2fc0b9a4b1ca985ab8e2ac24efb534ac4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 11:16:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8cde6695be146e78376a9595dc92ebe\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/45cf52ec95344b69a07b3c95024d22cc : FAILURE in 2h 06m 21s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0d99879d8f88476d8ddcecdce8869a3c : SUCCESS in 1h 40m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/621fdbedc99644b7a8762d6a62bd3a42 : SUCCESS in 1h 54m 09s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/58a1494567994ff0b00a1d834e4e867a : SUCCESS in 1h 20m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a18edc5dc886463c843028a9b8be41e4 : SUCCESS in 1h 42m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2dedff941e6f4405b77c0f67e4a9ca2c : SUCCESS in 32m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87fcf282d81b4b40beb1d279933177f9 : SUCCESS in 40m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c59b2b6b023648d1bdb9f3abd662bef7 : SUCCESS in 9m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f5c4552ce7a4161915283c506fdd9c9 : SUCCESS in 29m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/808dfcb6dfbb4c1788ff289d8b0147ff : SUCCESS in 26m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff64eb5c41df4a0eb9f748e96684d5dd : SUCCESS in 26m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dd422ba8c96495dbcec826eccd67f8b : SUCCESS in 10m 00s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/8df693786e08492a87692bb50f457426 : SUCCESS in 26m 08s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/50c8f351effc40199d4ff09d5b432fac : SUCCESS in 56m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/65092e015a51471eab392c0c82d4cfee : SUCCESS in 1h 42m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0046d55e5d7d4c249e0de903ef51a77d : SUCCESS in 46m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/50dd93660e7f41e2a7aec6839bdd3a78 : FAILURE in 52m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e456cc693e3a422da30466b2ed5c0d8d : SUCCESS in 1h 35m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c8a8a414326e45d98f29b0cbaad1c88d : SUCCESS in 1h 06m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3af07678c34244acaa21cf8c3750c1eb : SUCCESS in 57m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a9489e36b736472ea2e8e73543fa908a : SUCCESS in 1h 18m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dc8dd42f26144525a1e6fbeea38b96e8 : SUCCESS in 1h 39m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fd0a99ea6cf64c13a46e4cea1afeb29f : SUCCESS in 1h 20m 40s","accounts_in_message":[],"_revision_number":2},{"id":"37793900641de8714481e93db2539a2e437be55f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-06-14 12:12:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"498298c44958d045cde616676f16491c43d8de4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 12:42:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6076f4cf975424995b63dbcda00b706\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/75f4b2cc48c74a8886a6f12e15c25af2 : SUCCESS in 25m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92767a0927c4425eb229a2f9aa1ea7cf : SUCCESS in 28m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2a3f175deac64c0e822a67805010656e : SUCCESS in 25m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e9947992a727abd6e8786b19be5c0d403a6fd892","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-06-14 13:01:17.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f321b645a41ae3418ea693960b1023af061b6f88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-14 14:22:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76469dcdfea14f9ca68e33a58d5dd46c\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d4d45daa55b74d7d9d8f9f982d90b2d2 : SUCCESS in 1h 39m 10s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/179bac932cae45feadcf94edbcab6b6f : SUCCESS in 1h 40m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/04193eed6bb247ce815a2743e0bf6083 : SUCCESS in 1h 37m 12s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/76a9d21cc1ef474d831cb2036687fd34 : SUCCESS in 1h 35m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/82d73400fa9f4a51b14937e3c775268f : SUCCESS in 1h 32m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6a9489445f3947f1be8e088831700942 : SUCCESS in 31m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c43b0a0fa6e4b939e21d172a0aa918c : SUCCESS in 39m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83e56c95d7124f4681979e78eedf16e6 : SUCCESS in 11m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98b91fee3ad147fe94799c2d9ecd26d8 : SUCCESS in 26m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d4fc10cd5e34e87a12e3a7118f13959 : SUCCESS in 26m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/521191ce1a934c519d81719f7f4083ff : SUCCESS in 25m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e1d79708fc48b3883d745d8fb253c5 : SUCCESS in 11m 32s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/f215870d7a374dbf9ab470e23d05b250 : SUCCESS in 23m 04s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/41457aac2855459a9f3c231796a9d087 : SUCCESS in 59m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/02269bd6cee14d24b02055d1e4614bcb : SUCCESS in 1h 58m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fc133ecc895d41d5a14fe63053b58c8d : FAILURE in 46m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7017c29fceb64957868bbdd9978700ae : SUCCESS in 46m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b0dcacfb6cfa4c06ab60937bc6f35126 : SUCCESS in 1h 28m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f5f9406f688c4e6cb1241071bf520122 : SUCCESS in 1h 08m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c04a6c57fda4449797abc4faa543a44a : SUCCESS in 1h 15m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a0dc77fd3fc4471aad1c649dbc556c49 : SUCCESS in 1h 50m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d74f9d231dab4c6ca7e7ca5397c4fd2a : SUCCESS in 2h 00m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/942e3830241042ef8bddcc9c55e85f15 : SUCCESS in 1h 27m 50s","accounts_in_message":[],"_revision_number":2},{"id":"aba82f2875fcdc73422e7424f51770f5f341c2c5","author":{"_account_id":33983,"name":"Luigi Dino Tamagnone","display_name":"luigi","email":"ltamagno@redhat.com","username":"ltamagno"},"date":"2023-06-14 15:15:35.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e157b35ccf8fceb65569db545a862ba3ea0ed047","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-06-14 19:54:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d1032b7bbdd11ce389093ea52ff1eb3913ff1c6d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-06-15 07:00:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cd26f522a42de32a8c78b175f700a1fe3c0f04ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 07:50:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47aaf7eacf0a4040a8f5f20cbc9a09e6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e32f79ce84b84439b3e579908ca70cad : SUCCESS in 37m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b5e442309e484a788c18f17fd6bdbc1f : SUCCESS in 47m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f521dacb2924ce4b52ac49e6794ff77 : SUCCESS in 45m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0cf6bb85e3d4681d477124cd0d9e71670bf20a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 09:39:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9cc871d65d040f48f3b50bb41dc9a39\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/7aa5d365111847d38ef0f40dbde59340 : SUCCESS in 2h 38m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a6d7352cdd794fd8af3c5866a3a7a31f : SUCCESS in 1h 56m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a6e10845606f4f599723152b3b2c3725 : SUCCESS in 1h 41m 25s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/740c4e81279f49c9baac7edaf06052fd : SUCCESS in 1h 42m 15s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/56e30dd9cf434f57875891c2f7169939 : SUCCESS in 1h 47m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bd6bfc82e8fe4808ba697ab616672883 : SUCCESS in 40m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e20f8a3a2a54432ba70fe20b4f2dfaf : SUCCESS in 25m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3010594c2b049c5bc22ff7520eae444 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a28660072ed40508db81d1474039cdb : SUCCESS in 19m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d135a713587647bfaed7dc6ace41aa3b : SUCCESS in 20m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72604f5941ae4fd38f6eafd2d4d6396f : SUCCESS in 19m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82835a96a6a141d588c2ad84882be54d : SUCCESS in 12m 18s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/cde3cd4b32a44904a51fe7b8b90228dc : SUCCESS in 25m 58s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dcfed06dfc5d4eca830eb2464c374a06 : SUCCESS in 50m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d27a397bfd44f749dda892cadf101ab : SUCCESS in 2h 01m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/31842191ec4c4e01bb1418483293e97d : FAILURE in 2h 25m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/87489d45f8bd4cd1b433a312f3c5549f : SUCCESS in 1h 05m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d2704add4a4c4684842e0b11e52f7261 : SUCCESS in 1h 01m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a11a9aab64884b03a172cf775d18944f : SUCCESS in 1h 07m 00s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5a8c76d31bfb4eb69decc44d9d6e9e8f : SUCCESS in 56m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/114e878987fc477c85cad8fa2ccf2209 : SUCCESS in 1h 16m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/97700217cdd548d0a28e4f260dabe403 : SUCCESS in 1h 32m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/117e11a042a7413fb5e80f031aaee3af : SUCCESS in 1h 31m 27s","accounts_in_message":[],"_revision_number":2},{"id":"9fe1f2ddf7441a8f3380748e4c8e98ccc00520db","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-06-15 12:29:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ab8272ed7df5414fdc3a1a3743d880c1aea59dbf","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-06-16 08:38:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"259196233c2d0d73adffb3977dbe47e95184071b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 09:23:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea29b45be1764ed292bab3dd79b0de9c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5e61ea55d634e6d90f060f87f54b4e4 : SUCCESS in 19m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5cbab756ea24b13b27cc320782d950c : SUCCESS in 19m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4998d098cdac4414899785d96b34f8c1 : SUCCESS in 43m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0bfcb9d92fede10670f5cc0eb4b561955b82b001","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-06-16 10:02:14.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce9fa3be2654c753b003e303c241bd0774f76815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 10:53:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5e5b29eee4046ed9d19d0b420b6c7c2\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d55a5cf36a0648b1be52bb0e9bcda271 : SUCCESS in 2h 00m 48s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a4eb60353ac945cda5d15c13dd28990d : SUCCESS in 2h 00m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9550d62919b442af90bdd9cfa1b7725a : SUCCESS in 2h 11m 20s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c5257da8464a4fd98bdaa317a903eba1 : SUCCESS in 1h 40m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b3a19a6feffd48eebf5fb254dda53448 : SUCCESS in 1h 29m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/26c71db4fe444ae88c8d72181ca78923 : SUCCESS in 32m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef7a999fcbd54d84afaeaf212c4098b1 : SUCCESS in 32m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e722f996b6847cd8db8f67d1957fcb2 : SUCCESS in 11m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41807524184c4b0980b09b54a1a7de8b : SUCCESS in 23m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14fec41754c64368a86c190ca11d782d : SUCCESS in 23m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78bb50a879f346c098401843005bd503 : SUCCESS in 23m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2838046646714b329b76666708ca81f4 : SUCCESS in 11m 18s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/11f64703acbc4de982a721e07100f866 : SUCCESS in 23m 42s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b9ee6c0102a1477a9e74f5a86915d2aa : SUCCESS in 47m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/130e83381ec54e59a890ed06ec6daa2f : SUCCESS in 1h 55m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f77bee9a0d024442b68708e4ed5064e8 : SUCCESS in 39m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b769d13dbea440918dec0ca99f15449d : SUCCESS in 38m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dff090e2c2954322968544bc11a00fe9 : SUCCESS in 1h 22m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d519865b1bac4f57be03a02beb1a5ad5 : SUCCESS in 1h 02m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9024bf3c26a9440ea8f9e7c3b8b23dc8 : SUCCESS in 56m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3fa80c2b69d74e99be02c136e0fdf51e : SUCCESS in 1h 35m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a0abc261f2ca4acfacf7f16a036ee379 : SUCCESS in 2h 01m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f0e2ac600b7b4717ae977dc2ceafc8d6 : SUCCESS in 1h 19m 42s","accounts_in_message":[],"_revision_number":2},{"id":"ec2066006696a6ed0760ff663a88e5a12459cc20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 10:53:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"83e38048f75b3e37e8916534cfef565e8ce9e05a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 13:02:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d070ab575f564ec3b96ba53c73fca51e\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f869f1648014472883828d8692dfb17c : SUCCESS in 1h 58m 40s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7976356911d84d22bdbda39b4ee7ccab : SUCCESS in 1h 47m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/24ee2062635347ac85adfdfae78951d8 : SUCCESS in 1h 52m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a94bd4be26c7490cbdd9238d39d2802d : SUCCESS in 1h 43m 49s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/50d3ec2d1bb6489cbdef250a2fc1d0cf : SUCCESS in 1h 33m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16feb74ca3ac4fd191d305ad10e4ecc1 : SUCCESS in 13m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ef49e12cd58445995e0f472eaead03f : SUCCESS in 26m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e6f1a52e721401d8d7eaf2886f5e03f : SUCCESS in 26m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3530b1ce9be42cdb1b5b8c7fbe70826 : SUCCESS in 29m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e5bc52d8b8e43e8a5bb695039614347 : SUCCESS in 11m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d6d2f7df5b3433ea2d437ad5560d8d7 : SUCCESS in 1h 01m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dac689e22e4542b0a5fc5f7ab3b332fd : SUCCESS in 2h 01m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0a815973c9ca4116b37fdc5e57845d3d : SUCCESS in 1h 32m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5facab985b52479895c19544e08b1410 : SUCCESS in 37m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5a88eb321fac4787854e4ee510c5d130 : SUCCESS in 36m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/367cfb2059154128a0f46d06758f8c2d : SUCCESS in 47m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/11ee9a9cc4e6401296923e69d335e6ec : SUCCESS in 1h 20m 50s","accounts_in_message":[],"_revision_number":2},{"id":"4e9c2b6e8255728fe42bae6530399a1197faa3d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 13:02:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"54ef7fa5158ea25fa9330cd4e4a6630a99563cbb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 13:03:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/222ddc0de2e842b5b12e49cc5ca42c0a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5163f7fd71cd4f8188c3646ec1592fc8 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"afa20faec3c37bd06346360cadbad0d69e9925f0","revisions":{"062280c161e3ebe0c6a0dd47c5e38079a0f164eb":{"kind":"REWORK","_number":1,"created":"2023-06-13 15:04:06.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/05/886005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/05/886005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/886005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/886005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/886005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/05/886005/1"}}},"commit":{"parents":[{"commit":"6626fd9c9c9b75ac28b5f85e6a92a24f55e9c44e","subject":"Merge \"Allow Multiple External Gateways\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6626fd9c9c9b75ac28b5f85e6a92a24f55e9c44e"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-06-13 14:53:54.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-06-13 15:02:52.000000000","tz":60},"subject":"[OVN] Improve Hash Ring logs","message":"[OVN] Improve Hash Ring logs\n\nDebugging Hash Ring problems can be difficult challenge given that prior\nto this patch the logs were very limited.\n\nThis patch improves the logging for this feature as follow:\n\n1. Log when a node is added to the ring\n2. Log when nodes are removed from the ring\n3. Keep track the number of offline nodes and log it upon loading the\n   ring\n4. Improve the \"Hash Ring is empty\" exception with the number of offline\n   nodes found (if 0, means the ovn_hash_ring table has no entry in it)\n\nCloses-Bug: #2023670\nChange-Id: Ic90432b5ddea8cf176de159ec7eaafd5fd7bdd6e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/062280c161e3ebe0c6a0dd47c5e38079a0f164eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/062280c161e3ebe0c6a0dd47c5e38079a0f164eb"}]},"branch":"refs/heads/master"},"afa20faec3c37bd06346360cadbad0d69e9925f0":{"kind":"REWORK","_number":2,"created":"2023-06-14 09:07:24.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/05/886005/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/05/886005/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/886005/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/886005/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/886005/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/05/886005/2"}}},"commit":{"parents":[{"commit":"6626fd9c9c9b75ac28b5f85e6a92a24f55e9c44e","subject":"Merge \"Allow Multiple External Gateways\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6626fd9c9c9b75ac28b5f85e6a92a24f55e9c44e"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-06-13 14:53:54.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-06-14 09:07:09.000000000","tz":60},"subject":"[OVN] Improve Hash Ring logs","message":"[OVN] Improve Hash Ring logs\n\nDebugging Hash Ring problems can be difficult challenge given that prior\nto this patch the logs were very limited.\n\nThis patch improves the logging for this feature as follow:\n\n1. Log when a node is added to the ring\n2. Log when nodes are removed from the ring\n3. Keep track the number of offline nodes and log it upon loading the\n   ring\n4. Improve the \"Hash Ring is empty\" exception with the number of offline\n   nodes found (if 0, means the ovn_hash_ring table has no entries)\n\nCloses-Bug: #2023670\nChange-Id: Ic90432b5ddea8cf176de159ec7eaafd5fd7bdd6e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/afa20faec3c37bd06346360cadbad0d69e9925f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/afa20faec3c37bd06346360cadbad0d69e9925f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33983,"name":"Luigi Dino Tamagnone","display_name":"luigi","email":"ltamagno@redhat.com","username":"ltamagno"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
