)]}'
{"id":"openstack%2Fneutron~886213","triplet_id":"openstack%2Fneutron~master~I271c109a597eb0aa088a7a9c785e8631bfaa01d7","project":"openstack/neutron","branch":"master","topic":"bug/2024044","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-06-30 14:50:39.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-07-01 01:19:51.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-06-30 10:59:04.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-06-30 12:15:41.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I271c109a597eb0aa088a7a9c785e8631bfaa01d7","subject":"[sqlalchemy-20] Remove redundant indexes from some tables","status":"MERGED","created":"2023-06-15 19:58:48.000000000","updated":"2023-07-03 07:25:36.000000000","submitted":"2023-07-01 01:19:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886213-bug/2024044","meta_rev_id":"ae8eabc728adc339308c92ff8cfa9153aac73dae","_number":886213,"virtual_id_number":886213,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-06-23 15:34:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-01 01:19:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-30 10:59:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-06-30 12:15:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-06-30 12:15:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-15 20:37:52.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-15 20:48:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-15 23:06:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-16 08:21:38.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-06-21 14:59:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-06-22 13:43:26.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-06-22 14:28:48.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2023-06-30 10:19:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"1f9a6f1e6105c26a61b4f21b1b8063f3a8ba9f9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-15 19:58:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e4f5d8f96762af1f3a0c77029b6d79dbe89a068","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-15 20:37:52.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/13/886213/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"85d3c959b8f7343e0ed05710f0f55834c7fde416","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 20:48:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96ed08aa9e0f4ead8b734e8693e6e643\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7810071600de47aa92f03effdbdce2ae : SUCCESS in 47m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87c0bb17ef5c4330927506df7c8b6a99 : SUCCESS in 20m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2567e89d40b6402382a6aeb03f80c7a8 : SUCCESS in 47m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02f186c83e49b6619ea77c6c23b8a468ef3e0898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 23:06:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2091d500f8714510913b6554ad3f87c4\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/87489f62b118413d8789d7279975f0c0 : SUCCESS in 1h 57m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c78cfb5cfee54da1b9353bb77c9138fb : SUCCESS in 1h 34m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fb930b003c0841e8a720af56ca95387d : SUCCESS in 1h 41m 48s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fab122f953f44f1c8ba7620bd60f304b : SUCCESS in 1h 18m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f457ec59143d4fe9af8da33fe1f143ab : TIMED_OUT in 3h 03m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b1f4d7613b4a4c1c980553a3784a7ca0 : SUCCESS in 36m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34067a66601247578738d179c719b92b : SUCCESS in 40m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d668e15efb843a195499bd9e43cde9b : SUCCESS in 9m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/764ffb203b7c4ef1bd44080aa3f19617 : SUCCESS in 29m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81a8d19176174043b1958999c5ea3e25 : SUCCESS in 28m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af8cc587ce0c4500be830def94408247 : SUCCESS in 29m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8072b1dd31d74619baf87017f83849be : SUCCESS in 12m 45s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/08aec441a78442048592cc43d7b32cbd : SUCCESS in 23m 22s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c1883cb09e14c9e809fea4bd20f798e : FAILURE in 1h 00m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3381e5b50ffc4d1d9ee619f3afcfbb43 : SUCCESS in 1h 39m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2138cce9dc2a46828721cd06bb898cf4 : SUCCESS in 59m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1b1522c53d2545b2960a74c82f03d4da : SUCCESS in 42m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6926c4e85f8943aa854b5fba3c6e7eb0 : SUCCESS in 1h 35m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0a69e623299c4f65adca235a0a72f207 : FAILURE in 2h 12m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0adce2da173a4eb9ba18a6b689d87550 : SUCCESS in 1h 14m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/573c7542eb7640bca27f14d339ee61b6 : SUCCESS in 2h 06m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/270cea905f844b8cbd23d85b969412a3 : SUCCESS in 1h 36m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4088aee22f254173b36d12a07db4c4c5 : SUCCESS in 1h 24m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3ec86e6583fb13c2f2fd4e7a1b2462d79aaa2ee7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-20 16:08:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dba62e900996a86693ca6db4545123e461adbb88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 16:33:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0beb0032e1204f6f8b267b3e18dd0b58\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/73ac495ec5284ef386143174a51c9505 : SUCCESS in 21m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d1f34ddbba2a42d6833ca3675a0aa717 : SUCCESS in 24m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e038053c601c4c1ab80338be6b2f3651 : SUCCESS in 21m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"19bc8d5d5df77dfd43bc445d71ea420823ff299e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-20 18:30:44.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/13/886213/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"a2dfc1a7f40d36257e9cd05281e94649feb58749","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 20:06:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9242b45531646caba8809cb1940f19c\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/90e06c464aa74a7ba39e1eb323422a63 : SUCCESS in 1h 40m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/059bafdfbe05434b8b14167e3fce0210 : SUCCESS in 1h 25m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fe70bfb1cd0d40ebb66318804300e98c : SUCCESS in 1h 32m 20s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/760b4b47d3c74aa1bf673c93105513e2 : SUCCESS in 1h 41m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cae018e70b49417b935c6423bbc4b59d : SUCCESS in 1h 22m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fc32f8c7c8234608a137e4c586b6f1a8 : SUCCESS in 36m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c89d46e3353480e8de9c2f3fcd5731c : SUCCESS in 39m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d2b4c91a2a40fc8a7aa57c54e74786 : SUCCESS in 14m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdc2fd45eed643928c3aac59dffff175 : SUCCESS in 34m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4280e6fc237b43fc8d6eb42bb2258399 : SUCCESS in 27m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9917fdccf82c417790148440fb8e917f : SUCCESS in 26m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2b0764370bb4c75b7b735bd3e51e688 : SUCCESS in 11m 07s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/2f764f3619164294b161964752c42647 : SUCCESS in 25m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/500158f9f16f42c09ff9a57959be04e5 : FAILURE in 1h 11m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee65f1abfc964de5963152799a3b0a11 : SUCCESS in 2h 05m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/21e65f10caab49528b7d42fad5b02f5d : SUCCESS in 59m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/68e222faceb545e8a924c1834804c360 : SUCCESS in 55m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5a4db2d21d504493b0e3f17198a6df29 : SUCCESS in 1h 49m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f301c1a78ed44a29993942ad57d5fdd5 : SUCCESS in 1h 24m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/78786b00593441fc814c162e3fcc72fa : SUCCESS in 1h 08m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6f61122172984d4c91d83c0a34ab13d2 : SUCCESS in 2h 25m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/56d80da410984a159766520b64972a30 : SUCCESS in 1h 49m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5796838a5e094d84aad9bd629809f78f : SUCCESS in 1h 40m 18s","accounts_in_message":[],"_revision_number":2},{"id":"9a1a9c7163736d17458913f0f5ee967e083afc15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-21 09:58:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"75f2b69472d43659c3d154dca62576ebbdfc78b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 10:43:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/789062ecc5824175b76d4fb6b8906444\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/48a4e3b97f9a4979b4844c6383a2652c : SUCCESS in 37m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd4e42ba230d4191892744f14ac8e8d5 : SUCCESS in 38m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/52971e679ccf486589c696137ee794a3 : SUCCESS in 40m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"081a75d9fcdf0dec8fb7ce28e3fd85f61ba68325","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-21 13:14:08.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/13/886213/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"7e6057599345693e8b6096b914ade8b7e495e140","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-06-21 14:59:15.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"990787928214075d97e85900becc1676d01508d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 16:41:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/54f33f830ceb4003bdd6386286473f2d\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9080e8b8222640849806da1ba6defd81 : SUCCESS in 2h 02m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/762fe2d6d2994237a8257a9d7b8456a8 : SUCCESS in 1h 23m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d67c47e259cf463083cfd92f4eb36584 : SUCCESS in 1h 38m 25s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/499e166d2664419db6c0764ec953734b : SUCCESS in 1h 34m 34s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0a7f3049a30b4908a531dd7de59f9268 : SUCCESS in 1h 44m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/64fcfb5f4b814ffcbb66a2ffadfaf712 : SUCCESS in 33m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c52e0db6316e4c0fb241a56ce3f960cd : SUCCESS in 36m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9844011fee464884b1a6899b4e870605 : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1b28eaf0b8247808db897caac1ff2eb : SUCCESS in 25m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b93e3e109c99418a8133e14d1246970a : SUCCESS in 29m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db14869ca98f4ae2ba0142217f18b335 : SUCCESS in 35m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd98d12b7b24b77b2e2e79d9b9ef115 : SUCCESS in 11m 30s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/3228b5abd4394edb8aeec0138226814d : SUCCESS in 28m 47s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f03eeccdf9544abaa6dd491bc2f554d3 : SUCCESS in 1h 11m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6bbceff529b3486e85a9e4b2f74a6157 : SUCCESS in 2h 02m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/97e6279de211419f9da413985aec989e : SUCCESS in 1h 06m 23s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a0e7f4c1cabc4b8b85d9664478bbc506 : SUCCESS in 1h 00m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/22fac2fefaf64617abe27c03075f578a : SUCCESS in 1h 30m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/817ae2d47ff94261b6368dc3cca8ed85 : SUCCESS in 1h 22m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1a6ce86c6f994c088d60572a56f012ca : SUCCESS in 1h 30m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1fa433f2769f4517bd400a3141c4b809 : SUCCESS in 1h 49m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0ddb8fd0f61d468dbc47ca89b5b08079 : TIMED_OUT in 2h 35m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5d5da26dec4e432fb8a9c283137a0e69 : SUCCESS in 1h 47m 49s","accounts_in_message":[],"_revision_number":3},{"id":"db33c5116766c5804155f69429e205be257b62dd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-21 17:12:59.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"75800e5c8e3961065171322781a3017fa810d000","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-21 17:13:10.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a56860df5a9355632b6aa9c6cd8f304b322cfd79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 17:44:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f074c0879e984d4fbd064100a7071fc1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e12f3b732d54cde993b0ce1e5a35fa1 : SUCCESS in 23m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e841256f0aa34b42943b5ff58ddd8c93 : SUCCESS in 24m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f4b234dd52d24c4aac3009447e443639 : SUCCESS in 30m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2e71ca2d28a9137418b250a38d65f41a65e7fcad","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-21 19:06:02.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/13/886213/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"9152d19ee5d5f033644b77ea6683f656338957a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 21:12:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd9c386edc9a4ed695b20305ff8ab5ae\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/93c37ea026604bdab7dc650cbf629a03 : SUCCESS in 1h 24m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0b94fadad93b4fa89c37d66b06272e96 : SUCCESS in 2h 02m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a90dffa011244fdfa385bce3275dd2ad : SUCCESS in 1h 49m 24s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/af8a4ffe14734b21b7d5787e081c0aca : SUCCESS in 1h 29m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/281c88ce221b417e9122094c20867bf1 : SUCCESS in 1h 33m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a1c9285e84b84e9bb7cf5a3d2fd3f62a : SUCCESS in 41m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2038b532890c42668078b51f11aebd3e : SUCCESS in 42m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54e1c4cbb91b41868339d597a72cace0 : SUCCESS in 10m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e3e48c5ea294f51879348baae5ffeaa : SUCCESS in 29m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43b4eb56569c4ca1a032f8080fbe766a : SUCCESS in 29m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8222bf0fcc664d33badaa0f15892375e : SUCCESS in 27m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6706da3297b4fb7a59780137240bbf0 : SUCCESS in 10m 52s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/83c03b39f72e4b8ca6682c3901963871 : SUCCESS in 30m 00s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/709a34e1f02349f4a24e786a2b9b48ff : SUCCESS in 1h 02m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8519e9702df44881b7d8e25fceea48ba : SUCCESS in 2h 00m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/77e83ace7e5a4a6fb13755cc1188a1f7 : SUCCESS in 1h 14m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/05d98a0fb8704c7c95b44b579f6293e5 : SUCCESS in 1h 01m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e0f7f0a2fe4843c6bf406ff1509b373d : SUCCESS in 1h 19m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6911376f149d490fb9f01c59c95ec8dc : SUCCESS in 1h 34m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6e47a43f6bd44522bb8addd23ebfaa46 : SUCCESS in 1h 00m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e61248c4e54d4c66a4546ea92d14c8f8 : SUCCESS in 1h 48m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2557b2545c2544e9b181024b677c372b : SUCCESS in 2h 11m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a45a421b7c474fe28b0b619d195e3d95 : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":4},{"id":"801b7bb8f1f6239f67f8b1db4cd13a35de24f31d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-06-22 13:43:26.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f8dc6865a62ba1103bfa54cda0b17034f3b9cb51","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-06-22 14:28:48.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e7abcd211e68b859774a0eb89a8d9b0657453119","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-23 11:06:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e01aa5d0c3ed8a9c3b5838b8765f08c5fe6657e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-23 11:06:36.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"64be4ebbe74632808f4376e946ad1f8bbe397ce0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 11:10:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb29e080642142e38024f9668b3d06c9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/029ee77e180e4ad7bf451e29e2e0672d : NODE_FAILURE Node request 300-0021527026 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b071ba4ec06a4fbfbb6fd5c708e89408 : NODE_FAILURE Node request 300-0021527027 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f364505bedef4b00804b24c9e94b06a6 : NODE_FAILURE Node request 300-0021527028 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dd3fa96b256df89f6dcc1808514c4190a85b579b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-23 11:24:23.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"3bb1b3dd1c64a9907be6207aae86f83f8333adf0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 11:58:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61ef65557c3546f6976c5a00148deca2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/65cfbb1940904a03ba004aac6c33176c : SUCCESS in 25m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e1ad58d9cbd4d949e9464bed256d8b6 : SUCCESS in 25m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d7593855a4cf4498b0d3d9cf9f0cdaa3 : SUCCESS in 33m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8ea28cae786099d3691276e02c26b2a6813fa9aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 13:40:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dca2a07ac7404afc86744c93a592e6d0\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/cbf56de24bc04a6d902bd888b9933bf5 : SUCCESS in 1h 48m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1bdd1fd85f3441bc8266e0facf0a3199 : SUCCESS in 1h 45m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b5bc45bb37cc4d8581da65fdbda164b2 : SUCCESS in 2h 01m 27s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/44c3e6decc29425880f6e19214ea09a7 : SUCCESS in 1h 33m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cbb48405a0024defbde58cd223a8db0e : SUCCESS in 1h 27m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fdf1381dc7c34215ba705fa4b143236f : SUCCESS in 38m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f55c8111a2dd486585914b80ce0f3e55 : SUCCESS in 40m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b090c4c620ed40738782328e11e488b6 : SUCCESS in 14m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff18dfe7d39440098fb1b72f94304983 : SUCCESS in 30m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64f82598423242c68ab9d353ac80f068 : SUCCESS in 28m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aff72500984948f6a815c8346894a203 : SUCCESS in 25m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8001967f74844ac8b6a79cd9757488e : SUCCESS in 11m 51s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/97de0c4448144b2bb63b8819fc206e04 : SUCCESS in 28m 54s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a48952bb68074798ba888cde29d72d56 : SUCCESS in 1h 08m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a63a454bba14953b296ec5ff5019897 : SUCCESS in 1h 40m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/727608bcfd61455baf124bb2e47ff3d5 : SUCCESS in 1h 10m 38s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/747d51db9f3445f897f6a06667e7b454 : SUCCESS in 55m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c92c68b372ac46a9b9c223da86eab967 : SUCCESS in 1h 39m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/875f1fec170a431680f961277b1e6a45 : SUCCESS in 1h 25m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/05c19280e3a849478abd2c1e50e7fa9d : SUCCESS in 1h 12m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4e3521dc2d9d4c7a90e574f82e52addc : SUCCESS in 2h 03m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b040c82a8ab04f4c9b7d82011bed55bf : SUCCESS in 2h 02m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/96a98c858ec34be19baeba66ee8f2b92 : SUCCESS in 1h 42m 39s","accounts_in_message":[],"_revision_number":6},{"id":"01070289c5fd2040406b16219b35cd45223d8cd1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-23 14:56:32.000000000","message":"Patch Set 5:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/13/886213/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"30bd5eeda2968b72e3f223690a3cf48a02d959d9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-06-23 15:34:30.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2023/Arista_Tempest_Testrefs/changes/13/886213/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"87f4cf985e7b2a2f37c82b9977a9db0130582e3d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-06-30 10:59:04.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b6736e3713b8a9edbb816ea6365b8f4356633e0e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-06-30 12:15:41.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e3c1b282e3cdd18c7f8f3cc0e4ef7eba267e2c94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 12:16:13.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ac04987e4e015e0d0270e291e7f20feb41b7523e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 14:02:54.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/99c0cfcbfd4a4d1f8ad2fdce933ad95a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/29efde29b04540c1801697f29e813606 : SUCCESS in 1h 28m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1a36262ec92541c88280b8527fc78a20 : SUCCESS in 1h 39m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/da657fc8a7b441a4b82b4467724d6044 : SUCCESS in 1h 44m 25s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/07d70d9a480042a699173576cf586451 : SUCCESS in 1h 30m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/963f5d5315694d1782e7a29aec4f9089 : SUCCESS in 10m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7a99d03af824c9a87ffd73360ab15e0 : SUCCESS in 29m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fe7835c3aff4feabac03bbb947d9a70 : SUCCESS in 28m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c98310405314ba592049c5ea8f1693e : SUCCESS in 30m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b958248f2fd84937ba7615ee8465745a : SUCCESS in 9m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/487b4b7257b24b10b1231d58937a7678 : SUCCESS in 1h 00m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc879a6674724638b854ef03cd2fca1e : SUCCESS in 1h 43m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/90aabe725a054924a022532dd0f60779 : FAILURE in 1h 15m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/11963e0d2509435ea8fd4ce49135ac87 : SUCCESS in 56m 14s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9cefd4a5e32742b4919b09c901333d63 : SUCCESS in 44m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2f81a5786fe5471386297c22922f63ad : SUCCESS in 1h 12m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7eb1216f32d347deba9ec9b733d6110c : SUCCESS in 1h 26m 10s","accounts_in_message":[],"_revision_number":6},{"id":"505e1ff3efe394a7c81139874d40dcd221eac890","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-30 14:50:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"407cd2c9ef3955784c9ab208b99876b935b8cd59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 15:38:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1514e736da9a445ea0f83463e9b9cac2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/09babedb492d4629ba6d8e7943b01fb1 : SUCCESS in 26m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3619cad79db470799e19aed7507c830 : SUCCESS in 44m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4987c4124ea4cbb9d302a825cf35d18 : SUCCESS in 25m 17s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"74422fb5a7391a92a85e795ab179cc0f5dcfb6d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 17:45:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31b7b8dc46f549e2b26cafaf764cbad0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3e0beb94b8d04ed88d602db1534726ad : SUCCESS in 1h 11m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1bbd690570f844c9999c2b4bdc3cb103 : SUCCESS in 1h 43m 33s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bd93651bbe3949b9a8498378ff7588fd : SUCCESS in 1h 21m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8abee498289c436a95e55f8b46e15b08 : SUCCESS in 1h 38m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/86f0d75c459d490faa7c2d1f5e6bb3ae : SUCCESS in 34m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6822145ec00940c2ad1801068a60df5f : SUCCESS in 43m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a435a07598a4476ea728d1a031ae954e : SUCCESS in 12m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3958f4c73e2046b5b745eec6c60121bd : SUCCESS in 24m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f46533cac4174717a85ea1ca2a1b8e87 : SUCCESS in 29m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06618c38108f42b8846dadf06fd88c69 : SUCCESS in 26m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e18f747a960f4eadae3b75c2882dd2da : SUCCESS in 12m 19s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/617bc775fd96499fbc695349213c2682 : SUCCESS in 27m 22s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/79347f6702844c46aa79cfd6e6981d38 : SUCCESS in 1h 06m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/10dd0b5247cb44fc9dbf5bc9b371867a : SUCCESS in 2h 09m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/926c6a2503b74d588e057e9ebe2ac340 : SUCCESS in 1h 04m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7c4fd47244334e168ae6b272e3220a88 : SUCCESS in 47m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dd6a2df8814843d5b0f2bfd949a42daf : SUCCESS in 1h 36m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b8a05548a679465ab9ca1a0cd5449c7a : SUCCESS in 1h 15m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/adc27dee172b4c53aeeedb6568ef2eb1 : SUCCESS in 1h 16m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/efa0cf78d9c64a6cbabff8e2de7ff92c : SUCCESS in 1h 32m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/176d84ab0311488fa0a37fd64088e12b : SUCCESS in 1h 41m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f5ab988ace144c57a39443690eac38bc : SUCCESS in 1h 33m 56s","accounts_in_message":[],"_revision_number":6},{"id":"b6d77c77301db6ed09a6015fd61b5c266cd6df94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 17:46:13.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3d3910ec604767264faa55b32d78d7d5831b6a0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-01 01:19:50.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42ac492994054690bc87b5d3ae37b6ed\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/73b1113ca7b1467da7df80160ab31c09 : SUCCESS in 1h 28m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/672709d882d54f728876c4bc448d4f5f : SUCCESS in 1h 40m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/88fb0fbaa0b64d0795ef669f0e2ad7fb : SUCCESS in 1h 37m 23s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fa6ddffa01e2484b931e1e0aa6c8bb65 : SUCCESS in 1h 22m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9618bf7e9b0b45cebbd07967afd2c53b : SUCCESS in 10m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfa65d9d7c724c26ba6dbc4257389b99 : SUCCESS in 22m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41a6ed8b5401431fa023daea0c814f0e : SUCCESS in 22m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/866a5df2f8914d81b8b77cf2d65cca70 : SUCCESS in 34m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edb378e7ee48468494721014cae91ce5 : SUCCESS in 10m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/79d055da51bd4bad9b76bd4d11c027a7 : SUCCESS in 1h 01m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f90a03706d0f4fed992119eb5f8da207 : SUCCESS in 1h 55m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/87830886c2df4cd6afec7818e98497f3 : SUCCESS in 1h 32m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dedd2f0915694475854b037131e3dff2 : SUCCESS in 58m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/81347f7b34824b2d93afd1f3b852e5f8 : SUCCESS in 59m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cd1c6babe38a44e4bdcccedef9202c9c : SUCCESS in 1h 01m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/42e89cf1f025443c9b49ccd1740338e1 : SUCCESS in 1h 13m 05s","accounts_in_message":[],"_revision_number":6},{"id":"bba69d8b4d0f54df6b3a9e2f64f733aa33e0f871","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-01 01:19:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"b2708bcd2a4f504da315727b6db03a513391f0ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-01 01:21:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ca9bd2da62d4cf4b911c9d7a32173c4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d57d5ff749f74af6a65276ce64359bfe : SUCCESS in 55s","accounts_in_message":[],"_revision_number":6},{"id":"ae8eabc728adc339308c92ff8cfa9153aac73dae","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-03 07:25:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"08fe84f4438293b7000d0bcc829a8c66762c7cbc","revisions":{"24b9dea828a592bbc256024716445fe6b2f9cede":{"kind":"REWORK","_number":1,"created":"2023-06-15 19:58:48.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/886213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/886213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/886213/1"}}},"commit":{"parents":[{"commit":"db12fc400e926ee4901e3e50412992a1aeed79a1","subject":"Merge \"Send ovn heatbeat more often.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/db12fc400e926ee4901e3e50412992a1aeed79a1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-11 04:22:38.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-11 04:22:38.000000000","tz":0},"subject":"[sqlalchemy-20] Remove redundant indexes from some tables","message":"[sqlalchemy-20] Remove redundant indexes from some tables\n\nThe following tables and columns were defined as primary key and key\n(index). A column defined as primary key creates an index in the table.\nThe is no need to create a second one.\n\nTables and columns affected:\n* portdataplanestatuses, port_id\n* portdnses, port_id\n* portuplinkstatuspropagation, port_id\n* qos_policies_default, project_id\n* quotausages, resource\n* subnet_dns_publish_fixed_ips, subnet_id\n\nCloses-Bug: #2024044\nChange-Id: I271c109a597eb0aa088a7a9c785e8631bfaa01d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24b9dea828a592bbc256024716445fe6b2f9cede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24b9dea828a592bbc256024716445fe6b2f9cede"}]},"branch":"refs/heads/master"},"64c70b54a602f2b147241ea0fa36108cac4e509e":{"kind":"REWORK","_number":2,"created":"2023-06-20 16:08:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/886213/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/886213/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/886213/2"}}},"commit":{"parents":[{"commit":"cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a","subject":"Merge \"[OVN] Improve Hash Ring logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-11 04:22:38.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-20 16:08:23.000000000","tz":0},"subject":"[sqlalchemy-20] Remove redundant indexes from some tables","message":"[sqlalchemy-20] Remove redundant indexes from some tables\n\nThe following tables and columns were defined as primary key and key\n(index). A column defined as primary key creates an index in the table.\nThe is no need to create a second one.\n\nTables and columns affected:\n* portdataplanestatuses, port_id\n* portdnses, port_id\n* portuplinkstatuspropagation, port_id\n* qos_policies_default, project_id\n* quotausages, resource\n* subnet_dns_publish_fixed_ips, subnet_id\n\nCloses-Bug: #2024044\nChange-Id: I271c109a597eb0aa088a7a9c785e8631bfaa01d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64c70b54a602f2b147241ea0fa36108cac4e509e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64c70b54a602f2b147241ea0fa36108cac4e509e"}]},"branch":"refs/heads/master"},"6ca714417e8b0bebfa458b0490a7b1a829f090a4":{"kind":"REWORK","_number":3,"created":"2023-06-21 09:58:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/886213/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/886213/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/886213/3"}}},"commit":{"parents":[{"commit":"cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a","subject":"Merge \"[OVN] Improve Hash Ring logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-11 04:22:38.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-20 17:06:19.000000000","tz":0},"subject":"[sqlalchemy-20] Remove redundant indexes from some tables","message":"[sqlalchemy-20] Remove redundant indexes from some tables\n\nThe following tables and columns were defined as primary key and key\n(index). A column defined as primary key creates an index in the table.\nThe is no need to create a second one.\n\nTables and columns affected:\n* portdataplanestatuses, port_id\n* portdnses, port_id\n* portuplinkstatuspropagation, port_id\n* qos_policies_default, project_id\n* quotausages, resource\n* quotausages, project_id\n* subnet_dns_publish_fixed_ips, subnet_id\n\nCloses-Bug: #2024044\nChange-Id: I271c109a597eb0aa088a7a9c785e8631bfaa01d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ca714417e8b0bebfa458b0490a7b1a829f090a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ca714417e8b0bebfa458b0490a7b1a829f090a4"}]},"branch":"refs/heads/master"},"8ac4c7a54cdf09a18a6f2f2897285fd3b7622c7d":{"kind":"REWORK","_number":4,"created":"2023-06-21 17:13:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/886213/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/886213/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/886213/4"}}},"commit":{"parents":[{"commit":"cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a","subject":"Merge \"[OVN] Improve Hash Ring logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-11 04:22:38.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-21 00:20:16.000000000","tz":0},"subject":"[sqlalchemy-20] Remove redundant indexes from some tables","message":"[sqlalchemy-20] Remove redundant indexes from some tables\n\nThe following tables and columns were defined as primary key and key\n(index). A column defined as primary key creates an index in the table.\nThe is no need to create a second one.\n\nTables and columns affected:\n* portdataplanestatuses, port_id\n* portdnses, port_id\n* portuplinkstatuspropagation, port_id\n* qos_policies_default, project_id\n* quotausages, resource\n* quotausages, project_id\n* subnet_dns_publish_fixed_ips, subnet_id\n* segmenthostmappings, segment_id\n* segmenthostmappings, host\n* networkdnsdomains, network_id\n* floatingipdnses, floatingip_id\n\nCloses-Bug: #2024044\nChange-Id: I271c109a597eb0aa088a7a9c785e8631bfaa01d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ac4c7a54cdf09a18a6f2f2897285fd3b7622c7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ac4c7a54cdf09a18a6f2f2897285fd3b7622c7d"}]},"branch":"refs/heads/master"},"78005ee235dbe4b55baa9b94af155c824f1818ef":{"kind":"REWORK","_number":5,"created":"2023-06-23 11:06:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/886213/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/886213/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/886213/5"}}},"commit":{"parents":[{"commit":"cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a","subject":"Merge \"[OVN] Improve Hash Ring logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-11 04:22:38.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-06-23 11:06:36.000000000","tz":0},"subject":"[sqlalchemy-20] Remove redundant indexes from some tables","message":"[sqlalchemy-20] Remove redundant indexes from some tables\n\nThe following tables and columns were defined as primary key and key\n(index). A column defined as primary key creates an index in the table.\nThe is no need to create a second one.\n\nTables and columns affected:\n* portdataplanestatuses, port_id\n* portdnses, port_id\n* portuplinkstatuspropagation, port_id\n* qos_policies_default, project_id\n* quotausages, resource\n* quotausages, project_id\n* subnet_dns_publish_fixed_ips, subnet_id\n* segmenthostmappings, segment_id\n* segmenthostmappings, host\n* networkdnsdomains, network_id\n* floatingipdnses, floatingip_id\n\nCloses-Bug: #2024044\nChange-Id: I271c109a597eb0aa088a7a9c785e8631bfaa01d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78005ee235dbe4b55baa9b94af155c824f1818ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78005ee235dbe4b55baa9b94af155c824f1818ef"}]},"branch":"refs/heads/master"},"08fe84f4438293b7000d0bcc829a8c66762c7cbc":{"kind":"REWORK","_number":6,"created":"2023-06-23 11:24:23.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/13/886213/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/13/886213/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/13/886213/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/13/886213/6"}}},"commit":{"parents":[{"commit":"cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a","subject":"Merge \"[OVN] Improve Hash Ring logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-11 04:22:38.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-06-23 11:24:22.000000000","tz":0},"subject":"[sqlalchemy-20] Remove redundant indexes from some tables","message":"[sqlalchemy-20] Remove redundant indexes from some tables\n\nThe following tables and columns were defined as primary key and key\n(index). A column defined as primary key creates an index in the table.\nThe is no need to create a second one.\n\nTables and columns affected:\n* portdataplanestatuses, port_id\n* portdnses, port_id\n* portuplinkstatuspropagation, port_id\n* qos_policies_default, project_id\n* quotausages, resource\n* quotausages, project_id\n* subnet_dns_publish_fixed_ips, subnet_id\n* segmenthostmappings, segment_id\n* segmenthostmappings, host\n* networkdnsdomains, network_id\n* floatingipdnses, floatingip_id\n\nCloses-Bug: #2024044\nChange-Id: I271c109a597eb0aa088a7a9c785e8631bfaa01d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08fe84f4438293b7000d0bcc829a8c66762c7cbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/08fe84f4438293b7000d0bcc829a8c66762c7cbc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
