)]}'
{"id":"openstack%2Fneutron~886590","triplet_id":"openstack%2Fneutron~stable%2F2023.1~Ibb1285f61c172d2b6e281f41315415ce8b8526e7","project":"openstack/neutron","branch":"stable/2023.1","topic":"bug/2023171","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-06-26 08:26:28.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-06-26 08:26:28.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-06-23 16:46:21.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-06-26 06:57:33.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}},"hashtags":[],"change_id":"Ibb1285f61c172d2b6e281f41315415ce8b8526e7","subject":"Delete the \"Chassis_Private\" register when deleting an agent","status":"MERGED","created":"2023-06-21 10:08:39.000000000","updated":"2023-06-26 08:28:06.000000000","submitted":"2023-06-26 08:26:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886590-bug/2023171","meta_rev_id":"e70a68dc8494773f682631a6478bddb66776e350","_number":886590,"virtual_id_number":886590,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-26 08:26:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-06-23 16:46:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-06-26 06:57:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-06-26 06:57:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-21 10:11:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-21 16:48:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-22 09:30:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-06-22 09:30:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-06-22 09:30:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"bfba482cb9e5bedfa291168a1ec383e5e338a803","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-21 10:08:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1acd5b52d13ee7b2bf4ee27c37110ec07bb759f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 10:11:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb23ca4dbad415ea556084224ae7cd9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d266d128821f41bf840406352e85de2b : NODE_FAILURE Node request 300-0021505570 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9904c30cfb2544939cedd80217c534fa : NODE_FAILURE Node request 300-0021505571 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"46035c04a1c20f46e38dc28f48934903e29a41d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 16:48:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1ee5edb493244a3b9a27bc052b48008\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ee7ac44dbd7a4da49425e7fae3537c77 : SUCCESS in 1h 32m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f36ea7b3ce924310bdcbadca7cc8b5db : SUCCESS in 44m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/060e61b88f44472794aed8b44d90b3fe : SUCCESS in 11m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/977a7ab4a8be420892e685a060ef2de8 : SUCCESS in 26m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ef40771cd4545a290067f005e06834b : SUCCESS in 22m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7e04ac0df494a2986d0eae252bc9ef6 : SUCCESS in 10m 48s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/19649c7818944a47a33bfdb1878dfa9c : SUCCESS in 25m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e5bf399944e4c1691cbfaf2c858e09d : SUCCESS in 1h 16m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/115eb7a14a804e63931c192038728589 : SUCCESS in 1h 07m 23s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4e4e750de0b243258d7c56af548a52d9 : SUCCESS in 1h 22m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e2800b292eba43c69ac2657df6a1412b : TIMED_OUT in 2h 12m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/087191234e624fe0b5b4cf1d91ef7712 : SUCCESS in 1h 30m 54s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/3a3538f0e1524debbd743715bcb19ede : SUCCESS in 1h 20m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ed90d999ec1073dc3712a05a0ffe6800dde2a6b7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-22 08:01:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a0bc75743907eb2904ea3c15be16c555d4b7799","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 08:33:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c79556e49e148258784083d5260dc5e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c01b3d36e38c483e84fd12007ff3fa9a : SUCCESS in 21m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8b1fec268ae94ce3881c8fb50116c6d4 : SUCCESS in 28m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f31016793ccfe0f7aa7a1229819494212a6e2ee0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 11:15:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7201eb4b20a44e3db42c31ff1c6afaba\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c1bb047c120c4ae4a69c782dc6816331 : SUCCESS in 1h 36m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2901f6ad722d4358972629711d0326f3 : SUCCESS in 44m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cb4084c29224b3281184c83117f1290 : SUCCESS in 15m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1c480f582e8474caee566f62799ed7f : SUCCESS in 32m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60a1cc3c7300474c8c5fb2b61a86f96a : SUCCESS in 26m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5e683c07ce4435888563e5c4fa81227 : SUCCESS in 11m 24s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/2e659df549114fa58eb670572f572462 : SUCCESS in 20m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b527bcd3f0d8471799a498051b758f34 : SUCCESS in 53m 14s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6ac161e44c0444828711692544273a1c : SUCCESS in 1h 30m 19s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bce702bc70fe4f5b8e2c35e542fb59dc : SUCCESS in 1h 30m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4b569429fc5046c59a842258ed52addf : SUCCESS in 1h 42m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1eff393c318242e2a8ee186d3717cfa4 : SUCCESS in 1h 28m 37s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/300ca407a50645a69819ac7be6821afe : SUCCESS in 1h 07m 13s","accounts_in_message":[],"_revision_number":1},{"id":"25f9bc7f0e616c9205046f76bedfcccdf0938b82","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-06-23 16:46:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a63e35ee526ac428d4b270920f4db5c2cd719c27","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-06-26 06:57:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"50cde77eebe78f4e7f120ef182701c52c0efe80b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 06:58:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bff7461dda9242e8fbce2ed0ed706ae009f3f2a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 08:26:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96b5c1d3f7d844ceac28d3cf68a2fc69\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a9cdc00159924eca8cc12ed6f6385696 : SUCCESS in 1h 27m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45be798bd56545d8bccb7cabde8dee34 : SUCCESS in 13m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b27e605513542e8a6eae57759a01a6b : SUCCESS in 25m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/820ca572c3634b54a3170db51817c09a : SUCCESS in 23m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/488ee492005a485abe4a869ce615d5af : SUCCESS in 11m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/515b1e1c87e74692aa026d9ebb5321eb : SUCCESS in 1h 09m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/82c848cfa33a47fab15c6c1ff22eb995 : SUCCESS in 52m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c6336da430214b6babb400a1ec2ae541 : SUCCESS in 1h 20m 00s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/1a9b4fa9cf55416ab85c661cf90da38a : SUCCESS in 1h 18m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6f2cb6c221248a2d8258d5a881cba1269543b848","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 08:26:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e70a68dc8494773f682631a6478bddb66776e350","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 08:28:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2fa9f0febea4619b3ea324db8a70945\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf5b4c09e84d4ada805cc3d450fd9e96 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ddb08cd5a3c9221ae1e6e4a9224bc78d6066b086","revisions":{"ddb08cd5a3c9221ae1e6e4a9224bc78d6066b086":{"kind":"REWORK","_number":1,"created":"2023-06-21 10:08:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/90/886590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/90/886590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/886590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/886590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/90/886590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/90/886590/1"}}},"commit":{"parents":[{"commit":"137d313706f96e2254e4ed0e469ea32ac050cba6","subject":"Merge \"Make DB migration creating indexes in RBACs conditional\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/137d313706f96e2254e4ed0e469ea32ac050cba6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-09 11:52:38.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-20 17:16:09.000000000","tz":0},"subject":"Delete the \"Chassis_Private\" register when deleting an agent","message":"Delete the \"Chassis_Private\" register when deleting an agent\n\nThe method ``delete_agent`` [1] now deletes both the OVN SB\n\"Chassis\" and \"Chassis_Private\" registers when called.\n\n[1]https://github.com/openstack/neutron/blob/ad81687e2955f494c28e2122763a2332afd3d26d/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py#L1400-L1401\n\nCloses-Bug: #2023171\nChange-Id: Ibb1285f61c172d2b6e281f41315415ce8b8526e7\n(cherry picked from commit 9ca0e34a5edeb0419d86117dfaabb3f7842bb347)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ddb08cd5a3c9221ae1e6e4a9224bc78d6066b086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ddb08cd5a3c9221ae1e6e4a9224bc78d6066b086"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
