)]}'
{"id":"openstack%2Fneutron~887250","triplet_id":"openstack%2Fneutron~stable%2Fzed~Ib3c6f0dc01efd31430691e720ba23ccb4ede65fa","project":"openstack/neutron","branch":"stable/zed","topic":"bug/2016158","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-07-19 10:49:36.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-07-19 07:31:05.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}}},"hashtags":[],"change_id":"Ib3c6f0dc01efd31430691e720ba23ccb4ede65fa","subject":"[OVN] Remove SB \"Chassis\"/\"Chassis_Private\" duplicated registers","status":"MERGED","created":"2023-06-29 07:52:06.000000000","updated":"2023-07-19 10:50:59.000000000","submitted":"2023-07-19 10:49:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887250-bug/2016158","meta_rev_id":"3c3257ce898602c47efa32cbdf9cc61b0f1f14b9","_number":887250,"virtual_id_number":887250,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-19 10:49:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-06-29 14:52:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-07-19 07:31:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-07-19 07:31:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-29 07:56:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-29 09:22:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-29 14:52:36.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-06-30 08:19:41.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"6679828549ac493a82f75170e02285e50cdb43e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-06-29 07:52:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"100fd17ca2fcbb002d289de980f0f4418cfb9570","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 07:56:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99446f7c912f4cc485d47c1756c23411\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47c8804127d04f7a95bacd8fff091eae : NODE_FAILURE Node request 300-0021573605 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6b429fcb5e8464188ed018cf57565e6 : NODE_FAILURE Node request 300-0021573606 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63d1e27e7a0bed7037ffacca071cfb85b96a4f61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 09:22:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e18430b1c5447ec8841645092440e6c\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/b840535d0384477c994d15d6d9230256 : SUCCESS in 1h 27m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f70d56654c045ab99b4f7c6bcefe8bd : SUCCESS in 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d47d6f9c8ed74cebac65a9d75de1821e : SUCCESS in 10m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34ce263cfaa347f292e81ed1478b941f : SUCCESS in 25m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d0aa014df8d4a939d37c2f77ba82e51 : SUCCESS in 30m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1887b44a20b04089ba1800e28f2d761f : SUCCESS in 28m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c78ac55ec8ed4b0e85052374bb5ac3bc : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d3ee8e15ae242728eb222655b183bf8 : SUCCESS in 11m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/76547df19843439fa2cff9357c47a9e1 : SUCCESS in 40m 49s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/812a500369804801ad23286656b0dd09 : SUCCESS in 1h 18m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/38b4f747f3884bb9a5626292dbfa26db : SUCCESS in 50m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03fd3c530b8e7f81d97c64bfb25b5da5491cf662","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-06-29 14:52:36.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d9a9f92737b6f2b38301fd2e54fa6a67c01f177d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-07-19 07:31:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"711bd069c48d6b849acfd2c16bd2408a46163e16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 07:31:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c6b4c541129fb6e6b42f4911d83b159f538dde7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 10:49:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be56475eddb94a0c9be32cdb8d3f1f9f\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/980cd43aed08480b8fef4f3af49528ac : SUCCESS in 1h 52m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f7c5fada2cf4a489eccaf0a408cb78a : SUCCESS in 13m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa183671762f40d3b46f650f2afc9769 : SUCCESS in 26m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00fd1141832a48e38d7f2737fb3a5ae6 : SUCCESS in 26m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/570c9a5e03b64dc398cb9d4e958727ba : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7e79b480b3a46ca9321b3f7afc72015 : SUCCESS in 9m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/34e3d1f364fc465da0e7e034eaaddd7f : SUCCESS in 1h 00m 38s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49d3905b5e174945839977120c4c2cdc : SUCCESS in 1h 08m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c8b5f3958e02d34446615e724cb6e3b57a749a3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 10:49:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3c3257ce898602c47efa32cbdf9cc61b0f1f14b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 10:50:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7a18dbcf8a74904a0496c41612fab88\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43c4b64ccbbe47869d40a1b134051626 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8963f69734934fd68ec700ff6f29ab4a : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1547abb99fd4fb27b7c317a07db6f9bf2988b58","revisions":{"a1547abb99fd4fb27b7c317a07db6f9bf2988b58":{"kind":"REWORK","_number":1,"created":"2023-06-29 07:52:06.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/50/887250/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/887250/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/887250/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/887250/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/887250/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/887250/1"}}},"commit":{"parents":[{"commit":"d539460ab59e13e7ac9eced80e871b5055912dc3","subject":"Merge \"[S-RBAC] Fix policies for CUD subnets APIs\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d539460ab59e13e7ac9eced80e871b5055912dc3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-09 16:18:09.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-06-11 08:50:16.000000000","tz":0},"subject":"[OVN] Remove SB \"Chassis\"/\"Chassis_Private\" duplicated registers","message":"[OVN] Remove SB \"Chassis\"/\"Chassis_Private\" duplicated registers\n\nA new OVN maintenance method is added. This method lists all existing\nOVN SB Chassis registers and checks if any of them has the same\nhostname. In case of having duplicated \"Chassis\"/\"Chassis_Private\"\nregisters, the maintenance method will remove those with older\n(lower) timestamp, that is stored in\n\"Chassis_Private.nb_cfg_timestamp\", leaving only the newer one.\n\nConflicts:\n    neutron/tests/functional/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_maintenance.py\n\nCloses-Bug: #2016158\nChange-Id: Ib3c6f0dc01efd31430691e720ba23ccb4ede65fa\n(cherry picked from commit 9d9f47c20c6c269db264d656a1026e0926f390fd)\n(cherry picked from commit 7fc12decdf60eeb162fe87f7a67746184023135e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1547abb99fd4fb27b7c317a07db6f9bf2988b58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1547abb99fd4fb27b7c317a07db6f9bf2988b58"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
