)]}'
{"id":"openstack%2Fneutron~887464","triplet_id":"openstack%2Fneutron~stable%2F2023.1~I9e8ddfa26c5402fefd573b0e2ea5f3a57983ca35","project":"openstack/neutron","branch":"stable/2023.1","topic":"bug/2008712-stable/2023.1","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-07-06 13:56:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9e8ddfa26c5402fefd573b0e2ea5f3a57983ca35","subject":"Delete sg rule which remote is the deleted sg","status":"MERGED","created":"2023-07-04 12:23:36.000000000","updated":"2023-07-06 13:57:47.000000000","submitted":"2023-07-06 13:56:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887464-bug/2008712-stable/2023.1","cherry_pick_of_change":884505,"cherry_pick_of_patch_set":3,"meta_rev_id":"f6885dd38ab66815a33f89494597105b572e73b6","_number":887464,"virtual_id_number":887464,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-06 13:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-07-04 14:55:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-07-06 11:36:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-07-06 11:36:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-04 12:47:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-04 14:36:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-04 14:55:21.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-07-06 11:36:37.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"9bcfd7b9fff80da404a02f58983ec4d3a7c91909","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-07-04 12:23:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"94ec3cf0165640fab6e9d867343994381b34b0ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 12:47:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf3c3fa1b7e64438a75d239a3853e11d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a7b482cea824103948782377d7a0aa3 : SUCCESS in 19m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0c039e6e05fb45a98cbe8f3de4cacf3d : SUCCESS in 22m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"011a89f8d0d69edfe14deccbe5f9cbb17b7545b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 14:36:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20d5a09e0628421b85b72b705efafd15\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4ff5209a81154094a0c58c1f6b881bee : SUCCESS in 1h 20m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b44a9d0e7e2a4af790a4566e204749c9 : SUCCESS in 1h 34m 51s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a18afe41473249018e67d49cceadcc79 : SUCCESS in 1h 21m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/32665f9e1e53445ab1462077024b2bc2 : SUCCESS in 1h 52m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e60938fd2ddb41cc96c4154bbcf14cbe : SUCCESS in 36m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0154c0bb8bb4e1abd61aeee1c88f266 : SUCCESS in 47m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df4b9f72f0444b43ad45943f6974f28a : SUCCESS in 12m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef2c49020b8749e6a061d37db94b3f9c : SUCCESS in 33m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66475d71f16d4220aac279c8437ff68c : SUCCESS in 25m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/474ed6a861a24fec9efea15b4f4e81ff : SUCCESS in 11m 24s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/c45f6a2224174517a32a5d654c37ccca : SUCCESS in 22m 06s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/916bd3fb14b043fbbe6b6d4eee1603f5 : SUCCESS in 57m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9d41b5ec9b7043dcad0e40c1670a1a56 : SUCCESS in 2h 11m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5bc1e04170fb4b9db35bf491d4df9cd4 : SUCCESS in 56m 27s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9639c1203dd94428a9db56e54842aff3 : SUCCESS in 54m 48s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/477f5e31e0ab436693393ed4dc52d77a : SUCCESS in 1h 04m 03s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b48cae76185d4d0ebf79c1653f9502b6 : SUCCESS in 55m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8807ba6370d648fea66f4a7684fa2f21 : SUCCESS in 1h 26m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2bf270f3e6c74db783d1e67f4eb0ec4d : SUCCESS in 1h 36m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dba7a6f117f342eab3863f3fad5875b4 : SUCCESS in 1h 05m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/47488bc1717e42ac863984896d9cdff1 : SUCCESS in 1h 37m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c47a76953d3442ddb2f7db07c98c4eed : SUCCESS in 1h 43m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8a16f8b9d10d415abd71f1c2acd0fb55 : SUCCESS in 1h 25m 10s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/b86eabc0b63747febea748b1c829ba72 : SUCCESS in 1h 05m 06s","accounts_in_message":[],"_revision_number":1},{"id":"cb47d6891435f78bf818c6586c0132896dfb1d41","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-04 14:55:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a22d60101c3146e28e6c63e8ad847d654b9586c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-07-06 11:36:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76e90cda8da4c977c9dbc4cb99412f83cd07da0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 11:37:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33c2d31f5e85b8f46c25ab782c8b5c36f5a67d6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 13:56:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2e48dccac5e49fe9a83467d3e2c5136\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/aaf77eb4b2eb41a9b1e4edbfcebe83cb : SUCCESS in 1h 21m 38s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/daf9134212334c5892d0517bc0a4f8d9 : SUCCESS in 1h 42m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/67dd78f1d1cc4b4bab26c3669220717d : SUCCESS in 1h 32m 21s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a3404d8f65b942249d741e50155f1f1b : SUCCESS in 1h 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64416ff0f0fe482aa9c83d1ab0ee87ed : SUCCESS in 13m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05f711ef6a48401f8da68695b6145098 : SUCCESS in 26m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35bbd90d28c545aeb340c2a055597064 : SUCCESS in 22m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a657f3682f2e441a8cefb8cc2a160c38 : SUCCESS in 10m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/636742bef0614c7c9f82af3e6b4f8f74 : SUCCESS in 1h 00m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab81a0ba60fb47f5844ddcc6c13c0424 : SUCCESS in 2h 16m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9f49a81cc1224d079abd32bdb6460492 : SUCCESS in 1h 33m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/462a7444b7ce4cd1be7280e315c97d6a : SUCCESS in 1h 12m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ded464f3dcbb42e591059b2f509ea392 : SUCCESS in 1h 10m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c7f026c7cd75468b80cdfa976508b741 : SUCCESS in 1h 27m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/29da71e6d0484ebfabb7ba912d104338 : SUCCESS in 1h 33m 49s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/f06c41e9b7d24687a5e0fd6e6816e8f9 : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c1de111b52dc1be823e12d412ea0a19e646bb39b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 13:56:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f6885dd38ab66815a33f89494597105b572e73b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 13:57:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19d4ef3311084fd280b7db1d5aaaa5fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97f7bc78848d4a7184419dcd3dc3d338 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"287f4231a0b4b0e7b3fdaa0fcd415d6a4aeb2930","revisions":{"287f4231a0b4b0e7b3fdaa0fcd415d6a4aeb2930":{"kind":"REWORK","_number":1,"created":"2023-07-04 12:23:36.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/887464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/887464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/887464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/887464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/887464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/887464/1"}}},"commit":{"parents":[{"commit":"9060971b73b406fb0125254ee389878a2e5129e7","subject":"Merge \"Improve Hash Ring logs\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9060971b73b406fb0125254ee389878a2e5129e7"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-05-26 10:04:49.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2023-07-04 12:23:36.000000000","tz":0},"subject":"Delete sg rule which remote is the deleted sg","message":"Delete sg rule which remote is the deleted sg\n\nBased on bug #2008712 if we have a security-group which\nis the remote group of a 2nd security-group, the backend\nnever deletes the rule of the 2nd group which\nremote_group_id is the original security-group.\nBy AFTER_DELETE event for each rule that has the\nsecurity_group_id as remote_group_id, we can make the\nmech drivers do their work and delete these rules in the\nbackend.\n\nOne version of this fix was merged:\nhttps://review.opendev.org/q/I207ecf7954b06507e03cb16b502ceb6e2807e0e7\nand reverted due to #2019449:\nhttps://review.opendev.org/q/I077fe87435f61bd29d5c1efc979c2adebca26181\n\nThis patch is based on\nhttps://review.opendev.org/c/openstack/neutron/+/876716/1\n\nCloses-Bug: #2008712\nRelated-Bug: #2019449\nChange-Id: I9e8ddfa26c5402fefd573b0e2ea5f3a57983ca35\n(cherry picked from commit 67a0b0728788207cee27adb586880fabb8da6f25)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/287f4231a0b4b0e7b3fdaa0fcd415d6a4aeb2930"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/287f4231a0b4b0e7b3fdaa0fcd415d6a4aeb2930"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
