)]}'
{"id":"openstack%2Fneutron~887583","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I1fb6a9ff94376ab5f695d311b6fa8034d73cf556","project":"openstack/neutron","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-07-11 07:08:18.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-07-11 08:37:34.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-07-07 12:46:46.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}}},"hashtags":[],"change_id":"I1fb6a9ff94376ab5f695d311b6fa8034d73cf556","subject":"[OVN] Expose chassis hosting information in LSP","status":"MERGED","created":"2023-07-04 08:42:12.000000000","updated":"2023-07-11 08:39:58.000000000","submitted":"2023-07-11 08:37:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887583","meta_rev_id":"0ac838e1b31be8f9dca23162585a0cf6b8eca489","_number":887583,"virtual_id_number":887583,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-11 08:37:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-07-11 07:08:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-07-07 12:46:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-06 06:12:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-07-11 07:08:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-04 09:11:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-04 10:55:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-04 16:17:49.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-07-06 06:12:22.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2023-07-06 06:12:22.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-07-06 06:12:22.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"54c95b1abdd4336904020cbcc993d4e40471d7f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-07-04 08:42:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9579880332a79a0b7623f399feb90491840f0980","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 09:11:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e0e912b8915425989ef07f2cd067661\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fe3ddbe11fc468a8351dd74338f7445 : SUCCESS in 27m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05c5a1cc8fb24d12a7f9a65cdd2cdf10 : SUCCESS in 24m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"337ead10236ac62375e67f278ce2a5d6a705bef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 10:55:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/137d46203b7e46059eb74a0e69bc8539\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/2d20fa64c40e4fa39500278f86d45bef : SUCCESS in 44m 27s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/5a6ca8f817384b43b1d5697dd4085c0d : FAILURE in 11m 28s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/c70b0417e6b64d4f83525575dd2aa13a : SUCCESS in 55m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f1fe60f770d4d3bad8a737be34b0cc4 : SUCCESS in 44m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c059114e7ba440bfabe65cc73d2f33dd : SUCCESS in 10m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d718aea7ef24d69906fba5261c6105f : SUCCESS in 38m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49a92857d3c44d408874da04eb1e546a : SUCCESS in 25m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2f78d6198524213a56cfb49ab16a665 : SUCCESS in 26m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3111551c5c354e21b7fe25636ed56d16 : SUCCESS in 9m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/82e1934ba32348a5b1a2a7df81944e88 : SUCCESS in 55m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3722e1aebec84dc080d232c1e1d0a1d1 : SUCCESS in 1h 11m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4c7a4ff0b3874c2cb09d37efb2448a7d : SUCCESS in 51m 02s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ec9485c6b4624b74adbb7a99fdd1c2d8 : SUCCESS in 40m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9821dc29a7a2464dac4e7efc45bc2aab : SUCCESS in 1h 23m 51s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/77a301d4113c415b882eed489d21010a : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":1},{"id":"45531b6b3934535d386d70c178158abc48b4b694","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-04 16:17:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"daca1d617495e9c8c2334089a4e6aa353cda28aa","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-07-05 10:49:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3ae13a391565d1382a08c3ef47c0d9630c3f03c","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-07-05 14:46:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"747b079863099b5f4d4ca8ac1dd028d87d94c5cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 15:33:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c58e1da19e7b4c87a62350fc794269e3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/838a2893ec9c40b29aadc36ae6c09de3 : SUCCESS in 23m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47c8174163e547d78c5a0832159b5e40 : SUCCESS in 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00ae302d83bbf2e70731991a7e8b21fbcf2ef3ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 17:08:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/394ce1646bf54bed9c9a3b657864d40d\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/4cbecd0e3ec246e1a51bcbeaff777581 : SUCCESS in 45m 00s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/5980af148d4e422c88e37a470e016173 : FAILURE in 18m 33s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/0c66045ddf6a4d75bf9e14a5e28c6d11 : SUCCESS in 28m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/490d40c5c8dc41bd991da4b95d70180b : SUCCESS in 35m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05a85e2c24094641bc719d7428a2a853 : SUCCESS in 10m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ca595c25c004df3bd49c4dad83f6c21 : SUCCESS in 33m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdaa47bafe8f4fc5ad78e5029d6aa755 : SUCCESS in 30m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0261e25ea4854f4b9dbf6e7dcc5ea034 : SUCCESS in 26m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15d38ba0246042e5ae9bb4ae0c09d567 : SUCCESS in 8m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d38b533a9cb74f49b63ae3b974718815 : SUCCESS in 40m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/38c86d2346e34f2699653d1318022a1c : SUCCESS in 1h 09m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ae7f2c38426e4076acc6a53086058dc4 : SUCCESS in 1h 11m 18s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/bccc322a12c04ccbaafab7bb582e417b : SUCCESS in 24m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a78cf528943d498eaddd82fe6b5206c5 : TIMED_OUT in 2h 16m 27s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ae9301b98ace4859a9acdead7dc49ba2 : SUCCESS in 1h 08m 37s","accounts_in_message":[],"_revision_number":1},{"id":"84e32a346828254de433a26048fdac6cc98ec635","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-07-06 06:12:22.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"86140d31669dc8db366ee391485a7cf59e169bb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 06:36:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01a75e4306f7497e9562ac5babd10d01\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab9b5832b2e44cabb021994a677ad892 : SUCCESS in 18m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22153bb6ebe34ee18113a12a5a86f2eb : SUCCESS in 22m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb462f2938ef32dba07a0e574a2a402b32f48c2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 07:40:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/006d9fb05b6d465489fd84f5c840a846\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/e9bee1c5fad444bca9c517b84923db87 : SUCCESS in 41m 51s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/f13b220bbcc449ab966540faa0ef69e2 : FAILURE in 17m 11s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/9bb55602f7c84bff9d13d1ef85de0754 : SUCCESS in 30m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/644d0f29348c49359caaeece960ae6a6 : SUCCESS in 27m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebeb02ea421840548a07f1606cd49ed3 : SUCCESS in 13m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e77e7c7afec948e099f22da795d00f97 : SUCCESS in 23m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0661cdd78e2f44898092fdbc153a326f : SUCCESS in 20m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6637cdab0a9948d38d43a5f8346d99af : SUCCESS in 27m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7b232e0a6a44aed8690be34717be6f5 : SUCCESS in 8m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1faed322f4c34c638ea6d51b7cc53e8d : FAILURE in 42m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f9cbf72f1f947669d3ec18f0a28a35a : SUCCESS in 1h 04m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ce2b668186e4414888da968fa3d41577 : SUCCESS in 54m 09s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3602eeadf26e432ba9df8060b0cb8e0b : SUCCESS in 24m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f5e31c165bdb4a85bc98ee171c61c8ea : SUCCESS in 1h 26m 13s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8fc82acf57474e0997e4ba9878e04f74 : SUCCESS in 1h 18m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b24794ec30fc3ff63d6a4e83bef90dfed48b9723","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-07-07 12:46:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac6bd16227dd1d8bbe04fae634d0895c3fe7a44c","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-07-10 13:22:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3c9ab4a8e91ba0bcb0c35f733089e166c1e4b60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 13:47:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/285c85f8094c48cf9778713deac32bd9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/06cc4beab04049e39d444f5ab7271e1f : NODE_FAILURE Node request 300-0021653156 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/926892af583c46838c7840bd08d030de : SUCCESS in 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd0f69993f08a7d9d5ba2355313820e21dee6941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 15:44:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb1620d8d88147cebc24af04eddd84f0\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/ed8918d7edd64d3083274b0fd647c929 : SUCCESS in 48m 27s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/72a1d85081654b55b40a04bd3884b5e3 : SUCCESS in 1h 46m 15s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/b7df76a7333747929be5bea178d6c7c1 : SUCCESS in 30m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14a39e1a561648ff85a6ffc7d00a210a : SUCCESS in 38m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b1bd759b98b4586824372dc96b49427 : SUCCESS in 13m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1306ef207e9455abf9095eacabd62a5 : SUCCESS in 26m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3867b0e976174965a8ff40cc1b2591ac : SUCCESS in 28m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/615d3163e042419ab5f7971a4faf66ad : FAILURE in 28m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56b0ca6666b94d76b58d0160e086b96a : SUCCESS in 12m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/762c8e7801124de9bad893bb4985ad5f : SUCCESS in 1h 00m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e6ab71a647af41f4a1bfa3b4e2cdafee : SUCCESS in 1h 35m 39s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3ff2b344d1954c9eb528013026d75b69 : SUCCESS in 1h 29m 47s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/79ebbdf094d8467dbe234953ed3a6273 : SUCCESS in 49m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2359d9bf7b4e48e98558ba0590e05055 : SUCCESS in 2h 12m 23s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5f6d3e811d1046239bf7be6f5a22d02c : SUCCESS in 1h 33m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c01a2e3b676f86cddbc7f6da28389aea4354a2e8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-11 07:08:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7bce927df39dea227d8e7e170c2f3e0634fff445","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 07:08:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef7ebf45e6faeba19a5ff97e65bb295ebaef9ab3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 08:37:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50178d80b81041aa9a1a81b214fce7e8\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/5261c12d74af459db67d7760ab474322 : SUCCESS in 48m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bbe6b5865cc48028f80478df509ca74 : SUCCESS in 13m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a3644333b614c11a1caaad8bf799693 : SUCCESS in 35m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5333b89ce3642c1964c0e702452e1f5 : SUCCESS in 32m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32e844e920194e558134976e4fc6a60c : SUCCESS in 9m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf319f652c8f4325942e503eb085e161 : SUCCESS in 57m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/14e5ff3608db41f78b684d4409d4b07b : SUCCESS in 1h 20m 06s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/853acdea5eb64922bf7306b7b6d1ae1c : SUCCESS in 1h 06m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d1fe43df85584ce8a3e3c7b56526bac7 : SUCCESS in 30m 46s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1a88e80c798d4b05a7065998256d2d6e : SUCCESS in 58m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fe2f7bf0b61258042ca4c19ef08508788c20ed5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 08:37:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0ac838e1b31be8f9dca23162585a0cf6b8eca489","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 08:39:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f93da7c4a0b94f869f0290fa88c27794\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a219764a53a43c985137263cabcf778 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"384fe4bb8e436bd5b7598697f2f62c3032d825ec","revisions":{"384fe4bb8e436bd5b7598697f2f62c3032d825ec":{"kind":"REWORK","_number":1,"created":"2023-07-04 08:42:12.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/83/887583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/887583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/887583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/887583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/887583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/887583/1"}}},"commit":{"parents":[{"commit":"a2341136694e379d140f2d68bf7d333b6fb3802f","subject":"[OVN][Trunk] Add port binding info on subport when parent is bound","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2341136694e379d140f2d68bf7d333b6fb3802f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-05-09 12:36:54.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-07-04 08:40:58.000000000","tz":60},"subject":"[OVN] Expose chassis hosting information in LSP","message":"[OVN] Expose chassis hosting information in LSP\n\nExpose chassis hosting information via LSP\u0027s external_ids. This allows\nfor projects such as ovn-bgp-agent NB driver that only connects to the\nNB database to consume such information (avoiding more connections to\nthe SB database).\n\nAlso stop populating the \"requested-chassis\" options for ports of the\ntype virtual because it\u0027s ignored by OVN.\n\nSee LP #2020058 for more information.\n\nConflicts:\n  neutron/common/ovn/constants.py\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n  neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_client.py\n\nCloses-Bug: #2020058\nChange-Id: I1fb6a9ff94376ab5f695d311b6fa8034d73cf556\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 28926957d691c8b0897c71fee59a888783d15974)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/384fe4bb8e436bd5b7598697f2f62c3032d825ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/384fe4bb8e436bd5b7598697f2f62c3032d825ec"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
