)]}'
{"id":"openstack%2Fneutron~887765","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Ie09d9d3e5efa8a57ad11655c2eb31d2604bab326","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/2025202","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-07-17 14:41:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie09d9d3e5efa8a57ad11655c2eb31d2604bab326","subject":"Set result when lswitch port exist","status":"MERGED","created":"2023-07-06 07:49:53.000000000","updated":"2023-07-17 14:43:04.000000000","submitted":"2023-07-17 14:41:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887765-bug/2025202","meta_rev_id":"0c089a5c55c0f00e2e020d799daf85d4c37062ab","_number":887765,"virtual_id_number":887765,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-17 14:41:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-07-07 12:49:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-07-17 12:55:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-07-17 12:55:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-06 07:49:53.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"state":"CC"},{"updated":"2023-07-06 08:28:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-06 09:55:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-07 12:49:44.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-07-17 12:55:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"722a8fe3f04dc0a195b3e841c76dbb111d0c56b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-06 07:49:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30973f95c2407d82653a3d0934779c9902c58220","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 08:28:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1479e8a499ad42d7a72a765f01b0e684\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f3f3ca46f954d78a5da305769177f0b : NODE_FAILURE Node request 300-0021624697 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa9ce579f6b742c1ac6517a711cfb3c1 : SUCCESS in 37m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b33ebb5d84858406ba5c32872f2eacf646ed4d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 09:55:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94e7626f08ff4b9fbc537b2b5227a409\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/3668cc70aff04e648d082a169b5f17f3 : SUCCESS in 1h 11m 31s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/45e3423c88cb47efaa2805f7033361e1 : FAILURE in 23m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/511dadcb27a3446c8dd9516db9972766 : SUCCESS in 37m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/453764483edc4e30962fab94c6630474 : SUCCESS in 13m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce7856656b294e8c8e7b1c39376709be : SUCCESS in 35m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd0f777c46f1470e90f8c85179383a1f : SUCCESS in 31m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67bb33a7caa94429ab7bf416c2fa1bc5 : SUCCESS in 35m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440f5d58a56b455b8b002664913e84d5 : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1378d6ff94a849e3a04af88551f317c2 : SUCCESS in 1h 01m 24s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/041e0c6e34a3450c8c5557bdb364375b : SUCCESS in 1h 22m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/8afffe6cfaf445eeb4583717d6d463e3 : SUCCESS in 45m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4e31662230e24bcfabd06df6afd54b76 : SUCCESS in 1h 33m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/76d3cf03d6ef499088d1781815ddc35d : SUCCESS in 1h 25m 01s","accounts_in_message":[],"_revision_number":1},{"id":"1f39118a26b25f634bb752d2d1b48c00e96d9d7d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-07-07 12:49:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4c605615e9e2b5762347b3addd708e29e48299db","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-10 14:10:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"407b786d5485484f05828a87d65c3bc427852ae0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 14:38:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb3a3e8c294f4b14ad63853466207ee5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b802cd118a24c3a99308359158e56d4 : NODE_FAILURE Node request 300-0021653748 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/089891c097c844bab3053e6fea0ac24f : SUCCESS in 24m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0994f83d5c84dc5bb2877be24418510fd468908a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 16:26:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7297528cc52546db8db605f544b149a8\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/ebb658a0076440699421aaa1d603727c : SUCCESS in 46m 06s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/802a9c96fc3441adbb61b78ffa0c0ab9 : SUCCESS in 1h 51m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9883cf1c724041b99cfed1b7e938f3de : SUCCESS in 45m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa5b5b2b95b449838e047d8a7fddbd41 : SUCCESS in 15m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6d0ff3c2e904301acaa9a1a901569b4 : SUCCESS in 30m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a13dd062bd314795b39bb2c821fe694f : SUCCESS in 32m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1547b736bf01451bb73244e4a7dfa898 : SUCCESS in 29m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06c855fa06fa425396b5bd22c3fb1fba : SUCCESS in 11m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/818c6a3fbe7a4ba8b10b8ee6eede07c0 : FAILURE in 50m 42s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/7aaafc9853354778aab4068a6beb25b6 : SUCCESS in 48m 30s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7a63c23c7c77429fab3fe3e2c753da76 : SUCCESS in 29m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a1124f15ba234eb2b05a1f5e81462ba3 : SUCCESS in 1h 32m 15s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/22cc27f4235841279973342d68fe0074 : SUCCESS in 1h 07m 23s","accounts_in_message":[],"_revision_number":1},{"id":"92ced30dd4e49a22d84412841c2b12e9940cd7bd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-11 07:07:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7364a37a113b540ea9683ac42aba7b3418c3430","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 07:48:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4077b808c2f4495a82fa2580d970ee26\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/43328c9a923043ce9059158bb3cc813c : SUCCESS in 25m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e5a3c1cefbf41879b314a29cfbc8bcc : SUCCESS in 40m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5fc0d45f79502d75dfa7b447dc1adb4b1fc38f9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 09:09:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/466371cfc7864509b595b928fc6646c2\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/0238db58080a4730afbf015041f82d6c : SUCCESS in 1h 00m 49s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/a19395efbe1d472699c0c32c1ae85230 : SUCCESS in 2h 00m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f535e743d0214543925274315a94d6ee : SUCCESS in 37m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa838b49eb594bd18b143db03d8864cd : SUCCESS in 13m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e2ebc2fc87f4134aae2ecb0184c2932 : SUCCESS in 26m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8213d0969c9246c6b3b8086e32464850 : SUCCESS in 25m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f88ac0b1ff1d4e619a4e23daaf91f22f : SUCCESS in 26m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be41f75072e14d32b38a39889c60e4fe : SUCCESS in 9m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e7147f78515426cb71547d772e7c622 : SUCCESS in 54m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cf1cc9983b044193853a243683316030 : SUCCESS in 50m 17s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/a841e8664e2d4a42a258971d93c757b6 : SUCCESS in 42m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d93b6f0912204819a673269c113afe42 : SUCCESS in 1h 32m 47s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/231bb245e8134cea889efe48c7b40074 : SUCCESS in 1h 25m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e3712ea201bca9aeb886ae3cd3e71ff5f26d6792","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-07-17 12:55:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ecd40dd261ac1bf83eaa31e891230b45b0a18c0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 12:56:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b753c605be11c03083e3a187d36edfcad1589131","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 14:41:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c18d51aeb3b422abf3d2f7691ced71e\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/9fca8164730e4b79b958eeef7d054188 : SUCCESS in 44m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/944d3716bc5040978182fbe98aa4d867 : SUCCESS in 13m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a391a22608b48b19ba69cb401dfae67 : SUCCESS in 32m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40ce83325fd24c76909cd5a98dec4c73 : SUCCESS in 27m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbf46f9968af4ab899267e42c559f0ca : SUCCESS in 11m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5e935a2b5c743238e3085bd3f43d6bb : SUCCESS in 1h 01m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6fd37917060448248b49e6db460ad570 : SUCCESS in 1h 13m 43s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/41d41929223a4a88b5fed5b7dbecf24f : SUCCESS in 24m 50s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5f9f7f8565b64ecb91ec914f0b8c9591 : SUCCESS in 1h 40m 09s","accounts_in_message":[],"_revision_number":1},{"id":"2cdbe1e3e727a35b2f6e58f3a54b5ae69eddf11e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 14:41:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0c089a5c55c0f00e2e020d799daf85d4c37062ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 14:43:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b09ac27bdbf48729d3c901bf3e79ac3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d02533a99a5845caad54d12849b67385 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c103cfa25c98d6b7cdbb81bcfbd8937396987d9","revisions":{"4c103cfa25c98d6b7cdbb81bcfbd8937396987d9":{"kind":"REWORK","_number":1,"created":"2023-07-06 07:49:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/65/887765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/887765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/887765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/887765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/887765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/887765/1"}}},"commit":{"parents":[{"commit":"a2341136694e379d140f2d68bf7d333b6fb3802f","subject":"[OVN][Trunk] Add port binding info on subport when parent is bound","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a2341136694e379d140f2d68bf7d333b6fb3802f"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2023-04-28 02:00:53.000000000","tz":480},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-06 07:49:46.000000000","tz":0},"subject":"Set result when lswitch port exist","message":"Set result when lswitch port exist\n\nA TypeError was thrown during a synchronization\ncommand(neutron-ovn-db-sync-util) execution. From the code[1][2], it\ncan be seen. The result of the AddLSwitchPortCommand command will be\npassed as a parameter to the UpdateLSwitchPortQosOptionsCommand. But\nif the logical switch port exists, the result will not be set. Therefore,\nwhen the UpdateLSwitchPortQosOptionsCommand is executed, the port_id\nwill not be obtained, thereby throwing an exception TypeError.\nThis patch sets the result when the logical switch port exists.\n\n[1] https://opendev.org/openstack/neutron/src/commit/b71f7ceb3e97e021cb9aeda757a7ffdeeff80e8e/neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py#L488\n[2] https://opendev.org/openstack/neutron/src/commit/b71f7ceb3e97e021cb9aeda757a7ffdeeff80e8e/neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py#L505\n\nCloses-Bug: #2025202\n\nChange-Id: Ie09d9d3e5efa8a57ad11655c2eb31d2604bab326\n(cherry picked from commit 65bbbcee76fc2a7501d4c28e3cc716f1f9a6f763)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c103cfa25c98d6b7cdbb81bcfbd8937396987d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c103cfa25c98d6b7cdbb81bcfbd8937396987d9"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
