)]}'
{"id":"openstack%2Fneutron~887965","triplet_id":"openstack%2Fneutron~master~Ibdc0fff281b144161bbf22963f2b73c8a6ff0c9b","project":"openstack/neutron","branch":"master","topic":"bug/2026547","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-09-29 13:09:52.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-09-29 09:25:14.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-09-24 23:57:12.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ibdc0fff281b144161bbf22963f2b73c8a6ff0c9b","subject":"Drop ``AddressScope.shared`` field","status":"MERGED","created":"2023-07-07 14:48:36.000000000","updated":"2026-10-02 15:17:57.000000000","submitted":"2026-09-29 17:31:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887965-bug/2026547","meta_rev_id":"1c7d05a5bff0153b766294a824cd1102a9ec8d8c","_number":887965,"virtual_id_number":887965,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-09-24 09:43:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-29 17:31:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-29 09:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-09-24 23:57:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-29 09:25:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-07 15:11:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-07 17:22:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-07 18:28:42.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-16 14:53:42.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2026-09-18 11:32:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-09-18 11:32:52.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-09-18 13:31:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-09-28 10:42:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"93f91734127e88b0ae996a09882f0b02d138f4aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-07 14:48:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d0d4089abccc960240e6a8f34f40402d5ac6a56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 15:11:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ae52fea6e25451dbce0ee35332c2258\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39868a3d6efd463086813b4091acd479 : FAILURE in 20m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d603ff143ecd4a13b2f3b4b014c9d709 : FAILURE in 20m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d44821e2a20d4d518d2b4768360716ba : FAILURE in 20m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b8b66a47593e2632dbf1fe2b90b3fc785694b4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 17:22:47.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/458b91d612324ac0ad2060e73b4e04b2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/703f3cfd4e39485a93e0a525e10a774a : FAILURE in 1h 23m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5973819ca29548d29a253fb6387d8fcd : FAILURE in 1h 35m 27s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/909f76f79c3549d0be88d3468c3ec423 : FAILURE in 1h 25m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c6da83f7be324f81b993daf137e3b635 : FAILURE in 1h 20m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ee9630b2b59c4a3184dfb5d3133aaf73 : SUCCESS in 50m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e04bd50ddb634b75879a628713a5e54c : FAILURE in 37m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51d55687ba624914af33bf11cd644698 : FAILURE in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf232fc0a57c43d5b94c03e322e9dd8c : FAILURE in 29m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bae9efffe27149fa970dbf413e4b4ae9 : FAILURE in 34m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca9efe7522414b759ce545651cd9dccc : FAILURE in 26m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/763d69e2a95644b1a28587da18709fb2 : SUCCESS in 14m 46s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/8ce52b577b174849808dc1cb48911f83 : FAILURE in 26m 42s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/52c928e170264ffdaf1b44b2db42f9c3 : FAILURE in 1h 17m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fd6832e34c1441e383a12950e7a4c674 : TIMED_OUT in 2h 32m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/36fce5fb4a154b8fb8feec06bee5f215 : SUCCESS in 1h 08m 47s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a42788815fb14038a517754898735321 : SUCCESS in 59m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/81bc3c8a135b4a099bebc8c1a5bf57e5 : SUCCESS in 1h 37m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9a10d4587bcc42b783fd39eae8be065e : SUCCESS in 1h 52m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/91a7166da2e54442a83e2118c4cbc946 : SUCCESS in 1h 03m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0ee73271659a4c24be034123568dc75c : SUCCESS in 2h 08m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0a91ff53ab0d4862ad80b3d4bd8812b1 : SUCCESS in 1h 25m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fb8e506c9d1c4ccb8aa76c0ad6434817 : SUCCESS in 1h 32m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9e6a846b16b135c7b787aa6e261f9c495fbc98dd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-07-07 18:28:42.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2023/Arista_Tempest_Testrefs/changes/65/887965/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"44f5923167c666c7343bd7543a02b141a3fe1c93","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-13 10:49:52.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"e56d45148f0304fb54a0503352d510fda6d4a74d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-16 14:53:42.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1},{"id":"ff85a28b22cdb313f28c4e90283281b8fa2d44a4","tag":"autogenerated:gerrit:restore","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-15 09:20:57.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"021676883aa16f36c7c6244376d0f1dc2471da68","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-15 09:21:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"9c888223f5a419d4626445b769336529ece11f7a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-15 09:22:31.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"fa8b97763eecacd6f6240829afb48e669ca5e93e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-15 09:57:45.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/65/887965/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"b17be725c04eef91f6fee21b6f0823f925ce61d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 10:10:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7af1cba6572426d80a22fdc3c104d92\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ec9d8ad713a843b197e834d116ddef2a : FAILURE in 22m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ed7dcd514ecf4b0c88a5ef586022e5d4 : FAILURE in 24m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d0d7f3fc83ef4c119b3c4a2981d75754 : FAILURE in 46m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3218c2af3a9e9d35afcc02e9cd2a0cc93644e81c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-15 10:42:10.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"9a6ace708aa1d6c7947045da7c1b0c6566be57da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 11:33:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87b654808f224e5ca613fbdf454db351\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e909b061c9284cd08f729d5be5afb3e4 : SUCCESS in 33m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b451ec76c23442e7beec46abc1aaad74 : SUCCESS in 20m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0f8efbf793af49a6985f520d5ed7bcd0 : SUCCESS in 49m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"92acf7398b90c3d3c8e31a475469be3caf74468b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-15 11:46:04.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/65/887965/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"72abb22a25bcfc281b3479211cf14a1d5a8fdf03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 12:36:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/20d102ed667041d5a0668e321a98d577\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4ff70cf574554112a8cf453d6566b428 : SUCCESS in 57m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfc2da776c274e74baa391844ce0d2f8 : SUCCESS in 1h 03m 38s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ed2338b5828a4fdcb17f5541847e009f : FAILURE in 1h 05m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/712a08ad43da439e9499300f5b259fe0 : FAILURE in 1h 50m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/02e990734ccf4e4e96f971d23c6a74b4 : SUCCESS in 31m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15c598be29e944b18a75eb3e660278cf : SUCCESS in 1h 04m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36252996a3e48b29f13931763a690dd : SUCCESS in 10m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ad768ce8428342d0998448c456254133 : SUCCESS in 30m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b0a6687efc74aa08f00f249be4f9f03 : SUCCESS in 23m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e48974e5b31a4d0b97304b75f38b081a : SUCCESS in 33m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f02b0224fbc6404cb65313b2980546e2 : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0454389145643a0b4147bc785832a16 : SUCCESS in 25m 58s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/5d88903c00e542048410eca895a27967 : SUCCESS in 1h 38m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/33887e7f2aad489880f3a4a094ce2930 : SUCCESS in 1h 18m 52s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4f3a7277e1648f69f1167235f67d6b2 : SUCCESS in 1h 16m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3691bf75859f437fac707b062e6613af : SUCCESS in 1h 24m 28s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/57e29aa5de904d1da63c06c608eacd06 : SUCCESS in 1h 35m 00s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/edbd5934723847ee8b57027b5ef42de8 : SUCCESS in 57m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c40d83c3c45d4a138b3b7b7162de7851 : SUCCESS in 46m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4dedca90db40454eb9fe2424c182dd47 : SUCCESS in 1h 20m 17s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":3},{"id":"b1b6e914075a1f59ee9cea2e63dfd77e18e81e24","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-18 09:37:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c9462d84ad5a8be188dc86195d0c7a06f39016ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 10:39:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/447cfcbadda5426d9a29303eb2206906\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/09f0849db80141ec82561906abceb307 : SUCCESS in 43m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/03cb0df3e9214470837a4b81399768f3 : SUCCESS in 38m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e8c6e88c28624353ba236c73520a48de : SUCCESS in 57m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aedd8b008c17c904bd812c86a7906c909d82073a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 11:30:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c15b605b907045cb8c1aa6dc24e9238c\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aecabfc4ad7c41dc9404e94c83c47f3c : SUCCESS in 1h 03m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdd8697963bb46e5b43ee67295c2dd02 : SUCCESS in 1h 04m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8687fe40d7864f0cb1df2138918350b1 : SUCCESS in 1h 04m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8459fe22caa14f019a5bc072fc70a65e : SUCCESS in 1h 49m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c29a490c53f64873b3f92ae5904fd7d0 : SUCCESS in 31m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17a90bc1d5cf4ad0a4cdb029e3096332 : SUCCESS in 1h 28m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ca7633e587e4d18a632581257e87e3c : SUCCESS in 10m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56ef833883d641e5b8ab5a4e2218fac6 : SUCCESS in 33m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/273d3b9f74b84c4d98dce22b996a4e55 : SUCCESS in 25m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/46b5ad46d8db42ceb157067c71351791 : SUCCESS in 34m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cf04ffad3644d4dbaa51568bfee3dd9 : SUCCESS in 12m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca60b669c8b34d6aae6fd1e9d7637d6c : SUCCESS in 21m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a0c7ac2f77ad4c2c8bc0f0c25c9a532a : SUCCESS in 1h 39m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef3a455a618d4a7bb4e69c1ec3989ec0 : SUCCESS in 1h 07m 14s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/38707bd18ad64c35839d88a2e0bdba24 : SUCCESS in 1h 38m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9cf64cc0355343529f5fd628dec993c4 : SUCCESS in 1h 35m 57s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d21c29a5ccdc4319995023d7e8f1f9eb : SUCCESS in 1h 13m 58s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cc500405204f40798ee0360a0b354138 : SUCCESS in 1h 01m 21s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0795c8de01d145f791c49382326cfc15 : SUCCESS in 1h 02m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/22a20f6ed8d4428a9db91dd603792c24 : SUCCESS in 1h 21m 32s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":3},{"id":"3aeceea68833eeb4f0de475efc2a681da2b2ff1a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-18 11:32:52.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"43480bd2f6bf9cc4a4b2838391921ba1e77e6169","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-18 13:31:43.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0dee00d7e18da0d975933fad375db58e737629d9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-18 14:23:45.000000000","message":"Patch Set 3:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/65/887965/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"ef9e64209b6c14913fee62c703d7755a2511ef13","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-24 07:42:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0de46a09ad8fcdb90d3be14825d161e5e6e063d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-24 09:04:11.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"af561b0e756bdf5833dbd9e5dcb3028ae1a4a700","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-24 09:43:23.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/65/887965/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"b9f1a8d4f7c5025da3a4bf63d6ee599e4bb3c1bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 09:52:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc2f9b9306a14b5c903c00e3dd7262ca\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/973e05d211fb4ca9a40115e1d03539e4 : SUCCESS in 22m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/120b2983e022406facfb02e2977767e9 : SUCCESS in 46m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/234c9e559e8b4aeab782a0e3e7b26fb5 : SUCCESS in 30m 08s","accounts_in_message":[],"_revision_number":4},{"id":"09907223cbf4b5dc0f2f393a8df813a542720287","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 10:46:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55be02f455834cf5be846dd9dee9865a\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d79179b082ea446285a17f1ffe303420 : SUCCESS in 1h 14m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93b19bc7f1a446b5b8aa33cfdbd18492 : SUCCESS in 1h 05m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ef3b501cd82c4935a32d5fce5debbc8f : SUCCESS in 1h 03m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2a27d926a5e345e2980c7184d2f7a40c : SUCCESS in 1h 40m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/841c1712fb5c487b89c56311e41db33c : SUCCESS in 30m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e703b75763f4a57b37ace914305937b : SUCCESS in 1h 03m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11ad9b07b6384399b09906f577bd7070 : SUCCESS in 8m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/654571ba048f48718f305ad0800d1814 : SUCCESS in 25m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37ae95a950cb442bb2e552b3818c9b43 : SUCCESS in 17m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f5e14a1541384c22a4c98a26d84b1724 : SUCCESS in 39m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcd86ca61d034f23ac76f8f40055681d : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff3ee18c5bf74c1ab4ad738400783119 : SUCCESS in 11m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/fe92ba28c7bb42119a6a844cf63383e7 : SUCCESS in 1h 02m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/07e331a4ee7c48079d8360bd41a9ef04 : SUCCESS in 1h 08m 48s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4eec94bb0944721891730edd7d57837 : SUCCESS in 53m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5b806ba0bf994207a7f3d7d98e7a7e2d : SUCCESS in 44m 01s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f235678590fd492e868cda0b05aa0db3 : SUCCESS in 1h 13m 00s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3dee8486a6be434d97584ca96bdaf1b4 : SUCCESS in 39m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/04cdfcfb07984baf91e0ca594dae5e9a : SUCCESS in 51m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/27541e3ab9804c6abc7ddced231ff74d : SUCCESS in 1h 26m 03s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":4},{"id":"89d5e6fd8398560455cda6a23b8f58ed6a17107a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-24 23:57:12.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bfcc9ba302d24f09ef46eb0203ac0cf6e6bf306b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-28 10:42:35.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6002ee963662854429b9f32e5223f9a4cce81377","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-28 11:17:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0d8795ff5b60f76d4c0f6a3e62bbaaf2690e5a0a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-29 09:25:14.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3d1f6bcdf8d115ca2ac4ad30aff4c6de07c3a01b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-29 09:25:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"738069a063aadf4a7a7a38ba37181ce24b9a59ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 09:25:59.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5151512f5a8de3437b2b554dfd62508dc40047d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 11:43:00.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6388f73efe4f4f039c66b7026f0bcf47\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/20c3492da855416c99e4c244d4c05ab3 : SUCCESS in 1h 18m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a8dfdecfd72405cb14dfcb8555f6b67 : SUCCESS in 1h 02m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fb6626b4fa164449b59c46eea8392451 : SUCCESS in 1h 05m 29s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4c18348e1c264ea3af40cd98505b7f9c : SUCCESS in 2h 13m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2143aa35d734e24bf963b58650c5d7f : SUCCESS in 11m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b020e88775641568cff2125c6722785 : SUCCESS in 29m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ee35994d102341038a9ecd78c95d18c8 : SUCCESS in 35m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2c9fae585ea4ee08189b09ca0d9477a : SUCCESS in 14m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0f3d71c389541fda3dafa8ce0fa2fb3 : SUCCESS in 24m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6cceae0fe82b49b98ddf15e1ea07bc3b : FAILURE in 20m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8dcc50933943490591538f2fdf16d8d5 : SUCCESS in 1h 30m 10s (non-voting)\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/85aca03268374058b63c2c34e024cc23 : SUCCESS in 1h 12m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f5ad7d2d33b84fcb8a0a0345027e922e : SUCCESS in 1h 13m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/32d0103991b14250afa3bb11d0ea4dde : SUCCESS in 59m 35s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":4},{"id":"ac423f1f245311d115a5541f443da5eef5565800","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-29 13:09:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1e936d7b3793f7a3b4958aad0d092522b0487002","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 13:58:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f7be90d5a3c4a9daf4be0f4f0b75667\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/847eda266140468493f0453333eb2b39 : SUCCESS in 28m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0019b2ebeab640ec8c52ed870070c32b : SUCCESS in 45m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d5fc31388d9064828e14fc7844b1542911535ba7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 15:31:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95682a8de6514948aa276ed889222a98\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6cca341adf404e54a249f48c96998da6 : SUCCESS in 1h 33m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cbf5b1d97654ebd9bb0425052057298 : SUCCESS in 1h 03m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6d56b69906c449afb2b7a2a347411e19 : SUCCESS in 57m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5edd22b19c2043818027054fd88d89a0 : SUCCESS in 2h 12m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b3f8030892334582b6ce7d4bd92f7e13 : SUCCESS in 30m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9be6f33bff8d442dabb4834ea65ad85d : SUCCESS in 1h 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a22a484e9d241158b3b49536a842116 : SUCCESS in 12m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67458fc3b78c4940a89ecc3ec1d35d50 : SUCCESS in 31m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/60fa4c14522f46448418465171f83bad : SUCCESS in 33m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1284b01fb1842238bce851fc0b8e052 : SUCCESS in 13m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4fa04df945b4773a68ba69be5cd41c3 : SUCCESS in 28m 11s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d087902925f848e8b924e64552380ba0 : SUCCESS in 1h 46m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/74b826ee99a64332bccc519de268a15d : SUCCESS in 1h 22m 23s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/d65675d6b9a949a5b87105d3d102b094 : SUCCESS in 1h 01m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/de3af2132ef84eb89ddaf1d23f8b85a4 : SUCCESS in 1h 25m 15s (non-voting)\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f218f68ea0e34f88a8431cad42fc628f : SUCCESS in 1h 34m 39s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/43edce3a4c86466aa3fa18393b8c12a9 : SUCCESS in 57m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b962851cb4d045debea6a43d2612c3ca : SUCCESS in 1h 45m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0dc2f0b9b10149049a8def8dd701bc12 : SUCCESS in 1h 20m 01s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":4},{"id":"b47be78c14bb9358879125dde9052491e764871b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 15:32:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dc18dc6449165b64d4174ec4a59b7e3c32b021a2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2026-09-29 15:59:18.000000000","message":"Patch Set 4:\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2026/Arista_Tempest_Testrefs/changes/65/887965/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"41c5e932a442b9150ff819d6f004b40e21fae90f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 17:31:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d75e977cfd5946beb7ac26ed216b6112\n\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9e048fead61844eeaadf4a14cc42b99e : SUCCESS in 47m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75873a8bd8da45fa96a8e2228cf3ee5f : SUCCESS in 1h 01m 13s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/056fa736de274eb6ba45658fca5a43d9 : SUCCESS in 1h 00m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5db6b8d591e44a0db4abf76cca8c24fa : SUCCESS in 1h 57m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/875b5ab02e4c40e7b160cc33fc6c69fd : SUCCESS in 9m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/315042e769a2465591f8f8c564049129 : SUCCESS in 21m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/474ef281b9224403a95b33fef8b2c4ed : SUCCESS in 26m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c11335df3447b8960bc4edf0e52920 : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee1848f82cb34ed1ae48d35a5ed738e3 : SUCCESS in 20m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/acaf07a9f5094b6cbbd27c9ac85e3834 : SUCCESS in 1h 08m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b9a4103a8b8a4aa9b378e0a4bf919607 : SUCCESS in 41m 49s (non-voting)\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/e965f2ad23e245ff85c6407106ef5fd2 : SUCCESS in 1h 14m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ba6f042f5a44ef9b7a289d9ea0091ad : SUCCESS in 1h 11m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5cd449b606ad4fd08d27e538c880ae02 : SUCCESS in 59m 00s\n\nWarning:\n  Comments left for invalid file neutron/agent/l3/router_info.py","accounts_in_message":[],"_revision_number":4},{"id":"6f39dbe4d3574f734be938db26c0526332fdcebe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 17:31:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"04eb389cb0b1d5371ddf4926d30eaf06eb02031e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-29 17:33:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdd93e38d2444bc081a6ea394ebdde53\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a161ffd7974929972545e5debc7d80 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4af95377b454e668439d54146dee01e : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":4},{"id":"1c7d05a5bff0153b766294a824cd1102a9ec8d8c","tag":"autogenerated:gerrit:revert","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-10-02 15:17:57.000000000","message":"Created a revert of this change as I0af7768ef6e0a6db7932a26d1ef62b506a88ae4c","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9ba99540ec86aa3ce26d794250a37418fecaebb5","revisions":{"1aae052f8a7dc982fde76fb76e11b0a9eb2094c6":{"kind":"REWORK","_number":1,"created":"2023-07-07 14:48:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/65/887965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/887965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/887965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/887965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/887965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/887965/1"}}},"commit":{"parents":[{"commit":"eee897323ca5d202e9d8f00861db42191aaf3cc7","subject":"Merge \"[OVN] Prevent binding a virtual type port\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eee897323ca5d202e9d8f00861db42191aaf3cc7"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-06 00:11:30.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-06 00:17:51.000000000","tz":0},"subject":"Drop ``AddressScope.shared`` field","message":"Drop ``AddressScope.shared`` field\n\nThe column is not removed from the DB definition (this will be addressed\nin a future RFE for all RBAC controlled OVOs), but the \"shared\" field\nis no longer required during the creation or update of an address scope.\n\nCloses-Bug: #2026547\nChange-Id: Ibdc0fff281b144161bbf22963f2b73c8a6ff0c9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1aae052f8a7dc982fde76fb76e11b0a9eb2094c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1aae052f8a7dc982fde76fb76e11b0a9eb2094c6"}]},"branch":"refs/heads/master"},"5132e2588e8977034a2a436aca746f213e5c2810":{"kind":"REWORK","_number":2,"created":"2026-09-15 09:21:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/65/887965/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/887965/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/887965/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/887965/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/887965/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/887965/2"}}},"commit":{"parents":[{"commit":"51659eecb7f61db65ea0fe46d792155445a752c9","subject":"Merge \"[OVN] Fix segment bridge reconcile healing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/51659eecb7f61db65ea0fe46d792155445a752c9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-15 09:20:35.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-15 09:20:44.000000000","tz":120},"subject":"Drop ``AddressScope.shared`` field","message":"Drop ``AddressScope.shared`` field\n\nThe column is not removed from the DB definition (this will be\naddressed in a future RFE for all RBAC controlled OVOs), but the\n``shared`` field is no longer accepted during the creation or\nupdate of an address scope. Use RBAC policies to share address\nscopes.\n\nCloses-Bug: #2026547\nAssisted-By: Composer 2.5\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ibdc0fff281b144161bbf22963f2b73c8a6ff0c9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5132e2588e8977034a2a436aca746f213e5c2810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5132e2588e8977034a2a436aca746f213e5c2810"}]},"branch":"refs/heads/master"},"551c062c4f097407dfca0b1b08cf4a58d05e8abf":{"kind":"REWORK","_number":3,"created":"2026-09-15 10:42:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/65/887965/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/887965/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/887965/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/887965/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/887965/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/887965/3"}}},"commit":{"parents":[{"commit":"51659eecb7f61db65ea0fe46d792155445a752c9","subject":"Merge \"[OVN] Fix segment bridge reconcile healing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/51659eecb7f61db65ea0fe46d792155445a752c9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-15 09:20:35.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-15 10:41:59.000000000","tz":120},"subject":"Drop ``AddressScope.shared`` field","message":"Drop ``AddressScope.shared`` field\n\nThe column is not removed from the DB definition (this will be\naddressed in a future RFE for all RBAC controlled OVOs), but the\n``shared`` field is no longer accepted during the creation or\nupdate of an address scope. Use RBAC policies to share address\nscopes.\n\nCloses-Bug: #2026547\nAssisted-By: Composer 2.5\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ibdc0fff281b144161bbf22963f2b73c8a6ff0c9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/551c062c4f097407dfca0b1b08cf4a58d05e8abf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/551c062c4f097407dfca0b1b08cf4a58d05e8abf"}]},"branch":"refs/heads/master"},"9ba99540ec86aa3ce26d794250a37418fecaebb5":{"kind":"REWORK","_number":4,"created":"2026-09-24 09:04:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/65/887965/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/887965/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/887965/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/887965/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/887965/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/887965/4"}}},"commit":{"parents":[{"commit":"e8d8dd051b289183e6b4c61894df054e17b892d3","subject":"Merge \"Fix metadata/OVN LB HM port MTU set to \"None\" for non-ML2 callers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8d8dd051b289183e6b4c61894df054e17b892d3"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-15 09:20:35.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2026-09-23 17:58:50.000000000","tz":120},"subject":"Drop ``AddressScope.shared`` field","message":"Drop ``AddressScope.shared`` field\n\nThe column is not removed from the DB definition (this will be\naddressed in a future RFE for all RBAC controlled OVOs), but the\n``shared`` field is no longer accepted during the creation or\nupdate of an address scope. Use RBAC policies to share address\nscopes.\n\nCloses-Bug: #2026547\nAssisted-By: Composer 2.5\nSigned-off-by: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nChange-Id: Ibdc0fff281b144161bbf22963f2b73c8a6ff0c9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ba99540ec86aa3ce26d794250a37418fecaebb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ba99540ec86aa3ce26d794250a37418fecaebb5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
