)]}'
{"id":"openstack%2Fneutron~888039","triplet_id":"openstack%2Fneutron~master~If0f62a369f2d39b0f457c2ee4ccb09c081020d00","project":"openstack/neutron","branch":"master","topic":"bug/2012332","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-08-16 14:53:54.000000000","reason":"Change was abandoned"},"28056":{"account":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"last_update":"2023-08-16 14:53:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If0f62a369f2d39b0f457c2ee4ccb09c081020d00","subject":"[UT] Enable by default \"enable_security_group\" and set a firewall","status":"ABANDONED","created":"2023-07-10 14:01:40.000000000","updated":"2023-08-16 14:53:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"52c5f25614669e9d0f21287c52cfb554b8799af9","_number":888039,"virtual_id_number":888039,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-07-11 15:53:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-10 14:31:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-10 14:43:20.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-10 15:00:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-11 07:56:03.000000000","updated_by":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"reviewer":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"state":"REVIEWER"},{"updated":"2023-08-16 14:53:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"58fa6be030e25fd1ead04f7b3915af347e9d44ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-10 14:01:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"893d87cc4a557039bc82f0a04ecdea7754bba6fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-10 14:03:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5c2b58eb8177b173d16616ed490b3e19994ef4f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 14:31:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdff32f5fb374addb8fc985598c75f61\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b12306a8dda24c98b4d4f27c8f3a28d2 : FAILURE in 25m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5fda85699aa949318b2bb1616315a232 : FAILURE in 26m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d0b28754778144b798b9ec3f434ea0f0 : FAILURE in 25m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"913b3edcf8d5b460023dca27a7fa30de92a35b99","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-07-10 14:43:20.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2023/Arista_Tempest_Testrefs/changes/39/888039/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"bd971ada2a1a09b3d26d89373a2ea6900f0168cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 15:00:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7ade48614a04b4ca8a4924e87436008\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bfbd7db58374b3abf6b7359ea9f3366 : FAILURE in 39m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0afebfe2845949d79a6f2b029edb158b : SUCCESS in 11m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/985869d413ac4d85960cc5ee0bafa70a : FAILURE in 26m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/323e303fad174c3cbc68a6c166d2698c : FAILURE in 32m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94403a194a3a4fada8bf3eeb90107aed : FAILURE in 33m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30346cfca8cb403cb68734caa5763c62 : SUCCESS in 8m 59s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/f1d60cab74964ebf8500e4dcf80aae85 : FAILURE in 22m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c939da0e539908627b5ffcfdf03446467ab6b509","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-07-10 15:24:26.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2023/Arista_Tempest_Testrefs/changes/39/888039/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"00d093d48e66ac7b71942f66a3519aa59c13c22b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-11 13:18:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"34c5536642da6ba865e7ef995144ef7d58f83612","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 13:45:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e441e826a8e44f089ff531b8f3953d26\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef910adf239c4ffd90a1bda143c1871d : SUCCESS in 25m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1769a90851c54ae49d8ebc562f753c97 : SUCCESS in 24m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/259f17e4190e46b291a4786d9a85a018 : SUCCESS in 20m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"80a0fb7785d184f7d9a74e80720b835ac076638f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-07-11 14:00:18.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2023/Arista_Tempest_Testrefs/changes/39/888039/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"231ab6dfb09174caebce550a487a82d6d465c1db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 15:53:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d363c08d8a8c46db8d280b37e71caf90\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/61784f8f1d094663aadf9bbde232ecf0 : FAILURE in 1h 48m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5cd075bc2b4e438e80743891c9cdce1b : FAILURE in 1h 27m 41s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/25f9f571908c42e78f87a8b0eeed63d9 : FAILURE in 1h 17m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a850a88e4e9d4378a13f3ad53daa8902 : SUCCESS in 1h 30m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e5496b0c9734091ab2dd5cf0635ff0e : SUCCESS in 29m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a31488d0a2240c8ab9c4f567600afe3 : SUCCESS in 10m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e377618415e349afa260fff0a8a63941 : SUCCESS in 25m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1ec7c53a4244cd5b1c394b6176d5d9b : SUCCESS in 26m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a794989bb9f94fe6912c7fbd1dfe7bc8 : SUCCESS in 30m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c4cc43241e74ac192cd7d17d6a83ddf : SUCCESS in 10m 18s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/222f59582c414ac6b49cb8bdc5127c90 : SUCCESS in 28m 38s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b1adbd3e38484b20bfaf71373819b471 : FAILURE in 58m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/522f5be1e77a4af5b07c3fe239d85b1f : TIMED_OUT in 2h 32m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec39272510d945d39df02e06d5fc8870 : FAILURE in 1h 04m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b03dadb140dc4925aa9ed3237f518f41 : FAILURE in 1h 10m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6dc2733166034b86b16b6b5ce76b71a3 : FAILURE in 1h 11m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/afd38e308dc640bd883e4dd3cf1dda42 : FAILURE in 1h 13m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2daf8619039f48d4bb0fa048e80bdd65 : FAILURE in 1h 05m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/827d2be52d0947ab9e284f2bb8ec0643 : SUCCESS in 1h 54m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b3a0ac7201e545d091c6b939689775ee : SUCCESS in 2h 13m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c1a4d553e8fb4a6c9776b0dd1d3301aa : FAILURE in 1h 26m 25s","accounts_in_message":[],"_revision_number":3},{"id":"52c5f25614669e9d0f21287c52cfb554b8799af9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-16 14:53:54.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"eb57d7a34cb1978802e646d9882e9bd520ab5b3d","revisions":{"e4aa817c4e6b7ef1f0d11081dd8742e59d376617":{"kind":"REWORK","_number":1,"created":"2023-07-10 14:01:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/888039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/888039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/888039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/888039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/888039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/888039/1"}}},"commit":{"parents":[{"commit":"29ea825519fa9d4444fca2a386a7a4c43064b2e6","subject":"Add allowed_address_pairs add/remove atomic operations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29ea825519fa9d4444fca2a386a7a4c43064b2e6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-06 08:27:57.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-06 08:36:02.000000000","tz":0},"subject":"[UT] Enable by default \"enable_security_group\" and set a firewall","message":"[UT] Enable by default \"enable_security_group\" and set a firewall\n\nThe base testing class ``Ml2PluginV2TestCase`` now enables by default\nthe config option \"enable_security_group\" and defines the firewall\n``neutron.agent.firewall.NoopFirewallDriver``.\n\nRelated-Bug: #2012332\nChange-Id: If0f62a369f2d39b0f457c2ee4ccb09c081020d00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4aa817c4e6b7ef1f0d11081dd8742e59d376617"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e4aa817c4e6b7ef1f0d11081dd8742e59d376617"}]},"branch":"refs/heads/master"},"850484fee3d80c21483c1d30b51ff84a43f2aac1":{"kind":"REWORK","_number":2,"created":"2023-07-10 14:03:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/888039/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/888039/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/888039/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/888039/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/888039/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/888039/2"}}},"commit":{"parents":[{"commit":"29ea825519fa9d4444fca2a386a7a4c43064b2e6","subject":"Add allowed_address_pairs add/remove atomic operations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29ea825519fa9d4444fca2a386a7a4c43064b2e6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-06 08:27:57.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-06 08:37:45.000000000","tz":0},"subject":"[UT] Enable by default \"enable_security_group\" and set a firewall","message":"[UT] Enable by default \"enable_security_group\" and set a firewall\n\nThe base testing class ``Ml2PluginV2TestCase`` now enables by default\nthe config option \"enable_security_group\" and defines the firewall\n``neutron.agent.firewall.NoopFirewallDriver``.\n\nRelated-Bug: #2012332\nChange-Id: If0f62a369f2d39b0f457c2ee4ccb09c081020d00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/850484fee3d80c21483c1d30b51ff84a43f2aac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/850484fee3d80c21483c1d30b51ff84a43f2aac1"}]},"branch":"refs/heads/master"},"eb57d7a34cb1978802e646d9882e9bd520ab5b3d":{"kind":"REWORK","_number":3,"created":"2023-07-11 13:18:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/888039/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/888039/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/888039/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/888039/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/888039/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/888039/3"}}},"commit":{"parents":[{"commit":"29ea825519fa9d4444fca2a386a7a4c43064b2e6","subject":"Add allowed_address_pairs add/remove atomic operations","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29ea825519fa9d4444fca2a386a7a4c43064b2e6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-06 08:27:57.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-06 14:16:39.000000000","tz":0},"subject":"[UT] Enable by default \"enable_security_group\" and set a firewall","message":"[UT] Enable by default \"enable_security_group\" and set a firewall\n\nThe base testing class ``Ml2PluginV2TestCase`` now enables by default\nthe config option \"enable_security_group\" and defines the firewall\n``neutron.agent.firewall.NoopFirewallDriver``.\n\nThis patch is also adding the \"allowed-address-pairs-atomic\" extension\nto the list of extensions to be disabled in the ML2Plugin when the\nfirewall is disabled, same as the required extension\n\"allowed-address-pairs\".\n\nRelated-Bug: #2012332\nChange-Id: If0f62a369f2d39b0f457c2ee4ccb09c081020d00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb57d7a34cb1978802e646d9882e9bd520ab5b3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/eb57d7a34cb1978802e646d9882e9bd520ab5b3d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
