)]}'
{"id":"openstack%2Fneutron~888643","triplet_id":"openstack%2Fneutron~stable%2F2023.1~Iac1472f7f896ea434deacb6d236ab469f4f6ed56","project":"openstack/neutron","branch":"stable/2023.1","topic":"db_sync_sgs-stable/2023.1","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-07-24 12:29:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iac1472f7f896ea434deacb6d236ab469f4f6ed56","subject":"Fix ACL sync when default sg group is created","status":"MERGED","created":"2023-07-17 16:31:17.000000000","updated":"2023-07-24 12:31:29.000000000","submitted":"2023-07-24 12:29:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888643-db_sync_sgs-stable/2023.1","cherry_pick_of_change":875989,"cherry_pick_of_patch_set":4,"meta_rev_id":"b55a5035a26b03067a23f540f0e63a88e6287aef","_number":888643,"virtual_id_number":888643,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-24 12:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-07-19 11:27:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-07-24 10:26:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-07-24 10:26:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-17 16:31:17.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"CC"},{"updated":"2023-07-17 16:52:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-17 18:41:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-19 11:27:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-07-24 10:26:37.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"48db0412982b4d2d9562962bc131b30a4f2c2f5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-07-17 16:31:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e26f6a85917dcfc3ed394e109a7d047fcab34e08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 16:52:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9385381d062a4edaa512e0cf8ac464ea\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b115760739dc4372bbd8662c4029b07f : SUCCESS in 20m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b9e70e5d46b545659308ca0d652e0341 : SUCCESS in 19m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3de43fec8a097f230b9c5af03ad92e3a5c02e4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 18:41:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d1489c9c4a6466995ae2fbf26c1e09a\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/193e634f15d7405a8d99143002bb1be9 : SUCCESS in 1h 21m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79430e002a6f4c46987d4fcef020c0a1 : SUCCESS in 41m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/740dd335d1614bf89fd4ea78eb751c02 : SUCCESS in 13m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd24ae94f33c4828b259e37efe4306f1 : SUCCESS in 19m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c23223a9a6342eba0c25b0e9c0f0070 : SUCCESS in 23m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aba5461596643e59c6b2413066abee7 : SUCCESS in 10m 51s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/8f2b194dc3d34b2d8074af768e6d3d2f : SUCCESS in 17m 11s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0f2336819b7b4a6d99eed7263dc7531d : SUCCESS in 43m 06s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/701745c87cc4402985f1234d65854c01 : SUCCESS in 1h 17m 23s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8ec82642b7d940ada0f26380ab3bb615 : SUCCESS in 1h 06m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ed49d13c41894c18865af4008a1a4ebb : SUCCESS in 1h 40m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dff8004302864463b3ad2f0452f0d7f5 : SUCCESS in 1h 07m 08s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/1011916c43234f77ac855e84f13dcc7e : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":1},{"id":"726afba9c0e5a5864eab1b044a1da6fb03520424","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-19 11:27:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df760ce06a1f287b4284efd1986ee00fbe8311ba","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-07-24 10:26:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87d6e7d13354d1a080c28f6e79cb5d326a0f9128","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 10:27:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"171481f9a226dea0911a841d56860cea07f3eb5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 12:29:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/349a901f737349c28582e3ef80592564\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/21369b02520f4d3294a8f780b007b7ca : SUCCESS in 2h 01m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0819f2abc96c4b5fa1f7c017315089ac : SUCCESS in 13m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dd07f89901c4aae9f01799ec75226ca : SUCCESS in 20m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/893197dce0f845018cbd6f12679f840b : SUCCESS in 22m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9602da6357aa4e4dbe6ffc9a454e6e99 : SUCCESS in 10m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c5ab80d581e403ba2007961a8d8dbb0 : SUCCESS in 55m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/56fe33e3e15444a1a9c8b0eb0c502a2a : SUCCESS in 57m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0331771814a14a3986c3c0ab7f9e582b : SUCCESS in 1h 09m 12s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/a13697f7abfb4e0097c22f67786ea416 : SUCCESS in 44m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b36e101ea2094af88dcb5532a8ae71d4510b8b28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 12:29:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b55a5035a26b03067a23f540f0e63a88e6287aef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 12:31:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61ecf6a0d6a74a3c97979d66c9f7016d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91df86866b904dc7819e87a06bc26d00 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1f5bac4a2ab4fb34d56a5c7ccc118c82e133d5d","revisions":{"c1f5bac4a2ab4fb34d56a5c7ccc118c82e133d5d":{"kind":"REWORK","_number":1,"created":"2023-07-17 16:31:17.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/43/888643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/43/888643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/888643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/888643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/888643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/43/888643/1"}}},"commit":{"parents":[{"commit":"657fa25dae56c70d11a8eb524b168244db774cb0","subject":"Merge \"[OVN] Hash Ring: Set nodes as offline upon exit\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/657fa25dae56c70d11a8eb524b168244db774cb0"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-01 22:32:50.000000000","tz":-360},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-07-17 16:31:17.000000000","tz":0},"subject":"Fix ACL sync when default sg group is created","message":"Fix ACL sync when default sg group is created\n\nPort group not being available in NB DB during ACL sync\nis bit of a corner case but possible during the ML2/OVS\nto ML2/OVN migration sync. It can also happen in ML2/OVN\nonly enviroment. See my detailed description of both\nscenarios in the linked Bug.\nThe easiest fix is to just retry ALL port groups sync\none more time if ACL sync cant find a port group row. This\nadditional resync is really quick.\n\nCloses-Bug: #2008943\nChange-Id: Iac1472f7f896ea434deacb6d236ab469f4f6ed56\n(cherry picked from commit 33cf2cdc83a8cee9ee075eb371f779c3d356cf48)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1f5bac4a2ab4fb34d56a5c7ccc118c82e133d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1f5bac4a2ab4fb34d56a5c7ccc118c82e133d5d"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
