)]}'
{"id":"openstack%2Fneutron~888652","triplet_id":"openstack%2Fneutron~stable%2Fxena~Iac1472f7f896ea434deacb6d236ab469f4f6ed56","project":"openstack/neutron","branch":"stable/xena","topic":"db_sync_sgs-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-07-25 16:19:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iac1472f7f896ea434deacb6d236ab469f4f6ed56","subject":"Fix ACL sync when default sg group is created","status":"MERGED","created":"2023-07-19 20:43:16.000000000","updated":"2023-07-25 16:22:06.000000000","submitted":"2023-07-25 16:19:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888652-db_sync_sgs-stable/yoga-stable/xena","cherry_pick_of_change":888645,"cherry_pick_of_patch_set":1,"meta_rev_id":"1a8dbf9e3beb8073acbe629ec4608667903aec08","_number":888652,"virtual_id_number":888652,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-25 16:19:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-07-25 13:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-07-24 10:27:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-07-25 13:53:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-19 21:02:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-19 22:37:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-24 10:27:18.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-07-25 13:53:04.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"a81aedb4d6f282266530bf1ab8e5b628e40e4f4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-07-19 20:43:16.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"bfc7c3b1a165042b8956201b7fe3e4d569857c84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 21:02:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4d7e2ed362345c5918c7f6811796536\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/652946e8437d4953b3d2f2b2775f315f : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/caebf163d3f64cd0b90cefc5b7946cb4 : SUCCESS in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8822e42ef7c0134c2af63c25d8c8cd4bb1d0b5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 22:37:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba3bca5a5a54429f85e624c051e9eebe\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/7dcbf38f497044d49f79226ea1e6be82 : SUCCESS in 57m 59s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/66643f3072ed4a6f917ce50afdc2e143 : SUCCESS in 1h 48m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/015cbe7311664cd9b873a268b2a62267 : SUCCESS in 35m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a758ed285a0476e94be1541677d417e : SUCCESS in 13m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b34deb2450ee4b978f491f156ee41e16 : SUCCESS in 25m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57565cc6a858437c88d41803bae640e4 : SUCCESS in 25m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/284bf42b114b421fa8592e4d59a3c1aa : SUCCESS in 23m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ab3f30c692414ab29ec235554155b6 : SUCCESS in 9m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/73fd002c082a4c56b742a85a9c8aa23f : SUCCESS in 42m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/326d42c277c842d3808aac41119333f9 : SUCCESS in 56m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/17c1c81d05fa4c6f8ed57f46101261a8 : SUCCESS in 1h 16m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4a65353f36d3ac5ffa5aca294443aa2eae4c0ed","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-07-24 10:27:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5953b13917f795d1444066457fa6d8456fd6db0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-07-24 10:27:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"488aad77d465a96fba888b7e20d44c1c28e895f9","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-07-25 13:53:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2ab293cd12605d6bb934785378e64386d8b1b068","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 13:53:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dac5b3d62ce38fc52c763119691be2ae3f3ec60f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 16:19:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/133247b89f81455dbe3f844dbe7e4d49\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/a2ddc3b17d0745a59cf89e30cac56aab : SUCCESS in 1h 27m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fda9b6d8d944f098aa6bf1dfa83cca1 : SUCCESS in 16m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29b9ab5e440b4c239ad741f6860db65f : SUCCESS in 28m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e9128b65dc242aba8d9fd35ccc629b6 : SUCCESS in 26m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9d04864964f4b44b4890ffa44fc6d95 : SUCCESS in 9m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/162a5da88f784c2b93683249d7540cc8 : SUCCESS in 1h 03m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/633440a8d99947c3aca636c915c7a5a5 : SUCCESS in 1h 04m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5a8a08d6e5f746e25cf33e664124ebfbb35452f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 16:19:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1a8dbf9e3beb8073acbe629ec4608667903aec08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 16:22:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b6c41809d4047b1b4a0aecc53217c05\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3a9dd4cce1642d29e8a3e872b848171 : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc130493d2038c93665664110c4cc3e1da24a001","revisions":{"fc130493d2038c93665664110c4cc3e1da24a001":{"kind":"REWORK","_number":1,"created":"2023-07-19 20:43:16.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/52/888652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/888652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/888652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/888652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/888652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/888652/1"}}},"commit":{"parents":[{"commit":"b89ab6f5837656a7f2bcd6e21f11a0a36c712490","subject":"Merge \"[OVN] Remove SB \"Chassis\"/\"Chassis_Private\" duplicated registers\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b89ab6f5837656a7f2bcd6e21f11a0a36c712490"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-03-01 22:32:50.000000000","tz":-360},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-07-19 20:43:16.000000000","tz":0},"subject":"Fix ACL sync when default sg group is created","message":"Fix ACL sync when default sg group is created\n\nPort group not being available in NB DB during ACL sync\nis bit of a corner case but possible during the ML2/OVS\nto ML2/OVN migration sync. It can also happen in ML2/OVN\nonly enviroment. See my detailed description of both\nscenarios in the linked Bug.\nThe easiest fix is to just retry ALL port groups sync\none more time if ACL sync cant find a port group row. This\nadditional resync is really quick.\n\nCloses-Bug: #2008943\nChange-Id: Iac1472f7f896ea434deacb6d236ab469f4f6ed56\n(cherry picked from commit 33cf2cdc83a8cee9ee075eb371f779c3d356cf48)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc130493d2038c93665664110c4cc3e1da24a001"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fc130493d2038c93665664110c4cc3e1da24a001"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
