)]}'
{"id":"openstack%2Fneutron~888945","triplet_id":"openstack%2Fneutron~master~If4b03e42c5eaedad926e3b8efb8f72ed6872c9b6","project":"openstack/neutron","branch":"master","topic":"db_sync_sgs","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-07-24 16:20:37.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-07-24 16:20:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4b03e42c5eaedad926e3b8efb8f72ed6872c9b6","subject":"Follow up on a small nit from patch 875989 [1]","status":"MERGED","created":"2023-07-19 17:24:51.000000000","updated":"2023-07-24 16:22:59.000000000","submitted":"2023-07-24 16:20:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"888945-db_sync_sgs","meta_rev_id":"0414bb82b549c780999f66fe59c847c9826fd5cd","_number":888945,"virtual_id_number":888945,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-24 16:20:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-07-20 07:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-07-24 13:01:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-07-24 13:01:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-19 17:25:44.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-07-19 17:54:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-19 18:52:22.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-19 19:49:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-20 07:43:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-07-24 13:01:50.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"c777e872d30750571616777250b4f0613bad80ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-07-19 17:24:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71e47e2bfdc8e0d36e4b29a477097d1fae27cdcf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-07-19 17:25:15.000000000","message":"Topic pg_resync_bool removed","accounts_in_message":[],"_revision_number":1},{"id":"4228b9f1ae46d47b7208dcb46ab31968a2adaf35","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-07-19 17:25:23.000000000","message":"Topic set to db_sync_sgs","accounts_in_message":[],"_revision_number":1},{"id":"efc748adaacf64560aa0f0ba17c8d71b592a442b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-07-19 17:25:57.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8f05fda7ef32e62c1930de80d6608a006363b6e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-07-19 17:31:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"28d5d9ea2280cc0c4ebea55e834f0aaa51167657","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 17:54:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9243fc12629f40b3a44756120c996e12\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/065653dc2b3442a9b49a8daa20db79fb : SUCCESS in 19m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0cfd444fa9c84990a4c8352498cc785a : SUCCESS in 20m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ae2457fa98a541aa8fba012eec75f0dd : SUCCESS in 19m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c25c743ec3abd50592b473215ca1e7ed444935ff","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-07-19 18:52:22.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2023/Arista_Tempest_Testrefs/changes/45/888945/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"f677363e881df7b6e08ed85fceabc60f29dd308f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-07-19 19:24:49.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2023/Arista_Tempest_Testrefs/changes/45/888945/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"7c1604eb143ad27b18b1c937d647f11097ee6229","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 19:49:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b77aafe2d8834bf58ad33a90901dc00c\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7e9bd873f2784b8282b2ba313035645b : SUCCESS in 2h 17m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a35da5f6bda0436e935fc190ed572305 : SUCCESS in 45m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38a0606acd4046b5a92f06ead9eb214c : SUCCESS in 11m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a94698f9d28419493d3e43ab1228654 : SUCCESS in 27m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8ce893038ac4e07a538416cfb3c7859 : SUCCESS in 29m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84aca373b4c34b7b9e7b8c057fda121e : SUCCESS in 30m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bf1945a4ae142f4b589903d9b9b05b2 : SUCCESS in 10m 40s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/b1583cad33ca4697bb1fd6b61b96c75a : SUCCESS in 25m 46s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/724a5fd92af9473ba64a7b5cfc7a0483 : SUCCESS in 59m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/81fcc63f9c704706b5f16a4e260d7db6 : SUCCESS in 1h 04m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b6e0b57cbd1d400f83e2ea4b07ec8c85 : SUCCESS in 1h 38m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f339f585b4b54f838823dbaccdd7ab6b : SUCCESS in 1h 15m 11s","accounts_in_message":[],"_revision_number":3},{"id":"afd43fc0dc108084580a5682d557c2aa56ff8f44","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-07-19 19:57:13.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2023/Arista_Tempest_Testrefs/changes/45/888945/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"4fc0cde69359b7834401cd04080afb74cdf6702e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-07-20 07:43:24.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8015dc782272aa6888b17556b36fbe93f3b68362","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-07-24 13:01:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f556a535eab4e58648d7dc72b52b31cbac33a04e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 13:02:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c3e1248973476e846205a20108c0a71ff26710ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 16:20:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/192c957368be4fe184fe9934619c0393\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ed5ef3ab0451408a970a1027807a6cf0 : SUCCESS in 1h 54m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b67586f7664d4ced8a1a323fe207b3c0 : SUCCESS in 10m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b63d76073dc4937a56d4c5349abae63 : SUCCESS in 27m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60477bb7089845ed8e3cf7df3098933b : SUCCESS in 24m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6dcb1b7afe6140e5adde94efa9255c33 : SUCCESS in 30m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/082420e280684610a0c7fdde1ca0454a : SUCCESS in 10m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0540b2bf61864a408d45b39bc2fbacee : SUCCESS in 1h 08m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/36331512d956435594fcd1bd4d85bdaf : SUCCESS in 1h 06m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7dda2570b523426493ed94d34be59a90 : SUCCESS in 1h 13m 57s","accounts_in_message":[],"_revision_number":3},{"id":"4f57fdf3563a0ee7f8e6c048098cb4e39aaad0a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 16:20:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0414bb82b549c780999f66fe59c847c9826fd5cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 16:22:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/787c96376cbf4558881d641c80df8d9f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11c4de1fce434838b293ed4ce171d3a1 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"321182980d8122a82e5f2a75a693cb527ee7257b","revisions":{"f9946b864477493060a32d5929b5a1ab1addf06e":{"kind":"REWORK","_number":1,"created":"2023-07-19 17:24:51.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/45/888945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/888945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/888945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/888945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/888945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/888945/1"}}},"commit":{"parents":[{"commit":"321f7672e7573801ffe855d20d8dfe574ec960c0","subject":"Merge \"[Docs] Add recommendation about usage of cache in the neutron-metadata-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/321f7672e7573801ffe855d20d8dfe574ec960c0"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-07-19 17:18:15.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-07-19 17:18:15.000000000","tz":-240},"subject":"Follow up on a small nit from patch 875989","message":"Follow up on a small nit from patch 875989\n\nUsing boolean instead of instead of number for one time resync\n\n[1] https://review.opendev.org/c/openstack/neutron/+/875989/comment/e2d314d1_3b1ded43/\n\nChange-Id: If4b03e42c5eaedad926e3b8efb8f72ed6872c9b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9946b864477493060a32d5929b5a1ab1addf06e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9946b864477493060a32d5929b5a1ab1addf06e"}]},"branch":"refs/heads/master"},"ffa2b66f61dfdf41d396a677babc0005a0a0bf86":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-07-19 17:25:57.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/45/888945/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/888945/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/888945/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/888945/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/888945/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/888945/2"}}},"commit":{"parents":[{"commit":"321f7672e7573801ffe855d20d8dfe574ec960c0","subject":"Merge \"[Docs] Add recommendation about usage of cache in the neutron-metadata-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/321f7672e7573801ffe855d20d8dfe574ec960c0"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-07-19 17:18:15.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-07-19 17:25:57.000000000","tz":0},"subject":"Follow up on a small nit from patch 875989 [1]","message":"Follow up on a small nit from patch 875989 [1]\n\nUsing boolean instead of instead of number for one time resync\n\n[1] https://review.opendev.org/c/openstack/neutron/+/875989/comment/e2d314d1_3b1ded43/\n\nChange-Id: If4b03e42c5eaedad926e3b8efb8f72ed6872c9b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ffa2b66f61dfdf41d396a677babc0005a0a0bf86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ffa2b66f61dfdf41d396a677babc0005a0a0bf86"}]},"branch":"refs/heads/master","description":"Edit commit message"},"321182980d8122a82e5f2a75a693cb527ee7257b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-07-19 17:31:30.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/45/888945/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/888945/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/888945/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/888945/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/888945/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/888945/3"}}},"commit":{"parents":[{"commit":"321f7672e7573801ffe855d20d8dfe574ec960c0","subject":"Merge \"[Docs] Add recommendation about usage of cache in the neutron-metadata-agent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/321f7672e7573801ffe855d20d8dfe574ec960c0"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-07-19 17:18:15.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-07-19 17:31:30.000000000","tz":0},"subject":"Follow up on a small nit from patch 875989 [1]","message":"Follow up on a small nit from patch 875989 [1]\n\nUsing boolean instead of an integer for one time resync\n\n[1] https://review.opendev.org/c/openstack/neutron/+/875989/comment/e2d314d1_3b1ded43/\n\nChange-Id: If4b03e42c5eaedad926e3b8efb8f72ed6872c9b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/321182980d8122a82e5f2a75a693cb527ee7257b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/321182980d8122a82e5f2a75a693cb527ee7257b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
