)]}'
{"id":"openstack%2Fneutron~889133","triplet_id":"openstack%2Fneutron~master~Ia9d5f69d2274d0f864e66dd249797ce5a276be85","project":"openstack/neutron","branch":"master","topic":"tempest-job-timeout","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-08-02 23:35:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia9d5f69d2274d0f864e66dd249797ce5a276be85","subject":"DNM: testing Tempest run increased concurrency","status":"ABANDONED","created":"2023-07-20 20:54:58.000000000","updated":"2023-08-02 23:35:20.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0d045b1d55893ecb7bc7477d38c3154f575fdf5b","_number":889133,"virtual_id_number":889133,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-07-24 19:27:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 23:04:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5449f7fb55814fc87940ff83ea3d7ad4279005c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-20 20:54:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d04cddaae1e2527497fdf5482980697947053015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 23:04:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64cd4343f5414095b89c9086749cc85c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/42189d1a7bde471f931be55825e1cbcf : SUCCESS in 1h 12m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/57b5b44e9776455c9b6e6b45fba2f6a1 : SUCCESS in 1h 27m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f319a923e2c74f5185ae5b6bef3bdd45 : SUCCESS in 1h 46m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/db547ab3e8374605a3659b70acb01632 : SUCCESS in 2h 06m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fb70d4486f1f454d96d362c795fdaa0e : SUCCESS in 31m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/986e022926d94c57a3ea06ab28b31e10 : SUCCESS in 11m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89544f2cfd0a4a7598ec97109c5361b5 : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5ec71e8c9f1497794d0b9bf02828c93 : SUCCESS in 48m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea8ad6d460d9464d9468c878214c4fd5 : SUCCESS in 1h 42m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f2b7ef0056546dda65b1f32186a2aba : SUCCESS in 47m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ccba6b79cf6f4050a35834fe1af1d50d : SUCCESS in 1h 00m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/223948004fbe44c7bce2360e4b98a6d3 : SUCCESS in 1h 21m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/425b3f71719a4feeb8b7158b2be0f4d9 : SUCCESS in 1h 23m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/59c8abb4163146f590fa7fc085b1aaa5 : SUCCESS in 1h 02m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/85c79aa67703412d8a3001821f8737dc : SUCCESS in 1h 15m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/71d6bed5e62c4e2fa2f297e7a0465618 : SUCCESS in 1h 45m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/05261f8d88d64183aadda78a5fcd79d3 : SUCCESS in 1h 17m 29s","accounts_in_message":[],"_revision_number":1},{"id":"395e829d3a7125d2a27f1768fb6f6e6115d9bf10","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-20 23:14:24.000000000","message":"Topic timeout removed","accounts_in_message":[],"_revision_number":1},{"id":"7908e98917a071a14bad722fade4c60e9c7ec429","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-20 23:14:28.000000000","message":"Topic set to tempest-job-timeout","accounts_in_message":[],"_revision_number":1},{"id":"e48e6f26f0a6736ae4008f45de09c0cf58502c41","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-24 17:04:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d3b5193998992fd2db4830e33ef8e666e49be65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 19:27:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45687450db4d47faa7b68959cb2b1271\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1cb8e285dfe6465d9b61f6fd6796dd67 : FAILURE in 2h 03m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d9311633d6474fe6a059125866667f84 : SUCCESS in 1h 30m 36s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b2532a2a006741a08f188c926ab5af4b : SUCCESS in 1h 21m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/14e658a48e324dc5b534f9467a7d1d82 : SUCCESS in 1h 42m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/242380990f5b4307b7d1e248ceb925f5 : SUCCESS in 35m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab8c4ff1497a4a93b5ec8b064317d902 : SUCCESS in 11m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6674966a987643b081e33c210480fe37 : SUCCESS in 13m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6e219588bac43eb8c128280b0752a2c : SUCCESS in 52m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a274a330799e49a485327fe7478600ae : SUCCESS in 1h 46m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5cde60adadd4d4c8d1b89fd55bdea06 : SUCCESS in 1h 18m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f09f1fef8f9440d3a24690803259ef63 : FAILURE in 1h 18m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/25d1a7cd548b47e988ec62f906de9201 : SUCCESS in 1h 28m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b78cc0785d9047da9c6ebb2d0248c9fa : FAILURE in 1h 08m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5dd03174f6754f62a6026dd71dafff38 : SUCCESS in 1h 21m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bb8d5607b6824df4b65cf6f973495425 : SUCCESS in 1h 49m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aa327839bb9345028f94c71e25f320a7 : SUCCESS in 2h 08m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/071272b18c944838840becc0e3b6a930 : SUCCESS in 1h 20m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0d045b1d55893ecb7bc7477d38c3154f575fdf5b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-08-02 23:35:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9c1866f33416319446da0bd56c0fa6d13e9f46d","revisions":{"d9c1866f33416319446da0bd56c0fa6d13e9f46d":{"kind":"REWORK","_number":1,"created":"2023-07-20 20:54:58.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/33/889133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/889133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/889133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/889133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/889133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/889133/1"}}},"commit":{"parents":[{"commit":"dc1b7851af9a7197383c2bb7a1a21c699d870f4a","subject":"Merge \"[PostgreSQL] Subnet entity with ServiceType grouped by both tables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc1b7851af9a7197383c2bb7a1a21c699d870f4a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-07-20 20:53:17.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-07-20 20:53:17.000000000","tz":-300},"subject":"DNM: testing Tempest run increased concurrency","message":"DNM: testing Tempest run increased concurrency\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/887220\nChange-Id: Ia9d5f69d2274d0f864e66dd249797ce5a276be85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9c1866f33416319446da0bd56c0fa6d13e9f46d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9c1866f33416319446da0bd56c0fa6d13e9f46d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
