)]}'
{"id":"openstack%2Fneutron~890189","triplet_id":"openstack%2Fneutron~master~I5e011a2916e9c57a5b951a6ae77973597a4bb55e","project":"openstack/neutron","branch":"master","topic":"bug/1983053","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-09-01 07:03:43.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5e011a2916e9c57a5b951a6ae77973597a4bb55e","subject":"DNM just test for the neutron-fullstack tests","status":"ABANDONED","created":"2023-08-01 10:39:15.000000000","updated":"2023-09-01 12:58:56.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fedbc3513dfa40937db8339fe97a00ee0c128014","_number":890189,"virtual_id_number":890189,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-08-30 13:02:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-01 10:39:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-10 10:59:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 09:58:08.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"}],"messages":[{"id":"9b7f54264ce1a244e2b994e796b95ceb8b94e1a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-01 10:39:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac5e1bb1dfd65f8f511e4cbecb686f05b1c71cf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 10:39:40.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][1][\u0027neutron-fullstack-with-uwsgi-2\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-2:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-3:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-4:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-5:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-6:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-7:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-8:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-9:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-10:\n            parent: neutron-fullstack-with-uwsgi\n\n  in \"openstack/neutron/zuul.d/project.yaml@master\", line 5, column 3","accounts_in_message":[],"_revision_number":1},{"id":"6b2687ddc69e9c853128190f0f421b8ad556bb24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 10:39:47.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][1][\u0027neutron-fullstack-with-uwsgi-2\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-2:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-3:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-4:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-5:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-6:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-7:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-8:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-9:\n            parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-10:\n            parent: neutron-fullstack-with-uwsgi\n\n  in \"openstack/neutron/zuul.d/project.yaml@master\", line 5, column 3","accounts_in_message":[],"_revision_number":1},{"id":"5c484ce3e382850b09b05488d902deb44c0ee9b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-01 10:40:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ae827248cef29cc307184236584c4c4321eec187","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 10:41:21.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  Job neutron-fullstack-with-uwsgi-2 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-2:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-3:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-4:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-5:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-6:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-7:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-8:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-9:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-10:\n              parent: neutron-fullstack-with-uwsgi\n\n  in \"openstack/neutron/zuul.d/project.yaml@master\", line 5, column 3","accounts_in_message":[],"_revision_number":2},{"id":"02c938ba09978406a24dd55e2da79902a1d2229a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 10:41:27.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  Job neutron-fullstack-with-uwsgi-2 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-2:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-3:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-4:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-5:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-6:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-7:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-8:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-9:\n              parent: neutron-fullstack-with-uwsgi\n          - neutron-fullstack-with-uwsgi-10:\n              parent: neutron-fullstack-with-uwsgi\n\n  in \"openstack/neutron/zuul.d/project.yaml@master\", line 5, column 3","accounts_in_message":[],"_revision_number":2},{"id":"f33faa1bd95a0d8da4d5bcc47b81a5ded41e1b57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-01 10:43:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"87801f2088d015954a4f483a43fe7a8c71f1445a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 13:14:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c6273b08bbb44bdb4eddf3018761c6b\n\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b90e77b024184378bcc94d2eb1d2ebe0 : FAILURE in 1h 33m 26s\n- neutron-fullstack-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/aac652f767c04ea29ebef28758616236 : FAILURE in 1h 24m 54s\n- neutron-fullstack-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/d942aaba2c604cc68eed1409604035e2 : FAILURE in 1h 44m 13s\n- neutron-fullstack-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/d9562518cc3648c7b7ae17cfce630338 : FAILURE in 1h 30m 54s\n- neutron-fullstack-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/b9a1303572b6412a9b754278065f8e0c : FAILURE in 1h 43m 45s\n- neutron-fullstack-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/6b3a43db897649ae954cabc49db4bd2b : FAILURE in 1h 37m 12s\n- neutron-fullstack-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/bc64b457b8ec44a9bc915e5f82bab9f9 : FAILURE in 1h 45m 48s\n- neutron-fullstack-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/119f49ab616443bd92d06c6e60b7abcf : FAILURE in 2h 29m 56s\n- neutron-fullstack-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/dfcb83811ca04195b2f0424e091752a7 : FAILURE in 1h 43m 31s\n- neutron-fullstack-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/dca021f24e154b09b4fda711c21c7a8c : FAILURE in 1h 56m 47s","accounts_in_message":[],"_revision_number":3},{"id":"ccab3f9eb972c343f5460024c30ac86b8f26092f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-10 10:18:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"acf0c3660626a2c6c70c26d98b176c2941dea6ce","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-10 10:59:28.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/89/890189/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"ab35f0cbe8a0df026288a04e68d56b445ff1ad84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 12:58:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9d9fa445b5b4a218fb43a0d1d6e7f21\n\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/64546046d42a4edbbae13b85f3616806 : FAILURE in 1h 38m 48s\n- neutron-fullstack-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/c1efdba595f543cbb5fcc0a3fc9e5067 : FAILURE in 1h 57m 14s\n- neutron-fullstack-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/6ab46aaa25a64dd59c39da2aa92ff7e4 : FAILURE in 1h 57m 14s\n- neutron-fullstack-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/b9c1e16ddd7247d5a250fe9832a648fb : FAILURE in 1h 51m 50s\n- neutron-fullstack-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/bc089e2b2b574a3ea42b61c57b058f60 : FAILURE in 1h 43m 57s\n- neutron-fullstack-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/4695550032f6485a803eff5f7e1b2e05 : FAILURE in 2h 00m 24s\n- neutron-fullstack-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/e7b352f8af4e403996677f1a1383634a : FAILURE in 1h 34m 40s\n- neutron-fullstack-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/929387e01a68447abc0b9d00e25bcfbd : FAILURE in 2h 00m 45s\n- neutron-fullstack-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/a249685f00c94396852a957f40c57a8b : FAILURE in 1h 53m 21s\n- neutron-fullstack-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/d7e8be95d52b4a4dbd3d30c5519bf92b : FAILURE in 1h 44m 29s","accounts_in_message":[],"_revision_number":4},{"id":"1360f3d2e93013c52bee2c2f9a1c9cb0f948f0ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-10 14:47:52.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c9e4302d21d0ffd7716603a141f7842a4f96a921","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-10 15:29:58.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/89/890189/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"bca6d530dce9105419ee84fa1ea66f30e8dd1788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 17:28:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b6e30ed6c594216987cfb0a83c93b3a\n\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/119ec557bad242c38e25da4c23615882 : SUCCESS in 2h 28m 31s\n- neutron-fullstack-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/31a672d83a7d482e96c5a19c99440e03 : SUCCESS in 2h 17m 08s\n- neutron-fullstack-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/b56a5db1fb1b4f5a9a3c6cdb4ad01e5c : TIMED_OUT in 2h 32m 28s\n- neutron-fullstack-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/a7c4b2d8387542fab03c4dd6ea8362b1 : SUCCESS in 2h 03m 34s\n- neutron-fullstack-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/116a34e2c83f410fbadc3b5f9f04fa09 : TIMED_OUT in 2h 32m 24s\n- neutron-fullstack-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/f3e52e48042f4810928d6797df012ad5 : SUCCESS in 2h 16m 07s\n- neutron-fullstack-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/e0a8b57c9bbd4dd5aa02ee6a346e4545 : TIMED_OUT in 2h 32m 21s\n- neutron-fullstack-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/6313ed14054c44a8a18522751a2a5774 : SUCCESS in 2h 16m 01s\n- neutron-fullstack-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/c39e98a9e06a43229d930f30b7c404bb : TIMED_OUT in 2h 32m 24s\n- neutron-fullstack-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/f41edf8a35e34cffabf14d2c6899254d : SUCCESS in 2h 19m 42s","accounts_in_message":[],"_revision_number":5},{"id":"58cb9c0a83da02519bb8382a980c3f5baee97123","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-11 07:00:30.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"481f0a53439ad3246cfb904cddd584b92818aced","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-11 09:08:23.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/89/890189/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"186e4c27f916ed31e44564e9a4cdc0570df6468f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 09:14:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4ebfcf8b5c944cb9b4c00aca7ca6edf\n\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/39f84de3fbb5466f9ca9fa4308591085 : SUCCESS in 1h 49m 17s\n- neutron-fullstack-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/eb5b2b93fd2b468ba7ca41ab74c16b63 : FAILURE in 1h 29m 59s\n- neutron-fullstack-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/4ba8510d9033473f82a8352b8d343026 : FAILURE in 1h 34m 01s\n- neutron-fullstack-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/89624cc2886340d4b9f71e1f2118a256 : SUCCESS in 1h 42m 00s\n- neutron-fullstack-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/85e3fe74a72e4253a19ce01a80a901bd : FAILURE in 1h 48m 52s\n- neutron-fullstack-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/0fc998c85ae942f898ad6cf7b591a5c3 : SUCCESS in 1h 49m 38s\n- neutron-fullstack-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/e74580d0d87741aa9829df3472d81b54 : FAILURE in 1h 27m 23s\n- neutron-fullstack-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/d2985759e311416eb74d577feff23faf : FAILURE in 1h 50m 54s\n- neutron-fullstack-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/64b25ec3da7048b9a78ae61a4edf026c : FAILURE in 1h 47m 12s\n- neutron-fullstack-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/524503c7ec2a442387e02ac0976cdbc1 : SUCCESS in 2h 06m 43s","accounts_in_message":[],"_revision_number":6},{"id":"724b5a1b3b75f4b0a56fc31feb5141f377fc318c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-23 10:38:04.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"d5d9948718ef538c4c12447dc9fb9597e205a952","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 13:11:53.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/667a29cbb0394926baaddf2cbe2e9fff\n\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2328ccda46e647e4b8aefbe77b8e525c : FAILURE in 2h 02m 54s\n- neutron-fullstack-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/5f4ccb853e2146d791209b1627e270f3 : TIMED_OUT in 2h 32m 17s\n- neutron-fullstack-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/526b04e869554d3cbe6f6c39742c4cae : FAILURE in 1h 44m 39s\n- neutron-fullstack-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/d97ee62c3a7a4d278e0cb2a3a0086aee : FAILURE in 1h 37m 51s\n- neutron-fullstack-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/a2cf0b9caad4477da9ec0293e46cf509 : FAILURE in 1h 43m 59s\n- neutron-fullstack-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/80f65c24fe474a9fa14e83ded61ef47b : FAILURE in 2h 12m 52s\n- neutron-fullstack-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/8c986a694af547d1a818615bdcd7633c : FAILURE in 1h 50m 57s\n- neutron-fullstack-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/3f8d8c36f8524c3aa6d6d3f6a0ea4cb4 : FAILURE in 1h 56m 09s\n- neutron-fullstack-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/dd0a680997374563a2c4507dd90270f0 : FAILURE in 1h 47m 23s\n- neutron-fullstack-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/1ee42d0c61824ca1812080b8aacb7fb6 : FAILURE in 1h 47m 11s","accounts_in_message":[],"_revision_number":7},{"id":"370b503ba9169b84b902e0e6d1ed3f73b1a3287a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-29 17:54:56.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"010837cfb095573db30bfe36558bda2838725a43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 20:33:03.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b64c97163e543199ac5cca6c6d78710\n\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9eb5d9941d0e42b4afed05bdfa1f91bb : SUCCESS in 2h 27m 26s\n- neutron-fullstack-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/145b9442948844edaad291a701e17a4b : SUCCESS in 2h 31m 58s\n- neutron-fullstack-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/806b18aeedc7486fa170d26cbdce56f6 : SUCCESS in 2h 18m 17s\n- neutron-fullstack-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/2ec3839857eb416789a45c9dfc43ecb0 : SUCCESS in 2h 11m 15s\n- neutron-fullstack-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/968cfb63458c4f52b4d25b9e7cf58470 : TIMED_OUT in 2h 32m 21s\n- neutron-fullstack-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/93e982e5a13e44ccb3778b5e2122ce5a : TIMED_OUT in 2h 32m 29s\n- neutron-fullstack-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/7345a31e379240f681e4cf583d2105da : SUCCESS in 2h 12m 18s\n- neutron-fullstack-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/916aadc49301493b9a948c18f268faa3 : SUCCESS in 2h 17m 16s\n- neutron-fullstack-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/0df3f47f94e84dbf89755e773f14d37a : SUCCESS in 2h 28m 32s\n- neutron-fullstack-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/09f81785b7e84c978119687432ab599a : SUCCESS in 2h 24m 30s","accounts_in_message":[],"_revision_number":8},{"id":"5254cb8c481b39a40a7d1b153bc1226ef68a50e1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-30 09:58:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0af1a0abc2c082d2e12950a0afc3af9558358d4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-30 10:21:22.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"19edaa1a374370d8873f773e7b40ba0fb8cfda03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 13:02:51.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40bc8cbace294249b55a99f04383c2d0\n\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/782ec8c303964e23873da93db00301a8 : SUCCESS in 2h 27m 09s\n- neutron-fullstack-with-uwsgi-2 https://zuul.opendev.org/t/openstack/build/9fe68e31a858421f991552097d6644c9 : SUCCESS in 2h 32m 36s\n- neutron-fullstack-with-uwsgi-3 https://zuul.opendev.org/t/openstack/build/d5d45eef2ec14f98aed0be8bc231bba4 : SUCCESS in 2h 25m 41s\n- neutron-fullstack-with-uwsgi-4 https://zuul.opendev.org/t/openstack/build/3aad94f844ec47369e7e88586c12af89 : SUCCESS in 2h 33m 47s\n- neutron-fullstack-with-uwsgi-5 https://zuul.opendev.org/t/openstack/build/5b2cde4f38e347928a7ee7f6537b06b5 : SUCCESS in 2h 30m 08s\n- neutron-fullstack-with-uwsgi-6 https://zuul.opendev.org/t/openstack/build/0d7a11bca2834d6bac70f8484a655dc1 : SUCCESS in 2h 22m 04s\n- neutron-fullstack-with-uwsgi-7 https://zuul.opendev.org/t/openstack/build/6999a44fd20d41bba6016a9b747aa5e3 : SUCCESS in 2h 27m 44s\n- neutron-fullstack-with-uwsgi-8 https://zuul.opendev.org/t/openstack/build/d84429475d38491e9c866687af285889 : SUCCESS in 2h 07m 09s\n- neutron-fullstack-with-uwsgi-9 https://zuul.opendev.org/t/openstack/build/5ea77a937abe4ac399e0c262342e66f2 : SUCCESS in 2h 07m 24s\n- neutron-fullstack-with-uwsgi-10 https://zuul.opendev.org/t/openstack/build/98e0ae79217d4ae39c9022092f692514 : SUCCESS in 2h 33m 41s","accounts_in_message":[],"_revision_number":9},{"id":"2e6fa8b6084e53842d8b125da876f2646b1a14b4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-31 00:27:27.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/89/890189/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"b2db3101b92e17155272e077ffd109a1a722263e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-09-01 07:03:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":9},{"id":"fedbc3513dfa40937db8339fe97a00ee0c128014","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-01 12:58:56.000000000","message":"Patch Set 8:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/89/890189/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8}],"current_revision_number":9,"current_revision":"28313ce5fbcbc58187e4a3fecaeb28bbee2971cf","revisions":{"ffd06c3fc36a316939d51676dc0e982d3f920f2b":{"kind":"REWORK","_number":1,"created":"2023-08-01 10:39:15.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/89/890189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/890189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/890189/1"}}},"commit":{"parents":[{"commit":"625acf8dea224b966768038ecd56d2ae9f1df231","subject":"Default SG rules - use new rules templates to create rules for SGs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/625acf8dea224b966768038ecd56d2ae9f1df231"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-01 10:38:47.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-01 10:38:47.000000000","tz":120},"subject":"DNM just test for the neutron-fullstack tests","message":"DNM just test for the neutron-fullstack tests\n\nChange-Id: I5e011a2916e9c57a5b951a6ae77973597a4bb55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ffd06c3fc36a316939d51676dc0e982d3f920f2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ffd06c3fc36a316939d51676dc0e982d3f920f2b"}]},"branch":"refs/heads/master"},"cef0fa760d47320567eaa94b3e2d4169f278fc30":{"kind":"REWORK","_number":2,"created":"2023-08-01 10:40:56.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/89/890189/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/890189/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/890189/2"}}},"commit":{"parents":[{"commit":"625acf8dea224b966768038ecd56d2ae9f1df231","subject":"Default SG rules - use new rules templates to create rules for SGs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/625acf8dea224b966768038ecd56d2ae9f1df231"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-01 10:38:47.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-01 10:40:41.000000000","tz":120},"subject":"DNM just test for the neutron-fullstack tests","message":"DNM just test for the neutron-fullstack tests\n\nChange-Id: I5e011a2916e9c57a5b951a6ae77973597a4bb55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cef0fa760d47320567eaa94b3e2d4169f278fc30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cef0fa760d47320567eaa94b3e2d4169f278fc30"}]},"branch":"refs/heads/master"},"0be53469dbce3d1e2f47f44f1c03a19e3a2f9cb0":{"kind":"REWORK","_number":3,"created":"2023-08-01 10:43:26.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/89/890189/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/890189/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/890189/3"}}},"commit":{"parents":[{"commit":"625acf8dea224b966768038ecd56d2ae9f1df231","subject":"Default SG rules - use new rules templates to create rules for SGs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/625acf8dea224b966768038ecd56d2ae9f1df231"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-01 10:38:47.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-01 10:43:08.000000000","tz":120},"subject":"DNM just test for the neutron-fullstack tests","message":"DNM just test for the neutron-fullstack tests\n\nChange-Id: I5e011a2916e9c57a5b951a6ae77973597a4bb55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0be53469dbce3d1e2f47f44f1c03a19e3a2f9cb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0be53469dbce3d1e2f47f44f1c03a19e3a2f9cb0"}]},"branch":"refs/heads/master"},"dab770b012f93bdb16da792d6e37170771606d20":{"kind":"REWORK","_number":4,"created":"2023-08-10 10:18:21.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/89/890189/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/890189/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/890189/4"}}},"commit":{"parents":[{"commit":"625acf8dea224b966768038ecd56d2ae9f1df231","subject":"Default SG rules - use new rules templates to create rules for SGs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/625acf8dea224b966768038ecd56d2ae9f1df231"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-01 10:38:47.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-10 10:18:05.000000000","tz":120},"subject":"DNM just test for the neutron-fullstack tests","message":"DNM just test for the neutron-fullstack tests\n\nChange-Id: I5e011a2916e9c57a5b951a6ae77973597a4bb55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dab770b012f93bdb16da792d6e37170771606d20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dab770b012f93bdb16da792d6e37170771606d20"}]},"branch":"refs/heads/master"},"a11c71ea976e99cedebc3deb1bd791d2ba5fa0ec":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-08-10 14:47:52.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/89/890189/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/890189/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/890189/5"}}},"commit":{"parents":[{"commit":"578644998a0a4ad84bcd6843e74ae294a4e4a330","subject":"Default SG rules - use new rules templates to create rules for SGs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/578644998a0a4ad84bcd6843e74ae294a4e4a330"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-01 10:38:47.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-10 14:47:36.000000000","tz":120},"subject":"DNM just test for the neutron-fullstack tests","message":"DNM just test for the neutron-fullstack tests\n\nChange-Id: I5e011a2916e9c57a5b951a6ae77973597a4bb55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a11c71ea976e99cedebc3deb1bd791d2ba5fa0ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a11c71ea976e99cedebc3deb1bd791d2ba5fa0ec"}]},"branch":"refs/heads/master"},"85773388645f70728acc0128f38815a1c7b5bd24":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-08-11 07:00:30.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/89/890189/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/890189/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/890189/6"}}},"commit":{"parents":[{"commit":"35acd57db3fad582e7010580c4e6ae14a2d986fd","subject":"Default SG rules - use new rules templates to create rules for SGs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35acd57db3fad582e7010580c4e6ae14a2d986fd"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-01 10:38:47.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-11 07:00:12.000000000","tz":120},"subject":"DNM just test for the neutron-fullstack tests","message":"DNM just test for the neutron-fullstack tests\n\nChange-Id: I5e011a2916e9c57a5b951a6ae77973597a4bb55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85773388645f70728acc0128f38815a1c7b5bd24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85773388645f70728acc0128f38815a1c7b5bd24"}]},"branch":"refs/heads/master"},"f841c2c8db0c8e3f69369725231d1551452b0edf":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-08-23 10:38:04.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/89/890189/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/890189/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/890189/7"}}},"commit":{"parents":[{"commit":"16899bede2cf0059abd9eaaba70f5435059c0c76","subject":"Default SG rules - use new rules templates to create rules for SGs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16899bede2cf0059abd9eaaba70f5435059c0c76"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-01 10:38:47.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-23 10:38:04.000000000","tz":0},"subject":"DNM just test for the neutron-fullstack tests","message":"DNM just test for the neutron-fullstack tests\n\nChange-Id: I5e011a2916e9c57a5b951a6ae77973597a4bb55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f841c2c8db0c8e3f69369725231d1551452b0edf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f841c2c8db0c8e3f69369725231d1551452b0edf"}]},"branch":"refs/heads/master","description":"Rebase"},"bb73711fa37a9d15f5c205314497a93abc14348e":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-08-29 17:54:56.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/89/890189/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/890189/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/890189/8"}}},"commit":{"parents":[{"commit":"d98d8324bd065d17a5d5b0baa45ba847716af65d","subject":"Default SG rules - use new rules templates to create rules for SGs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d98d8324bd065d17a5d5b0baa45ba847716af65d"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-01 10:38:47.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-29 17:54:43.000000000","tz":120},"subject":"DNM just test for the neutron-fullstack tests","message":"DNM just test for the neutron-fullstack tests\n\nChange-Id: I5e011a2916e9c57a5b951a6ae77973597a4bb55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb73711fa37a9d15f5c205314497a93abc14348e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb73711fa37a9d15f5c205314497a93abc14348e"}]},"branch":"refs/heads/master"},"28313ce5fbcbc58187e4a3fecaeb28bbee2971cf":{"kind":"REWORK","_number":9,"created":"2023-08-30 10:21:22.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/89/890189/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/890189/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/890189/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/890189/9"}}},"commit":{"parents":[{"commit":"a4c8392209f7935cc6699c1cf9dc36d483b2f864","subject":"Default SG rules - use new rules templates to create rules for SGs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4c8392209f7935cc6699c1cf9dc36d483b2f864"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-01 10:38:47.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-30 10:21:01.000000000","tz":120},"subject":"DNM just test for the neutron-fullstack tests","message":"DNM just test for the neutron-fullstack tests\n\nChange-Id: I5e011a2916e9c57a5b951a6ae77973597a4bb55e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28313ce5fbcbc58187e4a3fecaeb28bbee2971cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28313ce5fbcbc58187e4a3fecaeb28bbee2971cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
