)]}'
{"id":"openstack%2Fneutron~890231","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I3ce026054286c8e28ec1500f1a4aa607fe73f337","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/2028795-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-08-08 13:11:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3ce026054286c8e28ec1500f1a4aa607fe73f337","subject":"dvr: Avoid installing non-dvr openflow rule on startup","status":"MERGED","created":"2023-08-03 12:43:52.000000000","updated":"2023-08-08 13:12:43.000000000","submitted":"2023-08-08 13:11:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890231-bug/2028795-stable/xena-stable/wallaby","cherry_pick_of_change":890230,"cherry_pick_of_patch_set":1,"meta_rev_id":"4d8b6045dad138bd27adbb331015ce26521815bc","_number":890231,"virtual_id_number":890231,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-08 13:11:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-08-04 09:52:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-08-03 13:09:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-08-04 09:52:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-03 13:09:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-08-03 13:42:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-03 15:24:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-04 07:33:26.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2023-08-04 09:52:35.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"9ca9210283c8fc065d979ec7d1faee1b8c7a5fc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-08-03 12:43:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"0049713b0adbf6e595483144c60e406e7ac43f90","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-03 13:09:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cce14e3d718045dfefe604b8156684f38c1d3b4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 13:42:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fbf561134444ce9ad3f52f9ae5bd59f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/312d1c0d4f9644ddb8a656dda2044a35 : SUCCESS in 18m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/441bfe11a2fc4cd7bc265bfd8c007788 : SUCCESS in 51m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"824609bfc29a3e41112c3c1e6a84caf82fa154e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 15:24:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c653b7a85e04e1da389773237379300\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/4d089613a3314b7fa350c6d0089ffc0b : SUCCESS in 34m 58s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/d55bf93621fc4771888567344b1e02f1 : SUCCESS in 1h 44m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/ccaa9e56dd3c411fb59159bfbbf04d39 : FAILURE in 2h 36m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4170c95e4a74461780c741965d996032 : SUCCESS in 38m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76d69a5b80254f46ac038c99d1179f48 : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/594a8050e88b439f95b5d3b0855ee26c : SUCCESS in 34m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00eef442d76347bc898ce931a87e858b : SUCCESS in 30m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30d928c2c72047aead07310a3d917c0c : SUCCESS in 28m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/756b0f7d33754eb094580e9855066bf4 : SUCCESS in 10m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b55215b7b7f4837bb1db9a9c356cfd9 : FAILURE in 45m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c4d2af4c93554fc89b427dbb68972725 : SUCCESS in 1h 31m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/aae2e28fa41f4ace9afddc5fbf484153 : SUCCESS in 1h 20m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb119192a13149de8f1645c4412ada39 : SUCCESS in 1h 10m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/20217a77fbdb4c258b142b09408f3cb4 : FAILURE in 40m 55s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1f25feb61cb44115869c88967502d0d0 : SUCCESS in 1h 25m 05s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bb838c6c1910459a9ebb0182402b911b : SUCCESS in 1h 24m 50s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/342f25daebed44978c91365527c560e4 : SUCCESS in 1h 48m 09s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/484db0500a6245219b42eeff0d18425b : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":1},{"id":"aadaff5b503615dc4ce5eb0c8638f91a637491df","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-08-04 07:33:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"12f01137d13abc61a43a7b9ada50a12340318d3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 07:54:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95c6c539c49247428ca9a8e307895931\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/90c503c4434c4ecea9790bc3bd4fd42e : SUCCESS in 19m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89a96233735442769cf723760312cfc0 : SUCCESS in 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"938880278d8ab6ed5b44502d3df46220d8643fbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 09:48:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0696dfedf344a93a0f8f21cd39bb8ea\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/d6f3aa6abacd4446959f016f80b43b07 : SUCCESS in 56m 18s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/4c9ea3cae51c4ca094cf1c662ff70b40 : SUCCESS in 2h 12m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/188a1520aa204feba3b9d7041e03a0db : SUCCESS in 1h 50m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bad4e45308234948a03b56585aa0fab2 : SUCCESS in 30m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1454ef67d1dc4883973d208955a3ad76 : SUCCESS in 13m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2422905a3eea4d81906d40b2a5cc44f6 : SUCCESS in 30m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0174a145112f42afa687b41c3cf4a9e7 : SUCCESS in 29m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6417e6c0b50049cf9ada20312c978e19 : SUCCESS in 30m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4755c71359334ca2a84250910933f9e2 : SUCCESS in 9m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/664448affce043f8874571f5589c9e5f : SUCCESS in 52m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/596d79dc93024c708f8af7d3aec28708 : SUCCESS in 1h 23m 08s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1db5de22dee9410fb991d867af22c970 : SUCCESS in 1h 41m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/254840846e8b4cf49515402779ad6067 : SUCCESS in 1h 05m 49s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4f707175bde64b05a399620cd35ea604 : SUCCESS in 50m 31s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fae75f5304d3439b980bd4a27968bdad : SUCCESS in 1h 19m 10s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3ab5b14ff9cd447ba0cbe3dae37b5e6f : SUCCESS in 1h 18m 59s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/282a63ac86c7430a97fd7fdceab96c0b : SUCCESS in 1h 42m 09s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6db275f17a9446fab79d3c635ef07b9d : SUCCESS in 41m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4165dd34a5bbf515cd402633aff6dbf510bc20c3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-08-04 09:52:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1081f1525b56ac770b8e138ed672562c9c6b9fac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 09:53:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7992965960ac2703d6c17cef53459f181e6af392","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 13:08:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ee864073c1e49e28a5b72f694087f0a\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/9a8f00698e6d474f996150b06deb1962 : SUCCESS in 53m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdcae40abe784365a228e47e5c3c4671 : SUCCESS in 11m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d128e66a69546749f68528d43bb5d54 : SUCCESS in 33m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f1279ecaa3a46b1b3c696e5dfcef34a : SUCCESS in 21m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b098fb1cacc343d983c1795ac0c212a1 : SUCCESS in 8m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1eaf056be22f4962be562cf9f4a452ce : SUCCESS in 56m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9fe51c04e8734559baf40ec769ac8ed9 : SUCCESS in 1h 23m 46s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0239ab610821407198e1aed7ad1058ed : TIMED_OUT in 3h 10m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6991403d29884d078c4ee61d90b894c4 : SUCCESS in 1h 01m 12s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/08a15e3bfa4a453984783574c847ae06 : SUCCESS in 1h 40m 55s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c44a5dbe459486ea96084d37055a003 : SUCCESS in 1h 07m 08s","accounts_in_message":[],"_revision_number":1},{"id":"bb5fb3911c984ff7d755f11b1f46785316a62264","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-08-04 13:45:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ddd69fcfc568e9512f565285913590e4c19d02c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 14:08:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/147b0e94715a484582380327a878dc30\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ba388e4f30b485eb50f2b2fb2aaeecf : SUCCESS in 21m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8cc3652861c04d90a7d64751010e7aa6 : SUCCESS in 21m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"136ea6190a1c6419db3ebfb960a155dd3dc658f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 16:37:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f43417f902514e9aa6d86e06661e2a8a\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/e43e5f8c9816485c8fb58fd2dd833b03 : SUCCESS in 59m 09s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/fc807f2f8c064b9f975603f93ad55da4 : SUCCESS in 2h 44m 26s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/4338d8f10bcd40ba957733a9677d3edc : SUCCESS in 1h 41m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e74d71687e34769a9558a6c11dd35b5 : SUCCESS in 41m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a456501640f948b3849e4fd4204936ef : SUCCESS in 15m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bff9170e676e443ca76a01cadc02ca84 : SUCCESS in 39m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3e0df130fac49789e838961834e3ac3 : SUCCESS in 28m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d57cb2d58d134411808597696e5e2b2e : SUCCESS in 31m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c79f41258a1347e2865c2456c23b48eb : SUCCESS in 10m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/84fc281faac44c4dac65bcaa7559a429 : SUCCESS in 51m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9f5b5a4b39a46bba7c00d098afa09b2 : SUCCESS in 1h 27m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/75749679decd4f568a013f11a464330d : SUCCESS in 1h 13m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e958faa5dab40b6a8b4a7493af18b2a : FAILURE in 47m 49s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/525acb3590e147be97117220c14f28f7 : SUCCESS in 1h 02m 12s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9f1c7062821a40a9ad7ac7843c466468 : SUCCESS in 1h 14m 03s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2527b60b0c954844bdedbcc0038161b1 : FAILURE in 1h 49m 04s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1240d1a07a51472a8de0ab81bdc4a5b4 : SUCCESS in 1h 28m 12s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4c49435576a4dacb1ae21ecc7845d05 : SUCCESS in 1h 00m 56s","accounts_in_message":[],"_revision_number":1},{"id":"7369d3926edd02d0073d5956a816f077d7f8f2e5","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-08-08 08:13:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aeab9c58a8e6153736b5476e79a587e11542cd0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 09:06:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57ce0aa404504756887f5a4be39721cc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ac1dceacb86465a85f3117065084481 : SUCCESS in 23m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f232fd7a114748d08bb578830ed5dc2b : SUCCESS in 51m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56bb0debc90c71c2e52cf68a620a3ed5caa4be52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 10:43:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b82ccc2a33994f3ca48e7e9cd70646b2\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/9f148f8f58bf4c3bb70f8d2653605797 : SUCCESS in 1h 06m 36s\n- neutron-tempest-plugin-scenario-openvswitch-wallaby https://zuul.opendev.org/t/openstack/build/af8b99c461b94fe18b48d1c107b09280 : SUCCESS in 2h 22m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-wallaby https://zuul.opendev.org/t/openstack/build/d4d7de62204a48b6a6fdcc33ea8895c1 : SUCCESS in 1h 59m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0508d461edb14dd687bc847a6389fee1 : SUCCESS in 38m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7619f3cb341e4c22afa5580d84ccd24e : SUCCESS in 11m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5cde9aeaf95439ba97e4340e249b7cb : SUCCESS in 34m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d40c73451e614409bfc06fe3b747e669 : SUCCESS in 27m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/910f7021023045ab9bb2a5617c5e16ff : SUCCESS in 32m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8a835072ec144f2b5a3724cd993cfa1 : SUCCESS in 9m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e2d16d9a48bf484d94b1af3e92972ebb : SUCCESS in 48m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2dc248bf098a4b7bb72b8b4932f67719 : SUCCESS in 1h 04m 25s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4c159e78361b422a813d4a1c1d61b523 : SUCCESS in 1h 38m 21s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b17fa9a54b85470190fc3f2ddcbf3c54 : SUCCESS in 1h 17m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5810c0bb6dec435ab773ec7a8105f5db : SUCCESS in 1h 00m 09s (non-voting)\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/81f8361299914478bbd38b820e18c4ce : SUCCESS in 1h 31m 11s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3d25d982e71c409381079dd95be81b4c : SUCCESS in 1h 35m 34s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ffc23d3ba0b44c55abec47a1363b614b : SUCCESS in 1h 39m 51s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2088b61bc634d75b75f68bdf1e7017e : SUCCESS in 1h 00m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7bfcda1973605934fd409ae11314d61cc83e0354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 10:44:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d022e9b823c75939b59c88c889af98271f4f2aa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 13:11:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b77fb9efeb24357bc4d817fc08e4bf0\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/4ea218de2aaf41cda5fd1e98de0ffa59 : SUCCESS in 44m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4abaebb37da43089ea0f629863790e3 : SUCCESS in 11m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2ff8846450a448683938b31d092a2b3 : SUCCESS in 29m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4c333d5319f4dc59fc1b0a2c658f5c6 : SUCCESS in 30m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83eafc3075464c02801c326e2a4f05fa : SUCCESS in 8m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef77215b0bd2493f814b40d28b7b6bcd : SUCCESS in 43m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a553b54717b466db01e50d9fd0eaec4 : SUCCESS in 1h 32m 49s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d7631a83d9ab41db90107c89b8575060 : SUCCESS in 1h 17m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/abbf4c581ab841bcb1db6f9723845e9c : SUCCESS in 1h 05m 17s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/febcee1200f54046a6db85d72221e94d : SUCCESS in 1h 24m 10s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81334bc4309444c29af4c3e36149e3cd : SUCCESS in 47m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5f0c63470204ccf2ada2c036e1f7897a660a250e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 13:11:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4d8b6045dad138bd27adbb331015ce26521815bc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 13:12:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05c237b81dac4fddb025ad12478494ee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e428a7831bcd4508b1412504330ec382 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"494f0e01e8fd5c58239eed580eec607c6db18f1e","revisions":{"494f0e01e8fd5c58239eed580eec607c6db18f1e":{"kind":"REWORK","_number":1,"created":"2023-08-03 12:43:52.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/31/890231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/31/890231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/890231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/890231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/890231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/31/890231/1"}}},"commit":{"parents":[{"commit":"564a58581d3c4991c66904b803951c16b36e5035","subject":"Merge \"Fix ACL sync when default sg group is created\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/564a58581d3c4991c66904b803951c16b36e5035"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-07-26 18:28:29.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-08-03 12:43:52.000000000","tz":0},"subject":"dvr: Avoid installing non-dvr openflow rule on startup","message":"dvr: Avoid installing non-dvr openflow rule on startup\n\nThe tunneling bridge uses different openflow rules depending if the\nagent is running in DVR mode or not. With DVR enabled initial rule was\ninstalled that caused traffic coming from the integration bridge to be\nflooded to all tunnels. After a few miliseconds this flow was replaced\nby a DVR specific flow, correctly dropping the traffic. This small time\nwindow caused a network loop on the compute node with restarted agent.\n\nThis patch skips installing the non-dvr specific flow in case OVS agent\nis working in DVR mode. Hence the traffic is never flooded to the\ntunnels.\n\nCloses-bug: #2028795\n\nConflicts:\n\tneutron/plugins/ml2/drivers/openvswitch/agent/ovs_dvr_neutron_agent.py\n\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nChange-Id: I3ce026054286c8e28ec1500f1a4aa607fe73f337\n(cherry picked from commit ba6f7bf83e6f17048a97f781aa16bf4a643a53d2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/494f0e01e8fd5c58239eed580eec607c6db18f1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/494f0e01e8fd5c58239eed580eec607c6db18f1e"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
