)]}'
{"id":"openstack%2Fneutron~890233","triplet_id":"openstack%2Fneutron~stable%2Fzed~Ie2382540d7c0a8813f093ddf51d82fe530026d71","project":"openstack/neutron","branch":"stable/zed","topic":"bug/2028285-stable/2023.1-stable/zed","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-08-04 09:06:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie2382540d7c0a8813f093ddf51d82fe530026d71","subject":"[UT] Create network to make lazy loading in the models_v2 possible","status":"MERGED","created":"2023-08-03 14:06:06.000000000","updated":"2023-08-04 09:08:18.000000000","submitted":"2023-08-04 09:06:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890233-bug/2028285-stable/2023.1-stable/zed","cherry_pick_of_change":890232,"cherry_pick_of_patch_set":1,"meta_rev_id":"d47220999b1bcaac05d89f2a05354a4a6c3cd97d","_number":890233,"virtual_id_number":890233,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-04 09:06:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-08-04 07:50:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-08-03 17:14:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-08-04 07:50:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-03 14:42:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-03 17:14:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-08-04 07:50:02.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"3262a836737db6691c8ccb7c936a451cfe1d1517","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-03 14:06:06.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"f4f0ce1da0e1551846aab5f55197ec4e5111f050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 14:42:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0233d6f546b546bcb4666126150af3a5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf68212ca5384321b9be292a97741586 : SUCCESS in 31m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7a041d4de824ab99c204d4047c198a1 : SUCCESS in 10m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5447044ae2294a659a4ef87c50b854ef : SUCCESS in 28m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd08d377a1a4498eadb88ac686785b7a : SUCCESS in 20m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba26230093404443ae9e9ac2ddc1e4d8 : SUCCESS in 25m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2daf1dec042848b9bdd48e30193b8e74 : SUCCESS in 12m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c7e3d293876aabc8341161827c797ffabc3fbb4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 14:56:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/397fa3ec14ca400bad4a97fee38991c2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a421a75c03d34c49a124420b1bd02efa : SUCCESS in 48m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fad968b0ed1a4d909143ed5c31eb9f03 : SUCCESS in 20m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d25281f01d08720cd659fac08ea364ecc751100","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-03 17:14:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0fcd2dee7bf2ffe5be2b490f58c9eec395d9e474","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-08-04 07:50:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e64397ba3241bcf5380004afa565338d0dae7c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 07:50:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb64c1b9aa8077eb0e89c12ce4a6aa90ced689ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 09:06:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47c5e36b095242819693bc78eb0537e2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3b58a21eb1147479fec373d6194ac69 : SUCCESS in 13m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ebbd8e4bbb84e7995c2ac1dde4f21f0 : SUCCESS in 23m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ceaf2db7f304968bc537eb7358867df : SUCCESS in 26m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b9a3abddbf34a95a4aa7bc0b8c3a592 : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":1},{"id":"bfd895f18bb2b40581da4f523a2a6b3d91d7d572","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 09:06:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d47220999b1bcaac05d89f2a05354a4a6c3cd97d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 09:08:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0214b490d6094c37a296a68064a29508\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/819ffe1aa9054e37981556347bc21d8f : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e50aef67aa7dfcdd7e35432e0d34b065d3e3266","revisions":{"2e50aef67aa7dfcdd7e35432e0d34b065d3e3266":{"kind":"REWORK","_number":1,"created":"2023-08-03 14:06:06.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/33/890233/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/890233/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/890233/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/890233/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/890233/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/890233/1"}}},"commit":{"parents":[{"commit":"2badef930a17f4adbf8ebbd7c4d8f2229bd67d59","subject":"Merge \"Fix ACL sync when default sg group is created\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2badef930a17f4adbf8ebbd7c4d8f2229bd67d59"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-02 14:27:50.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-08-03 14:06:06.000000000","tz":0},"subject":"[UT] Create network to make lazy loading in the models_v2 possible","message":"[UT] Create network to make lazy loading in the models_v2 possible\n\nIn the test_port_deletion_prevention_handles_missing_port test which is\npart of the classes:\n* neutron.tests.unit.extensions.test_l3.L3NatDBSepTestCase\n* neutron.tests.unit.extensions.test_extraroute.ExtraRouteDBSepTestCase\n\nit is needed to call pecan application to e.g. create network to make\nall models to be ready. Otherwise things like port_forwarding attribute\nin the Port class, which is loaded in lazy mode isn\u0027t available and test\nwas failing when was run in the isolated environment.\nIt wasn\u0027t failing in the gate as there were other tests run by the same\nworker before this one and then it was all initialized properly.\n\nCloses-Bug: #2028285\nChange-Id: Ie2382540d7c0a8813f093ddf51d82fe530026d71\n(cherry picked from commit 14b2f4f60f1f1c00f41545394bd9761f57b6e37f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e50aef67aa7dfcdd7e35432e0d34b065d3e3266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e50aef67aa7dfcdd7e35432e0d34b065d3e3266"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
