)]}'
{"id":"openstack%2Fneutron~890645","triplet_id":"openstack%2Fneutron~master~I1fcc4676dd1b58c40cc8e725d4bc65f4aa73c8ae","project":"openstack/neutron","branch":"master","topic":"dnm","attention_set":{},"removed_from_attention_set":{"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2023-10-04 07:57:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1fcc4676dd1b58c40cc8e725d4bc65f4aa73c8ae","subject":"DNM test https://review.opendev.org/c/openstack/devstack/+/890644","status":"ABANDONED","created":"2023-08-07 12:21:40.000000000","updated":"2023-10-04 07:57:52.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d72d8fc2a3cd2d5404c875266b49dcc15fb0137e","_number":890645,"virtual_id_number":890645,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-07 14:44:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-07 13:12:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-07 14:44:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-04 07:57:52.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"bcf3e1cf18a865e97257f343aa79516c3720033c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-08-07 12:21:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a973956656ca2fc17bce06d0285f511a4b37311","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-08-07 12:22:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b15fa2f74282231c8cec20ec67b7e0a9dcf58d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 13:12:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b18fdbfc31d74bc89e788d7504cf2c6b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/19237e260acd4616988d8302ddefdde5 : SUCCESS in 20m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a42816ce99e34373afc17d59cfd5646b : SUCCESS in 49m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a0fb686a1aec40e18bfbfbbf2f5731c8 : SUCCESS in 24m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b326e589bd5b13eee4f197c2dcf4b200824825a0","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 14:44:27.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/149bad629d804c7aabc76b9a82c8451b\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/3d6b9af8b8ac44a4ac328a0db8795f12 : SUCCESS in 1h 32m 56s\n- neutron-tempest-plugin-openvswitch-distributed-dhcp https://zuul.opendev.org/t/openstack/build/30bb4e80f7d64905a817f32f68f621ba : SUCCESS in 1h 23m 42s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-distributed-dhcp https://zuul.opendev.org/t/openstack/build/9ddb336266544221b3123410ef8846f3 : FAILURE in 1h 26m 19s\n- neutron-functional-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/6f89e34ca3bc4ba197e48e2ddffee1a4 : SUCCESS in 54m 22s\n- neutron-fullstack-with-uwsgi-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/0eb010ca46ca42d29678b978096211d6 : SUCCESS in 2h 19m 06s\n- neutron-ovn-tempest-full-multinode-ovs-master https://zuul.opendev.org/t/openstack/build/e88e37bccb724661b62c1fc32b297558 : SUCCESS in 1h 09m 22s\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7f9109d87e642de9f3d8a05190e052b : SUCCESS in 1h 05m 23s\n- neutron-ovn-tempest-ovs-master https://zuul.opendev.org/t/openstack/build/dc1815aa92b24625be65607604962328 : SUCCESS in 1h 24m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/78ab24a85cad4030a681dc6d650d3a35 : SUCCESS in 1h 18m 23s\n- neutron-ovs-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/f9050bc28f554035ad15e6929bdb142e : SUCCESS in 1h 16m 16s\n- neutron-ovn-tempest-with-uwsgi-loki https://zuul.opendev.org/t/openstack/build/c23dd504302c4e27a2b30720148af484 : SUCCESS in 1h 37m 46s\n- openstack-tox-py310-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/a0eb69276ff94bbaa8c0953cd05df560 : SUCCESS in 29m 17s\n- openstack-tox-py310-with-oslo-master https://zuul.opendev.org/t/openstack/build/625a8a3fd41b41aab9927f9751aee3c5 : SUCCESS in 35m 41s\n- openstack-tox-py310-with-ovsdbapp-master https://zuul.opendev.org/t/openstack/build/6bc8d9f354ed4a23824ba38e4f80ef10 : SUCCESS in 36m 08s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/04a68ab57a5f4100a3a30e829902a093 : SUCCESS in 25m 16s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4e4e34bb3ec74b04ba52090ebb873112 : SUCCESS in 1h 03m 01s\n- neutron-functional-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/6a824c96c52b4018a079ed6056487f0f : SUCCESS in 55m 09s\n- neutron-functional-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/cab8969fbe4a4f27b648e0004c33279f : SUCCESS in 46m 09s\n- neutron-functional-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9297f8051d3c4900989290c699b6e6da : FAILURE in 52m 21s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/afd04f2856684ef3baccf6eaf57dbd1c : SUCCESS in 2h 07m 42s\n- neutron-fullstack-with-uwsgi-fips https://zuul.opendev.org/t/openstack/build/47a7659ff37b431d8318a29134920d3b : SUCCESS in 2h 01m 16s\n- neutron-fullstack-with-pyroute2-master https://zuul.opendev.org/t/openstack/build/16d7600839bc47889201462cc097526b : SUCCESS in 2h 02m 41s\n- neutron-ovs-tempest-slow https://zuul.opendev.org/t/openstack/build/29b4e9c0ffc843d9b17cc8a6b9b8b9cb : SUCCESS in 1h 14m 37s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/de4eb68a4f114d06a7999216963caef5 : SUCCESS in 2h 05m 30s\n- neutron-ovs-tempest-with-os-ken-master https://zuul.opendev.org/t/openstack/build/0068c57d0b1b4e5480f93cf5b7f5a9b1 : SUCCESS in 1h 19m 01s\n- neutron-ovn-tempest-postgres-full https://zuul.opendev.org/t/openstack/build/f85b3364e4834dd28bb8a30b9e0736a3 : SUCCESS in 1h 21m 42s\n- neutron-ovn-tempest-mariadb-full https://zuul.opendev.org/t/openstack/build/bac84b56c6044e7990212aa9370aca86 : SUCCESS in 47m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-master https://zuul.opendev.org/t/openstack/build/2a08fc3554d14a558102abee40683c08 : SUCCESS in 1h 01m 17s\n- neutron-ovn-tempest-ovs-master-centos-9-stream https://zuul.opendev.org/t/openstack/build/c578c37dfba04934937a8bb567d63bdf : SUCCESS in 58m 52s\n- neutron-ovn-tempest-with-neutron-lib-master https://zuul.opendev.org/t/openstack/build/24a383c9da184704a9371141f4e877a1 : SUCCESS in 1h 27m 04s\n- neutron-ovn-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/379b5a26e10d43b680a419928bf9d587 : SUCCESS in 58m 20s\n- neutron-ovs-tempest-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d1e2232ad36945af977289b7b4a29d64 : SUCCESS in 1h 04m 27s\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/4a5ea157a2fa40e2bc4075fd441499b3 : SUCCESS in 1h 11m 19s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/bf8ce7d0d81641beacaa98eedab6f5c8 : SUCCESS in 1h 09m 39s\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/72f9360824dd4e77b0ad2af6d29631d9 : SUCCESS in 55m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/02abab5a50e74692947ded20d13c094b : SUCCESS in 55m 13s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/00a97dc97e184f52ad03a546fb6dcd18 : SUCCESS in 1h 17m 04s\n- devstack-enforce-scope https://zuul.opendev.org/t/openstack/build/3edf4df7a5204604a508f954a237a2f1 : SUCCESS in 17m 26s\n- openstack-tox-py39-with-oslo-master https://zuul.opendev.org/t/openstack/build/f9dce6be0e8d4b5faf2e47b9ef8d92ca : SUCCESS in 42m 18s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/756d603e0a4946e88daefa2e82e62bff : SUCCESS in 31m 42s\n- neutron-functional-with-oslo-master https://zuul.opendev.org/t/openstack/build/4ae1e16b1f5c406e8a5ffafeb44b8e69 : SUCCESS in 54m 04s\n- neutron-ovs-tempest-with-oslo-master https://zuul.opendev.org/t/openstack/build/5679957657e7494ca778fa144dac210e : FAILURE in 1h 09m 49s\n- neutron-ovn-tempest-ovs-release-with-oslo-master https://zuul.opendev.org/t/openstack/build/81f2059ea2414f8c9d737365bcc5b249 : SUCCESS in 1h 15m 41s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e6dbe00309de42ba8ef0a952c1350678 : SUCCESS in 41m 06s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f0c967695a134370af74913983e87f43 : FAILURE in 1h 13m 31s","accounts_in_message":[],"_revision_number":1},{"id":"4ca02def310b7286afd319d7bb5ad70532b8a9e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 14:44:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3b8ec3a5f6f4842b3c7ec6ae5c13af3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a1f3f66cc7714b97b65e8ee5d98f1636 : SUCCESS in 1h 24m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e16768220e14412ab6e62bc2c5cde7c4 : SUCCESS in 1h 44m 30s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/28cb36b61b054e99924428093aded241 : SUCCESS in 1h 28m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3e6e76d9a2134f7693631f7602804548 : SUCCESS in 1h 29m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0249beaf17f1403caf66308a2b4c1534 : SUCCESS in 32m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa08c9c1bc7a43a28af3bae4a332fe3e : SUCCESS in 41m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/177490e7e3e5410da92043f5aafca93f : SUCCESS in 11m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04c2f8ec032a4bb6b7a9f3d51c29fbe8 : SUCCESS in 37m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4dc534511024fcd98f888b00e92867a : SUCCESS in 27m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58d810267860446da231dbbae20fc54c : SUCCESS in 25m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1c54bc882d44487815b0e86758cab1e : SUCCESS in 11m 24s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/5541295ab4a54e8a81cddcb73aa89104 : SUCCESS in 26m 18s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/47975bbfc4b74875b04eb4740d8f50a1 : SUCCESS in 1h 06m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a64542f5fb649a6bb8fc89ffb1f1c41 : SUCCESS in 2h 01m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5cb770e22f604979bd586ef7ee8f556a : FAILURE in 1h 29m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/391d813d28404c3db47a18eceec68718 : SUCCESS in 1h 04m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0f2a27c4912646b0811357b0372eca1b : SUCCESS in 1h 33m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bec3e03b750b41dfa891275067a32ad1 : SUCCESS in 1h 13m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/99b208c4909b4c92bfcb815b49f975a5 : SUCCESS in 48m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b68f1e29ba7e490e85cdc0152d8927f5 : SUCCESS in 1h 28m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2d99626ec58649acb326a21f67c61f88 : SUCCESS in 1h 36m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/63439af1386e45c3a38fd4691e565c44 : SUCCESS in 1h 14m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d72d8fc2a3cd2d5404c875266b49dcc15fb0137e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-04 07:57:52.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3487e5b4b325428328c0629f3b8f9c9c537f9d8","revisions":{"c3487e5b4b325428328c0629f3b8f9c9c537f9d8":{"kind":"REWORK","_number":1,"created":"2023-08-07 12:21:40.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/45/890645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/890645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/890645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/890645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/890645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/890645/1"}}},"commit":{"parents":[{"commit":"2ea0d0bd1bb00558d12cd4e834af33d1a71f8bac","subject":"Merge \"Update Cirros to 0.6.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2ea0d0bd1bb00558d12cd4e834af33d1a71f8bac"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-08-07 12:18:06.000000000","tz":120},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-08-07 12:21:37.000000000","tz":120},"subject":"DNM test https://review.opendev.org/c/openstack/devstack/+/890644","message":"DNM test https://review.opendev.org/c/openstack/devstack/+/890644\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/890644\nChange-Id: I1fcc4676dd1b58c40cc8e725d4bc65f4aa73c8ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3487e5b4b325428328c0629f3b8f9c9c537f9d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3487e5b4b325428328c0629f3b8f9c9c537f9d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
