)]}'
{"id":"openstack%2Fneutron~890795","triplet_id":"openstack%2Fneutron~master~I54dec6a6a6ab0989205898bab509ec69151b6dfb","project":"openstack/neutron","branch":"master","topic":"feature/more-logging-for-lrp-scheduling","attention_set":{},"removed_from_attention_set":{"35969":{"account":{"_account_id":35969,"name":"Tobias Fischer","display_name":"Tobias","email":"tobias.fischer@cloudandheat.com","username":"tobfisch"},"last_update":"2023-12-02 19:46:53.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2023-12-02 19:46:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I54dec6a6a6ab0989205898bab509ec69151b6dfb","subject":"Add more logging for lrp scheduling on gtws","status":"MERGED","created":"2023-08-08 13:40:12.000000000","updated":"2023-12-02 19:48:08.000000000","submitted":"2023-12-02 19:46:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"890795-feature/more-logging-for-lrp-scheduling","meta_rev_id":"442cd1e21def55002c14e2ae4374731962ca2b86","_number":890795,"virtual_id_number":890795,"owner":{"_account_id":35969,"name":"Tobias Fischer","display_name":"Tobias","email":"tobias.fischer@cloudandheat.com","username":"tobfisch"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-08-09 09:20:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-02 19:46:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-01 22:06:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-08-10 09:28:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-01 22:06:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-08 14:23:13.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-08 14:43:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-08 16:18:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-08 21:47:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-08-10 09:28:51.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-12-01 19:00:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2023-12-01 22:06:23.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"67b27ec87d75e3c7ad15751a271b0b197af04123","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35969,"name":"Tobias Fischer","display_name":"Tobias","email":"tobias.fischer@cloudandheat.com","username":"tobfisch"},"date":"2023-08-08 13:40:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2faf18b4fc371b5f8374844cfe699a49708eba78","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-08 14:23:13.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/95/890795/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"7fd1d8250e212842fbd98da7932592cbe45e962c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 14:43:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/265d885009c44413ba423bf66e3445fd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d55a6431ee341f38ff4911184d34b13 : SUCCESS in 58m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31c74a8e8d1e4052a3b3ed1741c65ac0 : SUCCESS in 1h 00m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a3480a4c7b314422b15f35981792e877 : SUCCESS in 1h 01m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66399e0a98d977d618dadfe687f2c82b062fc206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 16:18:12.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1873a032b52d4c3080ece8040a82fd2a\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/680b46c563404052aedfe29566c41330 : SUCCESS in 2h 06m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e04ecdb6308f4de2a3e355cc8126d168 : SUCCESS in 38m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3345a52717fb4e6e95502c4c75861854 : FAILURE in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7117344a2e89481ebdf7ff95fd581b79 : SUCCESS in 33m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/214de80781684368a44307f34ca13a62 : SUCCESS in 31m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69ec18b98635499e9b2a6cd5f595da97 : SUCCESS in 34m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/295bb558557e4ef9bd8ab11d6b6248f4 : SUCCESS in 9m 14s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/6d2e161b096747288da81d535f6f78cd : SUCCESS in 27m 08s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a16e8d6e96249fa826e0aeb8802e66b : SUCCESS in 1h 05m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c7e5ae3947624c8688ecd892289b9474 : SUCCESS in 55m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/541c68fb8c404f65a9b26d8d924f177e : SUCCESS in 2h 20m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4be154d4c92a4a0abe4af85bf20f20e2 : SUCCESS in 1h 33m 48s","accounts_in_message":[],"_revision_number":1},{"id":"bc6814d8e4cd8a142fde2cac4e644ea5ee464231","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35969,"name":"Tobias Fischer","display_name":"Tobias","email":"tobias.fischer@cloudandheat.com","username":"tobfisch"},"date":"2023-08-09 07:52:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e60a5516c2b1fec29e5d723e9a4fdcb3c842534d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 08:58:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c9de57c8fb14a9db20f73ec1638ccf1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/081f9eef382c45edbd817dc0c5d20eab : SUCCESS in 1h 03m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8a48520380344a7bdd3c6a13529a31f : SUCCESS in 1h 04m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/54dc989392654a83925723adb3a09411 : SUCCESS in 1h 00m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"60b1f3a03af40dbda16216c8df89a0e600644989","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-09 09:20:19.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/95/890795/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"6b4a756d7ce30385c6e2075f6adb776c8972d6c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 09:49:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dbe2eb8119c44d399b61a9a12fc3d67\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dd45ca9bd37347358b3c5385876ab100 : SUCCESS in 1h 55m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d0349f0324a42deb889b2e46ea5327a : SUCCESS in 40m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1ef8525373a4e0bab20fb915d433c44 : SUCCESS in 10m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/784932001f6c4f0f941036a02d6226ec : SUCCESS in 25m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64a0306838f54a378b957e173a3d7904 : SUCCESS in 23m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b877a7b5ba84c14b87442ecd6fc91d1 : SUCCESS in 23m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f3707cb4fbe4a5db7554cf36d4094c7 : SUCCESS in 10m 55s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/daf72c7b197345e6bc3c115d5c69f7c9 : SUCCESS in 20m 42s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e82824b0984d47e5b4270b92e7b88c21 : SUCCESS in 43m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/13ee8905cf634f14bde9ccb9dd565797 : SUCCESS in 1h 03m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b2ecc0784a884d95bd5e57f9a62a669b : SUCCESS in 1h 38m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b42a88c9b6b6491a826457b09011d049 : SUCCESS in 49m 54s","accounts_in_message":[],"_revision_number":2},{"id":"4997650ea65b3442cea00f123b867d4cba192246","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-10 09:28:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"70ea09a8f43a55d2129136221199c8930fde79d5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-01 19:00:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a869e225f905abc5bccd6bdba241a95600e324f6","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-12-01 22:06:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1adc2a6651d96f832907cf0018e7c8bbba37d1c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 22:06:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"baa5a965facd7e71865d121816c3c77ad0b36e83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 00:55:58.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a289c898672a47a28298d6374ae7638f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/36fc2b9c4db942a8b2ff545eefe4744b : SUCCESS in 1h 56m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e063d2ddc86433a8b96c199c1d1cdd9 : SUCCESS in 11m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdf4db7526364a37931c59aa10d5ad60 : SUCCESS in 22m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2ceb469bf9d463ba269fc5dd490d44f : SUCCESS in 24m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/864de32aa5f64b5d896c9e6cf2438646 : SUCCESS in 11m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/500d08785f9a4d64a1acc527dc123402 : SUCCESS in 52m 50s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/beeaba9fa51d4871abc0e496ad8fda42 : FAILURE in 43m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ad7b557a553d4b948af2d3148e14b485 : SUCCESS in 1h 07m 33s","accounts_in_message":[],"_revision_number":2},{"id":"22447ca6ac89d8f6f83e46f8f7b8571dcd478fb3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-02 15:59:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f84255513995532ddc88440d3e7a5e35f82dada1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 16:22:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cd40022ab35414da56e5878f969eefc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8bea9e0de6334727bdd69f1ddc34cf7f : SUCCESS in 21m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/507cdcadcd1b4e3abe3d5d15757b7ece : SUCCESS in 19m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b359bf25cb144e0fb31f68f9a40bedbf : SUCCESS in 19m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"076d15f14e5e817d16701225ce5e16791e99b64d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 17:45:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00d658a530694418b0c95548b2e35dcb\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/33598501fd9446afa590ccf12fd10022 : SUCCESS in 1h 42m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edf2950984c84f5d8c6aefc9bbeeaae6 : SUCCESS in 50m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c88c5b485c24d1bbccb10ef1b974958 : SUCCESS in 11m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c5ea78a9cec47a5b60cf9b2042f1030 : SUCCESS in 28m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b89a986fa0cc4ebda686c25134c8eadb : SUCCESS in 23m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd3e862c6e4949fcb1b01be88490b072 : SUCCESS in 11m 48s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/123b431e5191499d82d497973adf8d58 : SUCCESS in 20m 15s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fe0113442d1743aa9fc3b4e2822c8d0e : SUCCESS in 1h 01m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff258a03e01d460e852576bf6c54f3f1 : SUCCESS in 59m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8e32efa7b6df4903a73baddbac40eb6c : SUCCESS in 1h 00m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dfb60188a45943d4b7576ba52582db8b : SUCCESS in 1h 39m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b7bd71bf105a47cc8081a79783b9cf7f : SUCCESS in 1h 18m 36s","accounts_in_message":[],"_revision_number":2},{"id":"bcf97d47ef25b5d08bb8258d78e0066b25c77d64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 17:45:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"da0898e20db4913dadf78f156f99dfd831a56ae9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 19:46:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bc388f3ff594a16aa11c7fae36d9ac4\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/27d4cbd1ad834f2295cab4887e64ecd7 : SUCCESS in 1h 25m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3de20475fa3419e8bbbe531ec405e11 : SUCCESS in 11m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eee00a6fb4d44be7bb0bd05f2bb1272b : SUCCESS in 26m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/060f3664b6dd4e20af77280c91d0fe2c : SUCCESS in 17m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/233c973909bd4ee4947a3f307e783e2f : SUCCESS in 8m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3493510f294e4f799d06c10074f70f0c : SUCCESS in 45m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9547e1d159574cfdac887b2648e10672 : SUCCESS in 41m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/938fd173b2bc49fd853b91300c097274 : SUCCESS in 1h 54m 50s","accounts_in_message":[],"_revision_number":2},{"id":"301e0d48b0e2588a57a584868a59180dc54e5386","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 19:46:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"442cd1e21def55002c14e2ae4374731962ca2b86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 19:48:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bf8f7244d3143d492fbe62597005bd5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d1c7e80bade4b01b6f3cb7222c14657 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ccf2efdd3eaf3b3a5e56ef53a106ddca9fc8d771","revisions":{"cb5fef996423f5c543484965ba6f63278a49d737":{"kind":"REWORK","_number":1,"created":"2023-08-08 13:40:12.000000000","uploader":{"_account_id":35969,"name":"Tobias Fischer","display_name":"Tobias","email":"tobias.fischer@cloudandheat.com","username":"tobfisch"},"ref":"refs/changes/95/890795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/890795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/890795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/890795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/890795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/890795/1"}}},"commit":{"parents":[{"commit":"c453813d0664259c4da0d132f224be2eebe70072","subject":"Merge \"Add NET_OWNER_MEMBER and NET_OWNER_READER policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c453813d0664259c4da0d132f224be2eebe70072"}]}],"author":{"name":"Tobias Fischer","email":"tobias.fischer@cloudandheat.com","date":"2023-08-08 13:35:53.000000000","tz":120},"committer":{"name":"Tobias Fischer","email":"tobias.fischer@cloudandheat.com","date":"2023-08-08 13:39:38.000000000","tz":120},"subject":"Add more logging for lrp scheduling on gtws","message":"Add more logging for lrp scheduling on gtws\n\nCurrently it is not entirely clear from the logs of the neutron-api,\non which gateway chassis and with what priority a Logical Router Port is\nscheduled in OVN.\nThis commit introduces logging to have better visibility where a LRP is\nscheduled and which gateway gets what priority.\n\nChange-Id: I54dec6a6a6ab0989205898bab509ec69151b6dfb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb5fef996423f5c543484965ba6f63278a49d737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb5fef996423f5c543484965ba6f63278a49d737"}]},"branch":"refs/heads/master"},"ccf2efdd3eaf3b3a5e56ef53a106ddca9fc8d771":{"kind":"REWORK","_number":2,"created":"2023-08-09 07:52:04.000000000","uploader":{"_account_id":35969,"name":"Tobias Fischer","display_name":"Tobias","email":"tobias.fischer@cloudandheat.com","username":"tobfisch"},"ref":"refs/changes/95/890795/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/890795/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/890795/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/890795/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/890795/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/890795/2"}}},"commit":{"parents":[{"commit":"c453813d0664259c4da0d132f224be2eebe70072","subject":"Merge \"Add NET_OWNER_MEMBER and NET_OWNER_READER policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c453813d0664259c4da0d132f224be2eebe70072"}]}],"author":{"name":"Tobias Fischer","email":"tobias.fischer@cloudandheat.com","date":"2023-08-08 13:35:53.000000000","tz":120},"committer":{"name":"Tobias Fischer","email":"tobias.fischer@cloudandheat.com","date":"2023-08-09 07:51:08.000000000","tz":120},"subject":"Add more logging for lrp scheduling on gtws","message":"Add more logging for lrp scheduling on gtws\n\nCurrently it is not entirely clear from the logs of the neutron-api,\non which gateway chassis and with what priority a Logical Router Port is\nscheduled in OVN.\nThis commit introduces logging to have better visibility where a LRP is\nscheduled and which gateway gets what priority.\n\nChange-Id: I54dec6a6a6ab0989205898bab509ec69151b6dfb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ccf2efdd3eaf3b3a5e56ef53a106ddca9fc8d771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ccf2efdd3eaf3b3a5e56ef53a106ddca9fc8d771"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
