)]}'
{"id":"openstack%2Fneutron~890799","triplet_id":"openstack%2Fneutron~master~Id7bf5a6aa547795ba78724eed59ba9d4fb74f758","project":"openstack/neutron","branch":"master","topic":"bug/2030773-ovn-db-sync-router-ports","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-08-15 09:45:47.000000000","reason":"\u003cGERRIT_ACCOUNT_6773\u003e replied on the change","reason_account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-08-10 15:46:56.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2023-08-14 14:38:07.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}}},"hashtags":[],"change_id":"Id7bf5a6aa547795ba78724eed59ba9d4fb74f758","subject":"[OVN] ovn-db-sync check for router port differences","status":"MERGED","created":"2023-08-08 14:25:47.000000000","updated":"2023-08-15 14:12:04.000000000","submitted":"2023-08-15 14:10:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890799-bug/2030773-ovn-db-sync-router-ports","meta_rev_id":"10313a2a835ed6f99320c07d226d02ab9cc77dff","_number":890799,"virtual_id_number":890799,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},{"value":1,"date":"2023-08-09 14:52:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-15 14:10:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"recommended":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"all":[{"value":1,"date":"2023-08-10 14:28:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-10 15:46:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2023-08-14 14:38:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2023-08-14 14:38:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-08 14:40:51.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-08-08 15:09:16.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-08 15:46:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-08 16:05:10.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2023-08-08 17:41:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-10 14:28:41.000000000","updated_by":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"reviewer":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"state":"REVIEWER"},{"updated":"2023-08-10 15:46:56.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"b0949a58e1e27a84f680c5bdf7d7852b9092c881","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-08 14:25:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68ad43db91273137cab74946fa56886409a5b062","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-08 14:40:51.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"81f6bfbb0eac9c552a952b431281ff93c1b1a851","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-08 15:04:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7590896e61d64e0e492896e89d9babed3264ec4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-08 15:06:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"df53ccbe5dc1d239639a79ad4600c6ca5fb282df","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-08 15:09:16.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/99/890799/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"2ca536024b782176daf6b4ce54a2bee0c7ffc9a9","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-08 15:18:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b624462d60b235e69f84b00877bfa1d206194df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-08 15:22:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a6b2b876558d61877c3c7abcdd36e3e54c495a78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 15:46:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35d7c6ebdcb2405f99493287e2fa97b8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b044c6c514646adb4d18cb3f1cbd548 : SUCCESS in 22m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/569a4dc6987c46cd9a2870906fb118f1 : SUCCESS in 22m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f8328439086e40e78fe107e8646152a5 : SUCCESS in 22m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"539f90d0b58ec0fc6531d85cb43a27d53e3ec30f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-08 15:52:15.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/99/890799/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"59e2493067971f086dcdf8224a75f3b33d671bbd","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-08-08 16:05:10.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f9a29951c6d98fd8234828ee8253c2ecdb352aa3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-08 16:35:18.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/99/890799/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"5f9748e35d655e0bdbabce2a8b96ff18011a461d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 17:41:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a113e019546f4e7a966c7ca8d73bfcc7\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/20b7fc9fdf1846e3a9bb558ea9275372 : SUCCESS in 1h 33m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9e6d3d791814756b75a25d8fb7c05d2 : SUCCESS in 40m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/909db9ec582a4b54becbf186e9ee5024 : SUCCESS in 10m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aab34321d1744a79a05d6de2e22bcee3 : SUCCESS in 23m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0c4e27d71154aac95cda11acd0cdebf : SUCCESS in 27m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2864eaf04f74071b287da5dc5b88b90 : SUCCESS in 24m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bbda7c942eb4134921ff83003e14c1d : SUCCESS in 11m 57s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/bedefd7d60914981ae6e0a38f09a90d0 : SUCCESS in 26m 02s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6539f1e548d642fb863f572a9a0ffb02 : SUCCESS in 52m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/004a5132aaa14d29b25127070b67dd0e : SUCCESS in 1h 01m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6352abb750bf4999a0e80b09a7be5539 : SUCCESS in 2h 10m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/40bb2d71b3da46a0a35af1c84514742d : FAILURE in 1h 22m 22s","accounts_in_message":[],"_revision_number":3},{"id":"9390a35a7d2d1dcf740404af01a9a83da1b352d0","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-09 08:31:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"66f5414fabc6cc12e0ca9c20dd146d3e0a0c994a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-09 08:32:01.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"11b9d2cb0e829456486f43effedb8083ad2ecd1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 09:31:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3bf51e238954d45a529aa41e128e2a3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3ca823a7f63400db1eaeea2be1bd088 : SUCCESS in 57m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b49d838eff54439b2796a210c4c504c : SUCCESS in 57m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c7fad5ea03eb480398413456805e62cb : SUCCESS in 51m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"59ce9b0178ab5e7fcff13348234bef9beace052f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-09 10:01:54.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/99/890799/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"ecfd007f99690382aee4473ae2a582474acce330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 10:46:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0634b5275c564ddaa7389a718952de6b\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/75726f75414342cb81579f98b81c613e : SUCCESS in 1h 40m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/891b75df20a94238a6bb6ad134737eb8 : SUCCESS in 37m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10c3d39b40cc46db8c6e403525d9ee12 : SUCCESS in 9m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4068691316a74eaab75a2601c085079d : SUCCESS in 28m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b4c3a3ab6a44e959d246ffbd9bb5103 : SUCCESS in 26m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9fb0f9a25b8436a951e57975251178f : SUCCESS in 28m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53476321dbbe4663a4f95d4f52b36392 : SUCCESS in 9m 19s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/7a5fdb157e314df0bff506dee37cab03 : SUCCESS in 25m 35s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/86f837d499b34fa580229f21f9da4d52 : SUCCESS in 1h 03m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9ae78b0f2ee841f586381a6cb1b1b8d9 : SUCCESS in 1h 20m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5b2404dc50e448bfa4b001bc66b21095 : SUCCESS in 2h 08m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fde08971af4b4c5d8e2ffdaa9fca0f12 : SUCCESS in 1h 24m 16s","accounts_in_message":[],"_revision_number":4},{"id":"a46476955c5948882d308afd7736bcbae1373cea","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-08-09 13:33:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"807300325802b342437331f4760b060981610108","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-09 13:59:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"36ee620bd7f02e748f1ccd26c2e40efdff114723","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-08-09 14:05:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ca1739fcb5b62a8b69cefed698c0fdfebb929729","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-09 14:10:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"f77bf917ad6796d74f59210234d27f6a6f841bea","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-09 14:52:07.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/99/890799/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"8881bcf7efd9183d7f55c9ade088fde4c863e9d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 15:02:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d75670413124240b78fb8d966eab6b5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f4a2bacd800442f18aab203cba89edcd : SUCCESS in 26m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e736ace3f4cf45a994289c3375eec7c0 : SUCCESS in 50m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c6a7e75fa6f54c2b8c05719cec59ac82 : SUCCESS in 20m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9953a86bd0e1ed42e528e4604269773f4c48087b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 16:31:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc7a8a4ef4a94e83a808e842a0357333\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8841b2b160c44a3e83be69936b03b7d1 : SUCCESS in 1h 45m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16d112694e16463ab8a5ded87b30b57a : SUCCESS in 32m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f21fd3bc1fd49679aaafd0f15498cce : SUCCESS in 10m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2161d7b1aa12465fa842ac2f8704d869 : SUCCESS in 28m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4afc5366fba43d58bde846a059221b3 : SUCCESS in 22m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/deff93846b6e4e1fb6c3a7e97c6b3a82 : SUCCESS in 26m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a331fa61ef6d4683b620e7e400ccce5b : SUCCESS in 10m 55s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/174b174837554168aae8bc4534b0f8d5 : SUCCESS in 29m 22s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ad6dc4e671a42678163f297118fe216 : SUCCESS in 1h 07m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1be1741a8ab8464f9ca2603b91ab5f47 : SUCCESS in 51m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ff8bdacd432c4043b28935f339eb4c64 : SUCCESS in 2h 15m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f0668f46f1ce461e9b77efbc4f887fe5 : FAILURE in 1h 11m 01s","accounts_in_message":[],"_revision_number":5},{"id":"5956694ba3f25872865dcd7f7a0d8c59f35efb98","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-10 09:01:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2a48273f0c1674387ba8d80dd4d5e55a71283849","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-10 09:02:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0de3615390d7a1529a911d3d74a56d538ab44e44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 10:05:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c5c0f702eae466287c7732847d928df\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/693ac0763bbf435ba20561a51d57548c : SUCCESS in 59m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/291ac7b2eef24c61b1396c9321ee9ba1 : SUCCESS in 59m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/64d74fac475a4f76ace21d37fdca2515 : SUCCESS in 1h 02m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7840266904fe32410a02643fd1e2119059c431bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 11:16:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0362b59ffb494b9d8c7567896b38881e\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c971b5c7ca8843e2a604974fced3c2d2 : SUCCESS in 2h 11m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/071668f29f824c69888b940e742e3956 : SUCCESS in 36m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6be53b4ea5524c5da0423a79aa2a5aa7 : SUCCESS in 8m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16dcfb5a26274fefa235a90a7d29b841 : SUCCESS in 28m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c82e439c0b40498bb5aeb787e1a08e60 : SUCCESS in 28m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/109a2ea168a34cc09bf28bf4cdf39fd0 : SUCCESS in 27m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7765aaf00ae45deb153e3e27075297d : SUCCESS in 11m 25s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a041c57137fc4c07954e223c4ac2d3aa : SUCCESS in 21m 34s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1dc44f0043914b179b7adfa786214b9c : SUCCESS in 1h 02m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/431cce94839d4c85881bc535050fa8e6 : SUCCESS in 1h 06m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/97e5b4af66b844869c6027f0a060e5b2 : SUCCESS in 1h 57m 36s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cf4ad5e965074723b571578a7efe87a8 : SUCCESS in 1h 13m 37s","accounts_in_message":[],"_revision_number":5},{"id":"9028ae57f1cddbf51aecbd03f4017a981df0add8","author":{"_account_id":19234,"name":"Alexey Stupnikov","email":"aleksey.stupnikov@gmail.com","username":"astupnikov"},"date":"2023-08-10 14:28:41.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"abd01cec221c35c9d20babfcb321282685a9dd6b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-10 15:46:56.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"dd0085d54c1e4f38261d15732b5ec3f22a97c074","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-08-14 14:38:07.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bcdaef7bf607cbe38aa4b7b3cceb76b67726f314","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 14:38:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"64b04c02e3c1c4c230656b50aec39806e9f6ef10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 17:06:16.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48a6b574be6b4b698ce6866609dd23a3\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5135952ce41244a6b3ce3103ffe5611f : SUCCESS in 2h 00m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e72319f5b2ca425b86e4414cf1eec4ff : SUCCESS in 11m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f1f67f304794bb7af7458519c47d872 : SUCCESS in 27m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c8fefba2121431c80d4a643a28cd065 : SUCCESS in 25m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7b895b2cdfd4a95a6e2e0e6f8bbe4a6 : SUCCESS in 25m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a1e1383ba01473dadc0ba20867f0d87 : SUCCESS in 11m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b7dd29f8894d431d840a5894cbf5c2c0 : FAILURE in 48m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/186558b337b24b02b606876876aed555 : SUCCESS in 55m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e4fb6e1239ba4170a8d51b29a03a9731 : SUCCESS in 59m 56s","accounts_in_message":[],"_revision_number":5},{"id":"c086eaae6ac70bb761911d813373c010e888ac2f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-15 09:45:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7201efba0934dae81150f97ef5e8297557a99a26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 10:22:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb9956ae37c84b55992e9a6c8b4be4a0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/757f7a3a0b7d4910809b8fd386315f98 : SUCCESS in 28m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe436769f9f14e979b842b1df02a1b98 : SUCCESS in 28m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10bd09bb88294a92978f75d2fe83c191 : SUCCESS in 36m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c095d2957479349e05eb79ff238289856c95ce01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 11:56:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35983afb75764acaa33d055197733499\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ece3f90d8b9e498bba361d3ace0a8044 : SUCCESS in 1h 34m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/051710eced2e4736b8558f97266c7088 : SUCCESS in 31m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0a25daa1bc9487ead25bea5b07ffbe3 : SUCCESS in 10m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a41339d913a244c4ae3bfbe88893a546 : SUCCESS in 25m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a27e7d19e4524b9481ee8f99cc695372 : SUCCESS in 22m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7354461064f474cbf51a8ff9603dafd : SUCCESS in 20m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5d5dbc9977f47f397861bfcc9d78e06 : SUCCESS in 8m 58s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/67bb6f3b78c0402e810aef841f351d74 : SUCCESS in 25m 36s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fdb885d7f7664a20bd73e49e4b96b556 : SUCCESS in 1h 00m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8a93f1097a324fcab52708e8cfbca54d : SUCCESS in 45m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ccfd278c7dfa4ffbbfa836599e0e6cc5 : SUCCESS in 2h 05m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6f1abfd117584d33826b596c9e02fb8d : SUCCESS in 1h 22m 15s","accounts_in_message":[],"_revision_number":5},{"id":"2ad5ad507428968428aaaf6b96044c6becdf2c3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 11:57:18.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d235981515ae132442f8aadfd71e670cb1843972","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 14:10:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41168d68f2344fd2997debec77bb1a25\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b5f9447332d84a928613703371d1c7c1 : SUCCESS in 2h 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5475d0a08d904f0c9cd2627802bb0fbe : SUCCESS in 10m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61d6de8728c840cd8e516a24850e39ce : SUCCESS in 21m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50b338c8d55f402197e24472efbabdad : SUCCESS in 25m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1df6c003a0294a7bb4f580ba8b2cbb29 : SUCCESS in 22m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/915a9f0dedb943678fe1461e5ca86701 : SUCCESS in 9m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/980f2d343a0e477c94f2559c616f316f : SUCCESS in 45m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/192d8bc49db343028ef27a706663008d : SUCCESS in 44m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4918fb9dd4234371b7ed7f0a296651f0 : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":5},{"id":"e5bd709ea2226d98424d1146e630acdb8dfa7d7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 14:10:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"10313a2a835ed6f99320c07d226d02ab9cc77dff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 14:12:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4260f7db7bae424880213ca09391fd95\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50f33846bbcb4fbabf9a47c7d1e6706c : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4693836a1b58b477298e51cb47622222e3556752","revisions":{"8928e5c4f6f1036b0bc8770d0891e7d122a529f2":{"kind":"REWORK","_number":1,"created":"2023-08-08 14:25:47.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/99/890799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/890799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/890799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/890799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/890799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/890799/1"}}},"commit":{"parents":[{"commit":"c453813d0664259c4da0d132f224be2eebe70072","subject":"Merge \"Add NET_OWNER_MEMBER and NET_OWNER_READER policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c453813d0664259c4da0d132f224be2eebe70072"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-08 14:17:40.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-08 14:25:34.000000000","tz":60},"subject":"[OVN] ovn-db-sync check for router port differences","message":"[OVN] ovn-db-sync check for router port differences\n\nPrior to this patch the ovn-db-sync script did not check if the router\nports were actually out-of-sync before marking them to be updated. This\nbehavior introduced irrelevant information in the sync report (specially\nwhen ran in \"log\" mode) making the user think that the databases were\nout-of-sync even when they were not.\n\nThis patch adds the code checking for differences in the Neutron Router\nPorts and OVN Logical Router Port entries prior to updating them.\n\nChange-Id: Id7bf5a6aa547795ba78724eed59ba9d4fb74f758\nCloses-Bug: #2030773\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8928e5c4f6f1036b0bc8770d0891e7d122a529f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8928e5c4f6f1036b0bc8770d0891e7d122a529f2"}]},"branch":"refs/heads/master"},"e76fc931079e2f2f5cbb1c65c6e6792df4f5ae3b":{"kind":"REWORK","_number":2,"created":"2023-08-08 15:06:25.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/99/890799/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/890799/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/890799/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/890799/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/890799/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/890799/2"}}},"commit":{"parents":[{"commit":"c453813d0664259c4da0d132f224be2eebe70072","subject":"Merge \"Add NET_OWNER_MEMBER and NET_OWNER_READER policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c453813d0664259c4da0d132f224be2eebe70072"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-08 14:17:40.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-08 15:00:45.000000000","tz":60},"subject":"[OVN] ovn-db-sync check for router port differences","message":"[OVN] ovn-db-sync check for router port differences\n\nPrior to this patch the ovn-db-sync script did not check if the router\nports were actually out-of-sync before marking them to be updated. This\nbehavior introduced irrelevant information in the sync report (specially\nwhen ran in \"log\" mode) making the user think that the databases were\nout-of-sync even when they were not.\n\nThis patch adds the code checking for differences in the Neutron Router\nPorts and OVN Logical Router Port entries prior to updating them.\n\nChange-Id: Id7bf5a6aa547795ba78724eed59ba9d4fb74f758\nCloses-Bug: #2030773\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e76fc931079e2f2f5cbb1c65c6e6792df4f5ae3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e76fc931079e2f2f5cbb1c65c6e6792df4f5ae3b"}]},"branch":"refs/heads/master"},"bd89f4a7a77123373dd8bab2c78252bfdb7dd48a":{"kind":"REWORK","_number":3,"created":"2023-08-08 15:22:52.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/99/890799/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/890799/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/890799/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/890799/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/890799/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/890799/3"}}},"commit":{"parents":[{"commit":"c453813d0664259c4da0d132f224be2eebe70072","subject":"Merge \"Add NET_OWNER_MEMBER and NET_OWNER_READER policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c453813d0664259c4da0d132f224be2eebe70072"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-08 14:17:40.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-08 15:22:32.000000000","tz":60},"subject":"[OVN] ovn-db-sync check for router port differences","message":"[OVN] ovn-db-sync check for router port differences\n\nPrior to this patch the ovn-db-sync script did not check if the router\nports were actually out-of-sync before marking them to be updated. This\nbehavior introduced irrelevant information in the sync report (specially\nwhen ran in \"log\" mode) making the user think that the databases were\nout-of-sync even when they were not.\n\nThis patch adds the code checking for differences in the Neutron Router\nPorts and OVN Logical Router Port entries prior to updating them.\n\nChange-Id: Id7bf5a6aa547795ba78724eed59ba9d4fb74f758\nCloses-Bug: #2030773\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd89f4a7a77123373dd8bab2c78252bfdb7dd48a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd89f4a7a77123373dd8bab2c78252bfdb7dd48a"}]},"branch":"refs/heads/master"},"ec4c4d6fcc86f6cb995e21209d973ab2d8493d53":{"kind":"REWORK","_number":4,"created":"2023-08-09 08:32:01.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/99/890799/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/890799/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/890799/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/890799/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/890799/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/890799/4"}}},"commit":{"parents":[{"commit":"c453813d0664259c4da0d132f224be2eebe70072","subject":"Merge \"Add NET_OWNER_MEMBER and NET_OWNER_READER policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c453813d0664259c4da0d132f224be2eebe70072"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-08 14:17:40.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-09 08:31:54.000000000","tz":60},"subject":"[OVN] ovn-db-sync check for router port differences","message":"[OVN] ovn-db-sync check for router port differences\n\nPrior to this patch the ovn-db-sync script did not check if the router\nports were actually out-of-sync before marking them to be updated. This\nbehavior introduced irrelevant information in the sync report (specially\nwhen ran in \"log\" mode) making the user think that the databases were\nout-of-sync even when they were not.\n\nThis patch adds the code checking for differences in the Neutron Router\nPorts and OVN Logical Router Port entries prior to updating them.\n\nChange-Id: Id7bf5a6aa547795ba78724eed59ba9d4fb74f758\nCloses-Bug: #2030773\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec4c4d6fcc86f6cb995e21209d973ab2d8493d53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec4c4d6fcc86f6cb995e21209d973ab2d8493d53"}]},"branch":"refs/heads/master"},"4693836a1b58b477298e51cb47622222e3556752":{"kind":"REWORK","_number":5,"created":"2023-08-09 14:10:14.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/99/890799/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/890799/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/890799/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/890799/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/890799/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/890799/5"}}},"commit":{"parents":[{"commit":"c453813d0664259c4da0d132f224be2eebe70072","subject":"Merge \"Add NET_OWNER_MEMBER and NET_OWNER_READER policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c453813d0664259c4da0d132f224be2eebe70072"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-08 14:17:40.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-09 14:10:05.000000000","tz":60},"subject":"[OVN] ovn-db-sync check for router port differences","message":"[OVN] ovn-db-sync check for router port differences\n\nPrior to this patch the ovn-db-sync script did not check if the router\nports were actually out-of-sync before marking them to be updated. This\nbehavior introduced irrelevant information in the sync report (specially\nwhen ran in \"log\" mode) making the user think that the databases were\nout-of-sync even when they were not.\n\nThis patch adds the code checking for differences in the Neutron Router\nPorts and OVN Logical Router Port entries prior to updating them.\n\nChange-Id: Id7bf5a6aa547795ba78724eed59ba9d4fb74f758\nCloses-Bug: #2030773\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4693836a1b58b477298e51cb47622222e3556752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4693836a1b58b477298e51cb47622222e3556752"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
