)]}'
{"id":"openstack%2Fneutron~890832","triplet_id":"openstack%2Fneutron~master~I6060745aff0f5bc0037fb74568d81d27f3d91313","project":"openstack/neutron","branch":"master","topic":"ovn-db-sync-update","attention_set":{},"removed_from_attention_set":{"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2023-11-20 15:25:46.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-11-20 12:45:16.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-11-20 15:25:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6060745aff0f5bc0037fb74568d81d27f3d91313","subject":"Update OVN db sync code log messages","status":"MERGED","created":"2023-08-08 21:12:23.000000000","updated":"2023-11-20 15:28:01.000000000","submitted":"2023-11-20 15:25:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890832-ovn-db-sync-update","meta_rev_id":"54a50e0ca814b99ee6127f2e87ee348b5e998347","_number":890832,"virtual_id_number":890832,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-10-17 23:00:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-20 15:25:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-23 14:39:31.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2023-11-20 12:45:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2023-11-20 12:45:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-08 21:40:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-08 21:57:01.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-08 22:47:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-14 07:32:41.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-08-30 05:42:05.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-10-23 14:39:31.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"}],"messages":[{"id":"8eb0707f79e7be58773865c846bf349d4bcff46c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-08 21:12:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54c888a16a8d547769680a4f1c0e98422e38aa5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 21:40:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/199e93bf702b4986a6ee2c12bf0845f5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/66afadaffbf94744919d2289da791289 : SUCCESS in 26m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96cd105cbc82472ab817b7ab55acd3e1 : SUCCESS in 20m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3627c37530a2426b852a38ea21579af7 : SUCCESS in 19m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3f03b6bc21bdfbd03b1b8a09cd0f8a41fed3e82","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-08 21:57:01.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/32/890832/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"daac298f6cc95103bbf133309392b67c51a11f19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 22:47:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edee6ed455c94f5e853eb3e73023b894\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fc662ea07ef3491c8f35e9b37326dac8 : SUCCESS in 1h 33m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48448913e6214434950f8c327268bc77 : SUCCESS in 36m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1523274aedc74dbdb6e296112749bdbb : SUCCESS in 10m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa707ba0bdf348259d725a72e98fa2fb : SUCCESS in 23m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/286d5616cede4dda9b451847c3316e12 : SUCCESS in 24m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/821bff83f7c84122aaccd15fbda9ccab : SUCCESS in 25m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ca8f036e1747baab735a36448ebf8f : SUCCESS in 11m 10s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/c73ab5785f114e6094eb4d09cf133af6 : SUCCESS in 26m 41s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/75b92b20b4484777b8aaa70ed49b176e : SUCCESS in 1h 02m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9bdba908a4fb442e922da88b0656bfc8 : SUCCESS in 56m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/98644c969bf946deb750a89e2f6a6758 : SUCCESS in 1h 34m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/af72a8ebed5541369bc2682607de2969 : SUCCESS in 1h 10m 59s","accounts_in_message":[],"_revision_number":1},{"id":"51361c4ed1a22131bcc756e0a96c234592777fa7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-14 07:32:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06b04bbbdddc03d4f41b41238d799f76c268ab36","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-14 21:00:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fce4642e7aed52150d969727f51c0f2badf12c1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-14 21:01:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3e713f4084119f8f203d8c91daed734a337a5abb","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-14 21:19:27.000000000","message":"Patch Set 2:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/32/890832/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"e693187abc2957b09c7b525351fca3214772ce59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 21:52:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4405480f8d38498d8a1d6884c0bf7478\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7a7722a43984d1d8185a3c2452cbe28 : SUCCESS in 49m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/baa8ffa37aeb43e8a697282f030f7c96 : SUCCESS in 22m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2bfa0481c23d49cfa2168e43db5215b1 : SUCCESS in 49m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e45fe1200d18059ed200e060f6fd5d5fd3d3e8f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 22:46:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a337ae35b4ec4bffa0febd2203a4f9ef\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/327c64e0487b461eb360a99ae3ff9f15 : SUCCESS in 1h 42m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e6a97f5b3054340bb0fe1b30dbbea6f : SUCCESS in 33m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fb20110e4f04c4aa6ce84f2a7bbaf7a : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d41cac6e2ea4779be36a8f583edd5ab : SUCCESS in 20m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0dba56edca44d6e8f1be5a26d99469e : SUCCESS in 26m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/712caf71c734422eba4cc2d14922aa6b : SUCCESS in 26m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/262899730ca842c3a08e79cd7dcdac87 : SUCCESS in 9m 15s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/0868cce03da849a1b360da9763efdeb2 : SUCCESS in 23m 38s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/22c497ecd55c4d69945b11ee6a689bb7 : SUCCESS in 57m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/74c58ab4fb1b495e93f4cfdde0f451ce : SUCCESS in 1h 38m 35s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/44fe823e3fd04ecba8f0a7b864ea9131 : SUCCESS in 50m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/25f12848407a4c5ea40db10b4bc53e2a : SUCCESS in 1h 34m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/33c41b4c05744d82a0121aa9b8c78d79 : SUCCESS in 1h 13m 10s","accounts_in_message":[],"_revision_number":2},{"id":"5e655493fc6ff3b8e4ce66c391016b114648e85f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-17 15:16:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f46c103d5611fe9dc7d8979086aa2b5dd3e047df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-17 15:22:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d94c3834ac997e61c31b72faadbe9dc27a777fc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 16:20:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a632fcd9f59246fb9c796270ed06ae6e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/184729f76a9048b18024700b651f9262 : SUCCESS in 55m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cfefa73ba5f3451a8acac5c9f16b7677 : SUCCESS in 23m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1333216c3fd442ed8123ca8af6a9872a : SUCCESS in 24m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"40635d25272e61dc734004e28aed15e919b3d95c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 19:26:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdfe29436abc425f937f50334b2bd493\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/119cd2d5990f40de8970a77ab74b5452 : SUCCESS in 1h 52m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/939e0199e16242a8bb7cc20c157cca77 : SUCCESS in 43m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63df6284713c4c90a26e840655ee2de8 : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be8cd5405a434a3fb7ead383946d8233 : SUCCESS in 24m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73fbb7f1614145caa783a15bc66a872c : SUCCESS in 25m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa66e52c66d24c1ca7d436aa8e5afe94 : SUCCESS in 27m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60cd4e2a7d0f4dddb19f2283871eb834 : SUCCESS in 11m 15s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/bf377313e953451193c6afcf269a073c : SUCCESS in 26m 01s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/316aa231ecad4f9488a633976f04a3af : SUCCESS in 52m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8044be89a208470b9a4f8820430d7a9f : SUCCESS in 2h 00m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b3290b89b258456aadef4c4b7349ab38 : SUCCESS in 1h 11m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c06da385abbb4e3e8640957dae75fa41 : FAILURE in 34m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5250cb3006d647fab8aad77e17a32dc5 : SUCCESS in 1h 57m 26s","accounts_in_message":[],"_revision_number":3},{"id":"2ca474690cddb5fd341699acfdcb5b5f71ce1ea0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-17 20:23:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"63493d5578c7cfbb07ab46bb6d94beb190342d0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 21:20:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ca52452e7bb4527bdc9c1bf3dd41df6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89e5fd77022b459dac0554a60a10d5e5 : SUCCESS in 53m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f8d63b0e7a344cfad327f2b8eb737de : SUCCESS in 54m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e2e7c0edd26a47f6adcd7f168764afe6 : SUCCESS in 48m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3c0a9d507b36476791fd10d8d31a8356 : RETRY_LIMIT in 2m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"06dc5de62bf6948a04a4a697a5d1562d66ba5aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 22:08:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b501b33d0f774f22b1d970473f85483b\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6d1bf94cc33448c5bb523e02a01c4034 : SUCCESS in 1h 44m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcf6e5820caa452c86f1aad72951ae12 : SUCCESS in 44m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2139a88ceba5445fa76314cdd31c56e7 : SUCCESS in 8m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b6ba09cf8ab40d581731bb2b4fdf870 : SUCCESS in 31m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/312a327dc90b4fd8a971f43d2b01f788 : SUCCESS in 29m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/494dd44911e24dd68fcadf0071c6e0bf : SUCCESS in 25m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d9b26c88f2d14518be1d673b931ac237 : RETRY_LIMIT in 2m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cc35e4524ea46e5ae09605749b60d87 : SUCCESS in 10m 38s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/ae17f28cb91d430981b5da4d23279eed : SUCCESS in 25m 29s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2bb800e588e84c2d8d784895dd3701e3 : SUCCESS in 1h 03m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4e60ceb4716469dbc39ce07a4bd3820 : SUCCESS in 1h 37m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4beafe0d73d64a88a0bbf26d483be5eb : SUCCESS in 58m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/389a22b4abad4c04a2ec0932de5d14f0 : FAILURE in 21m 59s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c593db2804114246b1ec687f7ea3dc31 : SUCCESS in 1h 07m 05s","accounts_in_message":[],"_revision_number":3},{"id":"085116d8aa6c5c920893e9994b6436168554cb5c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-18 15:03:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cd39f7b268485cda9b15a0ceb1bba34ab641556b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 15:35:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25a6d8a9cb274369994e42824b2e943c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/500aeab4f25b41b485b5f7562838bd8e : SUCCESS in 25m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d1cda16631364eecb3dd58b443e8aaa0 : SUCCESS in 26m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5618a7407dda4531bba8786eac19fd82 : SUCCESS in 31m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4d63e24b182a47a0b4e34058457e31ef : RETRY_LIMIT in 1m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e479f99395ac5190231b4a26c3690cad38b529de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 17:10:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a74ea838ac8641e5a45390cd6ec8f17f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4535b474b1a04a7e893e0995a231f03e : SUCCESS in 1h 50m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fe7cbcd5abd4c4d9203389a5ba3754f : SUCCESS in 36m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4511e6f427274f1bb9e0d88b0ef9d3a1 : SUCCESS in 9m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/528df09589054667be44c78b0690ddcc : SUCCESS in 24m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a1003c5c6b342dab0796d7dab0d9a02 : SUCCESS in 27m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2fe6d8c1b0cd48f8aa9e01029b309040 : SUCCESS in 30m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd3eccd632b74286a50095eea7a07651 : RETRY_LIMIT in 2m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac001ee39cc3480284a1ba0f4576b221 : SUCCESS in 11m 17s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/e1f43a4295ae4d44aa713a4d1c1a984a : SUCCESS in 25m 20s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c1f8d148ea546d3a7d19dbb5617a5c0 : SUCCESS in 43m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/53c31d33c6004928927a0db18f26783b : SUCCESS in 1h 59m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/77f33ee8a2ce442b9bfef2a41e739d08 : SUCCESS in 58m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e2731c59f5144857a8adf0d959450c03 : FAILURE in 32m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c03f0b89a8c14bb6b0e8de44283170c5 : SUCCESS in 1h 09m 10s","accounts_in_message":[],"_revision_number":3},{"id":"56b4c2a92130b42b40bcbeb261d44a42c1d51a36","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-18 22:25:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"622593231bbe9ce8d6258df891175804d624b121","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 23:17:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/512704b80c7e414ba53bdecee1438757\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7749bac69a9e4a33a32ae74a599b9542 : SUCCESS in 48m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6997d2df549a4988ac9638e27318d898 : SUCCESS in 20m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2c36ee91697143b1ad53a2a3add75744 : SUCCESS in 50m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/082798b1c4b442cfad0604e3bab94075 : RETRY_LIMIT in 1m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e48304b5819b79143447c0280110618ed0e9f2c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-19 00:34:28.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e9cb1788683456fa01db8e884b86c6e\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ae88df6668444e74a2b12dacaebbb00c : SUCCESS in 1h 44m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1b774c4a81346259ef5d91fa068c018 : SUCCESS in 43m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de71b52d15ec48179790d227e6325619 : SUCCESS in 11m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/283b1e8935c34a49968d0fc87e1a8863 : SUCCESS in 26m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/702a891b588f4feda4b3900d4730dace : SUCCESS in 23m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b246d973cb6447d89544a2082f84d5d : SUCCESS in 20m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/53b55bad286c49a28a7010cf9826cfe7 : RETRY_LIMIT in 1m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beb8e58e561640b5bde5222f3ae0a5dc : SUCCESS in 10m 23s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/ce27bafa183f49089d64e7b8a98ed143 : SUCCESS in 25m 19s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c2678ddc6d7649a1a10d0692d3e95202 : SUCCESS in 59m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/483547b79b8649239048c7792069ec4c : SUCCESS in 2h 07m 14s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2549649336ef478a9bdf2b70d29386ef : SUCCESS in 58m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0743bca4dec84c6399ba8727980f4667 : FAILURE in 19m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4ffc32766f7b4e758dd47fa3f17f84d3 : SUCCESS in 1h 15m 04s","accounts_in_message":[],"_revision_number":3},{"id":"fa989e795e1ae941104df4743cae93b28e802866","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-21 08:03:28.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0cc4bb6ee132add5b8a0a7067dc8e72cd33e66ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 08:25:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/055e1da3a3ee44759e7808e8f79143d5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/80e1896acd9b4748a8d066b34902c68a : SUCCESS in 20m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0cd1e99e6ea4c4ebfee202c6a93c102 : SUCCESS in 21m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c9e152685d31416097e8cc11791e55e2 : SUCCESS in 20m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8af27207de3b43e9a211d5dd94354168 : RETRY_LIMIT in 1m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"82dbb4363cfa59460172db239eec610ff43190fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 10:16:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34d8783c2f784948b26f38d47c1209ca\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8485601e3a9549348733bfeeb7fe14e7 : SUCCESS in 1h 37m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a57bb6ca6fe495e8f2bf1a85aa5055a : SUCCESS in 46m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae23d05594db431e800775c402dc1c65 : SUCCESS in 10m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d3817d542914289ac9bef5dbc9925dc : SUCCESS in 26m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd7d338feed14a5b95c1ffd617e6ea62 : SUCCESS in 26m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9012633865c43bc8e452e5c699cca39 : SUCCESS in 31m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f983b0396dfe41ab99c59953648a1a74 : RETRY_LIMIT in 1m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5325c21166da41bf95ce88d9d717faf7 : SUCCESS in 11m 09s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/6e71f26095934034b29aebb4f7e72b9a : SUCCESS in 25m 08s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8d90a35f93d467a9a1e58567311fdc1 : SUCCESS in 1h 02m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d6b683e9b20428190d7b700b646dccb : SUCCESS in 1h 49m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a09d6700e2ba4504be041bd967bff7e2 : SUCCESS in 1h 03m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a543ebc4014d46b89e081a4caa167172 : SUCCESS in 2h 07m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f2647062b5f84872beddce0b780e311c : SUCCESS in 1h 17m 14s","accounts_in_message":[],"_revision_number":3},{"id":"1759c3f9e6c0d671b06e40b8592e7a7f1bf17e9e","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-08-30 05:42:05.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b8ecdc2be51e9611b31f010de5c7c8bba289e2da","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-30 09:40:25.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/32/890832/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"aa01b52263df0ee5f8001feae8368747f43d47b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-17 22:26:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"557dd364183e13b02cdae1fdb4db4400b8e790c2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-17 22:28:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6528af2bdb7dd0209f63de3c40a16ecdc2cbca92","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-17 23:00:01.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/32/890832/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"383a6d88824cbbd9b1e25f85678866f8d80f7453","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 23:46:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/819d3a7af3f1466086b60b387088f124\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/652182b35cab47c380b51c815bf7c3a8 : SUCCESS in 1h 10m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cbbb0ba7e963432a99c47741bf7c9dc5 : SUCCESS in 1h 09m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f9012299ad149bf9930dec3b5e09014 : SUCCESS in 1h 10m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/37f85e0e46284c4a86789e943a3c951a : SUCCESS in 1h 04m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6ff34d256d0d3cb50405a8d814981180a52a8d20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 00:59:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5012451dc5ed4c77837e2079e7e9cccd\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/815f0b398c0945489edb76ddb9468666 : SUCCESS in 1h 43m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85395b4e95a7418baf4fbaf6657f1750 : SUCCESS in 36m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f8cd0f609ef49f085b71950c445b505 : SUCCESS in 11m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab8623934c6e45f3b03df1cf3af97e4e : SUCCESS in 38m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85d449e8ec464881b5b2815b15f3b71c : SUCCESS in 26m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ebe157c5848431583137ea4bf0be490 : SUCCESS in 39m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d16caad418a2476185e47bbf045d7630 : SUCCESS in 24m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8888153040a41efa1cc20959153c8a9 : SUCCESS in 11m 54s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/72827552059c4efab87d4bab3cadc5aa : SUCCESS in 23m 09s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/36ba366e1c4d48828d8c42f411df8568 : SUCCESS in 53m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1f16a73ba7894bd8a7c96f5e2d032215 : SUCCESS in 57m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa7d4f15b4e64f7d85575a922b5d8da9 : SUCCESS in 2h 27m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0623c2577df74c10922caabd464ced9f : SUCCESS in 1h 03m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e68e28f0f100467faa4d124e455411e5 : SUCCESS in 1h 38m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ebfb4de6ea8e49508e077e24c534ebc1 : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":4},{"id":"c938f4e402f724b44265e80259197ef8d16cb200","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-23 14:39:31.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1d440de90e2565dc711730122115023f4d634db5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-11-20 12:45:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"99e0a97355064a46cc1eef7c047617fd827b5bdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 12:45:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"253cde57e9435a3d5a5bdb5fc921898ed9aefa6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 15:25:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2475ee207a614261807c52f3f1e9ed5e\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7a117e26c6a14958bc0afc14b4eed746 : SUCCESS in 1h 27m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3d707bee5ea41ac9cd03ffd5d34567f : SUCCESS in 11m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7802626a66844dbc8f0b43003109b808 : SUCCESS in 27m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f0acc682c88469d876f0b3b9390413b : SUCCESS in 24m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41579499f94242e08205ba8fe2508d9d : SUCCESS in 11m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a56857bf6ec3425cb59d2653562bf5df : SUCCESS in 1h 01m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/28e95ce88f0548dfb070d110725fd77e : SUCCESS in 1h 55m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b049b4257aab486486933677f69a0ea1 : SUCCESS in 59m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/fd60df737b194116a08193b471467a54 : SUCCESS in 1h 23m 53s","accounts_in_message":[],"_revision_number":4},{"id":"f51399afc48b4f98b8af34d42bb8bd847f55998a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 15:25:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"54a50e0ca814b99ee6127f2e87ee348b5e998347","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 15:28:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48be3076546c4966a5ee503ecbe234d1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c66c525813a4884aacc7983395c88fb : SUCCESS in 51s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"73cd6dca7442a33ac96fed157d41a4b5ec99915f","revisions":{"f68e044ec665a03b4666f0d943969557e34e3dbd":{"kind":"REWORK","_number":1,"created":"2023-08-08 21:12:23.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/32/890832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/32/890832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/890832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/890832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/890832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/32/890832/1"}}},"commit":{"parents":[{"commit":"c453813d0664259c4da0d132f224be2eebe70072","subject":"Merge \"Add NET_OWNER_MEMBER and NET_OWNER_READER policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c453813d0664259c4da0d132f224be2eebe70072"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-08-08 20:56:10.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-08-08 21:08:06.000000000","tz":-240},"subject":"Update OVN db sync code log messages","message":"Update OVN db sync code log messages\n\nWhen running the OVN db sync tool, the log messages are\ninconsistent between functions, attempt to sanitize them\nso the output is a little more structured by:\n\n1) Always logging a start/end time for each major section\n\n2) Always use LOG.warning when fixing inconsistencies\n\n3) Be consistent using \u0027OVN NB/SB DB\u0027 in messages\n\n4) Try to use full name of object being updated, for example,\n   \u0027floating IP\u0027 not \u0027fip\u0027\n\nAlso fixed the following:\n\n1) Only drop into blocks if in SYNC_MODE_REPAIR and there\n   is work to be performed\n\n2) Random text fix-ups\n\nFunctionally the code is unchanged.\n\nTrivialfix\n\nChange-Id: I6060745aff0f5bc0037fb74568d81d27f3d91313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f68e044ec665a03b4666f0d943969557e34e3dbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f68e044ec665a03b4666f0d943969557e34e3dbd"}]},"branch":"refs/heads/master"},"1424a8170cd1c5ba6ab75797d156fbc62421785c":{"kind":"REWORK","_number":2,"created":"2023-08-14 21:01:28.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/32/890832/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/32/890832/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/890832/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/890832/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/890832/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/32/890832/2"}}},"commit":{"parents":[{"commit":"c453813d0664259c4da0d132f224be2eebe70072","subject":"Merge \"Add NET_OWNER_MEMBER and NET_OWNER_READER policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c453813d0664259c4da0d132f224be2eebe70072"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-08-08 20:56:10.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-08-14 21:01:11.000000000","tz":-240},"subject":"Update OVN db sync code log messages","message":"Update OVN db sync code log messages\n\nWhen running the OVN db sync tool, the log messages are\ninconsistent between functions, attempt to sanitize them\nso the output is a little more structured by:\n\n1) Always logging a start/end time for each major section\n\n2) Always use LOG.warning when fixing inconsistencies\n\n3) Be consistent using \u0027OVN NB/SB DB\u0027 in messages\n\n4) Try to use full name of object being updated, for example,\n   \u0027floating IP\u0027 not \u0027fip\u0027\n\nAlso fixed the following:\n\n1) Only drop into blocks if in SYNC_MODE_REPAIR and there\n   is work to be performed\n\n2) Random text fix-ups\n\nFunctionally the code is unchanged.\n\nTrivialfix\n\nChange-Id: I6060745aff0f5bc0037fb74568d81d27f3d91313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1424a8170cd1c5ba6ab75797d156fbc62421785c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1424a8170cd1c5ba6ab75797d156fbc62421785c"}]},"branch":"refs/heads/master"},"060ffa3d4439c03fc6136ba5c2694fd546cf224e":{"kind":"REWORK","_number":3,"created":"2023-08-17 15:22:41.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/32/890832/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/32/890832/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/890832/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/890832/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/890832/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/32/890832/3"}}},"commit":{"parents":[{"commit":"c453813d0664259c4da0d132f224be2eebe70072","subject":"Merge \"Add NET_OWNER_MEMBER and NET_OWNER_READER policy rules\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c453813d0664259c4da0d132f224be2eebe70072"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-08-08 20:56:10.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-08-17 15:22:18.000000000","tz":-240},"subject":"Update OVN db sync code log messages","message":"Update OVN db sync code log messages\n\nWhen running the OVN db sync tool, the log messages are\ninconsistent between functions, attempt to sanitize them\nso the output is a little more structured by:\n\n1) Always logging a start/end time for each major section\n\n2) Always use LOG.warning when fixing inconsistencies\n\n3) Be consistent using \u0027OVN NB/SB DB\u0027 in messages\n\n4) Try to use full name of object being updated, for example,\n   \u0027floating IP\u0027 not \u0027fip\u0027\n\nAlso fixed the following:\n\n1) Only drop into blocks if in SYNC_MODE_REPAIR and there\n   is work to be performed\n\n2) Random text fix-ups\n\nFunctionally the code is unchanged.\n\nTrivialfix\n\nChange-Id: I6060745aff0f5bc0037fb74568d81d27f3d91313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/060ffa3d4439c03fc6136ba5c2694fd546cf224e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/060ffa3d4439c03fc6136ba5c2694fd546cf224e"}]},"branch":"refs/heads/master"},"73cd6dca7442a33ac96fed157d41a4b5ec99915f":{"kind":"REWORK","_number":4,"created":"2023-10-17 22:26:44.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/32/890832/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/32/890832/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/890832/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/890832/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/890832/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/32/890832/4"}}},"commit":{"parents":[{"commit":"849930ba5bdf4bd1675577f48f3fc3ec1d65cef5","subject":"Merge \"Bump skip-level lower version to stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/849930ba5bdf4bd1675577f48f3fc3ec1d65cef5"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-08-08 20:56:10.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-10-17 22:21:06.000000000","tz":-240},"subject":"Update OVN db sync code log messages","message":"Update OVN db sync code log messages\n\nWhen running the OVN db sync tool, the log messages are\ninconsistent between functions, attempt to sanitize them\nso the output is a little more structured by:\n\n1) Always logging a start/end time for each major section\n\n2) Always use LOG.warning when fixing inconsistencies\n\n3) Be consistent using \u0027OVN NB/SB DB\u0027 in messages\n\n4) Try to use full name of object being updated, for example,\n   \u0027floating IP\u0027 not \u0027fip\u0027\n\nAlso fixed the following:\n\n1) Only drop into blocks if in SYNC_MODE_REPAIR and there\n   is work to be performed\n\n2) Random text fix-ups\n\nFunctionally the code is unchanged.\n\nTrivialfix\n\nChange-Id: I6060745aff0f5bc0037fb74568d81d27f3d91313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73cd6dca7442a33ac96fed157d41a4b5ec99915f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/73cd6dca7442a33ac96fed157d41a4b5ec99915f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
