)]}'
{"id":"openstack%2Fneutron~890937","triplet_id":"openstack%2Fneutron~master~I13bcd36c237555e0ca60d5ffc214b8f3bf6e478f","project":"openstack/neutron","branch":"master","topic":"bug/2030904","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-08-11 08:49:50.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-08-10 20:44:26.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I13bcd36c237555e0ca60d5ffc214b8f3bf6e478f","subject":"Initialize logger for rpc-server and wsgi script","status":"MERGED","created":"2023-08-09 19:01:29.000000000","updated":"2023-08-11 08:49:50.000000000","submitted":"2023-08-11 02:39:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"890937-bug/2030904","meta_rev_id":"59a48daa6c80e8d8909771b08f6b87428dcb7df7","_number":890937,"virtual_id_number":890937,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-08-10 04:07:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-11 02:39:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-11 00:22:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-08-11 08:49:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-08-10 20:44:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-11 00:22:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-09 19:47:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-09 19:56:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-09 20:01:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-08-09 21:28:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-11 00:22:54.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-08-11 08:49:50.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"3b656dfa70bc5206c76d79c2ee2799d8e7b5f17f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-09 19:01:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1061988f0af6218a25854e65d68d7987a7a855e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-09 19:47:28.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/37/890937/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d64ecbb204b6e688632ac6455a8b705c1df1d35b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 19:56:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c739a706d252486ebd92d7ba0143e793\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/85a2308bd5914f8093a27a0daac91d02 : SUCCESS in 52m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1d16e32d4194981ae9e5c83eec7574a : SUCCESS in 21m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7c7d740baaf341d5947782179f28c4cc : SUCCESS in 52m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"336b7314addb6f161ac5154a91d327aa447327ee","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-09 20:01:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ccd672647101b74a5ed4af4d4aa87bc72be48a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 21:28:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/119b823563b648928409c544d94e205d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e0249215b0b64f00b71183bd1394f6f8 : SUCCESS in 1h 25m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2ae356ba0d27434388b0658d13aec0c0 : SUCCESS in 1h 59m 31s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/aa2f9d17933743bf90dab903b0786d94 : SUCCESS in 1h 53m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5963114e7f224fe0ace9619798f65086 : SUCCESS in 1h 44m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f2d108222606433e8fbb12af43077508 : SUCCESS in 35m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09d9b9e28fe7403e8baf178ca3d80e47 : SUCCESS in 31m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbb308967d0543479716c6f682d7134f : SUCCESS in 11m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/000a922f0ddf40ed996a0c8b7914df00 : SUCCESS in 33m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b79d74e9619644d7a0028a1759d1f676 : SUCCESS in 28m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07ebf333d10d4285af1be490d5020dc4 : SUCCESS in 31m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acfb0b6faa434ca8b48f79b14eaa9f4f : SUCCESS in 10m 23s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a4ab267beb944154bc6096cd7717d8de : SUCCESS in 31m 36s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f6ad36d5db4842feb55899993cd7f292 : SUCCESS in 1h 09m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4bc9f34fe736400da29d565ad27767e3 : SUCCESS in 1h 39m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc5bd6fbd8754f1ba3575892079e1de2 : SUCCESS in 46m 13s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1dff7d7fbff54828ba646238f7134dc2 : SUCCESS in 44m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4eba9b4c57994a0d8b890ae53b9e89c4 : SUCCESS in 1h 18m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0cb4d8431d124f4db252e8e3bc005d1a : FAILURE in 1h 14m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0c78991fce85424e8ee41617de3a5431 : SUCCESS in 1h 19m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ad915a861e524befa40f43f1ec44afb4 : SUCCESS in 1h 33m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/52963a1f8f404dd2be8d73c3824dd828 : SUCCESS in 2h 19m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/402fba981f55429da0859c4256b6b83a : SUCCESS in 1h 10m 56s","accounts_in_message":[],"_revision_number":1},{"id":"37dc1d7d95ea81717ae94b3466af968fcc6bb634","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-10 02:02:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0e9159f3abb74c33f2bcbb185cc6c9604ed3229f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-10 02:03:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6264c0bbdb18af1e10bc70621d9d8d139bd838f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-10 02:03:18.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bc7257829779d2c3e3461a112fe59c49bee4b99c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-10 02:03:53.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"ae3a9c819763fab54e97c089000565873117254a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-10 02:44:45.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/37/890937/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"6b93cd65b2c2247ed3d5cc512a0ae316eb319b11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 03:10:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc66fb972e5c42cfbd170a6f8e215a0f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81e3aaa0ad5441abb7e608d94a5e4dce : SUCCESS in 32m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e77147727ea54be9bc4e05c7d11dbc12 : SUCCESS in 53m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9e6afcb0a9d242588f80b213cd698eef : SUCCESS in 56m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bfccd2485b3c714a9600bfe3e3db4bd3475bd792","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-10 03:26:13.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/37/890937/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"ead22919d246260eff762546e8f12fa1888da1f2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-10 04:07:15.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/37/890937/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"61313f8430370a74b3dd25b33c5f96ced9b52c02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 04:52:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b0708174e344f47bfe74e798e1e1c06\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6004590dc27f49e5ae6532af46501570 : SUCCESS in 1h 24m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c7e4a5663d0a4cdfb92bf3aec6d16365 : SUCCESS in 1h 28m 44s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b3d61fdc70e54e3abf7bb7a836138c9f : SUCCESS in 1h 17m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a5747500786d465c8dfadfdc40e92e90 : SUCCESS in 1h 37m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/913be3511a3b48bca37899b745b80907 : SUCCESS in 33m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/412fb27389eb40f8970ea839700c5477 : SUCCESS in 47m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dbf206bd7e34614a37f6b06921af1b5 : SUCCESS in 11m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47c449eff8314b04a1454a368194e799 : SUCCESS in 33m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba7ef64811ee47cdba08a9c26493651a : SUCCESS in 27m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecedcd7306e940149412eded30f6cf75 : SUCCESS in 33m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e559380bd8b84bdb82979cde749a2271 : SUCCESS in 11m 50s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/ad8e0bdda8734167938eeb8227dd375f : SUCCESS in 29m 48s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d68691d13ec4bf28e75c53df17a294c : SUCCESS in 55m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e198adb780547e1be2db04d0e106e4f : SUCCESS in 1h 52m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4e8bce2de3241088aa5c0467063d1c2 : SUCCESS in 55m 50s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cd57846c3686435d8184249043ee6f5e : SUCCESS in 55m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/670973a9731b43a9b2ee18f4ea9bae7b : SUCCESS in 1h 16m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c6dfd575a78f46cf82e798fea9e8b889 : SUCCESS in 1h 37m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/670109937c4d41b6b9fdb959be574451 : SUCCESS in 1h 05m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f1d3d448d2fe4de8b121d97542103d81 : SUCCESS in 1h 54m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/631b2d7177c94633b1f2aca29792f929 : SUCCESS in 1h 36m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2d0b0192ffb74ee7a75dca93c083f895 : SUCCESS in 1h 23m 12s","accounts_in_message":[],"_revision_number":4},{"id":"16266941b11bd5db3b022f58b75290a856d94654","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-10 20:44:26.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5b4cdd590b55af11da39b48ff07cb95eb880a216","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-08-11 00:22:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"beb10f1433fc5c25fdb288992b3b5ccefc5d86ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 00:23:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0a6e04c07d82dc108bc2b5a17f6af433eccaeb6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 02:39:52.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45532c813dc44e1e8d3c80fd0875098f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b6d6a51dd89b434d92670e1455ea6530 : SUCCESS in 1h 38m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ce1b7a2696f3474ca335af5f7d587447 : SUCCESS in 1h 44m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2801f20800594fefbc9c1f1ca206a87f : SUCCESS in 1h 50m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/46412342bcec4c21a97d0c4bfef418fa : SUCCESS in 1h 33m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68de1fe264aa4311b34fd4cff23f7e11 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0bcb31a549649d1a1929ab82d58b90a : SUCCESS in 25m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55780cc3c2d14c368accdaf56f2d45af : SUCCESS in 20m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e0c6add3bdb4dcf97d42c7ba94f242c : SUCCESS in 25m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/577e1350104141b5b28903b76ce49f97 : SUCCESS in 9m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cfe2c20b3f6141ccad805e664052264b : SUCCESS in 57m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b327478606af49f5bcd63d458ec06ba0 : SUCCESS in 2h 09m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0843c94f02114831ad5df886193dec48 : SUCCESS in 1h 14m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/377ec13fd8414711af1208af1ea89efc : SUCCESS in 1h 03m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d9f75dea92544c57ad5f168c38f3dac1 : SUCCESS in 1h 01m 52s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/109410d79b11437b92b21b5c2b61280f : SUCCESS in 43m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f21ccffa8dc04aaab9b6fb9567eaa447 : SUCCESS in 51m 14s","accounts_in_message":[],"_revision_number":4},{"id":"42f110e2180ea9299b8f8b36baaa794c5781bd79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 02:39:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7787386f56012e0973b93faf893d731ee5ccb881","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 02:41:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d4bbd80e6dc46848226352013f5a787\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b0dc523f4fb4170bb4d735de647d636 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":4},{"id":"59a48daa6c80e8d8909771b08f6b87428dcb7df7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-11 08:49:50.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"afcce6d7493faaa2acf94c01268a2551a1300d31","revisions":{"4859bfb5b97e00fb47946006fd9be3df29a4eb3a":{"kind":"REWORK","_number":1,"created":"2023-08-09 19:01:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/890937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/890937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/890937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/890937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/890937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/890937/1"}}},"commit":{"parents":[{"commit":"cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a","subject":"Merge \"[OVN] Improve Hash Ring logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-09 18:54:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-09 19:00:09.000000000","tz":540},"subject":"Initialize logger for rpc-server and wsgi script","message":"Initialize logger for rpc-server and wsgi script\n\nPreviously logger was not properly initialized according to config\noptions or process arguments when the neutron-rpc-server script or\nthe api wsgi script is used. Because of this neutron does not generate\nthe log files as expected when we deploy api using WSGI and use\nneutron-rpc-server, instead of the monolithic neutron-server.\n\nThis ensures the logger is initialized even when these two scripts\nare used.\n\nCloses-Bug: #2030904\nChange-Id: I13bcd36c237555e0ca60d5ffc214b8f3bf6e478f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4859bfb5b97e00fb47946006fd9be3df29a4eb3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4859bfb5b97e00fb47946006fd9be3df29a4eb3a"}]},"branch":"refs/heads/master"},"0de2551b1a7e39aed146d3ac599cf275bd2fc44f":{"kind":"REWORK","_number":2,"created":"2023-08-10 02:02:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/890937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/890937/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/890937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/890937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/890937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/890937/2"}}},"commit":{"parents":[{"commit":"cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a","subject":"Merge \"[OVN] Improve Hash Ring logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-09 18:54:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-10 02:01:46.000000000","tz":540},"subject":"Initialize logger for rpc-server and wsgi script","message":"Initialize logger for rpc-server and wsgi script\n\nPreviously logger was not properly initialized according to config\noptions or process arguments when the neutron-rpc-server script or\nthe api wsgi script is used. Because of this neutron does not generate\nthe log files as expected when we deploy api using WSGI and use\nneutron-rpc-server, instead of the monolithic neutron-server.\n\nThis ensures the logger is initialized even when these two scripts\nare used.\n\nCloses-Bug: #2030904\nChange-Id: I13bcd36c237555e0ca60d5ffc214b8f3bf6e478f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0de2551b1a7e39aed146d3ac599cf275bd2fc44f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0de2551b1a7e39aed146d3ac599cf275bd2fc44f"}]},"branch":"refs/heads/master"},"3cc0bd8a82bfa3d997238010873d0bbd5dbc3559":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-08-10 02:03:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/890937/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/890937/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/890937/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/890937/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/890937/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/890937/3"}}},"commit":{"parents":[{"commit":"cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a","subject":"Merge \"[OVN] Improve Hash Ring logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-09 18:54:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-10 02:03:18.000000000","tz":0},"subject":"Initialize logger for rpc-server and wsgi script","message":"Initialize logger for rpc-server and wsgi script\n\nPreviously logger was not properly initialized according to config\noptions or process arguments when the neutron-rpc-server script or\nthe api wsgi script is used. Because of this neutron does not generate\nthe log files as expected when we deploy api using WSGI and use\nneutron-rpc-server, instead of the monolithic neutron-server.\n\nThis ensures the logger is initialized even when these two scripts\nare used.\n\nCloses-Bug: #2030904\nRelated-Bug: #1693691\nChange-Id: I13bcd36c237555e0ca60d5ffc214b8f3bf6e478f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cc0bd8a82bfa3d997238010873d0bbd5dbc3559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3cc0bd8a82bfa3d997238010873d0bbd5dbc3559"}]},"branch":"refs/heads/master","description":"Edit commit message"},"afcce6d7493faaa2acf94c01268a2551a1300d31":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-08-10 02:03:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/37/890937/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/37/890937/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/890937/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/890937/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/37/890937/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/37/890937/4"}}},"commit":{"parents":[{"commit":"cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a","subject":"Merge \"[OVN] Improve Hash Ring logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbb89fdb1414a1b3a8e8b3a9a4154ef627bb9d1a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-09 18:54:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-10 02:03:53.000000000","tz":0},"subject":"Initialize logger for rpc-server and wsgi script","message":"Initialize logger for rpc-server and wsgi script\n\nCurrently logger is not properly initialized according to config\noptions or process arguments when the neutron-rpc-server script or\nthe api wsgi script is used. Because of this neutron does not generate\nthe log files as expected when we deploy api using WSGI and use\nneutron-rpc-server, instead of the monolithic neutron-server.\n\nThis ensures the logger is initialized even when these two scripts\nare used.\n\nCloses-Bug: #2030904\nRelated-Bug: #1693691\nChange-Id: I13bcd36c237555e0ca60d5ffc214b8f3bf6e478f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/afcce6d7493faaa2acf94c01268a2551a1300d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/afcce6d7493faaa2acf94c01268a2551a1300d31"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
