)]}'
{"id":"openstack%2Fneutron~890986","triplet_id":"openstack%2Fneutron~master~I6da1b135c83ecfc41ec91e907ebf8500325a7a80","project":"openstack/neutron","branch":"master","topic":"check_haproxy_running","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-06 14:23:48.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2023-12-05 23:20:05.000000000","reason":"Change was submitted"},"34118":{"account":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"last_update":"2023-12-05 23:20:05.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-09-13 18:32:43.000000000","reason":"\u003cGERRIT_ACCOUNT_34271\u003e replied on the change","reason_account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2023-12-05 23:20:05.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-05 23:20:05.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-12-05 23:20:05.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-12-05 20:54:11.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-12-05 23:20:05.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-12-05 18:08:52.000000000","reason":"\u003cGERRIT_ACCOUNT_8655\u003e replied on the change","reason_account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}},"hashtags":[],"change_id":"I6da1b135c83ecfc41ec91e907ebf8500325a7a80","subject":"Metadata: handle process exceptions","status":"MERGED","created":"2023-08-09 21:35:22.000000000","updated":"2023-12-05 23:21:22.000000000","submitted":"2023-12-05 23:20:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":66,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"890986-check_haproxy_running","meta_rev_id":"eb58550e32730342689d3afde6f461ecb604bea8","_number":890986,"virtual_id_number":890986,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-11-29 21:51:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-05 23:20:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"value":2,"date":"2023-12-05 20:54:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-05 18:08:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"value":1,"date":"2023-12-05 20:54:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}],"reviewers":{"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":35499,"name":"alisafari","email":"alisafar1212@gmail.com","username":"alisafari","status":"Software Engineer"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-09 21:58:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-09 22:17:32.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-10 00:41:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-15 17:32:54.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2023-08-15 17:32:54.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-08-21 13:18:23.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2023-08-22 16:04:20.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-08-22 19:12:10.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-08-29 07:45:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-09-13 18:32:43.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REMOVED"},{"updated":"2023-09-15 08:57:12.000000000","updated_by":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"reviewer":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"state":"REVIEWER"},{"updated":"2023-10-23 13:43:14.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2023-10-30 20:33:01.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REMOVED"},{"updated":"2023-11-21 02:56:11.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-11-21 13:52:32.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"CC"},{"updated":"2023-11-30 10:55:14.000000000","updated_by":{"_account_id":35499,"name":"alisafari","email":"alisafar1212@gmail.com","username":"alisafari","status":"Software Engineer"},"reviewer":{"_account_id":35499,"name":"alisafari","email":"alisafar1212@gmail.com","username":"alisafari","status":"Software Engineer"},"state":"CC"}],"messages":[{"id":"e53bf97efbc7171d2e61190f45a57aa8b75bd48e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-09 21:35:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a1bc956d176946d778b055420391417a19d922b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 21:58:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc08d605dac4427db7a6d4af231d7d84\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/56612c233e29456496a6326caa0ca019 : FAILURE in 22m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9fb40c15c19646a1aeb05454115d5c0b : FAILURE in 22m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a02b2940b8704dc29759a5a2b2e0af3d : FAILURE in 22m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc593b15f88f6cbf0307c03e084795902e7bef7b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-09 22:17:32.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/86/890986/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"81ba1aeb8b998c68f54566130c3c1a4eb49db368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 00:41:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7ebd8055c2f4401b316ceca6281bdfa\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/14f159fb74764c0bba369efa5153b963 : TIMED_OUT in 3h 03m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35e20f07501a41fea3a0dcd3b2f0029c : FAILURE in 37m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/606098da822a4e86919431d0b48d19dd : SUCCESS in 8m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0467fd9e9c054c81b87daedc82a9060e : FAILURE in 26m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8498683de24c4f108e0870eef449ac34 : FAILURE in 30m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/986e4cad9f184f2db684bf6da459cb5e : FAILURE in 24m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de34d1fe462c40338c59ce03b3ef9edc : SUCCESS in 11m 06s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/af4a3ff39ad9402aa0a8ff4a34b4e955 : FAILURE in 20m 09s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cb60bcf51d914d4baae9d6b4905c7318 : FAILURE in 1h 00m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0baf69eecb6a4f53a3bec0e3f40de200 : FAILURE in 55m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/60feef5caa364790af1f990b9d3dc9d5 : SUCCESS in 1h 59m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8929b20abb9d4b82a790f7444ea54b26 : FAILURE in 1h 09m 37s","accounts_in_message":[],"_revision_number":1},{"id":"d3d302b4e95178179d7fb135c62c9b0392526db8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-10 20:01:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bdeb593bbd8bb916c189e71e30db9689675852b1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-10 20:43:12.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/86/890986/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"adfdbeb92bb6e75f84971ade0af9464d588a8f7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 20:58:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad644dadc96f440ea25bc8b496dd64a4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4190acae5b3a4bfeb0a87a76be48e211 : SUCCESS in 55m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/006223184a014e6a9aaeef587d399a4e : SUCCESS in 55m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d7e553cc955c4c81bd6153fd8125492e : SUCCESS in 54m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"04b5af472dbc2806f67dde44e91c172d62b65a11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 21:54:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecd4b7594baf40879c07a414e3fbaa43\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/417d1401e0e14a67ac1149428690cb91 : SUCCESS in 1h 51m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55e9fdf79a364c36b2f04c66438c30fb : SUCCESS in 43m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e66d214d1914e20963539194baec3f2 : SUCCESS in 9m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c82b3a1916de468281a76a1aa91df46f : SUCCESS in 22m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e87df553ec5414e8a2b068ec9aae72d : SUCCESS in 24m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b55523b723674fa49a77914fac8fec27 : SUCCESS in 23m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c80aa6b2fa6442db9df8ceab8593e4d2 : SUCCESS in 13m 00s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/ca437382ee5247fab87b0e1fec009d09 : SUCCESS in 21m 46s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0174b1e5a88c47c0af42bdc4d25d1c75 : SUCCESS in 46m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/deea026123d840229f7a5aa8505cbc31 : SUCCESS in 1h 00m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/99e522a732324607ae3f12aa4a9a1336 : SUCCESS in 1h 34m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/800081bddd5d4322956b11d5245084c6 : SUCCESS in 1h 14m 54s","accounts_in_message":[],"_revision_number":2},{"id":"41f2d5e98d9abf2d32471c24cbdd2c819dc8de17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-15 17:21:39.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5c2384e4e10c0e81f1bde90da116ed71accf5bea","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-15 17:49:33.000000000","message":"Patch Set 3:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/86/890986/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"2b9d5ef91c97b469530b86a1f7c27f340172ed10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 18:30:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d93b446036242ebb1a01ca6061d2fed\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfe431f092ef4e9b9a9b00c98521ba05 : SUCCESS in 36m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d4c388b3d44434686182c9c9d6d82bb : SUCCESS in 23m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/04c1ed3f2b51424cac148b982e8bec1c : SUCCESS in 43m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7a4be05a87d3f4ab46c7461595f8372a2c43cc14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 21:14:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/867458cf21154588a967e440e8159c00\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3c5d02b8f4034069bed507a3ad11c4f0 : SUCCESS in 1h 44m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9d089088ce942b3ac9ae2890463d1dc : SUCCESS in 44m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bf3068a10b44f709ebec1457fea2646 : SUCCESS in 13m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44d293ac22184968920b52313d840b35 : SUCCESS in 33m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/303ce1381e1b4aefb0878d8a3af4a0d4 : SUCCESS in 28m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0494335fd276487faa0a05b06472bf8f : SUCCESS in 24m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8057a9beca1462aa37a1d057f5b1307 : SUCCESS in 11m 41s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/b3473e7fb1f346e7a389bccfdadba9e9 : SUCCESS in 26m 00s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/99f713c63ba74f6a8eb743b459e5d9a0 : SUCCESS in 59m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f881825f7e13449d8902f319dce9c09d : SUCCESS in 53m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/29664e809f60472d8c091cf03b6ff4c7 : SUCCESS in 2h 26m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a3b9a10ac06e48f3bd11ea424b02fc4b : SUCCESS in 1h 27m 54s","accounts_in_message":[],"_revision_number":3},{"id":"3576382b3830d66d3a19a29ec171c21a3eeb6e66","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-16 17:00:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e4e215fe1e5e36a342a460998273056d9f52a6a1","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-16 17:00:51.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"07825f4ca3cfec71ff32e99fd4296eb340c0c06b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-16 17:37:54.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2a4b07158d0f82ab7c967af961bae93049cf7b91","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-16 17:38:10.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"36288d05a157fbb47bd92469f029bfe45e204ca2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 18:25:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c96c0f1061841fb9c4c52aef7d938b0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ce882f0a63e4a22b12ac270db6f8df2 : SUCCESS in 20m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cffab839d24e4f36a1c58a4c579db8c8 : SUCCESS in 45m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a647a26aca53473abbefcd134a88dadd : SUCCESS in 27m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e9e7579758cd2687882644bc06607b475f2f3cf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 19:43:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/215aa744240a4fb7883ae9d9e0a434a8\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a039a7b5b3344cb1829d653c238f0874 : SUCCESS in 2h 03m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/916b96cd19cd46bba97dc176ee35230f : SUCCESS in 34m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a66639700364fd38bd5f83afbecbaa3 : SUCCESS in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c64ff1ffabb643e481553dc918a21852 : SUCCESS in 22m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0ecd607c3ea45ab844402cee425fdc4 : SUCCESS in 26m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a66d84e0bcf84829b76351df814f5403 : SUCCESS in 29m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19a41ac107ba4276aadc632d12d578e0 : SUCCESS in 13m 40s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/7c6821fee3bd476982492a0b45a28119 : SUCCESS in 26m 53s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eeebdecffb894711ad2db4f2db676302 : SUCCESS in 1h 06m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3f6383457b3c428b84a82804661aa13a : POST_FAILURE in 58m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6ca9dd6bfaac429098112c4ba4cac5f9 : SUCCESS in 1h 44m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1c087620da804be491f35d425c2062d0 : SUCCESS in 1h 22m 36s","accounts_in_message":[],"_revision_number":4},{"id":"51f6eca16fdb562cdaf13796cd7dab1913eb9c00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-16 19:50:55.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"1a4e3c647ae476204efb4731f82cb0f15c4f24a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 20:20:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e7e317b6a634ae7b18e13400732020d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb276ea1f3a54dab8bc37ff0885accdf : SUCCESS in 19m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7aaf0ac582f45d1ae2334196266a5cb : SUCCESS in 20m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cea1dacba5c84ea78cdfd7bd6ed7c2c8 : SUCCESS in 28m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5603543a3f1e1fedaa241d91d5ff344aeb8ed583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 21:31:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76150b91cbdb4202a157069adaab50ad\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9473c71bd76145418987a98dc5b5db5d : SUCCESS in 1h 38m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55408a7e68594427ac2e52f798a2e101 : SUCCESS in 49m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/380b94f9ba144451bcbdedd2131d0248 : SUCCESS in 10m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d120f1f1f944e028aab7034baf687c4 : SUCCESS in 27m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48e306e69f824c37913ddc0fc94d965e : SUCCESS in 24m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac999da917be48209b9057ea3b4ae5e6 : SUCCESS in 25m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f2137fc972444e6a4ec8d4303f95d3d : SUCCESS in 9m 26s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/bcf96db982dc4ea58e6a41e8b53ac667 : SUCCESS in 20m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e09ef15f784942ba9c5c0148c2d5aced : SUCCESS in 1h 01m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/43f40343d09f4829b39898ff06b1e56d : SUCCESS in 51m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e144912d45a4458f9e9a3b2b2a93d0a8 : FAILURE in 33m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bb55582cabd945a8afb9186d62612959 : SUCCESS in 1h 15m 42s","accounts_in_message":[],"_revision_number":5},{"id":"fa974d7c6c2af19ea43dec5ea988e3c4fcf56cb7","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-08-21 13:18:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2195b4ef3ff173c5d9b766d20ebeb7fdaebbea55","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-21 13:20:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1a9cd59984bf59190bf0468b297ea4cbc36de80b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 14:07:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8818ae8085344fb9bca4fa6dd8276e07\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/135e15402eb64d1ead64f8cc1ca50ec9 : SUCCESS in 27m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/440b01fdbbbf4bf4a30744b70b9cbedc : SUCCESS in 23m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10484900d77c4e0a8f6ffd00d6ec1691 : SUCCESS in 46m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/26710d32ba954bfe989a3c16234babf4 : RETRY_LIMIT in 2m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"13c91d6ab0adbfbf4dc48ca0234bf918a36d71db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 15:35:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4260ec509d3943bdb8fe47f2ae363546\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c67729e5d4544caab6b1775206f0ba37 : SUCCESS in 2h 15m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41e6672f3077471bbffced747cd30d14 : SUCCESS in 41m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00bc120da54641428b3f76d1e51b8c60 : SUCCESS in 9m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f6c37b93c7946b49bdea7197bb31364 : SUCCESS in 28m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4661c99f0df2407b851adab809fda3ff : SUCCESS in 29m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/220bd176ae7a443e9629c198e5202a9a : SUCCESS in 26m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50f681a5a6744430a5043fc720fcefbb : RETRY_LIMIT in 2m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a14a39344cd4a20b0463f261ddeb840 : SUCCESS in 10m 17s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/46beb1992031492cab241645c5e8eb82 : SUCCESS in 19m 44s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b679a8a78b9d429fbc49127cfaa1ddd7 : SUCCESS in 1h 04m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/04d28f64a6564de0a630a6577d5fbbc6 : SUCCESS in 58m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ce0fbd73405641018c02894d465e8c08 : SUCCESS in 1h 34m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7b5da0e610f34b8fa7e2e72448dea462 : SUCCESS in 1h 19m 37s","accounts_in_message":[],"_revision_number":5},{"id":"be497f6712ecb896cc2443ac346a3f841cabb2de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-22 13:10:51.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"583ec04adb53a2d6b74c73a96eaa9c06f641fc6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 14:58:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24d369fadf3840f0808556159c1cd55a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37c47bd7b87b4c73b6ebc13ed3c862eb : SUCCESS in 55m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5119ac01820d4856bceb0d6b7bf8e95c : SUCCESS in 28m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4ff86bf24c464eb3bd8dccf08fbe0193 : SUCCESS in 24m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a16a361694b142b1904cd4226a16ca9b : SUCCESS in 20m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2d58f579676bb7a6f5c797c8ab9c0276d7277664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 15:01:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8865037e96c643eeb987aa6e42486914\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/62922b488e9146dd8a5af07812c95fb3 : SUCCESS in 1h 28m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68c1d52298ac4c5e80f9278ca61d8cb4 : SUCCESS in 40m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69fb7b07c90644a3aa815d7728bdc281 : SUCCESS in 11m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90d9b420fc6246dc8c9689401a89acd9 : SUCCESS in 24m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fde5258997049d5831c6e60bac223d2 : SUCCESS in 27m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e29449554e5c4fba98597d8d5f6f4ed6 : SUCCESS in 29m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/763816d833624a1faa1e4ef4adeed39c : SUCCESS in 27m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/089bf8757bce4d16860db7e88ff1b17e : SUCCESS in 10m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/110f1faea7714ef39aa5ecfc83e001c7 : SUCCESS in 1h 07m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5440a360a39c4bd480ecf0f20d602b14 : SUCCESS in 50m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6befc320b6794639b194e570a089dc06 : SUCCESS in 1h 34m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/71a3934ac1124bdab5b3d45d2e208646 : SUCCESS in 1h 27m 33s","accounts_in_message":[],"_revision_number":6},{"id":"8bc6c7299b821791ad3970d616823ab437637562","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-22 15:09:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9744fb3deaaecbd2d81b4721035fa7ff5dc003fb","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-22 16:04:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"79895a1e6b5677e5a0b595fb386e65f1efbcc714","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-22 19:12:10.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0bd841f54b0faf82440d302e555f72769d5e460e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-24 17:47:09.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"00988fe38028c60e1f773d1a52ff2ad5427d7000","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-24 17:47:25.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"4b0e2c3cdabfde2287429729b9073fde91174f19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 18:29:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c5ab3406c284aec851f8c37761b6e20\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4293a83af3dd4db6a2b89d95852c98c4 : SUCCESS in 29m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a393b6b0a1b4673a918348509f882f8 : SUCCESS in 34m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d799ff452d064bd4be6d1b9b0a2aec95 : SUCCESS in 27m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3ea46deac886494ab8b9c119eb89c834 : SUCCESS in 40m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ff0a03f7c4e064136659fbd548db4565686c9703","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 19:29:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10f16f48bd28423594644985920c7ede\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9b9e4f34368f40f1ab4203e668eabffb : SUCCESS in 1h 36m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6aa150d316a545c5b683390ef081c352 : SUCCESS in 29m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eba141f3160d41119548f18d11d0e727 : SUCCESS in 11m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a953697a8bdb42469430dfb668fa7857 : SUCCESS in 28m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d98c33db7c824652826f96b99f5f40e8 : SUCCESS in 30m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0957fd9e882e4ff790991ca5e1a458b6 : SUCCESS in 23m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb4d4c1e680547a8abe52b1eb416c241 : SUCCESS in 26m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb65996ce19469080e0f6e0328493f7 : SUCCESS in 9m 56s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/93e6c0a8315543368e5574ddadfcbc3a : SUCCESS in 31m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c32ff9312374c3992a467b52a955bab : SUCCESS in 48m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/cf965a2bfc8241cfa2a41bbea1253429 : SUCCESS in 1h 05m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/53e40a53b53b4053aa70ac3cad5009e6 : SUCCESS in 1h 39m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2cec1c1d0feb4cd1b21d51079e9d293a : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":7},{"id":"204aebdcabe210f50ac4fa3c740df7557132cacf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-24 20:19:22.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"39df1d56e73c71709e3ef8746d443d342f06b426","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-28 16:47:29.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"b99868beb753679f43785dada6a7f0ddb33b7f0b","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-28 16:50:10.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"53ba7e0a4791497863b677dbf13a359f651cd8c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 17:43:10.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ea1a5dc7a824ab4b693d24d9cda2ee4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6811a6eaa05640daa2db5f4a9e595287 : SUCCESS in 53m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63918bf8d64f48abb5261bdbbdc6d581 : SUCCESS in 52m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/231491f7b883466b95e146cffd2953e7 : SUCCESS in 53m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/99ba767cdcd442aba917989f98c70bad : SUCCESS in 42m 44s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"45624fa89cd9bb7ecb31c68c500444b0dd925ac1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:31:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d686caf17b9b4b728c664f9461d9851f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/01d7e8b9f01f496fa7deafebfb6e4315 : SUCCESS in 1h 29m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e753d96c37c478ea309df7b8a2e58cb : SUCCESS in 36m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18725532f1664ce7ba3532aeb407ffa3 : SUCCESS in 9m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fa1e130e6fc4915843f1e9092a87404 : SUCCESS in 25m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03acc16798b94c8c9f2b13640c06a38b : SUCCESS in 29m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a3ee7aa2f1846558c94713a4de3d47a : SUCCESS in 26m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94abae0563ba4fb18506ed94c8fa322b : SUCCESS in 31m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e7dd1c3c7214e4a85862a8fd908ec40 : SUCCESS in 13m 32s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/feb2a11b37a64caeb196814416b0a9d4 : SUCCESS in 30m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0e42cace806e4639a379e61026f2589a : SUCCESS in 1h 01m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/aa9f160dd6134ffa9ac083a54792120f : SUCCESS in 1h 07m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6b2cf0979423454097822433e57af0bc : SUCCESS in 1h 40m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/12c7c6841f054731bd972f9c8101f07e : SUCCESS in 1h 20m 46s","accounts_in_message":[],"_revision_number":8},{"id":"90d1dfd113521c1b3a31eb3ea01d13d50f882e18","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-28 20:20:33.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"069b180332f86f78b44d548cf8c92776d80e7c92","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-29 07:45:37.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5f01f949ccd4c30a04d0c2974dcb852ed8d5f566","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-29 15:29:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"22a2cf8d1b8114f35f234ba2a00157ecc42ba34a","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-29 15:34:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d0c1a62001c553d6c7a700e464419ad07ebcfbd3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-30 06:03:08.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/86/890986/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"553877e532d52674081cd68e718693279ca4789f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-30 06:38:30.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/86/890986/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"4e79fd779beb66800673e87958150b244ef46602","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-30 21:22:18.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/86/890986/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"de861cb8afeb1d5ca91066eb1bb69d9ab1773416","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-31 10:00:26.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/86/890986/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"f203d4c774c2cbbb8b2a7ad1e946830dc8b9a041","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-01 00:04:09.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/86/890986/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"c958c4673a88bfb8f84ccee70190f914ea13cb8c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-06 14:23:48.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"41c790421ffea56ba889a16a58898dd2fcf7eeab","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-06 16:36:16.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"5b1d991f51d67116407dfbd14ac3b95ce7c58591","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-06 16:54:10.000000000","message":"Patch Set 8: -Code-Review","accounts_in_message":[],"_revision_number":8},{"id":"2b5d1a19925abd87d2e1dbb0de599f886ea80c4a","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-06 18:25:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f779caf938d4b73858cdfba35e2bc1d751dbf0a7","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-07 09:04:54.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_16688\u003e\n","accounts_in_message":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"_revision_number":8},{"id":"8d91d8507b5c6e3f896fdda6b2a85c1ed17fda03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-13 18:32:29.000000000","message":"Patch Set 9: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"e74100c1d0bb91591793073a35664626d2ed3b81","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-13 18:32:43.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"38913c30e8671210d12918a788f3a25f6384d2e6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-13 19:07:28.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/86/890986/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"242f8e570d08bda16d4990b3018758f813606ff2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 20:42:27.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff108a84dd1844df8420a42dd7c63e8b\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5bfca9924c1643c5bee4b00f8a4824bf : SUCCESS in 1h 40m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/107770532d47412abbaba9f50175f122 : SUCCESS in 37m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e69b38cfd5a4e2eaa89e0ff93b1199f : SUCCESS in 11m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1fa40649f0d44a4837536d2a521a4f8 : SUCCESS in 28m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ff89595e95a42a1837638e77c48f704 : SUCCESS in 27m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43f56b7f22e24c35a7764894412da588 : SUCCESS in 25m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a22e7a330ce8472888ecd8724769ec27 : SUCCESS in 24m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c882bcb734f345c686dc7373c3be4476 : SUCCESS in 11m 46s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f8bd78c6f4a44594b287db3b4413bdbd : SUCCESS in 30m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d503f6b9839247f28eefd58829c47a21 : SUCCESS in 1h 03m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fd2b89aed5b048f3857b64f0f80a8c6f : SUCCESS in 1h 05m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5573411954ff4e98a9749fa1f6f94be7 : SUCCESS in 2h 04m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0f4d5ba0ab11493bb8ab4c902c00c13f : SUCCESS in 1h 17m 34s","accounts_in_message":[],"_revision_number":9},{"id":"21ab9f737dc92222cc27698a5cf6ce071bae97df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 21:30:24.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/839bf8b01136442ea8367fd8f6efba1e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac646303b28844d68ac23a33cedb80e9 : SUCCESS in 1h 01m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fef493a60f114f31bd5bd70bf2ec6292 : SUCCESS in 1h 00m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/70eb05dfefcb4b39b3b87168fc3f6be9 : SUCCESS in 1h 00m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3741dcae1bfd49db9025d5913c2bdf9e : SUCCESS in 37m 25s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1eff110abef4f4512c91a20fb5eccd36629dd206","author":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"date":"2023-09-15 08:57:12.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"0c78b1767bc42349f1ed7c4002f1319e5393903c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-18 14:32:25.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d42b496e27b60b1e58014ee0f8a180f5df879e3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-18 17:37:01.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"b3d526065fed338d766b695fd1a789bb7a1ce31d","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-18 17:38:22.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"d937b4ad3fe59781f400d6e0c2e8efcad7d56fc4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-18 18:10:53.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/86/890986/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"208d9ff61b0e710066e181e9ca18b37db14fec3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 18:33:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49bc4256b67c448e9bf6a9112fd9d344\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3f923558a9646b6be3380a4f4b7a7c9 : SUCCESS in 53m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f658b885a32415282b35c5c927a47d7 : SUCCESS in 51m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/662ca5e9361548cda54f9fa3fd8c4224 : SUCCESS in 52m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4a11892a17dc4d6c9ca5e6d3fa827694 : SUCCESS in 51m 44s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"8e5c6993b4d7f7506a962e611847d237563489a3","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-18 19:22:19.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7dd3587f60607bf1f806618ffad19dd64eec7db1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 19:37:10.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b51ae17b7120404ab5fc193a709baae8\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3104407765ec4e3c80fd6f2f4eb2f4c4 : SUCCESS in 1h 57m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42687b12d0e14bdea6d480e0847d3050 : SUCCESS in 35m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3425107d384149c2b2d2990fc9376afe : SUCCESS in 10m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed331c60eab74f4882737f6dbd0045cf : SUCCESS in 28m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c6bcf2a37424adda4e3f70ee86fb5d1 : SUCCESS in 21m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fe86dcd57814e7496ec59d0627e2c41 : SUCCESS in 28m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1eec61b4605e4c39b251c6ce22bf328a : SUCCESS in 24m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db4979c7138042349ac9f9bc4777159c : SUCCESS in 10m 22s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1284305644a04f7792d23b952e13e7a6 : SUCCESS in 29m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4bf1c82712fe4997ac9d10932918ab4e : SUCCESS in 41m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/df503fc1d69e4b0c96baf83bb760d790 : SUCCESS in 43m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b433d3b8e95f44c38ad1b2345e81f07b : SUCCESS in 1h 27m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/676a88b775794579843a75760138110c : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":10},{"id":"4c0484fadeb1428dd899a8bcf7d6a38f30b73927","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-19 15:12:51.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"182bc7c1f427b76e055a660e71853ca1d4fbc971","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-19 15:13:17.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"0c0ffad3fd36b03f41a683b0782673f46424d15e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-19 16:12:56.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/86/890986/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"adbd245518ff474e0b9d6b1ec68ca2e493cc9faa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 16:50:20.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fffbe4098c674a3dac7dc77a8ff057f4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e92eac1c0fbc48029b307e1258a482a0 : SUCCESS in 44m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2d4653404d64b9d9990f3526e1bd420 : SUCCESS in 42m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4914edcf2de74bb28c4e0ea60e760c4e : SUCCESS in 46m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3faab48161cb41369cfe7975381e5714 : SUCCESS in 42m 54s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b61e083b25b076adac4d148c9381de61e45dcda3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 17:38:38.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d629ab268ef4c77840df63a4c59f39d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/41e46cab370e424fb02a5833397a7ab7 : SUCCESS in 2h 20m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f017a217e494fb0b0a5889944e2b5d4 : SUCCESS in 36m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93d82d8b1600464f89760e048714cf38 : SUCCESS in 11m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90e5312594f449f9a8cf065d947f79a1 : SUCCESS in 25m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc68f4d85dd044e4b8f9b479f633ba27 : SUCCESS in 24m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea1b8da2f7254c238add5d81c4baac08 : SUCCESS in 28m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1fef39a89cf04ae0b50e0257f86173d7 : SUCCESS in 23m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ca4502643544955aadff35902c246cb : SUCCESS in 12m 22s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/464a8fd915f04cb39d3271a20617c4d1 : SUCCESS in 28m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d87fbf13a84e48ea99531823701a348d : SUCCESS in 1h 04m 04s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ad7b352acd4e42cd971f6192ca1e4389 : SUCCESS in 51m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b2c34ebea24a42b09891b87e2c30b97e : SUCCESS in 1h 34m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/100714e683d845e98ef4e33f751efab9 : SUCCESS in 1h 08m 30s","accounts_in_message":[],"_revision_number":11},{"id":"54540a6831e1507cd14c4f6ed75d6299f1525100","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-19 19:07:50.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"2b80a7596a59556b3cdceff678eb4fa34e619c0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-19 19:25:14.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"7a78dfe4ec87b67da2d1591fe6006900ad0dfd4f","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-19 19:29:19.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c5e0f8b95170eb727d105cc622d648b4f0d051ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-19 19:33:09.000000000","message":"Uploaded patch set 13: New patch set was added with same tree, parent tree, and commit message as Patch Set 12.","accounts_in_message":[],"_revision_number":13},{"id":"5732c6d13b87a5d0494bd257f5e3a9ab076a8233","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-19 19:59:38.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/86/890986/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"67563c947db75e5c7bb7819bf0935c6c6647e85d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-19 20:33:35.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/86/890986/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"6dc9772b11a1c242fe34cbb8115028100b7fc2c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 21:19:25.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7398b0ef75b348eebacc3feb79692270\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bec1d217c7e74a4fbd4b540ea9bfafe4 : FAILURE in 54m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f097d5a019474260859658da71143fde : SUCCESS in 52m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4d926f709d034ce8b3f9320377517dfb : SUCCESS in 40m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b8d7f0a88d0141e1bea6b220e92238d9 : SUCCESS in 35m 28s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"30156839636268dbec6aa4298a483b201f1d9274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 21:48:52.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/113a691f04eb402db6b738d78d2b81bd\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/10ff44e0fca242e5b4a51d9c92b38cc2 : SUCCESS in 1h 31m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5785e8cd515f4927a65ac269fd617eb6 : SUCCESS in 47m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32e775bab0d34987a7ea394256b5d0d0 : SUCCESS in 11m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58f2c65d8d8c4adca16de1b4e2dbf99e : SUCCESS in 26m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/399138b1eec544709b5bbb8ce28f2f2a : SUCCESS in 21m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e778c26dc32f4a8f8c69b8770c31bf0f : SUCCESS in 25m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8fe4b69eee8b40359c95ed829fb9f0de : SUCCESS in 25m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c08229a6bda4d95b1f846c81c34ddb3 : SUCCESS in 12m 05s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2bc5113c286b456a870a8d01905cdd1e : SUCCESS in 27m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b38ead624ce4608b5acb71679714d5d : SUCCESS in 50m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3a3f2b7fc539448dad6c5f815172e9bd : SUCCESS in 1h 17m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0b20171a09ca4844884a3f73a29a2b04 : SUCCESS in 2h 06m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2cb21197c8674ae98f3064e6091c7646 : SUCCESS in 1h 01m 28s","accounts_in_message":[],"_revision_number":13},{"id":"7ba72cca54212980b42e5876bf733e99d432c568","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-20 14:50:01.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"b4d1bcb8643f2f33640f6b7b63d23bda674c2478","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-20 14:50:07.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"429707607aadd9a1ce074b909e859d33e76b2c5d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-20 15:25:15.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/86/890986/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"723f592ce7d096320dade7abaf904d323d206dee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 15:48:49.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc639b52c99a4dd493725de5eabd5431\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e42fefbec134593a816c5acaa8e829b : SUCCESS in 43m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c80a2d0286041838804db1ed4f0880d : SUCCESS in 51m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/75f9cd3b6e844e7eb0fd7ef2b52d422f : SUCCESS in 41m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/63706a38a47d44048cd4147b3b90c2ef : SUCCESS in 48m 19s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"ccafb0d3efe44cab5a2354d5faa5ded061fc9525","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 17:07:44.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ffd565baad042c5bbd3b64e7bc94ef9\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4728a759320c42a3bb945ee96bd9304a : SUCCESS in 1h 45m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afccd28850ac4df694e1e9672d5f7249 : SUCCESS in 30m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4345851974b7420f99e9970c377550f4 : SUCCESS in 8m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9330f5bd010e4ee9b633224afa50407c : SUCCESS in 22m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d4a900e73e24b93a3867261732d5aef : SUCCESS in 29m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/802ccd36adf54786885fc1dc0ec69bec : SUCCESS in 26m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/660bb404899b4089a7a7149fad83187e : SUCCESS in 27m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382145725f4143d788d05dd9c237abd5 : SUCCESS in 12m 21s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2ea91530f6614f8faab772a3be554177 : SUCCESS in 43m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c771be04ccf4839815bdacd2c32dc6a : SUCCESS in 59m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4410b302d19c4be8b12df6e059ddb3c1 : SUCCESS in 1h 03m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8aeb95eabc4045169419b40be0cb02fe : SUCCESS in 2h 11m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/dbd8e4ae113a4a38ab7dc151370555ab : SUCCESS in 1h 12m 52s","accounts_in_message":[],"_revision_number":14},{"id":"aea58e15585ef7f3aea0fb887fba03b279ba4a4b","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-26 18:38:17.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"2f7437ce47920a4d15b188d749da21813befb645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-10-02 15:07:23.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"9ad8c4b4561e9399153f9e52e0f01b556b783082","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-10-02 15:07:35.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"a88b23c518352ff51121eab91c340e74ed5bbdd9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-02 15:40:07.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/86/890986/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"d76a3a27baea1c70553aa861c6dd3e91b5ddc07a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 17:19:38.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70a4825b889f47eda7acb5276f373d21\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d305f0f11f5c4d9faf725bc68ea76129 : SUCCESS in 2h 10m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78fff0e67b7241e4be29c6df34633615 : SUCCESS in 28m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b67ffa194fb44f189ae94a529385eb61 : SUCCESS in 8m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/905a9f89ad8c4350937b078303fd16a4 : SUCCESS in 27m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/504aef1d9f5d4600b964fe8270d7210b : SUCCESS in 25m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c89cf712cb454c34bd0097d50de880a0 : SUCCESS in 24m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/532f45c1b1f74e6e9e855273c36c0c95 : SUCCESS in 27m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6717347e963a4e5197dcac4c029fd1c0 : SUCCESS in 12m 17s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2bd3efb62d8549e9ba0ae4a789fa8439 : SUCCESS in 28m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac67fd0255394a488dca41b14e9226e4 : SUCCESS in 58m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/74aca383f76d41098813de65a1e0ac70 : SUCCESS in 54m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f627c9f44e97492699d1b9d94dabda68 : SUCCESS in 2h 05m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/05b468a567e940c69b5408112d014aee : SUCCESS in 1h 18m 31s","accounts_in_message":[],"_revision_number":15},{"id":"e4f14452632194c9edddef550f38d8b6854da39c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 20:40:29.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c37d72fb63a84c07b9f8e8868418f6ea\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2ef34b1e3b844b9b47b4990bf28e65c : SUCCESS in 1h 00m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f1a6f819d9c44a4833a699dc4b2312b : SUCCESS in 56m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/77d9b51e209440d299b12037e276df04 : SUCCESS in 58m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b0bbd4fabd4c497aaddf0795c0fdf811 : SUCCESS in 52m 47s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"e0457ab913a6587939cde199ef7884ec02750d71","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-04 18:04:19.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"e0926ff5e7600b8227a1c8aed8b0bbc612f2000a","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-04 18:05:37.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a60d2b5ec90c0784100a001cda0d3234954b8a41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-10-04 18:47:58.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"aa15fbdcb8d8187813aef05a037fa83fa2cbcdef","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-10-04 18:48:06.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"0ed8c7a70dda508f62f31bbcb7513d6e8507656e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 20:34:19.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a40b36163c7b486e84d87942d782cc17\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/35b83b8a615342e19be3957bfce3b7ea : SUCCESS in 1h 28m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f9c654d325f47bcad1fd17323653059 : SUCCESS in 40m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4886110f23a146e08d9a02afba4702fd : SUCCESS in 12m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c77e417251a48ad99b439f8b8f3f163 : SUCCESS in 25m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bacef00429d4837b62844600196c5ae : SUCCESS in 21m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0225cb41eeee4ec6bc1b21e1197121fe : SUCCESS in 23m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e13f9589761142d19286bdabb5c0264b : SUCCESS in 22m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f6bbfb073b34529bb65fae75d195a3f : SUCCESS in 11m 02s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/56e8dd5e07e54a849f33fbb23d187522 : SUCCESS in 22m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c4a3df9b9c0c415398a40e42bf47f86b : SUCCESS in 55m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/47817f7f3d1a44869ef6cdc92e85fad4 : SUCCESS in 42m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/618998c6d61f477aa879f14822f94877 : SUCCESS in 1h 33m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d3c2c1a6ec7644628790d56a082c1f5f : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":16},{"id":"182933408e1224a9aa9668c2f82c331271e6eaec","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-04 20:43:07.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/86/890986/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"ffceb01fb9c8a58a08654112ef123dd0aa632d72","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-04 20:43:39.000000000","message":"Patch Set 16: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"e504396794efc8501b43ccb94a5497981f62e5da","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-04 21:25:18.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"2415d843b5c0402cb2935f3e9b76ad27f4ca3f7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 00:12:06.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3adf32a173a499399569d9d461341d9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fc508ba654c4bcbafd8ef27593b6f5a : SUCCESS in 56m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54be7854eb704b96a00c9b19262ae8a8 : SUCCESS in 59m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1fdb6fa333224359ac1a4d5a6d527a7c : SUCCESS in 43m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/df3f0dd94e6f44c78dc649430a4776e7 : SUCCESS in 42m 34s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"7dcd3e81ab451b8317b7e8fdbdaa7286a1e3ecb1","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-10-06 03:03:50.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"3f35754b567fee1500aef1258b3d9a84afb6e159","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-27 01:01:15.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"1a562258cf0af6c115824bf606ee5a194106fc90","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-27 01:01:45.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"13bb4314cdde809d9edcbc3b6424e36d13ab4b97","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-27 01:33:39.000000000","message":"Patch Set 17: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/86/890986/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"94ba1fafcc8d510f719d4101694e03ae3c630587","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 01:52:38.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf332a9fd65a4268bb1961ae8ee3bb22\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/487dd77008da474fb4debee246545374 : SUCCESS in 19m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7943c7f943ba4df7843774b9e96b1feb : SUCCESS in 49m 00s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/11971fb2ff5a47f085bcb87937eefb24 : SUCCESS in 18m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/afe376c97bad4ea1af77ee8e1a84b26a : SUCCESS in 44m 41s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"8c6a4092c97b7810e42ee168e49f9ddc0dbd5840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 03:37:01.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87070fd356834ce8a83c5d641ec12f86\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/77e9aa76aa454218bc28bed777b7f7e4 : SUCCESS in 2h 09m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e1b9fe5261764fa394a958e8c9b631ca : SUCCESS in 1h 54m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/2304f5a3257b444fa07f2e7d43d02cae : SUCCESS in 1h 43m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b44c32240ffd416a8fc8682ac9243340 : SUCCESS in 1h 30m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e8de160bc48476781aa4ae6d03d03f2 : SUCCESS in 39m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/135e721717c94de28c0a45795bb716d3 : SUCCESS in 11m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a098ab3080f421fa8ed719eedca9dce : SUCCESS in 26m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d8fb6224ce8410db86891bd8d12c4d8 : SUCCESS in 26m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b17c87e67404158989c5bdc1bdc9b49 : SUCCESS in 21m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0fe623b1698840b58c34248096eb6abf : SUCCESS in 24m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec1dfba4aa1e40a0a3f008d0908edabd : SUCCESS in 10m 33s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c0f2dfa29fb44ed29933d1ad291d2b20 : SUCCESS in 31m 40s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b49e37cccc7849a19fff9e257ac7bc70 : SUCCESS in 1h 02m 07s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a126e0f2b4ef410ebdeba0bb1fdd3542 : SUCCESS in 1h 17m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/09d42917124944589f85ef3be7fec1d3 : FAILURE in 40m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e9ae8cf940b47b297f26df33f9f5111 : SUCCESS in 2h 30m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/30eb908c1d4b4322ab3aac536afb0977 : SUCCESS in 42m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/948c30e730cc4f75bdd55bcb5585a638 : FAILURE in 44m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d5510e632c9b47048acb9bbe09ae09f8 : SUCCESS in 1h 35m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9c768b1a23df4c25bce5c3f56b799236 : SUCCESS in 1h 32m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/63ab0429ab2c4adabb03b298cc10d7f4 : SUCCESS in 53m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8e549fc727ce47d5878234309511e321 : SUCCESS in 1h 30m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0c67136217f8470795d44bf7e4c06d08 : SUCCESS in 1h 23m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b146146031684dec98f6287968e4c915 : SUCCESS in 1h 09m 19s","accounts_in_message":[],"_revision_number":17},{"id":"8fc9b32cbec491d021c933cf1096264da3490647","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-27 15:42:07.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"23600482b83067ca27007b9937059c3d064ca2a3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-27 16:22:42.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"294a952e0b185082587f7109c7127f92491c32f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-27 16:22:47.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"d046b061504681e200951614986f6bfb1bf9f74d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-27 16:54:31.000000000","message":"Patch Set 18: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/86/890986/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"c7c851f49f85d7dd98e877618116d6ca4de45178","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 16:55:30.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/298beb9fb9ed4fc59068779f268e3b70\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c79275466a2a49cabb67e63638a906dc : SUCCESS in 21m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/497c4cc5aaae41c2b565f21114b72cf7 : SUCCESS in 18m 51s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0778fa7bcb704856abf9328fe447f71a : SUCCESS in 18m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9f37a92d691b46f28ada4258da70af04 : SUCCESS in 31m 35s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"e87360ca672ec39d0140bc2627c4758f2d9a182f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-27 17:31:31.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"05df88a989b47deef808fe32971afc5f0e5f24a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 19:26:49.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/918aeb89a34542e59a1c095c56a9ba10\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/07cbd050e7fa48a7a2d48bf5e2e5a01e : SUCCESS in 1h 32m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/64db7cea7d7c44cd9ca861d208272d99 : SUCCESS in 1h 57m 53s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/453116cf1cdb453da204a3a448d5e3d6 : SUCCESS in 1h 43m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1596186a3202411192864f3ec9646067 : SUCCESS in 2h 01m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14ba68be058244509c328e45dbb3ef2f : SUCCESS in 37m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d8757a6f7d949c39f28cf83f55ab506 : SUCCESS in 10m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3f75351c53145ef97075ea7eefa7d1a : SUCCESS in 30m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d2909e5ff764fe4acaf3665c9db4f03 : SUCCESS in 28m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e04f39af9ef149baba458f551f494230 : SUCCESS in 28m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5eba2e3de3a49769eecc59403873006 : SUCCESS in 23m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9796503527664225adc638f3db52ba35 : SUCCESS in 14m 28s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3e808694d3c148c1a9182dfbcb514300 : SUCCESS in 30m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e4778b9e91d54b5296de0c0bbe85669a : SUCCESS in 53m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9d1f96b3b2664e2d855fbb019a5b7a13 : SUCCESS in 1h 01m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9434117fc1564bc59f16512036e35a2e : SUCCESS in 1h 02m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b5017ffd88b42988b02edc98ad74fcf : TIMED_OUT in 3h 02m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/efb0b06126aa460eb7c1b333fcb78e8f : SUCCESS in 54m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e15613ec7e2841748f5c94d5477d22e2 : SUCCESS in 55m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/845c34bae169472e97c0cdf68c067f37 : SUCCESS in 1h 28m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5d4a364ef0f4406980e7f494632284be : SUCCESS in 1h 11m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c1a363ad70a946b69b3037e4734e189c : SUCCESS in 1h 00m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4c80e3ef23044e91956442c720e18672 : SUCCESS in 1h 45m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/802a6a33efe741d499e53ce1cc04d346 : SUCCESS in 1h 34m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ccf771e6ed9440e9a88a6a69606654e1 : SUCCESS in 1h 12m 30s","accounts_in_message":[],"_revision_number":18},{"id":"82404168a6fd95cdf5b9125bbf5a73a6be6ba7d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-27 19:42:53.000000000","message":"Uploaded patch set 19: Commit message was updated.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"5371993dd0fffa043e2d6f3f3ed7988e0bdf43aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 20:17:43.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eda8eb9bf4304d9fb3e60f9727f3d199\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0d974a4c4804b90b8eacc2dee8a5edb : SUCCESS in 21m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e2eb676b93c48e890c4e7e6b3744365 : SUCCESS in 21m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/af9adfdd3c994527aa70ee61adf705da : SUCCESS in 30m 16s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"18719b69c15a7594e1e1035d5701255e5b43e900","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-27 20:46:14.000000000","message":"Patch Set 19: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/86/890986/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"71ed261a43777aeada43efd922ab6622542920cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 21:56:33.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f21b69122211406a82d33823dc4eaf06\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/20f97d49cc31496baaa661b261757ad6 : SUCCESS in 1h 31m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d8739846de68441395ff92b46bcf5b85 : SUCCESS in 1h 30m 35s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/881ded7c506a4dffaa46eb3c85268034 : SUCCESS in 1h 22m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0c9203849ad143018c1423702aec98de : SUCCESS in 1h 21m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8391feb260842a8ac332ba9c553f4dd : SUCCESS in 34m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56841f1909f64250b6ba5b85205eb17b : SUCCESS in 11m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f686f9ec87774c0295fcd9edfa432344 : SUCCESS in 20m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d948a4e37af94096893241ab5fcdf179 : SUCCESS in 21m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/626b61e1d3284268bc5c957f86ec63ee : SUCCESS in 10m 15s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f10f9854c9714e809465043ddca1974f : SUCCESS in 23m 21s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/da2522f427a547eb9cd768e1b8d54c45 : SUCCESS in 51m 04s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3ba85ec41b8541c9bbd6304f689e8807 : SUCCESS in 48m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f55ab4a41ac7486c9f6b4cfc8c0ced62 : SUCCESS in 43m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/80de9d0381744abdb1285d5fce97b0e6 : FAILURE in 1h 42m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/805315243c81462b990f2681c2c88441 : SUCCESS in 44m 04s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b8d8e4e681da4ad4b02a9f4bc18b4f41 : SUCCESS in 57m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/db64db6c25ad4b85b773462d428b2e9b : SUCCESS in 1h 09m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9ed67448415d45649f21b1de139ddbd4 : SUCCESS in 1h 09m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c8d7d1b0e7e349b0b8ab06ad04e72b28 : SUCCESS in 42m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f678344895494284bd531d1c1867170a : SUCCESS in 1h 34m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/924ea71bbff849ea8d0db7b55528a2e4 : SUCCESS in 1h 45m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c6f8d0cdbb2e4bd69e6f03c966dba676 : SUCCESS in 1h 08m 06s","accounts_in_message":[],"_revision_number":19},{"id":"b9b64670c8a24daa20a0f28c3f8e9cccf2965435","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-29 00:14:44.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"ee903398ed3349fac9193a74c5fff55b65a917d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-29 00:54:00.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18e33bed403a427d81e8368a056f0284\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/528a0b0fd43f46f98f302e303f910184 : SUCCESS in 33m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a8caa86c974a4cfa944317da842e745d : SUCCESS in 19m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e3f7d759daaf4844aa388cb2db784d2f : SUCCESS in 36m 33s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"cbe2987209bcdb8bbb8ebd54a3612014a6d5b701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-29 03:18:22.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/50b937211d2a47cc90b74046e50ed695\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a346ba27994d42b39f9466da048fefdd : SUCCESS in 1h 44m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1dec5f5ad56c4b4fbec4bd2a261c2815 : SUCCESS in 1h 38m 04s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/6d647039a0e84e3e85f5249afe3759c5 : SUCCESS in 1h 21m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ad45a9d9267a4df39615adf7ff93b5e8 : SUCCESS in 1h 21m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/334361d9c1194115a2359f8df7b3ccb4 : SUCCESS in 28m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac4c8871e9c24e2cad71e7187771e9e3 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/217223c1acf645b5bda8a24708c59c13 : SUCCESS in 21m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b26f59b67c9848ac946c02c55e6241bd : SUCCESS in 17m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f09a59641e74428880da0620f0aca8ca : SUCCESS in 12m 00s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/867f9231e42044bdb5f35d75645a397e : SUCCESS in 22m 54s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a6429e6f2efb4f38b92f093ec3fc8067 : SUCCESS in 34m 17s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4fc4217c2d6e4a22b813799e01e6802f : SUCCESS in 49m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f910869b22a4231b1e389099dc51183 : SUCCESS in 43m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/26cc3d338add498cabd728e5add8e077 : TIMED_OUT in 3h 02m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e85944e21264586a977165c71ba2b70 : SUCCESS in 40m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2157a923422d4f74b9152bdc80aea5a0 : SUCCESS in 45m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a949619bd5814573b87e6d793b9767d1 : SUCCESS in 1h 02m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e39dabf9e909448e89c62257d11d81b7 : SUCCESS in 1h 02m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1d63a3b85b2c4524a4047aacc2b4c975 : SUCCESS in 49m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/50c6bae1407542c5b9556cfe9e96d31b : SUCCESS in 1h 18m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/baea2184fc704c9f86029b71e5154c0b : SUCCESS in 1h 15m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/297f8f1736e14097b8ba25894397c362 : FAILURE in 39m 24s","accounts_in_message":[],"_revision_number":19},{"id":"b633299d73c1b99e354e02aa84857aaed42475a1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-29 04:05:00.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"892ba1775994e2d83a9b66a6dcc2e2f349d6f4ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-29 04:53:36.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54a261879adb46fc91347c7ec744f642\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c80197808fa541828e1567e77b5d3176 : SUCCESS in 45m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ef2794227f21405ba6dc7b4b6c4316e3 : SUCCESS in 19m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/be01307538834b58b2323b841bcbce76 : SUCCESS in 16m 31s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"e1933fec22af30814b3f9cf6fd22c4e28bd3c887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-29 07:39:08.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/54ab16f18c57492992d511b70e9a4267\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1db425617db340b2ad801b1c880db646 : SUCCESS in 1h 26m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/74784596b52b41b4845170ac068195de : SUCCESS in 1h 47m 50s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c35d8e2a706a4b49987f4afecd31bccf : SUCCESS in 1h 33m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1551527db64e4ab5a78334685a8549c4 : SUCCESS in 1h 42m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c46bf9df11f24167b1629dc0a067d365 : SUCCESS in 37m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05c87fd0c73a4ba2bfea7bbdcc996b2c : SUCCESS in 10m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0b74be0d10f438db484b0b1b250b853 : SUCCESS in 22m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c1e8c7e4eff4c2191663119869d9a50 : SUCCESS in 25m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce1f8e7de9942ed9dba236b1bc18bfa : SUCCESS in 10m 53s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/aadc32d732bf4cb4a53346a66ed247b7 : SUCCESS in 30m 47s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1ed1d5232d254d5b828c7b1eaf20e5fb : SUCCESS in 52m 47s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9f5326ce7c554cde827ae47db56a39da : SUCCESS in 50m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/be73b0c9f2a544ecaba619c1f57e3346 : FAILURE in 47m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac036b7d06b0484980792e38488bcbf8 : FAILURE in 2h 10m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/28c5849952a049628818ce5585cac32f : SUCCESS in 53m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d67214d8d7e74f9b9a687e89cea28799 : SUCCESS in 52m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/db29a2c76a834a2ea718b33181eae019 : SUCCESS in 1h 31m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/af3dadcb44e547779f9d5fdd7175a6e7 : FAILURE in 1h 01m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a7957dcee3034ea0b0503fe736c0ac51 : SUCCESS in 49m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8d1f240875f14bb7a82170a3c32daaed : SUCCESS in 1h 42m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2b298a7dbd924138aeeaa5b5de5c97d3 : SUCCESS in 1h 45m 09s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1614b7f9417f487db20f777f834e3691 : SUCCESS in 1h 12m 26s","accounts_in_message":[],"_revision_number":19},{"id":"d71bb243c11f423c035d6c6def55862491ad8bc6","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-10-30 20:33:01.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"158b7a089604b8e8f313f1455391381198a3bd21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 21:21:10.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/676512045b4c484c822a9e3ac2b68fc0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/274a38681287418c81ac6901a8226456 : SUCCESS in 19m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/940e4e61212c44acacfeb4d43ba6dbac : SUCCESS in 45m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/86b8649d9402430fb74588e9debe3ea6 : SUCCESS in 43m 04s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"7813e57e8115693de6f5eaf23c4df1226189c913","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 23:28:35.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8856d6d7523f4694b9ab4e4014649b2a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8b3c5e15d10d490884898f419d99c388 : SUCCESS in 1h 26m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e9bfcc39150e4f18bd7f8c1d87a3d3f2 : SUCCESS in 2h 01m 55s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/52797494c6d1463f8128a6784f29fe11 : SUCCESS in 1h 20m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ad2791784bf44cd8bc88c505c4fb86b1 : SUCCESS in 2h 18m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0572dd7e1c784922bc93bfa5de17f146 : SUCCESS in 39m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/979f6927c19a4772aa2c592b70846d60 : SUCCESS in 9m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b207a6212414c88a93af30a7d478647 : SUCCESS in 19m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b3c43d2bb1b45e481c234ce4127a6d5 : SUCCESS in 22m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bae6890c19604d41a503a34531cf31f2 : SUCCESS in 11m 57s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/bfcb743061da45959ea2763664ba43ff : SUCCESS in 30m 22s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8d490c5dffde4ca891f1dc4b5c62d680 : SUCCESS in 41m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1830603e984540daa5df6dcbf3e5a2d1 : SUCCESS in 54m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ebf0cfc89cec42a1b63e6c745e1528c4 : SUCCESS in 53m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d1ef313bc087480990382bff642dab22 : SUCCESS in 2h 54m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/42c0d6ce021f4cee853c55bfa46fa036 : SUCCESS in 53m 51s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1305aeea6e294bc7834ed0301941b1fe : SUCCESS in 54m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/000d935a19124a9e9b0adcb2c3a25958 : SUCCESS in 1h 19m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b8ebd2ad0e5249dcb99ed9ceec8ee916 : SUCCESS in 1h 05m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ef5062a0c4f84fb7a7730e3eaa2b15ae : SUCCESS in 1h 01m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/97111971fb144584afe4f3f64c872d32 : SUCCESS in 1h 21m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ef1498fd2bba49e1bb0dddc1302aaeab : SUCCESS in 1h 23m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3ea4469b5dcf495c85795e225841807f : SUCCESS in 56m 07s","accounts_in_message":[],"_revision_number":19},{"id":"32c7d540d611e29a05acd85912234018f0463294","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-11-07 02:04:19.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"789dd49515b9842629cf8af54c60a41149e39eda","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-07 02:37:36.000000000","message":"Patch Set 20: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/86/890986/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"e92bdbb1dc925e31fb3ed046b216b3fcc605fd7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 03:21:06.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ba7afd0fad744608f6815048b17b482\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/766c373b0b364cfb8f5877b7d5f8d826 : SUCCESS in 49m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7ac50236217040f989262d531d51907f : SUCCESS in 1h 08m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/67bfe2f1ce7f4b18823d6786c238e3d8 : SUCCESS in 26m 38s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"418384843cc13327cc829b1b559f93dce51a9879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 05:13:44.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5ae1be8adad42bd85c98ee43231b926\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/66b2d0edfd4b4a3cb24d5b3db94b5dd5 : SUCCESS in 1h 34m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b1ef8388d8b847cb8be776d7ec41afa2 : SUCCESS in 1h 40m 24s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/fc4ba7b92955448a961ee151e738bcb2 : SUCCESS in 1h 38m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dd26a60b08254417b1f989eb57cce4ff : SUCCESS in 1h 45m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1ab445bac7745358324721826a8832a : SUCCESS in 39m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77072ba334ab41549daf0a233cceb968 : SUCCESS in 8m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7931f1c6b65b40d7b29b3aac4f130cee : SUCCESS in 26m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1843ae9fa42e4a4ca834d6e6dba497e6 : SUCCESS in 24m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e220a5cca90432a83ebf68252fdb6ff : SUCCESS in 10m 34s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5ba8d9e731bf431c974ae27699c9ddac : FAILURE in 7m 36s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/74b106de865f48cb895bb1c3e7cfbfa5 : SUCCESS in 59m 55s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e3d2c230a4844892b5da4d1757594818 : SUCCESS in 1h 20m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ad0bbb054f64daebd6f017577145f54 : SUCCESS in 1h 03m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/088109dfa54b4292b7767ea3f14ae4b1 : SUCCESS in 2h 30m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ded3f6a2a02b441fbdbd37b0a9ac7591 : SUCCESS in 1h 10m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f6198de0dfcf4f6c97734217d4e790ba : SUCCESS in 1h 11m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ff37a2c2e92044ce8c9f083b6b6bfb71 : SUCCESS in 1h 42m 05s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/96ad24c285444d4fbcb243f5eb174928 : SUCCESS in 1h 25m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3b541dc656fb4183a1f771208246767f : SUCCESS in 59m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8f0fc0ef15b24df39ded03e3b812baca : SUCCESS in 2h 39m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ba401fd4caf84c468c2e6f5d29f5c07c : SUCCESS in 2h 29m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6c0df34330264651bdda684a2ad44e52 : SUCCESS in 1h 08m 53s","accounts_in_message":[],"_revision_number":20},{"id":"4f3eb50046b2756feb1d3ad206b5683e77e32824","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-07 23:46:28.000000000","message":"Patch Set 20: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"7ca3c61fa79c286f7fb151ea38026cf18314008e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 00:18:03.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a214be9b9b844e4d8fa6f4bd68190682\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f1739c9896214641854a707c418b2485 : SUCCESS in 19m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b68590a77293466f8251b046848b6c27 : SUCCESS in 20m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7cda0ea7e6fa4342b9444f3ff8706cb4 : SUCCESS in 20m 04s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"3761a0903ec2f3493f2b9afc581047a9f9a2e68c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 02:45:20.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d5ee1b78d5e470b904b8b4143da9c1d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d35bc99f83374c8490521e172dd27194 : SUCCESS in 1h 48m 32s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/07f5357939124786b5c273ce8c81bc40 : SUCCESS in 1h 40m 17s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/43abaebc5e9943d8bc4fc347a0902aa4 : SUCCESS in 1h 28m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1ff37fa923a34cf0943a64c14f18145d : SUCCESS in 1h 58m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58bc325ef5aa4d30bcbb9942302f1a29 : SUCCESS in 30m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/482dd7a1547d43599561e1d2d807f5f2 : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/003713628ea74c92802e6279ff4e831e : SUCCESS in 23m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbc0e23652c8478aa61cc2857a1ec861 : SUCCESS in 22m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b789b033e84e09bcd599c3340ad6cc : SUCCESS in 10m 14s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/db664e3a38e04c93ba39a821e026413c : FAILURE in 6m 46s (non-voting)\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/41aa064940094e93b53a0db752bd2b01 : SUCCESS in 41m 49s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8cfc8df1ea0b48198b73cb772260e936 : SUCCESS in 1h 01m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a416228e04c47179b24703e62af6551 : SUCCESS in 1h 05m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/53edda723d8440af97311008f6d8b94a : SUCCESS in 2h 52m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1964934621645be9e29061a4af02b8c : SUCCESS in 59m 54s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/45f0026b7ce1456f9464430f441c7d1f : SUCCESS in 48m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3f3e41be8ec7469988a567eeaa31d50c : SUCCESS in 1h 15m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e3c19e3247e84695b4fd40ad7b5295f6 : SUCCESS in 1h 26m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/00f42222cfb0443da79534b4a724b5eb : SUCCESS in 1h 03m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4a3b625be4534dfab74ad7dc89175c0c : SUCCESS in 2h 07m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c6cc64477f274957bc4a5fa4f40d3459 : SUCCESS in 1h 24m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/38e23561e28846eb9e6111c46d63a38e : SUCCESS in 1h 04m 05s","accounts_in_message":[],"_revision_number":20},{"id":"23e5e06a47976d5b8ed914bad83539dcfb24ca31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-11-14 19:40:07.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"b5d886fe97df9741d7fb67aee8aaa7060d023027","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-14 20:11:36.000000000","message":"Patch Set 21: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/86/890986/21 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"241f8da22a72b8b9c57fab6610b6ee6810418397","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 20:29:22.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/383b90ddad874748afa3dc42ff461086\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e52c68abf73f4d87b58614bbb3c13fd4 : FAILURE in 20m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/82fc29bd90994c40b05277a77e90a6e4 : FAILURE in 47m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c25053df5ac34087a963567a785330ee : FAILURE in 18m 07s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"1d3629968b0ceb978565b9cace90875ebaea76d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 22:01:09.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/faef6691c5cc4788aa10e62ff6bb90e3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5ee8ebf6fc804784a66f4b7090d78ac6 : SUCCESS in 1h 52m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/69290a98a9384a0a87bb7363d8314ce5 : SUCCESS in 2h 18m 28s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/07a6f74a736248518d12dda53d94b57f : SUCCESS in 1h 48m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cf7f035a977e4fe382789090099f4c05 : SUCCESS in 1h 54m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb9c7c6a9cb44694b31cce5b0463b08b : FAILURE in 36m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c05f3fd2fa54646a07938ddae2666f2 : SUCCESS in 9m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5616870baf6b4fe0a44d3c038bc309d9 : FAILURE in 25m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1596205820a4ef8927798602e340ab0 : FAILURE in 26m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41b56c7ec621431d908a3a88b0ba174d : SUCCESS in 13m 28s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e496005853154f4ca7f3aec9df96e2ff : FAILURE in 29m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3c4a6fa2c45340178e2ea0c9dcb8a9ad : SUCCESS in 1h 02m 21s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/627134a376d94baca7252958d4cd7116 : SUCCESS in 1h 09m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/54b6accade9143ab9b9a8948b1bee7b6 : SUCCESS in 49m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee1407a0739a4827ab50717ed650c59b : SUCCESS in 1h 59m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc084ca475a84328b0afecf092ea9555 : SUCCESS in 1h 15m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5cd8a63a02644a7696e0bb07dcd224a6 : SUCCESS in 42m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/11eac08908e84b3fa5b2b0ddea99df7d : SUCCESS in 1h 41m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/472285dec28d44be965e761effc24eec : SUCCESS in 1h 14m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9c0e4812d75c499f8d778065ebdd18a2 : SUCCESS in 1h 02m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/be66d203c22d475689b8f114412a6db7 : SUCCESS in 1h 55m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c585a9c5484742e59a9d16e790c4964c : SUCCESS in 1h 47m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d31721a820a44d4db405726e7246256f : SUCCESS in 1h 16m 59s","accounts_in_message":[],"_revision_number":21},{"id":"a3c8e59573b79f345e4774e904284cbb47a5f93b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-11-20 22:12:31.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"5f804377ad005cb805dd007edef5a08d3c297870","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 22:51:34.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d456100cc5fa45bb8f15c2e913385f28\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6f8ec203aca46a78a14888a317a619d : SUCCESS in 19m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a36f60d6fd4b4feeb5e41e16151c897c : SUCCESS in 20m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f46a0b6aeffb4da8bfff18b9a138a4de : SUCCESS in 36m 36s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"25b1a8c96484875010d354a5b87d9681f52c425c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-20 23:24:54.000000000","message":"Patch Set 22:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/86/890986/22 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"0fda7a3f23956c26bb71cb9c9529c7b61cf85991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 00:10:52.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c0bc3d0fef24a8589390628c94ead6b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a1d30dad414b4c7eab0b4e58aade617c : SUCCESS in 1h 31m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0170b62bf41342cf94581ada3fb683f0 : SUCCESS in 1h 52m 49s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/0b05ba0eeedf4c05877433c93ed72cee : SUCCESS in 1h 24m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ab070b4cba684b7284c804a40c3ad2f9 : SUCCESS in 1h 23m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c906cd3d85be46499ac8a3bca5dd7ac4 : SUCCESS in 33m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6365dd2ba5d04398b004d48a36e85eaf : SUCCESS in 10m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0835a4a62a94de3b2d9dd5b77eaf70d : SUCCESS in 23m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34a1b77bacb044d2958579af6c712538 : SUCCESS in 28m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c11e477619cd4101a3acf795045c3de6 : SUCCESS in 11m 38s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/69114d0112f644d5b78c91cf115b50ad : SUCCESS in 27m 01s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f49d7b637d3c45c88431b7b5ff8df682 : SUCCESS in 1h 03m 10s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/642f8fb9351b4701a8098c6316e4e5bd : SUCCESS in 1h 12m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0eb511cc445a48b291ba487951062129 : SUCCESS in 1h 02m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/524e30c863344b6f841574c08b019d73 : SUCCESS in 1h 57m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6f1fa5609294d63957e59e415c0652f : SUCCESS in 1h 01m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aba2539a650c4d549a4a7b386bfae25f : SUCCESS in 41m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9fb3480403584329899fd7a891c0a03c : SUCCESS in 1h 39m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4c1c407674234ec3af9064ac4a478544 : SUCCESS in 1h 03m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/24b426e89efa4951834e5d2a193af392 : SUCCESS in 1h 01m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/090d81665d8e4c51b31e6f0a376dc082 : SUCCESS in 1h 50m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0a0e32c044ec425ca67f1a973dc3ad30 : SUCCESS in 1h 45m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0f326846038f463badc77eaf0cf4c69b : SUCCESS in 1h 17m 53s","accounts_in_message":[],"_revision_number":22},{"id":"50059e803c91ba81cfca5f0328fdb83ed17cc387","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-21 02:56:11.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"5099f19238faca1af34af2f6ee63af22c70e9985","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-11-21 13:52:32.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"d30d65142bb13fb27e20c94bf4a11b3a8446de4b","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-11-29 20:23:30.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"3e738b59354d186152e63736da68e424f3e21519","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-11-29 21:19:09.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"3ff43cf640332df2fa578aca20714b398cca1f5c","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-11-29 21:20:23.000000000","message":"Patch Set 23:\n\n(3 comments)","accounts_in_message":[],"_revision_number":23},{"id":"cd6bd43d0f419a4ec063a777f5f62b58bcda26b5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-29 21:51:54.000000000","message":"Patch Set 23: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/86/890986/23 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":23},{"id":"115214a81cdca3df88dcf01c11f430be8217c2d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 22:16:20.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6565d3d199114a8988785f799a18bcdc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/de8aee01881d408a9d77ddef50d4ea1b : SUCCESS in 54m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e730196e3e664766bf1cc117db73d312 : SUCCESS in 52m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ef639fd71fb14191b312b6fa4edd65c2 : SUCCESS in 46m 44s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"c55cd044930bb69b0fc1991af23026b08331a870","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 23:26:26.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c08920aedf9342a3a27b05db64f9b68d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fff24347e8e64b528d0c027cf3f73809 : SUCCESS in 1h 21m 23s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/43bfccf878ba4c7ebdae3850ad48f7aa : SUCCESS in 1h 55m 01s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/b71172fc736e4a2ebada3a8dcde8de00 : SUCCESS in 1h 12m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b465d876db4f460f80f48d02d2277207 : SUCCESS in 1h 39m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46705253a37848239610eb3cd0292e57 : SUCCESS in 28m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d35dfe2f70848309d86da3e2dc7884d : SUCCESS in 9m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bdb3ab310c94a36ac22563588c7c8ed : SUCCESS in 20m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f40037a711446b99f3c2cb7bc6b15cf : SUCCESS in 19m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba2a3cea5904e3f8add4ee3b3579a69 : SUCCESS in 11m 15s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9d1e0de526db41c287f3b636d2c9a7f4 : SUCCESS in 23m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9630dbf269b84a81b18ac508bef6a90c : SUCCESS in 1h 01m 22s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bde902854af64bc293f58e426499f1b8 : SUCCESS in 1h 11m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a08058bc0784fa4a246146c13d75f75 : SUCCESS in 57m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b06598093e2e4949b041179e2143f6c4 : SUCCESS in 2h 01m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f9f13697aff42c5a8ab66fee5ced26b : SUCCESS in 1h 02m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3a8bc129f94047fbb6cb76ddc996e216 : SUCCESS in 57m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e047f4cbc9be4ac3bbdaf72bfe884b0b : SUCCESS in 1h 09m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c63b2ccf388d4b1eb158a33963c33027 : SUCCESS in 1h 06m 01s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2d362f2d9c0d4a759f11b951d830f7dd : SUCCESS in 49m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1dcd8f905a8e4293b83bf0d637a35fea : SUCCESS in 1h 21m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cb66ce7e5e2d47ca994c00417c2bfc53 : SUCCESS in 1h 26m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0905885dd7ac4f18b219987f19c430d5 : SUCCESS in 1h 03m 16s","accounts_in_message":[],"_revision_number":23},{"id":"7a24dbc7428bb6a891f3d455b3f30bd78e376067","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-11-30 18:27:21.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"9ca15280337583b900b8e086449e126a267d9f34","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-12-05 14:46:52.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"ff773024694c7f948bf678eeb4ceccffaef884f9","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-12-05 18:08:52.000000000","message":"Patch Set 23: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"c0916070632cf2eeba0bd3d7201d7ca2918a2e88","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-05 20:54:11.000000000","message":"Patch Set 23: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"5782e5432b5ed2d53f68cc7c3aa98c296a41c8ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 20:54:45.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"11765730ecf925bb58a5216be0163304262f6b38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 23:20:05.000000000","message":"Patch Set 23: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/368319ef21ca496c867f0bad8ae9da4b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a9e709b5aa0445ba934abfc46bf8a299 : SUCCESS in 1h 35m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9fccc29bf15a4f65a4431a69e76425f1 : SUCCESS in 1h 34m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1b9031f765214008b9845cd5d556d122 : SUCCESS in 1h 25m 56s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e26a1594e2034b59a2eefc09fe80fd10 : SUCCESS in 1h 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c2b0331e91c41f3bf0984f331b2f85e : SUCCESS in 8m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40e9d8e288ef4a77baf1b527236e459e : SUCCESS in 25m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0680f77058a74a348124745e30cbcc28 : SUCCESS in 24m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf11a8d90eb4b8db2e62769e30c7f5b : SUCCESS in 9m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/25445a6a456b48aa9c5526548e989fdb : SUCCESS in 58m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1f5f2ca2053442c5b9c0b35290f9d8d9 : SUCCESS in 2h 24m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/276ee146996d408689cb0fc90969dffa : SUCCESS in 1h 29m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5758ae8c93bf49c68f526210c009f684 : SUCCESS in 37m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a452cac909874232b985ed85f57cbf1a : SUCCESS in 1h 01m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f6718e760a154de8adba040e7d306446 : SUCCESS in 46m 27s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2d48973bf3dd427fa4b6d90ddc747131 : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":23},{"id":"81d5f49966cd68db353a787ef7bf65f8659dd105","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 23:20:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":23},{"id":"eb58550e32730342689d3afde6f461ecb604bea8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 23:21:22.000000000","message":"Patch Set 23:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b05a8a208c7b4743bbcdad4f11ee7a2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e83ee43a226c43dfbc3f1dc54deea446 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":23}],"current_revision_number":23,"current_revision":"ed0515737be75745933f4994b5e6b652f9e0a2be","revisions":{"13231e77b5a102a1215968f3292c0b3ccad58f3d":{"kind":"REWORK","_number":1,"created":"2023-08-09 21:35:22.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/1"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"subject":"[DNM] Check if proxy is already running","message":"[DNM] Check if proxy is already running\n\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13231e77b5a102a1215968f3292c0b3ccad58f3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/13231e77b5a102a1215968f3292c0b3ccad58f3d"}]},"branch":"refs/heads/master"},"05389cb38b975e21d298165434c04608cb9b233d":{"kind":"REWORK","_number":2,"created":"2023-08-10 20:01:04.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/2"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-10 20:00:50.000000000","tz":-240},"subject":"[DNM] Check if proxy is already running","message":"[DNM] Check if proxy is already running\n\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05389cb38b975e21d298165434c04608cb9b233d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05389cb38b975e21d298165434c04608cb9b233d"}]},"branch":"refs/heads/master"},"6fb6ee0f26f520f15d89dd1066c51e4ad6d81135":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-08-15 17:21:39.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/3"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-15 17:12:43.000000000","tz":-240},"subject":"OVN Metadata Agent check if haproxy is already running","message":"OVN Metadata Agent check if haproxy is already running\n\nWhen the OVN metadata agent is run inside a container the haproxy\nservice is spawned in side cart containers. These containers persist\nwhen the metadata agent container is restarted. The haproxy containers\ndo not need to be started again when the agent container is starting up\nagain. Just check if the process(container) is already running and return.\n\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6fb6ee0f26f520f15d89dd1066c51e4ad6d81135"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6fb6ee0f26f520f15d89dd1066c51e4ad6d81135"}]},"branch":"refs/heads/master"},"ab8aa9906f6cdc574ccebedda2a349662f8291b6":{"kind":"REWORK","_number":4,"created":"2023-08-16 17:37:54.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/4"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-16 17:26:43.000000000","tz":-240},"subject":"OVN Metadata handle process execeptions","message":"OVN Metadata handle process execeptions\n\nOVN metadata agent should handle process exeptions incase\nan exeption happens when spawning haproxy processes\nsuch that the agent can continue its operation for possibly\nother haproxy processes\n\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab8aa9906f6cdc574ccebedda2a349662f8291b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ab8aa9906f6cdc574ccebedda2a349662f8291b6"}]},"branch":"refs/heads/master"},"5ab44c52bbd62997b0f320c1f4b6baaac85add0f":{"kind":"REWORK","_number":5,"created":"2023-08-16 19:50:55.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/5"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-16 19:50:44.000000000","tz":-240},"subject":"OVN Metadata handle process execeptions","message":"OVN Metadata handle process execeptions\n\nOVN metadata agent should handle process exeptions incase\nan exeption happens when spawning haproxy processes\nsuch that the agent can continue its operation for possibly\nother haproxy processes\n\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ab44c52bbd62997b0f320c1f4b6baaac85add0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ab44c52bbd62997b0f320c1f4b6baaac85add0f"}]},"branch":"refs/heads/master"},"e6f4b3305dce1a1781bef304561cbef584af5e75":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-08-22 13:10:51.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/6"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-22 13:10:51.000000000","tz":0},"subject":"OVN Metadata handle process execeptions","message":"OVN Metadata handle process execeptions\n\nOVN metadata agent should handle process exeptions incase\nan exeption happens when spawning haproxy processes\nsuch that the agent can continue its operation for possibly\nother haproxy processes\n\nrhbz: #2229779\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6f4b3305dce1a1781bef304561cbef584af5e75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6f4b3305dce1a1781bef304561cbef584af5e75"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bb62d1bbb772e55831fe4fda0e6a358026ef5707":{"kind":"REWORK","_number":7,"created":"2023-08-24 17:47:09.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/7"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-24 17:44:41.000000000","tz":-240},"subject":"OVN Metadata handle process execeptions","message":"OVN Metadata handle process execeptions\n\nOVN metadata agent should handle process exeptions incase\nan exeption happens when spawning haproxy processes\nsuch that the agent can continue its operation for possibly\nother haproxy processes\n\nrhbz: https://bugzilla.redhat.com/show_bug.cgi?id\u003d2229779\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb62d1bbb772e55831fe4fda0e6a358026ef5707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb62d1bbb772e55831fe4fda0e6a358026ef5707"}]},"branch":"refs/heads/master"},"409a664935fe40cbe7030aca22bc579347b34981":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2023-08-28 16:47:29.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/8"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-28 16:47:29.000000000","tz":0},"subject":"OVN Metadata handle process exceptions","message":"OVN Metadata handle process exceptions\n\nOVN metadata agent should handle process exceptions incase\nan exception happens when spawning haproxy processes\nsuch that the agent can continue its operation for possibly\nother haproxy processes\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/409a664935fe40cbe7030aca22bc579347b34981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/409a664935fe40cbe7030aca22bc579347b34981"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6ce6f40cd6f1fd25620f54fac2b8e2e74e61cf15":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2023-09-13 18:32:29.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/9"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-09-13 18:32:29.000000000","tz":0},"subject":"OVN Metadata handle process exceptions","message":"OVN Metadata handle process exceptions\n\nOVN metadata agent should handle process exceptions in case\nan exception happens when spawning haproxy processes\nsuch that the agent can continue its operation for possibly\nother haproxy processes\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ce6f40cd6f1fd25620f54fac2b8e2e74e61cf15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ce6f40cd6f1fd25620f54fac2b8e2e74e61cf15"}]},"branch":"refs/heads/master","description":"Edit commit message"},"097fa507bde67fb238dc10b67044b635a5857a33":{"kind":"REWORK","_number":10,"created":"2023-09-18 17:37:01.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/10"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-09-18 17:36:51.000000000","tz":-240},"subject":"OVN Metadata handle process execeptions","message":"OVN Metadata handle process execeptions\n\nOVN metadata agent should handle process exeptions incase\nan exeption happens when spawning haproxy processes\nsuch that the agent can continue its operation for possibly\nother haproxy processes\n\nrhbz: https://bugzilla.redhat.com/show_bug.cgi?id\u003d2229779\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/097fa507bde67fb238dc10b67044b635a5857a33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/097fa507bde67fb238dc10b67044b635a5857a33"}]},"branch":"refs/heads/master"},"81bbe2e0ed20c85bb0e00fd4c1fd7390366d970e":{"kind":"REWORK","_number":11,"created":"2023-09-19 15:12:51.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/11"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-09-19 15:11:42.000000000","tz":-240},"subject":"OVN Metadata handle process execeptions","message":"OVN Metadata handle process execeptions\n\nOVN metadata agent should handle process exeptions in case\nan exeption happens when spawning haproxy processes\nsuch that the agent can continue its operation for possibly\nother haproxy processes\n\nrhbz: https://bugzilla.redhat.com/show_bug.cgi?id\u003d2229779\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81bbe2e0ed20c85bb0e00fd4c1fd7390366d970e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81bbe2e0ed20c85bb0e00fd4c1fd7390366d970e"}]},"branch":"refs/heads/master"},"90669276c7a8eb446de8c5f5e8f9e10e629c7dc6":{"kind":"REWORK","_number":12,"created":"2023-09-19 19:25:14.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/12"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-09-19 19:20:05.000000000","tz":-240},"subject":"OVN Metadata handle process execeptions","message":"OVN Metadata handle process execeptions\n\nOVN metadata agent should handle process exeptions in case\nan exeption happens when spawning haproxy processes\nsuch that the agent can continue its operation for possibly\nother haproxy processes\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90669276c7a8eb446de8c5f5e8f9e10e629c7dc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90669276c7a8eb446de8c5f5e8f9e10e629c7dc6"}]},"branch":"refs/heads/master"},"b9a1185809a6919b9816c746271c1f3c8005866f":{"kind":"NO_CHANGE","_number":13,"created":"2023-09-19 19:33:09.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/13"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-09-19 19:32:50.000000000","tz":-240},"subject":"OVN Metadata handle process execeptions","message":"OVN Metadata handle process execeptions\n\nOVN metadata agent should handle process exeptions in case\nan exeption happens when spawning haproxy processes\nsuch that the agent can continue its operation for possibly\nother haproxy processes\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9a1185809a6919b9816c746271c1f3c8005866f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9a1185809a6919b9816c746271c1f3c8005866f"}]},"branch":"refs/heads/master"},"502df507ed7c15d67e38776f54d55f90ba6395a7":{"kind":"REWORK","_number":14,"created":"2023-09-20 14:50:01.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/14"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-09-20 14:49:46.000000000","tz":-240},"subject":"OVN Metadata handle process execeptions","message":"OVN Metadata handle process execeptions\n\nOVN metadata agent should handle process exeptions in case\nan exeption happens when spawning haproxy processes\nsuch that the agent can continue its operation for possibly\nother haproxy processes\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/502df507ed7c15d67e38776f54d55f90ba6395a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/502df507ed7c15d67e38776f54d55f90ba6395a7"}]},"branch":"refs/heads/master"},"c4c220f014ee229a27ec10e4fbe3641dc662d59a":{"kind":"REWORK","_number":15,"created":"2023-10-02 15:07:23.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/15"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-10-02 14:58:32.000000000","tz":-240},"subject":"OVN Metadata handle process execeptions","message":"OVN Metadata handle process execeptions\n\nOVN metadata agent should handle process exeptions in case\nan exeption happens when spawning haproxy processes\nsuch that the agent can continue its operation for possibly\nother haproxy processes\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4c220f014ee229a27ec10e4fbe3641dc662d59a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4c220f014ee229a27ec10e4fbe3641dc662d59a"}]},"branch":"refs/heads/master"},"dab794c8bd20ad3f14f0e7d778057e83dcc95a35":{"kind":"REWORK","_number":16,"created":"2023-10-04 18:47:58.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/16"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-10-04 18:47:46.000000000","tz":-240},"subject":"OVN Metadata handle process execeptions","message":"OVN Metadata handle process execeptions\n\nOVN metadata agent should handle process exeptions in case\nan exeption happens when spawning haproxy processes\nsuch that the agent can continue its operation for possibly\nother haproxy processes\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dab794c8bd20ad3f14f0e7d778057e83dcc95a35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dab794c8bd20ad3f14f0e7d778057e83dcc95a35"}]},"branch":"refs/heads/master"},"c9e890ed237d69471e7903bb7af288ce0a9198c6":{"kind":"REWORK","_number":17,"created":"2023-10-27 01:01:15.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/86/890986/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/17"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-10-27 00:59:35.000000000","tz":-240},"subject":"Metadata: handle process execeptions","message":"Metadata: handle process execeptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9e890ed237d69471e7903bb7af288ce0a9198c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9e890ed237d69471e7903bb7af288ce0a9198c6"}]},"branch":"refs/heads/master"},"3e67b3c4b59f91ce572992dad59d356d61e39dc1":{"kind":"REWORK","_number":18,"created":"2023-10-27 16:22:47.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/86/890986/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/18"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-10-27 16:21:35.000000000","tz":-240},"subject":"Metadata: handle process execeptions","message":"Metadata: handle process execeptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e67b3c4b59f91ce572992dad59d356d61e39dc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e67b3c4b59f91ce572992dad59d356d61e39dc1"}]},"branch":"refs/heads/master"},"16f79c450607e4d893e48a690a43c1f21e3573da":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2023-10-27 19:42:53.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/86/890986/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/19"}}},"commit":{"parents":[{"commit":"176b144460570fcf2792a199d2b639b335822f14","subject":"Merge \"Add extra router attributes for ECMP and BFD\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/176b144460570fcf2792a199d2b639b335822f14"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-10-27 19:42:36.000000000","tz":-240},"subject":"Metadata: handle process exceptions","message":"Metadata: handle process exceptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16f79c450607e4d893e48a690a43c1f21e3573da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/16f79c450607e4d893e48a690a43c1f21e3573da"}]},"branch":"refs/heads/master"},"853191d8f05a90635414afed2cf48a939cb13ab0":{"kind":"REWORK","_number":20,"created":"2023-11-07 02:04:19.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/20"}}},"commit":{"parents":[{"commit":"f99da4d15c8efd74e7466e6585c17bbd441afc7a","subject":"Merge \"get_hosts_mapped_with_segments add filter agt_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f99da4d15c8efd74e7466e6585c17bbd441afc7a"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-11-07 02:04:05.000000000","tz":-300},"subject":"Metadata: handle process exceptions","message":"Metadata: handle process exceptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/853191d8f05a90635414afed2cf48a939cb13ab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/853191d8f05a90635414afed2cf48a939cb13ab0"}]},"branch":"refs/heads/master"},"f375b2b24d41946065eecb771de3d18cc662082b":{"kind":"REWORK","_number":21,"created":"2023-11-14 19:40:07.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/21"}}},"commit":{"parents":[{"commit":"0027928c348ecc437a460b059d1ade3a2d536d18","subject":"Merge \"Remove obsolete PID files before start\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0027928c348ecc437a460b059d1ade3a2d536d18"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-11-14 19:39:52.000000000","tz":-300},"subject":"Metadata: handle process exceptions","message":"Metadata: handle process exceptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f375b2b24d41946065eecb771de3d18cc662082b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f375b2b24d41946065eecb771de3d18cc662082b"}]},"branch":"refs/heads/master"},"c2820f1628f1688385513d5205042fc0ad7c20f6":{"kind":"REWORK","_number":22,"created":"2023-11-20 22:12:31.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/22"}}},"commit":{"parents":[{"commit":"d2b5ad4ac58340fbab7c9c223d8078565e1b94a5","subject":"Merge \"Make unit tests compatible with neutron-lib-3.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d2b5ad4ac58340fbab7c9c223d8078565e1b94a5"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-11-20 22:12:18.000000000","tz":-300},"subject":"Metadata: handle process exceptions","message":"Metadata: handle process exceptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2820f1628f1688385513d5205042fc0ad7c20f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2820f1628f1688385513d5205042fc0ad7c20f6"}]},"branch":"refs/heads/master"},"ed0515737be75745933f4994b5e6b652f9e0a2be":{"kind":"REWORK","_number":23,"created":"2023-11-29 21:19:09.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/890986/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/890986/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/890986/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/890986/23"}}},"commit":{"parents":[{"commit":"d2b5ad4ac58340fbab7c9c223d8078565e1b94a5","subject":"Merge \"Make unit tests compatible with neutron-lib-3.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d2b5ad4ac58340fbab7c9c223d8078565e1b94a5"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-11-29 21:18:53.000000000","tz":-300},"subject":"Metadata: handle process exceptions","message":"Metadata: handle process exceptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed0515737be75745933f4994b5e6b652f9e0a2be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed0515737be75745933f4994b5e6b652f9e0a2be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
